)]}'
{"id":"openstack%2Foctavia~693760","triplet_id":"openstack%2Foctavia~master~I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56","subject":"Fix filtering for provider capabilities list API","status":"MERGED","created":"2019-11-12 01:35:37.000000000","updated":"2019-11-18 18:19:59.000000000","submitted":"2019-11-16 03:24:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693760-1573874693209-7361ecdf","meta_rev_id":"5070b048010104119fefddd711bd341d13267f46","_number":693760,"virtual_id_number":693760,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-16 03:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-12 21:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-11-15 15:40:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2019-11-16 01:49:59.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":1,"date":"2019-11-13 08:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-13 06:34:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2019-11-16 01:49:59.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2019-11-13 06:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-11-12 23:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 21:40:52.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-13 06:18:25.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-11-13 06:34:30.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-13 08:31:57.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-15 15:40:07.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2019-11-16 01:49:59.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-11-16 03:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d57e3c54cff16364d26e2a31ddf58b8bf5f77275","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 01:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84325f320cb428de8c0093ec3cb2bf77da7c3582","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 01:35:59.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow-1\n\nNeeds some tests","accounts_in_message":[],"_revision_number":1},{"id":"61926913f7373354d0a15098e5408d5a0012ef4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 04:11:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbd00e1c2e1a4b4fa8df016f21238d3b : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e344f5cfc06a4d79906988ef6e28e51c : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edcb10e6636c4fbaad44437de856a84a : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e39a546443d491692f5329f3a12a850 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f76b21057cd443d8b1e1fe70f145f4d4 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9faffd3e538450f9e81a78636ee00d8 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eeba16bd9fd4d6d81ea92ef3cd37e9a : SUCCESS in 12m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/37aa496fcc6a461aaa758e780285dc90 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca7d8b18756d47638ea334122e1111d9 : SUCCESS in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0cc40341443643c2b7f8c7fab70ee167 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1126585d95d343ab95a057e86e6cf54d : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54fc34e2e2934f7a9fd8e4f0d6edd800 : SUCCESS in 41m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dc7e32364beb43c799c11d1e76fd3770 : SUCCESS in 56m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d112db2069f46b8a1d7be74ef10abae : SUCCESS in 1h 42m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/635878cbe27a4466bf72bda003e80ac1 : SUCCESS in 2h 00m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8d9771db196b46bb8058815c4d0b7b75 : SUCCESS in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b8b9377954542d2a7b4c6b6c8b23044 : FAILURE in 33m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d338bd98bb63457593ccdc795689712d : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0237584f90fd40e2a870ca6426e2bf1c : SUCCESS in 1h 33m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf982931854941dda1530f7c8319ccc5 : SUCCESS in 48m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b9d4a9255910465cb6087b7843a34269 : SUCCESS in 1h 35m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51746f3970d64379bb5d60baaa9970d3 : SUCCESS in 1h 10m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1e0af9d6aa2f4aa6bf7797bf5407394c : SUCCESS in 1h 24m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/80e066b80c104de88eb7bd5bd9a501a0 : FAILURE in 1h 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4e482931310be43395ae9a5358f94afdfd919b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 18:03:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"447430160420367655e642bfbb7e0f4e4e3f4b09","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 18:03:46.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"4f6b83b540fb7545af7c05fcac7c8f14a5bd473e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 20:29:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acd566cd09004008be45d19fda5c2c4e : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb3f99c40dcc4ab2af81076ca1879f54 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57dae0cc1a7461ca18261d1fa825a8d : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55a86919de5344bba420260c41d7bb2a : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b3d0ae95a674a6391050fed6075ff38 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aa3b6a382154d30bd0ed710a3a924ba : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37c8b3d29d94201ab94ca531102a79f : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/393e95a0d3844aaea7f7b9add869f8ab : SUCCESS in 6m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0e368873e99b49009667acc47191598d : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51ea1b1a88a84db88123e4ee1d7f8d3c : SUCCESS in 7m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bb6b13ccac6b4ace815d823d6bcc6f75 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93691fa3433e45e6b22c64eb48e732b7 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e8b94527be741b0abcbe6bc0a1556c0 : SUCCESS in 44m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c83cb1a272254885bc31f8a0dd440782 : SUCCESS in 55m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88213c95b7f5463dabb8f412a5e6617f : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/21580053cbed4d0d9a2b878abc7a263c : SUCCESS in 1h 41m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1ff89deab4094a0db44ebc2e12a27015 : SUCCESS in 2h 20m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80ea8cc5315c42438c58069ca3afe3be : FAILURE in 35m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/30bf74dd72c849b696c0e75bb8f4845a : SUCCESS in 1h 08m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c821b5043044e8dba642a50cbbee3c4 : SUCCESS in 1h 08m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59bb6973519442eba748cdcbb9b31fc2 : SUCCESS in 42m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/971ad259418d4dde994676329ae8279c : SUCCESS in 1h 23m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e5f14cad82b4643b82651222a1f5ed7 : SUCCESS in 1h 17m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ff89d59bd7044dcbae7af6fb676c964f : SUCCESS in 1h 00m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f9bac857e1514d2eb49626e09de04aa3 : FAILURE in 1h 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"632a992478d0adf29ca3ea533d5204f7c45873c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-12 21:32:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLooks fine, will rebase my patch on top and see if it passes, but looks correct.\nIf you do make another patch, the releasenote needs love.","accounts_in_message":[],"_revision_number":2},{"id":"1fc10949c4f536870c8941620db0fe4580063f03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 21:39:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"929ac87a7e4220c925b7d4183964d568e77e7c84","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-12 21:40:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1b0534908597f79bd8a761e046e884309dbb77f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-12 23:12:28.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"c1503e944e4b2dd0d1a55b883c8171273c2191c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 23:45:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/562424708988424b9fe216615928122b : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9c24aa215d34aab9da1b48543ce550d : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c07339700f49d8a67b1de048e9ee49 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d16e47e5db64fa98722ce5dd7c26a34 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c11a72afffed4738a240898a5d21132f : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c076e4198bcb4c55a9717c60ddc43966 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0991258a1ddd4423a7f84561b15f2f77 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3d1947090134c3bbeb2cea9ad6226fa : SUCCESS in 6m 57s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/78fdc4fddb1d42218b5f6ab3e77cce06 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/86a3b91a9ba94972952ad183e802d786 : SUCCESS in 8m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7dfc2ac9052d42f4b0e77919a614d726 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15d360097bf04e9396247488801259ca : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2729f42ff1224dca84c25b14eb83c79c : SUCCESS in 44m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0eb5efa2d677452bac4697f2230c00bb : SUCCESS in 44m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/221aeb5d34d54129a6ae8abea8f15460 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a616efe9c6634a35a0903daa5179c59c : SUCCESS in 1h 39m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/db9b2ba86fa5418ebe2a0abdfb5ed896 : FAILURE in 1h 57m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f4b021488e64f9d9081b6765f9f7b4d : FAILURE in 32m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a388ef7398948a7934296157d626d97 : SUCCESS in 1h 09m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/046aaffe440b4faab6c55ae7864e3a7d : SUCCESS in 1h 17m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6239cb24869d47e4a2b8df9d0028fbf0 : SUCCESS in 58m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/720f3bdefb50480d9164e934c808dfa8 : SUCCESS in 1h 27m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c15564cb5767493f84cb74142a56ea41 : SUCCESS in 1h 04m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ed5b7ea9cafb4c8d8a098b97ae388836 : SUCCESS in 1h 04m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/89b75cc258ba4db294e90396f12faee9 : SUCCESS in 1h 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2424484d5c32e30f2e079f39ff4aa99220e1d688","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 06:18:25.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"3c61f5ac7151145a3cb5e690429c766eee4e9edd","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-13 06:34:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ac0d9e6f1c3ec58ba6eec3a0a472f1326c2b8e3f","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 08:31:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fec149c495672501c5c26cacea567124b474c56e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-11-15 15:40:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5444feac7ab4c0c2eb6d930c15fc412f6902453c","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-11-16 01:49:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a159dda60c70237d4aba57f9ebe922cb5418959d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 01:50:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"030cd0221b42eb0b77a6b11d252ebe39f2925a6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:24:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b052813078a1498cb545b07e13d79708 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0776f45397a461599e6bf2ed3c2eae5 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a019d3e912d44b6e891f2e509083e36e : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c61571ccdcbc45b39caa2255e63fca87 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be87641c07ba4cf8a8a6a850c48864e3 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5770f48320d4486b57144b60073d7b5 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8b3cc6a461040fba62c2cd5057d77d6 : SUCCESS in 5m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/03074b0cd67b40f89edf8f54e15b1e44 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17b813fd272240c28e57e84b086db0d2 : SUCCESS in 5m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af46af7e26e945028fada57612850281 : SUCCESS in 35m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80151d794f534590b9c582fc12ea19a1 : SUCCESS in 41m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7cfcc6ca85445bead9e28b3554d2468 : SUCCESS in 1h 24m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/05149e8c857f476994af3c1a30000ed6 : SUCCESS in 1h 33m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a514bb7a3734dada752e302a54ff618 : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":3},{"id":"29a241f3c397c510b9fb0f2f192898009a4734ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:24:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"74e557f1967cf66ed7980feb3d5cee933cf86975","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 03:26:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8826052278354abf907d36205a65ae7a : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adeb4416988f45e683023095572e4ffe : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3},{"id":"5070b048010104119fefddd711bd341d13267f46","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:19:59.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 735de5e175b8ee85fb87a897c70bba1e3c9f2f2d","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"28c936ec7101a03496558bcda311edf205ee74b4","revisions":{"1c1a40882560b0ee440c12db2bcdd71afa54ea96":{"kind":"REWORK","_number":1,"created":"2019-11-12 01:35:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/693760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/693760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/693760/1"}}},"commit":{"parents":[{"commit":"51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328","subject":"Merge \"Fix typo in doc agent.py-\u003eagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 01:33:03.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 01:35:36.000000000","tz":-480},"subject":"Fix filtering for provider capabilities list API","message":"Fix filtering for provider capabilities list API\n\nThe \"Show Provider Flavor Capabilities\" would silently ignore\nfiltering parameters passed into the API. This patch corrects this\nsuch that you can filter the providers capabilities list.\n\nChange-Id: I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56\nStory: 2006817\nTask: 37377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c1a40882560b0ee440c12db2bcdd71afa54ea96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c1a40882560b0ee440c12db2bcdd71afa54ea96"}]},"branch":"refs/heads/master"},"c3fbb4c29a9635f8e602958e4e23ab1de87faf13":{"kind":"REWORK","_number":2,"created":"2019-11-12 18:03:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/693760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/693760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/693760/2"}}},"commit":{"parents":[{"commit":"51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328","subject":"Merge \"Fix typo in doc agent.py-\u003eagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 01:33:03.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 18:03:14.000000000","tz":-480},"subject":"Fix filtering for provider capabilities list API","message":"Fix filtering for provider capabilities list API\n\nThe \"Show Provider Flavor Capabilities\" would silently ignore\nfiltering parameters passed into the API. This patch corrects this\nsuch that you can filter the providers capabilities list.\n\nChange-Id: I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56\nStory: 2006817\nTask: 37377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3fbb4c29a9635f8e602958e4e23ab1de87faf13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3fbb4c29a9635f8e602958e4e23ab1de87faf13"}]},"branch":"refs/heads/master"},"28c936ec7101a03496558bcda311edf205ee74b4":{"kind":"REWORK","_number":3,"created":"2019-11-12 21:39:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/60/693760/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/693760/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/693760/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/693760/3"}}},"commit":{"parents":[{"commit":"51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328","subject":"Merge \"Fix typo in doc agent.py-\u003eagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 01:33:03.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 21:39:50.000000000","tz":0},"subject":"Fix filtering for provider capabilities list API","message":"Fix filtering for provider capabilities list API\n\nThe \"Show Provider Flavor Capabilities\" would silently ignore\nfiltering parameters passed into the API. This patch corrects this\nsuch that you can filter the providers capabilities list.\n\nChange-Id: I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56\nStory: 2006817\nTask: 37377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28c936ec7101a03496558bcda311edf205ee74b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/28c936ec7101a03496558bcda311edf205ee74b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
