)]}'
{"id":"openstack%2Foctavia~693762","triplet_id":"openstack%2Foctavia~master~I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de","project":"openstack/octavia","branch":"master","topic":"octavia-availability-zones","hashtags":[],"change_id":"I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de","subject":"Support creating an LB in a specified AZ","status":"MERGED","created":"2019-11-12 02:37:30.000000000","updated":"2019-12-17 22:52:42.000000000","submitted":"2019-12-17 22:49:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":63,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693762-1576622948702-716aec10","meta_rev_id":"e2ad4d1515a1859c0333c555309a10c9347ff57a","_number":693762,"virtual_id_number":693762,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 22:49:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-17 14:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-17 21:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-17 21:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 10:22:28.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-12-06 12:36:23.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-17 14:08:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-17 21:17:03.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-17 22:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44dc3dbfc99b43501c79e4335a1dfed0dc4702d9","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-12 02:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a90f81e610237ddb435810055afc736b9f52dfe","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-12 02:38:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0b3dcc76c79d85a99cfdd1a62cac52f004b0213b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 03:51:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eca666f8ccd4e7da56e3480b3bad2c3 : FAILURE in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae9cf6fdb854eb490864f1506139779 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01aa1bcdeab24ee6985fa205bcee0743 : FAILURE in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36cd204596dd43c8860758dccf9f8511 : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2dae2e83ac4448c904266ac7c76286d : FAILURE in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d7c405c1ff3472aac8da19e856918f8 : FAILURE in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20be66169f74c71b3e2469a4e132c28 : SUCCESS in 10m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93b20004018f49c49161b5316854dad1 : FAILURE in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fd3b4de04fa848dbba6783277177f106 : FAILURE in 6m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/47f7f50dcdc94b02978d73638f51bc6c : FAILURE in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da6c0c9bc0e647249b3a190b893be2c0 : FAILURE in 5m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f900f4fa972a44919939036885230dbf : FAILURE in 30m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b303d7cfea034c0bb8784183309368dd : FAILURE in 31m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b9fe41715a247bda67973afc895e66b : FAILURE in 41m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/31760873a54940f2898f9c6155cf29b5 : FAILURE in 47m 26s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ece26d272f484368b85e54f33bdb3fc5 : FAILURE in 1h 00m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fbfd0276149441693bef18a7537a681 : FAILURE in 34m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/35483c6f74f9470eb161aefd8b2fc1fb : FAILURE in 1h 02m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5807dc804b54416ebe71a49c8f69022f : FAILURE in 1h 00m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/65019b23e8354ccb8f9aa3027e7d8d45 : FAILURE in 45m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/abcfebf5a4f0422c95906054ca1e87dd : FAILURE in 58m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db8c7d180f39459888f002a7a6fd5d34 : FAILURE in 44m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ab2dfe554dc34226bd420156c672d1ac : FAILURE in 55m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fce4d00006204597b0b6e0ca191c983a : FAILURE in 1h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3997d3f128d6cefad9c551b4cb60f87b48975811","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-15 02:22:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8fd31e6c280c58d871d1159d7f9de55d3a0c6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 03:34:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1012a311d9c41818449f207024b52f7 : FAILURE in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/829e18e02a3f44ccbfe6c70a7bd0cbb1 : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762231bac18c4bfe9c144e272a05d3f3 : FAILURE in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a1953c1b64e485e9a6b66867f778f7f : FAILURE in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfdb7f430034632bc985b7f1848daca : FAILURE in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48c6123006504590b8e68eaf73236b7c : FAILURE in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658f891a83af495db6986d4ff732d664 : SUCCESS in 10m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b21ac51d9ca348c29e9bd5e24189c725 : FAILURE in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4721eead6ce04b0bae9931f06e1f764f : FAILURE in 6m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/088b0ec91f244c9a93d41df524d2363b : FAILURE in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b43cc57c798420ca3241d3d63e6c685 : FAILURE in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7044ffb971a4f85b4ef64cf058c6216 : FAILURE in 33m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b46a5cc20a9d4ab7945a890588d9d83e : FAILURE in 35m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b953ab2a243d4e9f83cb3ae4a2a41139 : FAILURE in 50m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d8b01364b79434690ef86dc6f2e4516 : FAILURE in 43m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/af9dcdad376840ab9c61e312a5d85f57 : FAILURE in 1h 03m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b64f51d48044a0e91d72a5f9b938e7e : FAILURE in 30m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6c729fbd56a4a71b357fc8dffbb89e8 : FAILURE in 1h 05m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d703d19e86a43b7bd4590e07e149d18 : FAILURE in 1h 09m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/558b4c9cfa8b4aaab16ddff8197c2cbf : FAILURE in 43m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9fa4dd5a50b447191578a748117e533 : FAILURE in 47m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cea78d1aad1b4b21b58c45e712418216 : FAILURE in 45m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1f56057e1f624cf0939f5c38081600bf : FAILURE in 40m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/492a6905317f44b3a58f7ff1d00d6412 : POST_FAILURE in 51m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"311c5160afbf0720ccc9acf29fd5e25597883c0c","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-15 05:50:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ac811ce1a4f5f6a780b2fe11df180e4458316cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 07:01:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c36186cd6ef346c3b6d248321cc61585 : FAILURE in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f446b47640848cdb690d007dc3e48e3 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d80fa7925a4badb532f4ee7b21407c : FAILURE in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8610b5c34ff84f788df2ab89d6b3dd4a : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332f5e1c5e7247faba0a25c0c17b4030 : FAILURE in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c628932ca5e64ae88cf884ef4974556b : FAILURE in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37937604d13047bba57b549ffd767580 : SUCCESS in 9m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6163631630e74180af47bba52edf1f0f : FAILURE in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb849f51e4ee423fb002bf5b3d6a43dc : FAILURE in 6m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0eba69a8f8dc4e18a061b3df9d8d9c1b : FAILURE in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39eb03cb9a9b4abc911da648baac9ad0 : FAILURE in 5m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caadd44383294a2aa5c330bf49aae43b : FAILURE in 36m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8099457621bc454a858861899cede712 : FAILURE in 32m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0953d10424c042d39a82afe1813250c8 : FAILURE in 50m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e4a1bffdd14743d4817ac8aa118b2728 : FAILURE in 46m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9d6acdea9a8b406e9b3c7cb31de1869d : FAILURE in 1h 00m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14dc360e8fb34e158449781fd41da999 : FAILURE in 40m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c58bcf757c634a3cb03083b9f91be78e : FAILURE in 1h 06m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5908fd6d21243c894bc4dc922652f80 : FAILURE in 1h 09m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c8accb5316a4216ba07fcab4a29d1ed : FAILURE in 47m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/559eaf3f1e1a488ab8c8116d79bdbb4e : FAILURE in 54m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d83f57ddee3349aab4b54e02ed0bb3d1 : FAILURE in 44m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/44d2bd9617144a649eff7a7edfa97aaa : FAILURE in 45m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc8e7b95ba414c769477bff8a1320f63 : FAILURE in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddd42590d53beeefff2ea738cdab15da17f15915","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-18 01:56:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8f3a44cc8015547b3de0ee569f083b2261c67ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 03:36:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a77cf354bc14b14ba7c1c9890e88c26 : FAILURE in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf7415edb6204c89abd12abe3b728f8c : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd3baa6d6754831b8e70d14d8395ead : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c85622c759bf4dc6afe777e3c3a6241f : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7add4c48307b4760b3313edbe4c99bce : FAILURE in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/299e5c1f5c8f4e18bdca0b0e0aae1ed9 : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955309c8e9394e28a0412f1455da3391 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26832a9477424a1ea382fa4e204623cc : SUCCESS in 7m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b2a0bf462f6242438832281c1915646c : FAILURE in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/45ffc37207f34e15a7cc2b7372e26fe3 : FAILURE in 7m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d41782de4f5745bcbc4f000b3fb6d3b8 : FAILURE in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f42dc16b51c445899f85224d2513413 : FAILURE in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0377b46859f44bb28dac9bf7b2ef25a1 : FAILURE in 32m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f9d9f5ac54bc4463aa1ff058e5bb7585 : FAILURE in 29m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/834a05b4af054723b88c47535780112b : FAILURE in 54m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a8610f99ba1146c7809994ba0ffe0d43 : FAILURE in 48m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/24f8bcbe96804f39abf278448be91f01 : FAILURE in 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5eee31c30e1c484e8d732fccf1ca3507 : FAILURE in 43m 33s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/240b23ce8d7e42bfb3ae7358b84ec4fa : FAILURE in 1h 01m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ef3d39c3bb44b8d961a1791051c39fa : FAILURE in 1h 16m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ecfc458a37334350a62144d2c0328a67 : FAILURE in 54m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1e3de68526bc4fc181801da5784f8a6a : FAILURE in 51m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94e53b7413004cd18d272a612c814aed : FAILURE in 52m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e30f6d1c8b104680b5e7f5255e8ad198 : FAILURE in 42m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f35cc567371045a1b7d942317d492cad : FAILURE in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3b9a5c73b1b53180655f4270800b1d17d135a87","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-18 03:42:41.000000000","message":"Patch Set 4:\n\nNeeds https://review.opendev.org/#/c/694707/ for tests to pass","accounts_in_message":[],"_revision_number":4},{"id":"afb191d780b71da455f943bd5b9670e322614fc9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 15:13:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3e847b41f8771306acc71b48beb7e4e36044a559","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 15:14:02.000000000","message":"Patch Set 5:\n\nRebased and added depends-on","accounts_in_message":[],"_revision_number":5},{"id":"38af7556dec9fae706762d94c434da6e478dda1f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 15:33:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"969f0ef6f88069bb48d57662d3123639b62dfc7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:40:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bae18161acab4ccf875f6900f8c4a3fb : FAILURE in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7ae27a482764e39a4a735c0d449b852 : FAILURE in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df5aba78ba64d70af99e26d647f7510 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab46caae72e04a508852b9c0b1e3e2d5 : FAILURE in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18c338fbeed449b4889a28057eb2b61f : FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/697872186565477d9f3cefbf17b3cb00 : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b9eb0551154668aa4b4889c492a6c1 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec7be696dab84799a5c69971be99b9ea : SUCCESS in 6m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/28a5a870daef43dea94a5654a77358c4 : FAILURE in 5m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99846d22fc72407f990420e5b035271d : FAILURE in 7m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a17db02ad1cc4fc0b0ff7bb42a53377f : FAILURE in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88ce6add80a245568984cef64ae623e0 : FAILURE in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d73be4e3860a4fd48fab227c9a39b69b : FAILURE in 35m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dbcbcc038dba43e995ff3c22cafddd8d : FAILURE in 45m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67dc5bc0c2ff40d58244f074f94a23c5 : SUCCESS in 1h 37m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b578e199d49640e38fad3c5140285d8d : SUCCESS in 1h 46m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c6ba3baed8fb469aa723ec8af38ab8c1 : SUCCESS in 1h 52m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/469eb9274d8d4b0ebd171423365910f4 : FAILURE in 47m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/53beae7c68fc4ed2a8afa7bbc71af625 : SUCCESS in 2h 00m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8238e28ae312455f92cd64c679730174 : FAILURE in 1h 12m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4d2d1bd45e540e5b12f2f0ac95f477d : SUCCESS in 53m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/78f2c5dbc74848fb99441b3587319581 : SUCCESS in 1h 32m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bbd75ff3e274048a758964d843f2ef4 : FAILURE in 1h 16m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/46a5342ba67847849b300fd90707433a : FAILURE in 1h 19m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/49f50b196cbd4bbb9ddf27fb04fa1162 : FAILURE in 1h 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3b27b641e301de046a51d51a467093caf7fcd287","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 20:00:56.000000000","message":"Patch Set 6: Code-Review-1\n\nWith the depends-on, the \"tips\" jobs will actually use the new lib code. So, you can see there are some actual (but lame) errors there that need to be fixed. They look like simple stuff (wrong expectation?).","accounts_in_message":[],"_revision_number":6},{"id":"808a5fcec556d1e51e58eda462ffd5a712de2758","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 20:17:02.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"69b3532ca28e21353dff7bd151988a59cd92a9f9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 20:20:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"38e33afa9789a3771084f0efbd1c035fc76a11d4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 21:23:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"610c1002d04e2c5f514ec70e5276f1c494426d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:34:03.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"231d5e79fc3f962359373a2221505c252a0daf5b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-18 21:37:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"afbd385314bcdee5906db2ea1cab413a1fc56ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 22:16:23.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"1a633143789ba702f72b1229c59605f4bf913555","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-18 22:35:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3ccc9a7cf656abd0e51d8e9c4ae9d42141a4f502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:20:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/674aa713deb8401983a80de75f771561 : FAILURE in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23284a6037894c87ab9af3f527bc0e44 : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc74af9c1a11436da3168400444b5c21 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4222f10878074e7f89d8b824665d3e88 : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99c1cd7673d043309b27559f290ffe39 : FAILURE in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2e31f258dfb450b99bd00c88c9d8d54 : FAILURE in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cd30fb06064007ad3e04a34fd8c638 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0f105b2dbcc42cbbd6955350427e4ba : SUCCESS in 7m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1e9d6cdfb9a4416cbe2c996f5e4a5909 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0591a5aaa9ba462b8079d07137c252d8 : SUCCESS in 6m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a1ab93c4195b463390e29a3fd2b86b4a : FAILURE in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a11aeb49cf647cf88038a1ced4d752a : FAILURE in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/262f43b105cc475c8b2471d26240ca4a : FAILURE in 34m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3524ddd1c8ae46b996ac317b238d72e4 : FAILURE in 17m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22a84a91dc4e4d42b38ee93d2d19c950 : SUCCESS in 1h 39m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/81004f4c42ae4f2ebcbdbc440a4c9a5e : FAILURE in 15m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/624c41fabaeb4f6c890522ce8e08b2f5 : FAILURE in 22m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3491a4e927e64d4892629cc3ee53120d : FAILURE in 22m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a854e35e691941129248a8e4bad8557b : FAILURE in 17m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82bc70dedf5e4f588d84f4f918e9ef47 : FAILURE in 1h 14m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6aac662a151b480e95fb2b3c25a60fc7 : SUCCESS in 52m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/26d0e9f125a049a3bb69b3217cba138d : SUCCESS in 1h 43m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9423e773fecf43d49e6aacc1bf614bd4 : FAILURE in 1h 05m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d3325408e51949b0a31c73d79e8c8675 : FAILURE in 18m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/02233bc821934cb683e7bb0647077734 : FAILURE in 23m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7337109ab2a37df9534228850bf5963a58f971aa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-19 03:07:54.000000000","message":"Topic changed from 693765 to octavia-availability-zones","accounts_in_message":[],"_revision_number":9},{"id":"dc4584610f99a4f720e31d7ae5bbafd20d3f0d69","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-11-19 04:07:22.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e66d924f74b799f7736a7a5d33cbf02d86422257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 06:03:55.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2519fe169e8496ba4cec05468915550 : FAILURE in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f9dfebf7cbc4b63a22bcf90b5f9096f : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d640aeddc708476fa52f4b8568ad18a0 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aed47133628e4a8494a89310acc20f8a : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe2870bfa2f4fcab22052fdae419f7a : FAILURE in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e8e882d6c9c4dd5a05a5165ef93b231 : FAILURE in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b12f573f9964a099c739d70357cace5 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf58aba9c524650a2ad6cbc9f6a08fa : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb18ed0c76fb4ffcb12a7a8b03d283c3 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/436d824c49d045bdb18faa6cd09d9aa7 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cecf81ecc569428b8e0d086631866e05 : FAILURE in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e15fc342d7342478b90c02f3324eb7c : FAILURE in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/380f1bf8d3e2429397e0f5456635267e : FAILURE in 34m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8f77677c5c7f4e1396e5fde27001f992 : FAILURE in 15m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03bcb125611446e0ad8ac798e23309f7 : SUCCESS in 1h 51m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7f6e43daa27f4a6595023c497b788d00 : FAILURE in 17m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b96abec550e04f11b127eb8e6def37e1 : FAILURE in 21m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/11cf590ce5ee4c26b56f9b0c9e9e53a3 : FAILURE in 20m 49s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ed869e375d44257b46127371df6da02 : FAILURE in 17m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23e5ac093c9f481b8f69eef23fdc75b8 : FAILURE in 1h 10m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03d5a87c18b5461f9970fe5469d5f0ca : SUCCESS in 45m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d9322cb4c5f14fafa988c14d7d69dacc : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9e6b2db611d41dbad1cc17d4e1c3516 : FAILURE in 1h 11m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/675e2005d1a0415380d506895c12f26f : FAILURE in 11m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9771ce341e174e399f36087731c55bce : FAILURE in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"377bb19d41c987e104b9edf5907b3d6cca105ab0","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-20 10:22:28.000000000","message":"Patch Set 9: Code-Review-1\n\nI think recheck will not work. Because your change are depend on https://review.opendev.org/#/c/694707/ as you needs changes in octavia/tests/common/sample_data_models.py.\n\nSo,  https://review.opendev.org/#/c/694707/ must get merged first, after that test case (octavia.tests.unit.api.drivers.amphora_driver.v1.test_amphora_driver.TestAmphoraDriver.test_health_monitor_create) will be passed","accounts_in_message":[],"_revision_number":9},{"id":"d1c7c2f44dfe03a020ae0cff3c6a9fd53c128454","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-22 00:17:12.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nA few comments. I will check coverage and look at the test deeper when it is passing.","accounts_in_message":[],"_revision_number":9},{"id":"8f6bde5591700fdc8fd40ccd1bef3e2c194dd553","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:10:50.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"88712b2156d96f73493f8276e1de0ccb5ff6daad","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:26:12.000000000","message":"Patch Set 9:\n\n(7 comments)\n\nFixed a bunch of stuff and rebased, but ... there\u0027s more work to do here, around spares pool handling. Fortunately I\u0027ve already done most of it in my other AZ patch, so I can port it in here soon.","accounts_in_message":[],"_revision_number":9},{"id":"cb1597d513b906160d76e13f1b04c94cc1e1073f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:26:27.000000000","message":"Patch Set 10: Code-Review-1\n\nNeeds to handle spares pool stuff.","accounts_in_message":[],"_revision_number":10},{"id":"c483008649ef20369f921d1946e0e5e246cb3be3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:38:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8f65c506302c2f26c8f7628ba4ea462d2c9e2907","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 06:38:49.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"205740b2451f46b88ce2684f4c473c9251c454c2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 08:10:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bcfd516839d366b190951595a6309c0296c9be0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:49:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89f336c591bb4ff5b4314cc84c1808ad : FAILURE in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83270830ac75496499f71ebd98711831 : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2bf250e0ee9450db3c3fab3230bd4d1 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3fd5fc9289a444d83d94311d01fa28f : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53489396efe34c9991e333eee8fcd9b9 : FAILURE in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/651ba40d6b484128b412e562c2728871 : FAILURE in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076803e8bebe4133bb3d4164b81c7b1d : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f57ce9c3e9aa44c184ae1112a4953e1d : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d8575e971a1481facc977e59a223fe9 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d6739a73fb694e5b81ebaf4ef2a8fae7 : SUCCESS in 7m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fa03f150525e4211a274ce824fd4795d : FAILURE in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98bb613fc0c441a78dce0ab7e49db514 : FAILURE in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f8294c6aa30342f7b98f9bf17c4c5751 : FAILURE in 39m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bf7abf2aea8e4b1e9cc116e871b29762 : FAILURE in 34m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84561218151244c5bd7644ab4ccc42c4 : SUCCESS in 1h 32m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/80e887cea06c4312a4d1bf604ce8c8d7 : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8baf3185d27e4dba990a6228ec4b3db8 : SUCCESS in 2h 11m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d22817453c4e44c5971f5d66e97a341b : FAILURE in 34m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ed836867d8944d8a6bac0055663eb65 : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5755c5ddea54ab2be349703af3a1610 : FAILURE in 1h 03m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/414a3da3ed0144eab2758667cbc042d1 : SUCCESS in 53m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6743f4f90fcc4bbcb06a9ae786a672b9 : SUCCESS in 1h 29m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2bfe1ec8a9848beb4590e1ffd46dff8 : FAILURE in 50m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77c4f009b24c40529504c55b91b5d0ee : FAILURE in 1h 00m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8bdaf8eeeb8a4f30b2f4b44b4e5b3370 : FAILURE in 1h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"076955dad2250f946f7067ba9fad66ac52fe5e21","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-02 00:45:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7cfe24f8313c723d9274e18958c596fd6f43f238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 03:15:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c21f3c42065846eb96bae50208f2d306 : FAILURE in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8bd4f0d72a34fb18469260175d7d851 : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6393d1f1180641969dc5d340d6ff24e9 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9952891a80c540a2bbfc193cb56b0c09 : FAILURE in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f45b33529d57439f871e8759f9252d2d : FAILURE in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b1b754c1aa74c59b0de1caa0d6701b9 : FAILURE in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d709b2c3c16477ca1b53b5b18e7d940 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bea95b58e45e4016811a68f0c6eb707d : SUCCESS in 5m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/49bdbc0acfd947789e39dde81954d1c3 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/08d6097f5aae43fcb32a7254d0669862 : SUCCESS in 6m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a29b287d123d44b19f10f96dcc74e91f : FAILURE in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/262eeb692df14676b611aaed8ddc016d : FAILURE in 4m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f886b2683174d499b337fbc922f0645 : FAILURE in 34m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bfcb7bf8eb4346fea93f3e93a3ad86b8 : FAILURE in 31m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e48d82fa146460b8af8cd72240b4871 : SUCCESS in 1h 30m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/632bf5d38b474c079520c938dfef8a62 : SUCCESS in 1h 27m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f0a6e398d43849ca9499dfacbfce4108 : FAILURE in 2h 26m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/acacfe0779ae48a09364e3375ed802a8 : FAILURE in 32m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/eaa97680cc864543a281f5dbff25c4c6 : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfe4b7cbd88f4ff992407055e25cb917 : FAILURE in 1h 02m 16s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e30c3a34c55941829d6388437ec9f077 : SUCCESS in 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/59721e756921495c930d5b816fe99d19 : SUCCESS in 1h 33m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8bd7a08f80ab4e2bacf881dcba639470 : SUCCESS in 1h 08m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/79ffb623d20d46fba705ccaceaca42ad : SUCCESS in 1h 11m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f41c69a803b24b4ea7f7f9e58f3832d3 : SUCCESS in 1h 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fbea848ff7f54861c8170ba77e405da528f1ffd8","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-04 04:38:13.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"ed816cf29d8c319c7b2234e7c2825642c25ec86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 06:19:37.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3a55a4cba9342dc8b949b64dd6e9511 : FAILURE in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3432621b77504a19817fde54fce2d145 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5ade19e09041ac865739cb730e699c : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e07432a8394912a10b62767cf9736f : FAILURE in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0ac28a862b542b2b3b30eca35fe70db : FAILURE in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad7aaa05de94077bb29b6de72e1bee8 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1a4a120cfb94b79991b32a804700945 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d6dde3087684b91af854e9f0ba0979c : SUCCESS in 4m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a390d9ea5f8c47718efe60542d19b9ae : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa621efd285a4456ab96962f164c0cb6 : FAILURE in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b832cd0078d46958803fc34afecf030 : FAILURE in 29m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27177802a4e94885baaa928710ffe478 : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4f766fa55be4877ace9215d2fdb4cf2 : FAILURE in 28m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0eb881cc7b14180a0b83d0341d168ab : FAILURE in 1h 00m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8b883abe0a1c4046b97113dd4da0c1b7 : SUCCESS in 49m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/369b15b55d554372a648a541068a4a35 : FAILURE in 1h 35m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/901a791909f246d79c1f2d8d853d0446 : SUCCESS in 1h 08m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6930bbad1948bc5df34300a7f7ba978d253719b9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 21:28:16.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"16268ffa0df7e8e92e7cf277242605a41176c870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 22:27:25.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"3c181c3eed512b2aba7853e2c20a8af7f5278794","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-05 05:57:20.000000000","message":"Patch Set 15:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"2f6002463329e60d1876b9796b0b6f869e1b0b2d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-05 10:20:25.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"0a1bf6aa1f15fdf2ca8d31cf18d047a1dfbafb34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 12:31:46.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bee7b84e4a5249b8885c3869a85bc492 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f763c5a19cc44a058679cbf064e58d6a : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75eb995aaf864f40861c15f9b9240343 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f466dda05f4d3482bf8aa3966f360a : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed19e36f09a14e969ff2a190829278ab : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fa1a17a1a040eeb0d00855ce4987bb : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9977a2cc49f476eb5b75ff9b3947271 : SUCCESS in 7m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4c845612e0734023871446f0a80d914d : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/80c1f917198c4f2d879622f286e4bc1a : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b71c04afc0ff41f0ab0f0896e260e8d7 : SUCCESS in 5m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d580af2fd404aafa364df3fa8e7b666 : FAILURE in 38m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5f9ca4e623648f3b9f80854ba773ca4 : SUCCESS in 1h 28m 24s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88654122c98a4325b70b7c2ebb5e1d8c : FAILURE in 11m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7d6a010f3264f31a791d9e583e67edc : FAILURE in 1h 00m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/357e72fac57449ebb4a5cd3125d443ca : SUCCESS in 46m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec6a77ec01094934875f3fc6f4b4314d : FAILURE in 2h 06m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8687c15b8c6747f8aeeb5a6de785d85a : SUCCESS in 1h 04m 41s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6efedb4a5d97ee218a2ec10f9b1fda5a29b9c1f2","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-05 23:29:04.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a695862a3ae8c25f50604688d23e24c5d1eed6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:25:20.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1d3c29d8b134a8183dada31a223fadc : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ccdde2abdaa4c6cbbacbad95855ce8b : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af73076271aa4094b694bccaeb9a888f : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4e3750487b404c984a18f3ff7f0ae0 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be1b1191ac1249b197369b52ec349165 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f069ea3a90954daba2e85f8aa0cfe3e8 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/928c2d91c30f4a32bad9ff1a9ed913b3 : SUCCESS in 5m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3c59f500afaa41be809214626ccc8901 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9bc9b03659e4a77a60d2bed1d936c2c : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92ff89abd0b44b94b421461ee9e44c65 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bddbda04ab294d3cabb299272e4448da : SUCCESS in 48m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82ebaa76415941729ac63dcdaf4f3d3d : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed8def1753ab4a6e8d745c3952d9ef22 : FAILURE in 11m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd61b46e7192446ead4f6857dfdd71fc : FAILURE in 1h 18m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/098ea314f8b947dd9fde8f9c50426a28 : SUCCESS in 1h 07m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b56b5ac6c0f04542aa2cd0bad2a6c2cb : FAILURE in 1h 43m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeac1598353a43a69f5a66f2ee1a40fb : SUCCESS in 1h 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ee2ecf98f8f89a39e3afcbd33385191944175a2d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-06 02:26:57.000000000","message":"Patch Set 16: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"9125314442b0e2d364cf59c367ed7f6f21a5dedd","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-06 02:54:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6db48f5446f8f642cbe9e8b693d5512ef29cb832","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-06 02:57:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"a0b76a7685de8426cb7655ce9f1a1b1423237d27","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-06 03:33:41.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d3b8c58711a0ed99d4d13dde1037273f5879731a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 05:17:03.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1140649d77e24a1aa89420fce9a0f7bf : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cae2dc9edcc24df0bc7a75bfff80a007 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcec22bb699242a6aa08cd98e2568ba4 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252eaf524fdc4564b2549cd1e1fd6a02 : POST_FAILURE in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/283206f06e1d4b42b53721736b9f3c41 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3e5529f236424bb5b25bfcc3368be4 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270ea111380645bc97a97a6ba16ce8a0 : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1c3f218949e54e31ad740ee57b82cf54 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8e2244de2a14d8db706817462dea13e : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e7625976b5047e5b80729ec9c77c43a : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db91292cc0f940a899616b5c06745b7d : SUCCESS in 43m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cf8568cbdff4e6fae49f2f500e85231 : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2094f6c470e44d94880d47e6beff58fb : FAILURE in 12m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/610844ffa5384829b5682fa2ae5cbad4 : FAILURE in 58m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a40676d817dc4eda85cc0dc21df40a24 : SUCCESS in 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bcc25ee7b313492abd02975eb5e4cfe4 : SUCCESS in 1h 36m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fde3ac5267934ccb9a8b566eb14905d2 : SUCCESS in 1h 06m 03s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6241d258919c18070ad69e8e0be40c20e2d51634","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-06 05:36:37.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"0f9f050231a25ec8e7ff8719b203e6117c579502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 07:28:44.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eb90a4b114d44979b05c5710f6ef4a0 : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f2db01b52704172bd3243a3015565c3 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc123c0a3a84ce085ac962718f7e67f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6088ccee75a54c67b4b9c7434494f786 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e92750123164410bb63e4489e7d8291 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40a335fc98e4c96b0050ac4520ef5f8 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a621e9b12aa34f4f98c99229e6fdde40 : SUCCESS in 6m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/af37463e800b4cf29a5a94724bbac135 : SUCCESS in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8cdbf33091324d65a78c13f4b6a04f59 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a736a74998a9433fbbee8d7e6fa0951d : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f163a443b14f4972ba7336824190c9ab : SUCCESS in 51m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79c239ce1fe648d7861f311445f5e317 : FAILURE in 1h 48m 37s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ea25b930b93d4461ae5e30579b71d90a : FAILURE in 8m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4524af87206c4013b23a318f8aeca00b : FAILURE in 1h 07m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9d7d0feac5f442ab8d3e2a2818e74f3 : SUCCESS in 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/78902e637ba9408ab430b1783cc7a224 : SUCCESS in 1h 39m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fafba341871f40e9ae1125dc8a71a248 : SUCCESS in 1h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"fba0a1bfb2b1d5db09ddd63c69ec2002b642cf1d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-06 07:34:00.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"94fdb2807a8dac6b8357ab847478e5302c5b2a99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 09:27:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2afc190b10d4a81ac887434da4cb7de : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3a1f2f85b9144c5a1225af4de917ec8 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f11f298b004a35b96c7541079f9981 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b72bfda9412e4bbcb9a55b117319ad0c : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c62fdae7fbf84d7aa4cffbcf8de97f54 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69404e8a75b64432968bf547c3a370b3 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2307c2cef6241d49ffd0a051eaadfdd : SUCCESS in 6m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0c67643727bc4a0ab0bc4296258075ba : SUCCESS in 5m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/00b36b36ae864d158c399557f847a317 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6f6d27affb34e5aa390c181b2fd8de5 : SUCCESS in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ba5d46bd217c4f278b5c608696ca141f : SUCCESS in 47m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1c62705e2a540e9ba3b7945011a0d01 : SUCCESS in 1h 45m 20s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/046f3fd6bc7f4c849fc2173201ef671f : FAILURE in 11m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0510b0d1db3e448fbde383265e05661a : FAILURE in 1h 08m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c05e2c3f0dd4390ab0882fb02043d07 : SUCCESS in 50m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/313fe103e75842e882c41f2dd88ec907 : FAILURE in 1h 45m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e14a45ae5c9842fb827e3876a95d1672 : SUCCESS in 1h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"9768448de4fc75dc3c1636d73038ca51dc176454","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-06 12:11:33.000000000","message":"Patch Set 19: Code-Review+1\n\n(2 comments)\n\nOk, this looks generally ok. I\u0027m going to try to figure out how the heck to test it in devstack, if I can... Meanwhile, we might want to try adding a test or something in tempest for the spares pool stuff? But ... I don\u0027t know if we\u0027ll have the time for the amp builds if it\u0027s not in two-node :(","accounts_in_message":[],"_revision_number":19},{"id":"fb2b34f983f2c875eee082fe428d795b7c215d82","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-06 12:36:23.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"c3f6e9d7a47cc5af6d9ac05fcdb84d28165e77c1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-06 16:57:27.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f4867a4153e42d034d816880be4f55c69c5fee28","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-08 23:55:57.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"c399ffb577a02fc4c4b94ef45bdcc25311aba84a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-12-08 23:56:15.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"a73337b8593b60404e00dd99ad9a6a1ba3cb7cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 01:49:25.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/948c6dd3387f4aad87cf16c24bd378ef : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8ce89cbedd4fd68448d07821b52233 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70b2df777b6458ab1c1e043c78a55cd : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ca379f13bda4a1a972fd1953eccf979 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2fb8db158cc418e9a5f67beea24e8a1 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516b4aefacb94876a545bb3f1b68045e : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37143593aa6d4d989eb654795394b331 : SUCCESS in 5m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/75ead40d022549338891b36b51984217 : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/da6aeff8975846278358edbff886112e : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3939ec8045524fee9cde4c89bef0b398 : SUCCESS in 5m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6e0761a389b444eaaa29ce0b61f7351 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87482e0ff09d4973918da9677de511c7 : SUCCESS in 1h 34m 51s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec0363c494c64cd3b4ec897cd4ed8ed6 : FAILURE in 11m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9f51747cbf040ab98c23f89eeb597a1 : FAILURE in 1h 02m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86eeaf4bf5db49bd9b4d886ccb94f84d : SUCCESS in 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/650e959abb984aa8b4c77cc7bd13979d : FAILURE in 1h 42m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9a5d41a35594713ab85f1b58f6d7946 : SUCCESS in 1h 02m 40s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d9586040dacc6f73a1fe1165cc45a6374979fbd2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-13 20:40:18.000000000","message":"Patch Set 20: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":20},{"id":"39bc2b04c0018dc95ca4ceed8b6374b3c90463a2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-13 22:22:52.000000000","message":"Patch Set 20: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":20},{"id":"20da998bc56bbe605f503fc2522deb6bcdd6210d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-13 22:42:31.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"eac8392a7c47fbe1989c897670736688f493695b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-13 23:48:58.000000000","message":"Patch Set 20:\n\n(7 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a9565ce463a1c89bb4731aa5603eca8bb0d0eff1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-14 00:13:25.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"00b937bbdffd1a743fb7e4708d134a98e51bd5fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 02:29:34.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/889e683539504fff9f6b651b32a3c921 : FAILURE in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7474b25695214d4f970fa3e2ba60c389 : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f5fb3b906b427284536fc8636ee994 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e640828a5fa84872a87bd2a0cd5a6a7d : FAILURE in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/224ee220a08242359145814260dad35a : FAILURE in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d3923ddf264426b6bf5649bb5a9cfe : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d2995c0a29b4f06b8663d4853ac7efb : SUCCESS in 9m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a754ce61468c4d1e8306e304b7b1c979 : FAILURE in 7m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c758b642fd6c4715942d95aaf72aa0f6 : SUCCESS in 14m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/64cb644af4e04d32850d3cb5a0f94ae3 : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5cf7f7a354614ab4be97f328a4006c0f : SUCCESS in 55m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59f8c727f0854780b47474d1721bc3ee : SUCCESS in 2h 06m 09s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4da3bc183c72417aad47600e38692a51 : FAILURE in 9m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3a315d7bd9674ccf8226e8456179077a : FAILURE in 1h 12m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/896e21526394496ba8870d7cb56572f9 : SUCCESS in 1h 01m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b72794d19bc84201b6821cc2c7655ca4 : FAILURE in 1h 49m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d65a6d1145244a179dabe5c1262408da : SUCCESS in 1h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"815ac4539a117af151653600e0da097e70a6c44f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-17 00:26:56.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e039a9f2badbe9a560dadde1843da1f42aa6a68b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 02:02:05.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1a9e697e8b24895899a17860d9bebf2 : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a20434ba6fc4509895fcb39e499a541 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14de757a5f9d41bd844058e453abaea1 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1af9de6220be4f79b25ce948b19cc472 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6532da7f44714d1694c8eda9af9fb2d9 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8c42fd44ce49508949dbc09c591b7b : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24204ed741344009a0e4da882f336f81 : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a5a3179b295467e89ddb24e32ca4cc7 : SUCCESS in 5m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/50470d72aab04fbebd832c24e075b33a : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe997d36c6824645af6e58312c5274d0 : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/434559ae9e8f4668895b9b49e2d6efb9 : SUCCESS in 44m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8559295c38d140fc8aa97d30d7b3294e : SUCCESS in 1h 30m 02s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fb184a90d634dfcbbe3162152edd9b1 : FAILURE in 8m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ba2bd4e2076460b8fcd4fb0d25deff4 : FAILURE in 59m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0becc03130fe47aa958added7788bff9 : SUCCESS in 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bbb9df3d86ed43ab9f98a9cb9a33d34a : SUCCESS in 1h 33m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4016deaa2a1478db9a6e041d30be336 : SUCCESS in 1h 02m 07s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"209289b937220e5cb822af76d4ef88bcad4c5b38","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-17 14:08:40.000000000","message":"Patch Set 22: Code-Review+2\n\n(1 comment)\n\nAt this point I think all of my concerns are handled for an initial merge.","accounts_in_message":[],"_revision_number":22},{"id":"fcc4c5870ab148caf22e9a0e2ae4a6be511ef3ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-17 18:11:19.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"591c3c8e5a375dc853cad08d03c955c3c4e210c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-17 21:17:03.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\nLGTM and the spares pool test passes.","accounts_in_message":[],"_revision_number":22},{"id":"a8e1a1d94f3d61a08bf1fa3acfee48af558d30fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:17:18.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"00668d2c5fa9e493c51fd25a49f30859bc839880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:49:08.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/957738c1f0b941d787b8b0c7eb5f3d4f : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945026a3713c478ca8d55573e360c4cd : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0d03d4addc49899e4c60829b8c7be6 : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5df237f2a5ac477893eb6ebb817f5bf7 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026cad0e5995468fa347fa7ae075f6c5 : SUCCESS in 13m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ecb933fa9074db5952fae6be3166524 : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce2988b8bc4848ed86d3d84636995cd8 : SUCCESS in 10m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/857fd04163894d229997b1ce7091ff53 : SUCCESS in 51m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fab406498aa4c25bf5cba7f582c93ff : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":22},{"id":"0c4909ffde16e98101dbfd4696260c3246c1f82a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:49:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"e2ad4d1515a1859c0333c555309a10c9347ff57a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 22:52:42.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62aaa18087e4351bb45786534a9363a : SUCCESS in 2m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e00c1a0064e249678b696ab409c43af7 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"3cce347129fe51c219d9630a96b568633eef0098","revisions":{"61dbc7c1f69866f3d414047bd6f3c8e2ca6d2bcd":{"kind":"REWORK","_number":1,"created":"2019-11-12 02:37:30.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/1"}}},"commit":{"parents":[{"commit":"51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328","subject":"Merge \"Fix typo in doc agent.py-\u003eagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51425dcda6fd4ea9f06c3b93a2b3fc6f47ae2328"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:37:27.000000000","tz":660},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61dbc7c1f69866f3d414047bd6f3c8e2ca6d2bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61dbc7c1f69866f3d414047bd6f3c8e2ca6d2bcd"}]},"branch":"refs/heads/master"},"fd62612afd57f1308c9936ba278e47d9b7ab1cd1":{"kind":"REWORK","_number":2,"created":"2019-11-15 02:22:43.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/2"}}},"commit":{"parents":[{"commit":"aa670584436c73f21c0ba216609f239fd0c3d772","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa670584436c73f21c0ba216609f239fd0c3d772"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-15 02:21:25.000000000","tz":660},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd62612afd57f1308c9936ba278e47d9b7ab1cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd62612afd57f1308c9936ba278e47d9b7ab1cd1"}]},"branch":"refs/heads/master"},"ecc199b83cd62d1fbb4dd1bce09081a5dcf2340e":{"kind":"REWORK","_number":3,"created":"2019-11-15 05:50:35.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/3"}}},"commit":{"parents":[{"commit":"aa670584436c73f21c0ba216609f239fd0c3d772","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa670584436c73f21c0ba216609f239fd0c3d772"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-15 05:49:18.000000000","tz":660},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecc199b83cd62d1fbb4dd1bce09081a5dcf2340e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecc199b83cd62d1fbb4dd1bce09081a5dcf2340e"}]},"branch":"refs/heads/master"},"623f206770fcab5137f17d065644428fe560c6c8":{"kind":"REWORK","_number":4,"created":"2019-11-18 01:56:50.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/4"}}},"commit":{"parents":[{"commit":"56d2dd925dda2f1164af4322df9fe29f500ac78e","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56d2dd925dda2f1164af4322df9fe29f500ac78e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-18 01:56:06.000000000","tz":660},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/623f206770fcab5137f17d065644428fe560c6c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/623f206770fcab5137f17d065644428fe560c6c8"}]},"branch":"refs/heads/master"},"2c28864c0ab3ffe49381cabd59ed6490ecccad88":{"kind":"REWORK","_number":5,"created":"2019-11-18 15:13:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/5"}}},"commit":{"parents":[{"commit":"143935db129adbcd8debd28e7f603994b4daf4b7","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/143935db129adbcd8debd28e7f603994b4daf4b7"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 15:12:57.000000000","tz":-480},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c28864c0ab3ffe49381cabd59ed6490ecccad88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c28864c0ab3ffe49381cabd59ed6490ecccad88"}]},"branch":"refs/heads/master"},"e5c55a54d56171926784ebef27203e064320764c":{"kind":"REWORK","_number":6,"created":"2019-11-18 15:33:33.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/6"}}},"commit":{"parents":[{"commit":"2c71eb42c46c00df2a51ea285aa28eaefdfd8761","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c71eb42c46c00df2a51ea285aa28eaefdfd8761"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 15:31:00.000000000","tz":-480},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c55a54d56171926784ebef27203e064320764c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5c55a54d56171926784ebef27203e064320764c"}]},"branch":"refs/heads/master"},"0b82a0bcc55ee8f1951c268fa5edd86b31af5815":{"kind":"REWORK","_number":7,"created":"2019-11-18 20:20:15.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/7"}}},"commit":{"parents":[{"commit":"2c71eb42c46c00df2a51ea285aa28eaefdfd8761","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c71eb42c46c00df2a51ea285aa28eaefdfd8761"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 20:19:32.000000000","tz":-480},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b82a0bcc55ee8f1951c268fa5edd86b31af5815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b82a0bcc55ee8f1951c268fa5edd86b31af5815"}]},"branch":"refs/heads/master"},"d7991e0b93146f4ea419aabf0f42345aedac3afd":{"kind":"REWORK","_number":8,"created":"2019-11-18 21:37:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/8"}}},"commit":{"parents":[{"commit":"2c71eb42c46c00df2a51ea285aa28eaefdfd8761","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c71eb42c46c00df2a51ea285aa28eaefdfd8761"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-18 21:36:58.000000000","tz":-480},"subject":"WIP Support creating a LB in a specified availability-zone","message":"WIP Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7991e0b93146f4ea419aabf0f42345aedac3afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d7991e0b93146f4ea419aabf0f42345aedac3afd"}]},"branch":"refs/heads/master"},"cf67143c24c6e32e4b11fd36a18a0bc5dae22827":{"kind":"REWORK","_number":9,"created":"2019-11-18 22:35:21.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/9"}}},"commit":{"parents":[{"commit":"2c71eb42c46c00df2a51ea285aa28eaefdfd8761","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c71eb42c46c00df2a51ea285aa28eaefdfd8761"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-18 22:34:06.000000000","tz":660},"subject":"Support creating a LB in a specified availability-zone","message":"Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf67143c24c6e32e4b11fd36a18a0bc5dae22827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf67143c24c6e32e4b11fd36a18a0bc5dae22827"}]},"branch":"refs/heads/master"},"520d5963f8f42cdd3e1fe4d344d22d38d66374c6":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-22 06:10:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/10"}}},"commit":{"parents":[{"commit":"ad34b788eb54bf9b58c2995cdda3c5895cd168c5","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad34b788eb54bf9b58c2995cdda3c5895cd168c5"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-22 06:10:50.000000000","tz":0},"subject":"Support creating a LB in a specified availability-zone","message":"Support creating a LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520d5963f8f42cdd3e1fe4d344d22d38d66374c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520d5963f8f42cdd3e1fe4d344d22d38d66374c6"}]},"branch":"refs/heads/master"},"679c4381342e4d5396070f5d526241d0a658232e":{"kind":"REWORK","_number":11,"created":"2019-11-22 06:38:49.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/11"}}},"commit":{"parents":[{"commit":"ad34b788eb54bf9b58c2995cdda3c5895cd168c5","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad34b788eb54bf9b58c2995cdda3c5895cd168c5"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-22 06:37:42.000000000","tz":-480},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/679c4381342e4d5396070f5d526241d0a658232e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/679c4381342e4d5396070f5d526241d0a658232e"}]},"branch":"refs/heads/master"},"eff91c08b8112151b4f3553692109f748ec4cfbf":{"kind":"REWORK","_number":12,"created":"2019-11-22 08:10:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/12"}}},"commit":{"parents":[{"commit":"ad34b788eb54bf9b58c2995cdda3c5895cd168c5","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ad34b788eb54bf9b58c2995cdda3c5895cd168c5"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-22 08:09:59.000000000","tz":-480},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff91c08b8112151b4f3553692109f748ec4cfbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff91c08b8112151b4f3553692109f748ec4cfbf"}]},"branch":"refs/heads/master"},"7b18705cc7720c497cc01e9a2679ed4bc5387a96":{"kind":"REWORK","_number":13,"created":"2019-12-02 00:45:31.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/13"}}},"commit":{"parents":[{"commit":"8ae6bc3697434aa2c8676afbff3d4938676a35a5","subject":"Availability Zone admin API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae6bc3697434aa2c8676afbff3d4938676a35a5"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-02 00:40:21.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b18705cc7720c497cc01e9a2679ed4bc5387a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b18705cc7720c497cc01e9a2679ed4bc5387a96"}]},"branch":"refs/heads/master"},"5c6505e22ac0cb4de1ba62ae529d4ff2922e5fc0":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-12-04 04:38:13.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/14"}}},"commit":{"parents":[{"commit":"d6dc2b9a3e673100ee42af79d7a4af1f65bbd9ce","subject":"Transition member flows to use dicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6dc2b9a3e673100ee42af79d7a4af1f65bbd9ce"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-04 04:38:13.000000000","tz":0},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c6505e22ac0cb4de1ba62ae529d4ff2922e5fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c6505e22ac0cb4de1ba62ae529d4ff2922e5fc0"}]},"branch":"refs/heads/master"},"08014d9c930b91e659cd1f7986af8988135f9534":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-12-04 21:28:16.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/15"}}},"commit":{"parents":[{"commit":"d115644d3eacbe472c98440148a0231006a0a94e","subject":"Adjust for new octavia-lib LB signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d115644d3eacbe472c98440148a0231006a0a94e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-04 21:28:16.000000000","tz":0},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nDepends-On: https://review.opendev.org/#/c/694707/\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08014d9c930b91e659cd1f7986af8988135f9534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08014d9c930b91e659cd1f7986af8988135f9534"}]},"branch":"refs/heads/master"},"c39a1016b41a9ebef5534988e2dad3df6299b3e1":{"kind":"REWORK","_number":16,"created":"2019-12-05 23:29:04.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/16"}}},"commit":{"parents":[{"commit":"d115644d3eacbe472c98440148a0231006a0a94e","subject":"Adjust for new octavia-lib LB signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d115644d3eacbe472c98440148a0231006a0a94e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-05 23:28:50.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c39a1016b41a9ebef5534988e2dad3df6299b3e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c39a1016b41a9ebef5534988e2dad3df6299b3e1"}]},"branch":"refs/heads/master"},"1aa412c5c3eae897a06baaaea8067642c5f2169f":{"kind":"REWORK","_number":17,"created":"2019-12-06 02:54:31.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/17"}}},"commit":{"parents":[{"commit":"d115644d3eacbe472c98440148a0231006a0a94e","subject":"Adjust for new octavia-lib LB signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d115644d3eacbe472c98440148a0231006a0a94e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-06 02:54:19.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aa412c5c3eae897a06baaaea8067642c5f2169f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aa412c5c3eae897a06baaaea8067642c5f2169f"}]},"branch":"refs/heads/master"},"a4576a69f90a3dbafbed4fb11b0b953ce51949c5":{"kind":"REWORK","_number":18,"created":"2019-12-06 02:57:43.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/18"}}},"commit":{"parents":[{"commit":"d115644d3eacbe472c98440148a0231006a0a94e","subject":"Adjust for new octavia-lib LB signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d115644d3eacbe472c98440148a0231006a0a94e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-06 02:57:27.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4576a69f90a3dbafbed4fb11b0b953ce51949c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4576a69f90a3dbafbed4fb11b0b953ce51949c5"}]},"branch":"refs/heads/master"},"5bacc06bbb2351caee55c109eeb6e850a680114b":{"kind":"REWORK","_number":19,"created":"2019-12-06 03:33:41.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/19"}}},"commit":{"parents":[{"commit":"d115644d3eacbe472c98440148a0231006a0a94e","subject":"Adjust for new octavia-lib LB signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d115644d3eacbe472c98440148a0231006a0a94e"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-06 03:33:30.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bacc06bbb2351caee55c109eeb6e850a680114b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bacc06bbb2351caee55c109eeb6e850a680114b"}]},"branch":"refs/heads/master"},"464ae2531a50f56fe361ede8fd5b319436ef2eff":{"kind":"REWORK","_number":20,"created":"2019-12-08 23:56:15.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/62/693762/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/20"}}},"commit":{"parents":[{"commit":"3740b67854bd9bbff791b215273e2e5fd7a75675","subject":"Add support for CentOS 8 amphora images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3740b67854bd9bbff791b215273e2e5fd7a75675"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-12-08 23:56:03.000000000","tz":660},"subject":"Support creating an LB in a specified availability-zone","message":"Support creating an LB in a specified availability-zone\n\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/464ae2531a50f56fe361ede8fd5b319436ef2eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/464ae2531a50f56fe361ede8fd5b319436ef2eff"}]},"branch":"refs/heads/master"},"89c6ec39d3a17ea75386b31aa8a094da2f6dd421":{"kind":"REWORK","_number":21,"created":"2019-12-14 00:13:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/21"}}},"commit":{"parents":[{"commit":"45cb1e4e161239ef93adaf545dd6db2c537c69e3","subject":"Merge \"Update ListenersUpdate for lb/listener dicts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cb1e4e161239ef93adaf545dd6db2c537c69e3"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-14 00:12:20.000000000","tz":-480},"subject":"Support creating an LB in a specified AZ","message":"Support creating an LB in a specified AZ\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89c6ec39d3a17ea75386b31aa8a094da2f6dd421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89c6ec39d3a17ea75386b31aa8a094da2f6dd421"}]},"branch":"refs/heads/master"},"3cce347129fe51c219d9630a96b568633eef0098":{"kind":"REWORK","_number":22,"created":"2019-12-17 00:26:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/62/693762/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/693762/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/693762/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/693762/22"}}},"commit":{"parents":[{"commit":"45cb1e4e161239ef93adaf545dd6db2c537c69e3","subject":"Merge \"Update ListenersUpdate for lb/listener dicts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cb1e4e161239ef93adaf545dd6db2c537c69e3"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-11-12 02:36:42.000000000","tz":660},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-17 00:25:25.000000000","tz":-480},"subject":"Support creating an LB in a specified AZ","message":"Support creating an LB in a specified AZ\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I55d6c1a0b3e6060d6dacc13ee67d87f0219ef7de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3cce347129fe51c219d9630a96b568633eef0098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3cce347129fe51c219d9630a96b568633eef0098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
