)]}'
{"id":"openstack%2Foctavia~694663","triplet_id":"openstack%2Foctavia~master~I723e025e9bfbfe71e4804de13e5625769827f7cd","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I723e025e9bfbfe71e4804de13e5625769827f7cd","subject":"Fix two bugs with in-use flavor deletion","status":"ABANDONED","created":"2019-11-16 08:07:53.000000000","updated":"2019-11-17 20:44:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"20df08e31b4afc755eba54f37c5cfecccfeef8f4","_number":694663,"virtual_id_number":694663,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-16 10:23:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-16 10:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec8877f420a8eec2b18d14d199dd79a9539232f9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-16 08:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"340d45ed5ac7d4c108bd4c092d2dd4e1020de6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-16 10:23:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1be6b0755aa44aae9b237dbf899a926a : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97b4b33ed70f4e4198f643a6c9ec3512 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8052d05b2aa44ff1af0013903be87daa : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dcf783589db4232867e1d9e18360818 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc979c3a32ae4bab950b975207cd02bf : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96acb0db2e904a6d8912f63b95f1c57f : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeaeceb74a514367ab89b48e73793e21 : SUCCESS in 10m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4d080f41a8ea4cc7a4157fabe9a0862e : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e3bff437531548c483b2169277a7d91b : SUCCESS in 6m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/751ab95ddd8547dea6df016676106f4e : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a5025089d454b5fb7896b6ac9ee24c1 : SUCCESS in 5m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76c98bde6d294862ace36f895d8d72d3 : SUCCESS in 42m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/03fa5d7309bd4fbaaaec290606899d8c : SUCCESS in 42m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efd897e7ab29473aa049cdd0cfbf8bf0 : FAILURE in 1h 41m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9c66895cb45d4c758b111d7a216cb142 : SUCCESS in 1h 35m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3fa42fae3b0a4069bb600fc91355c0d5 : SUCCESS in 2h 10m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aeaa2df3015f4c8cad3ab54ce5eeaae6 : FAILURE in 36m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb7de3ef714e4f8080c2cc94248f4786 : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d49d6c1879a4cf5a0968602be8e1b12 : SUCCESS in 1h 16m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25df05ca4c0541feb7d352343fce62a5 : SUCCESS in 48m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c398530968cb4eb9917123d225b84b6f : SUCCESS in 1h 45m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a747dfeb2f9465ca8d47244874881fc : SUCCESS in 1h 02m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5a34aee944bf4b479c4fd128a698c217 : SUCCESS in 1h 00m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/43557e0240494314b6257bd82078559d : FAILURE in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20df08e31b4afc755eba54f37c5cfecccfeef8f4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-17 20:44:11.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/692427","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d493e4412ffefd85703d08458985bac605dc81f","revisions":{"4d493e4412ffefd85703d08458985bac605dc81f":{"kind":"REWORK","_number":1,"created":"2019-11-16 08:07:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/63/694663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/694663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/694663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/694663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/694663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/694663/1"}}},"commit":{"parents":[{"commit":"057354ab7d672179f2c21662b78c93a98a249fc3","subject":"Stop allowing the deletion of an in-use flavor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/057354ab7d672179f2c21662b78c93a98a249fc3"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-16 07:56:23.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-16 07:56:23.000000000","tz":-480},"subject":"Fix two bugs with in-use flavor deletion","message":"Fix two bugs with in-use flavor deletion\n\nChange-Id: I723e025e9bfbfe71e4804de13e5625769827f7cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d493e4412ffefd85703d08458985bac605dc81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d493e4412ffefd85703d08458985bac605dc81f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
