)]}'
{"id":"openstack%2Foctavia~694823","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I8e449b19af72d72f6effd52e0a2debb5754a19b3","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I8e449b19af72d72f6effd52e0a2debb5754a19b3","subject":"Fix a potential race condition with certs-ramfs","status":"MERGED","created":"2019-11-18 18:17:29.000000000","updated":"2019-11-27 02:04:58.000000000","submitted":"2019-11-27 02:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694823-1574820201537-eeb6c3ec","meta_rev_id":"f1308d6660b95366b5fe1ae30b887fa960f26c7b","_number":694823,"virtual_id_number":694823,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2019-11-26 22:02:03.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 02:03:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-22 21:18:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-11-25 08:25:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-25 08:25:16.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 21:18:31.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-25 08:25:16.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-11-27 02:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab3318779766bae5aee78642f5dc676fafb459a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c03bc4c07e52d2ae37743c7cf68fe90dcbf4567","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:40.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c3f95915d2b281893bd321efadb4a93c440822b9","accounts_in_message":[],"_revision_number":1},{"id":"612221092638f44c8b557cc59d42f9243f20d70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:59:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af29b829a10c47a392bc075b5922afbf : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42e8f9356d8c4096b2d8dcac9eb3e5f1 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266be361914948528c33ac1720ce486d : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2b213c569614919ba329d7230368576 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf29c19d39b482dba176fb0d86160ea : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/980df1f07e144e9eaf5552f2a0b02ac0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473cb3b6850c45f98266d3ba550bcbf2 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11df8a9449a4184bba90b887391bbfa : SUCCESS in 6m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/79e2696d303948edbbd55298d93071c9 : SUCCESS in 4m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ced58e51068e41b0904c0ccdc8bf61e3 : SUCCESS in 6m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7061203794c144788d1380b8cd13ca66 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0c02feb603d4f7aa93216d675d31d04 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65b83e053f6b4beca0908bbafa201104 : SUCCESS in 50m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f2ef47d43fd847e29da8bd039f10474c : SUCCESS in 48m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d81846ff5be4c83b64a289d08686616 : SUCCESS in 1h 35m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6dac2297fd1247b8b3b619eb3fcc98bd : SUCCESS in 1h 41m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b12657974f18484faf7a226003d44ef7 : SUCCESS in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be0519787593420fa151d50df382f7ff : SUCCESS in 1h 23m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0eb96a6007c4d18bcd7c9eaf72988fc : SUCCESS in 2h 21m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a20d74231c54eebbc3e6e356407b093 : SUCCESS in 1h 20m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eed06c4ae9144199a5d879a17a481aec : SUCCESS in 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6741c14ec9594ed0a0d408ce1e04586e : SUCCESS in 1h 48m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5da3b9a83bf8454db17d4bb8bedee029 : SUCCESS in 1h 08m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9e01fb4f728e447692feb154352d1f07 : SUCCESS in 1h 11m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0b646ef49f8342dc8fef9e6bad1d79b2 : FAILURE in 1h 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32bb46c10fd5b32230ebfb8f32de9d623a9f1a14","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:18:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51f8767529251ef8c1a07b12109ac8a1f8b34009","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-11-25 08:25:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4870664d7cb57cdba34ce5936c429ad65382e52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:25:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f05bcf7b45182cbfa8fd80cc8522aff8b693710b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:58:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab268ae77931431eb28d4a9e0e6a4318 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c806d11e06bb430889332aba6e2b6d5f : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f796163132a945b1a1468755034f892e : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3663097c6b4eeb9439b203e43c85c0 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d23f43385e84c26907e638749a7f2f4 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547ef0aea1d14df38c8cdafb6e556052 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5769388a3075487fa68a438746333708 : SUCCESS in 7m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4adfa2bdd2354ffa8b228274a391008d : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a78335bb7869457d918f163f6f0fa9c1 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edd4a1fb4dc44926aca1d50dcd47dcd1 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7877d53ec5c840098b2096398ba617aa : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e01593ae66604809a23d6fc93582b0c4 : FAILURE in 31m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/56c39fcdbe8340f7ba643ea9adb03c8c : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24dca46c31d7422eb7aaba0158d33862 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"09d224599d294f12c856d38cbcbd5ed5c3ce7da7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 22:02:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e05577fda09dc46d68a5fc8e47743b9edad59ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 00:17:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96bc8f7a20d24b7684cee230f40297df : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1dcc00492cb4f00aaa17a6b4cff9749 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e620605efa364d529a5a6fee73e5750d : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27d0d3e2b5d645c48f9d54dfaf917606 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ceab0e72d14050bb5e81908c46e7c7 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1abe0c8270f14c8a87177179824d848e : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f456ec40344d729150ccfab5794734 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c3d0a2137f94fd3a242b3fd26ed47dd : SUCCESS in 8m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1dad4c43cd4d433589112868dc9f21a3 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/718345fe7ed34e539ef9ef57d084250e : SUCCESS in 9m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9a6f9f807c424028bfe37fc2cd7e6419 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a23c2de5e6d14da299ebee8301712a59 : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f36bbd7904b434d82442d65fb9c7ca9 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d1c3469d8bd14bd5b4113e40897feb30 : SUCCESS in 46m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b99498f4ad484560a2b86ff66f3c008a : SUCCESS in 1h 36m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f37f612b47ac416f8e762f3420a246aa : SUCCESS in 1h 33m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/83f9c7a8b26346bd98a02b53e07e262b : SUCCESS in 2h 09m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25b34c61a66e4abf80a4eaa88a54d77f : SUCCESS in 1h 58m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e376a94c78842fca7e85fffa9074fae : SUCCESS in 1h 47m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28df9fd861b043f784062c61e8cf16e7 : SUCCESS in 1h 12m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6e6b495ba5c4895a586767a5f81ab51 : SUCCESS in 59m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a8c808c148440a6b9c6752e34ce6901 : SUCCESS in 1h 36m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/056797f5eb58424392bbffa1e48404d2 : SUCCESS in 1h 13m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3c4c4e15bf6046cfa1578104fbd12ca0 : SUCCESS in 1h 04m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/592f5f1017d1451bbad8dc4c9645ef32 : SUCCESS in 1h 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"889f2088893a5b98eaa324b4852a531e91276207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 00:17:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eccbcd2da75328b2d1be00386bf9348b9d9438b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:03:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa740f736de546d4a83e4e4297cc0e14 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a933c9477db3497795c9cbd93aa5a718 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd0e95b2006b456db7d15d7a97d38666 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bd3ed6c1e3a4d55be7f5dd23ebb09b4 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ceea623e44c4706afd25c0103040da9 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3ff761e07d4c94a791110c4511155e : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6985f4b17e84f28a98ec464382babe8 : SUCCESS in 5m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/576afb19a021416e9aa77fe4045f1585 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93ad85f2ebc7429cb70c8b700561e189 : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9020c3849a764df4929ab021ccb2312e : SUCCESS in 41m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8b588dd611f0478aaa01c5076db50d88 : SUCCESS in 42m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a2ae358732f4b6aadaaeee14b77215d : SUCCESS in 1h 41m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bca335e6c7224a4984965ed74b05d146 : SUCCESS in 1h 36m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b11a465dd3984874ae8cb2f0ec1373ab : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"58708f2f26f216bc027d52ba48b58a663f944f3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:03:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1308d6660b95366b5fe1ae30b887fa960f26c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:04:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7936797c4fbf45a19c4a8bb91f3a949e : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838d3b3c7b684478a50d51d2de962b7e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c1ab033326559f27a489bade083a14164908e2e","revisions":{"7c1ab033326559f27a489bade083a14164908e2e":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:17:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/694823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/694823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/694823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/694823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/694823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/694823/1"}}},"commit":{"parents":[{"commit":"ce22677502c36245f695b734f09ddd682be67faa","subject":"Merge \"Fix issues with unavailable secrets\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce22677502c36245f695b734f09ddd682be67faa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-18 18:17:29.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c1ab033326559f27a489bade083a14164908e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c1ab033326559f27a489bade083a14164908e2e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
