)]}'
{"id":"openstack%2Foctavia~694824","triplet_id":"openstack%2Foctavia~stable%2Fstein~I8e449b19af72d72f6effd52e0a2debb5754a19b3","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I8e449b19af72d72f6effd52e0a2debb5754a19b3","subject":"Fix a potential race condition with certs-ramfs","status":"MERGED","created":"2019-11-18 18:17:40.000000000","updated":"2019-11-27 14:57:39.000000000","submitted":"2019-11-27 14:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694824-1574866477030-2b47506d","meta_rev_id":"6be925cb10246a971f6283b2f8286c2780203bae","_number":694824,"virtual_id_number":694824,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-11-20 00:11:33.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 14:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 12:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-22 21:18:39.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-27 12:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 21:18:39.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-27 12:19:39.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 14:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cec95e341f98ebb661c49dcb1e6a68c041f74fb9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"873dc2fda79b1f893499aae6b01940f464ee24a2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit d48427281867bd62077a65ce092c1645b3d44db9","accounts_in_message":[],"_revision_number":1},{"id":"2a0a5df7e7465a87c0fe05379fab96bfbe3cb966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b916a6b6772c46e1bfe7f202f40bd2c5 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed19691dcee49d68526519f0e9e156c : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e13eb687974868907fc5b3903dc76e : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59d6510106d046a386b1fca8f4ff8993 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c939bc02c467481d8cb931245cb738aa : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aabb917d56354610a0b2f41925c0e390 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0211482e8a94343847d5f1258aecb3f : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eddf1b82dfa437bbd57bd0f30537c62 : SUCCESS in 5m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/56d832e47baa40c9a023f48db5656ba0 : SUCCESS in 1h 20m 26s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0189a5b0888d4f809082bf2313ae81a3 : SUCCESS in 1h 30m 15s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/847278479b4a4f1485b5ae946afa2974 : SUCCESS in 1h 33m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/eac36a152b884e3891e210cdf3722563 : SUCCESS in 1h 41m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c7ab2c7d4b724b19acf487d6abfda9bf : SUCCESS in 6m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0cd0c5da32fd403aa146be0a05aa6ee7 : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/997757e6c1f04a138f41512f6bd8b0cd : SUCCESS in 6m 36s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8fd304993e7419e8bdbd0142b5c451b : SUCCESS in 1h 08m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e2b2014c0dca48feb237afdd6e6b81a3 : FAILURE in 58m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a2a47a984dc42888a62d8fe625471c7 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5e6e7247075e4443b1b88d39f36ab1e7 : SUCCESS in 42m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8ef831a44b1420f856dfbc879dac17c : SUCCESS in 1h 57m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9c9511f691874bc4a1f1740971caaf9a : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3a6fa5988ea44564b61961d86c097814 : TIMED_OUT in 2h 35m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6ffa7ceb825942a4b1e5034e8b44988e : SUCCESS in 2h 12m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ee4b3ffa43443dfae0e01dc8683f2c6 : FAILURE in 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9df115157ede40ed92e4ecd27d600aff : SUCCESS in 1h 45m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/39d61a5e00a94c70853ab80c9844e2b5 : SUCCESS in 1h 47m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/283800a0b1784e26b1b754aa21d29681 : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9f7766b263d4e988b56d696f7d71d19 : SUCCESS in 52m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"789cf400d9b3434fee0c4bb8daafa78fcc82bf6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-20 00:11:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"659cbda5bee2c25b88f1211a02f1fa191ef36bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 02:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fd7ed185e314d678f8bb3c89247cadf : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d0b0317b7454c968d9fba67a0556682 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619ae1e872424d4f8341ed7e04c68c6c : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/173e63988bc146d8beb449e88dac807c : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6afeca6b77e94310957de1f5bc4d155d : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21ca6aee4b904e30a5e88901563327e6 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f270fc81bd96487797789cc16cca3183 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34dee7e3b52047039d67050a3efe8c5e : SUCCESS in 6m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bbbcc5e617354e42af89a6c53f7c04c9 : SUCCESS in 1h 27m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/08a760d1c69c457c87dea467fb9aa8a4 : SUCCESS in 1h 17m 43s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51ce299502fd45559211c033b25bfaf4 : SUCCESS in 1h 41m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6d5b052aedfb41e78e5382388cbfa113 : SUCCESS in 1h 27m 10s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/70c23aa1258f4b46bc0ab477e1a242c7 : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0fd9919f1c604bd78613c7b53246e954 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de7d56daa6a44d628cb3051c20407751 : SUCCESS in 6m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/566fc86d2531464d83eb874496fe6bea : SUCCESS in 1h 11m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e147db6b539b430ba4f04b9edd6fb00d : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82128623255647d38f9b32e6b4936320 : SUCCESS in 46m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/348ee3b2059743919870f459f876f920 : SUCCESS in 39m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c483c48764624f5b9971b1033bc26e53 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5674bc5737574a07b1a5672f967df37a : SUCCESS in 1h 24m 11s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/59ccdf76ca3548f9b4fa0e9467c28c5e : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4eaa24a8ab934fba9872abf532a4cc25 : FAILURE in 2h 30m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6eacc19a4c15466581a0a36a96ef4d40 : FAILURE in 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db9153aacd8f4924a9aa31c8a7bd6583 : SUCCESS in 2h 00m 34s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/143188d0ed264127927fc23ff09efe58 : SUCCESS in 1h 43m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ed7db430171f4d46b45817e1a0418632 : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a495a6e2097e4f048185545af1a4f714 : SUCCESS in 53m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb0c08a000c76cf4a6c7b944a957b1d5ba9bed60","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:18:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82c0d5eec3fbd7847d16d7bec0036071fccf182a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 12:19:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7aa6c115d7831987c3daa44bd90bbb1562bd58d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:19:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87b44fc77ad7ee02c3ba5b52dda33ed023967cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:54:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d696850cea3f4f23a72bb02771cae3c9 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0160aeb4e58f40e09502c9fd89377b00 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fea4dd9f75d148d8a49c2369850fda3c : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/664b2e98aae947239b22d1e3ad97d7fd : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54cf90b98a174280a373c83e2073e547 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5984775e8e75438b8cbcd0dbb06eb5c4 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af3be733c30940048209ea4a15ddc839 : SUCCESS in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/676825532f7142cd9fbba569bd3a511e : SUCCESS in 5m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/084b8dd4372642bea82c383da368215c : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/152657948b0246adbccfd1954cca5d8f : SUCCESS in 6m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb11e28b0049478c912966b261fd07e6 : SUCCESS in 1h 09m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d87830f2f9654e7293c5d8fa5d405eff : SUCCESS in 1h 06m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8a56a0f3e537492c9d403ed9df6c707b : SUCCESS in 1h 22m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/71e47e3dd7ec40188f97bf88de1da1e6 : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e1e102d7ff194c16b893022e9fc163b9 : SUCCESS in 42m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/417d8128744b4246a6904fe32863d934 : SUCCESS in 43m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a91491d422f4930b67ebabb98e2a09a : SUCCESS in 1h 33m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eab9ffe80040494dafa28d8cc337113f : SUCCESS in 1h 33m 18s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/482836ac95284b679ac482af9a277ac8 : SUCCESS in 1h 37m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/960032c4abd44bf2a318239368444ee6 : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"148470912ede00b3734283b97270d67461b25533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:54:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6be925cb10246a971f6283b2f8286c2780203bae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:57:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fd23bc799d48558f826fe73b524df6 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d62b2993436421997b14cb0a70a5465 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3f95915d2b281893bd321efadb4a93c440822b9","revisions":{"c3f95915d2b281893bd321efadb4a93c440822b9":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:17:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/24/694824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/694824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/694824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/694824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/694824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/694824/1"}}},"commit":{"parents":[{"commit":"29e9e2a3c30a91525d97fcb231898a9289297639","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29e9e2a3c30a91525d97fcb231898a9289297639"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-18 18:17:40.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3f95915d2b281893bd321efadb4a93c440822b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3f95915d2b281893bd321efadb4a93c440822b9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
