)]}'
{"id":"openstack%2Foctavia~694825","triplet_id":"openstack%2Foctavia~stable%2Frocky~I8e449b19af72d72f6effd52e0a2debb5754a19b3","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"I8e449b19af72d72f6effd52e0a2debb5754a19b3","subject":"Fix a potential race condition with certs-ramfs","status":"MERGED","created":"2019-11-18 18:17:49.000000000","updated":"2019-11-25 10:30:59.000000000","submitted":"2019-11-25 10:29:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694825-1574677769848-fe519b9b","meta_rev_id":"35239a6bda3749fdf8d00324290abf5b41cc2b49","_number":694825,"virtual_id_number":694825,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 10:29:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-11-22 21:18:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-11-25 08:25:31.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-25 08:25:31.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 21:18:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-25 08:25:31.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-11-25 10:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd231381e7ca1be312e624f9a9faa04dc63f84d8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"48965bcc8689e3146befeb8bb75b05d2c8d8726c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:59.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit aa53b66fda3eb63e5777ec79b765fb9610cb409d","accounts_in_message":[],"_revision_number":1},{"id":"6c9bc69819e67b815174e568b58968ad4a19540c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:53:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e1661ef9ed3429cad2edd57b8779bf7 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78a28a012b1498fb4d6bc98fa8eac92 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91e60029b05d4ba08fa4b567583245a6 : SUCCESS in 6m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae3057deecc943c08a2ec20dc1496ddd : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47be9586d1944bcc93051f387b7b4e70 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d712a8c848b48daae2fc6b0cbbfc81f : SUCCESS in 6m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7759e05c3f7b418f9960736a4917163f : SUCCESS in 1h 17m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/45425d9ce6524335bc8e973890d439de : SUCCESS in 1h 19m 00s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ff97d3e5ee44998a758a75bd88c7d59 : SUCCESS in 1h 58m 26s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d13468761972466bb43a1ded5aa71f87 : SUCCESS in 1h 31m 01s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/62a53c337b12498db4c980a870fd660b : SUCCESS in 5m 57s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4a4923851e8f448c9bccc8d2eaf2d366 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e87721997504ae896731c9b9f3b2a04 : SUCCESS in 5m 41s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b02bc15bd9b84418b8ebcc96007335c8 : SUCCESS in 1h 07m 44s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a4873ff467f6431eb517c030d7328009 : SUCCESS in 1h 06m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b1c60f85a8494f80820a203f7fa851f8 : FAILURE in 1h 26m 55s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e20cd165755b42f39afd78c8b013ae4f : FAILURE in 1h 35m 04s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/309e3d07b8974275b6097dc95437639a : SUCCESS in 1h 10m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e46712207fa49bbbfd90bb2e9226880 : SUCCESS in 45m 19s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a2d54f9fddbf49baaec3ce41059572a0 : SUCCESS in 52m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55f98ffe0025497baa21c47d3c41de91 : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/94e1a323e1744f5b8562ee2e3ce150f7 : SUCCESS in 2h 19m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3ce01a0f8e34dda9e077fc365c72083 : SUCCESS in 1h 13m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b369e8eddd9417498f67562330615e7 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"69ac4a4187d40c8e878f94e7d5f5bde4d62e5bcc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:18:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e83a9a6d37d6c03fc6d7482f7e6d787d0c00c28","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-11-25 08:25:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04a1889930744b90455bb9fb83528d0a67f3c018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"820e20a38a18e17b4e855efca912306b7f804b19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:29:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50adebe27e86cd5b9039572395096809a46eb690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:29:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ee6f348a494c378714162c0a7638fe : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b583ceaf713e47d0824eb53558c110e4 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ae08b57d0bf42c882cd96cdd3c865fd : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f70d128b23e14c00b72f41401d3edde6 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7f27fdeaf0403a82fe5aeaea287e00 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8044b6777d134a6fb75e239170f84f47 : SUCCESS in 5m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f5767cbd56934b6389f33f9ec9598318 : SUCCESS in 5m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4a6ee963b6c948ce921b4df740d5c6b4 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c8896b430a84bdfa3bbdd5c3f18a691 : SUCCESS in 5m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0376dfd677743f1bb3caa1bedad3306 : SUCCESS in 59m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/42f791ee2e5d41c9ae508bc8fe9534d8 : SUCCESS in 56m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b83691b109824e329ed8c582d35faede : SUCCESS in 1h 14m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8bd22e9331f1445d9956e82972bb2716 : SUCCESS in 57m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/18eeb6428a6c47e8817fa023db6216ba : SUCCESS in 42m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79566ebb32a54c629802516947383656 : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f50f5908d5e14ec28a1066bea87f79ee : SUCCESS in 1h 28m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8bfba00accf4aa18e9eb172b29fc45b : SUCCESS in 1h 06m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9359687e2df64500bf6215f241e42f27 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35239a6bda3749fdf8d00324290abf5b41cc2b49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 10:30:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d1ef24b8d3411abebbad99fbc6dd7b : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1068f4e0a3048fa8ddab56f543bff39 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d48427281867bd62077a65ce092c1645b3d44db9","revisions":{"d48427281867bd62077a65ce092c1645b3d44db9":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:17:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/694825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/694825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/694825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/694825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/694825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/694825/1"}}},"commit":{"parents":[{"commit":"9aebc5407d13003462d1178c76c580430bdf595c","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9aebc5407d13003462d1178c76c580430bdf595c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-18 18:17:49.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d48427281867bd62077a65ce092c1645b3d44db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d48427281867bd62077a65ce092c1645b3d44db9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
