)]}'
{"id":"openstack%2Foctavia~694826","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I8e449b19af72d72f6effd52e0a2debb5754a19b3","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"I8e449b19af72d72f6effd52e0a2debb5754a19b3","subject":"Fix a potential race condition with certs-ramfs","status":"MERGED","created":"2019-11-18 18:17:59.000000000","updated":"2019-12-06 04:12:41.000000000","submitted":"2019-12-06 04:11:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694826-1575605487341-a858fa3e","meta_rev_id":"a6bc8da7b345b8c345e1ee08fc9e81c957756fb8","_number":694826,"virtual_id_number":694826,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-12-06 00:03:47.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 04:11:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-04 20:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":2,"date":"2019-12-04 20:23:33.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-12-04 20:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 08:25:40.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-12-04 20:23:33.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-04 20:23:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-06 04:11:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0f64aa61b1bbe6c745f22e210e9ab79d477d582","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:17:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"eccf8a44b8a3fe62f280c09bd61d49c8c9e2c0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:32:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d955c2caacc481f93f355466625f212 : SUCCESS in 9m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdd53a4ec42d401ebc42796ffc183148 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2b39e3ab2474bfb812ce69dc00f051b : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465ad8df2e474ff184bdf5dd8f3ac919 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dfec80fe6924545a4b2a691a830e718 : SUCCESS in 6m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/06d18c526cb54ec696cb8e228b19e550 : SUCCESS in 1h 16m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/71ade163d2d24b63aa46a287027dd61e : SUCCESS in 1h 18m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e34e439a9d09413880860cde869012b6 : SUCCESS in 1h 29m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8bc550bc4395408f89e9fc2396caec46 : SUCCESS in 1h 35m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2c8a1dcc9c94fb3973643b5cd8ba66a : SUCCESS in 8m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/00cbca67dfad470ea35d8a54a6e4a56a : SUCCESS in 8m 35s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/49dc1bd5594b44f998a0a8ae294e396c : SUCCESS in 6m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2539dc247cfb44e49d8cb1837129a857 : SUCCESS in 1h 03m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3dfef89567ae495c9db82b8cf9f1f27d : SUCCESS in 1h 04m 31s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/07c8138003d34aa3a7cee2e6764317e5 : SUCCESS in 1h 16m 34s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/25589001fe384794a9fcc1c6ad4c6bfc : SUCCESS in 1h 17m 04s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/f1f77d3309c14d3da74dc9fecf34ee7c : SUCCESS in 1h 14m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8dd3ddb7da7447069976286aee4a629a : SUCCESS in 48m 07s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/b76aa936267b4ae6ac7563f0f1bb49dc : SUCCESS in 50m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f37501ac929d4183bff9463ee20135fb : FAILURE in 1h 46m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1bedf77a40e348169feb40c1ac5808e5 : SUCCESS in 1h 46m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e01862703130c51b36972ce400622caec942d2d2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-20 00:13:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"de68c3fa35ab50fce6acbe9c71f40b03ec8e9757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:51:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a83a930d8d848f9bcaa1bc5a8d6f87e : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac2755cd0de64d08a4f91fc62a2ed591 : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7869ca5af9ea40259031145581d80722 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cfc3cb7fe7452ab7e267323fe46def : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b5f06e4c5a14f61b54f5902c5f78089 : SUCCESS in 6m 08s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/50f6de125197443483cc53ce0cef7a28 : SUCCESS in 1h 19m 07s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ae148d2ada9b44c8b60755c00b846ee1 : SUCCESS in 1h 16m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6aad1893f2634deaa4e0a018f178afc6 : SUCCESS in 1h 30m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e112e0e58d3e4f30aa25e825ef2f98a8 : SUCCESS in 1h 27m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3518c98f85d4e18935520aa45a9c141 : SUCCESS in 9m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd8d9ff1773f48c08940b6d7aaba8aa4 : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/904082712c654407a9e53fe1ed78060d : SUCCESS in 6m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5875ef10e95c47d3a078c566a69719b8 : SUCCESS in 56m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7fd60e8a51c5488a8786aa610d1da2ad : SUCCESS in 58m 37s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/6a732a1e33934eca98ca3aee5aaff9da : SUCCESS in 1h 03m 47s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a1f0749bd97548469743e32c85474e56 : SUCCESS in 1h 13m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8bab9c9dd0034c54b868a2319f6b502b : SUCCESS in 1h 04m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee0b1026a8fc42de895e2811683791d3 : SUCCESS in 45m 25s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/b2b6dfcd57af48ce92cd0b8db597e0f2 : SUCCESS in 45m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7e6b53355f4497abd4a3fb3718c4db7 : SUCCESS in 1h 12m 18s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/215f167919ec4375af794a6e76a321e0 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ea3b303db370196e80595c8d8e685b0f62146a81","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:18:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"638567a7596a92afeb58e61adb9bb550a597ea0a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-11-25 08:25:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab21b5aabcff9dd289295e5c449c27bd85ef11c8","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 15:51:10.000000000","message":"Patch Set 2:\n\nWhy does this use depends-on for a patch in the same repo? \u003e_\u003e","accounts_in_message":[],"_revision_number":2},{"id":"279ab4ce336705bc6a73645e3459f97b57abab10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:58:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37f79ace259b8d3ac2f57d004fb4e6017d8730e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 21:39:10.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"615cdc56cc63f4832fc18c5de0e13aeb1aadd578","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-03 01:57:39.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8f9bba15845502a5d0c7a38129b231ba2573c203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 03:42:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf6ac4b52e749d689338100b2c5f3a0 : SUCCESS in 8m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26bd17a853864731ab49d3e629f0f849 : SUCCESS in 5m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/25d9adc833e8438dab9b9bc901f5bc1c : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41960dbc54e4570a548afda17c4937e : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74980144770a4f1f9080f011db89f4e3 : SUCCESS in 6m 22s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f86a91ea8e114a1ba3749df0400095f3 : SUCCESS in 1h 21m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1184f4533908405e9bd3b63667083129 : SUCCESS in 1h 22m 31s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c4f403a7a3843ebb280ec140fdc1bea : SUCCESS in 1h 15m 21s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6a9dfd3a126d4fc49d4089f6174eb98b : FAILURE in 1h 26m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a1f3c4e5cfb483dab91e55b0b509b39 : SUCCESS in 8m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7a550958a1c24e728c55c9cd1b98c8d6 : SUCCESS in 7m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e6c394f4e63a4e38a53bfdcae7ba171d : SUCCESS in 6m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f6e582832d0468eb2191f33a835ad91 : SUCCESS in 1h 06m 25s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/986e6a0ff4e84868a795097e0386618e : SUCCESS in 1h 09m 56s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5e137989d37e4d20a7486d2bf2e78fc1 : SUCCESS in 1h 20m 28s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ea9614052bf64775a3e69d0577f60482 : FAILURE in 1h 19m 14s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/47ddce9a28da4730ba6951a45f88d87b : SUCCESS in 1h 01m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5105348537254cc799db94b37b3f0bcb : SUCCESS in 43m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/f034ff94283d4d9185b99e33b6ac5559 : FAILURE in 44m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e403405280de4df790954a42c5d9dcd1 : SUCCESS in 1h 26m 22s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/240a34991eb846e0936139ff46d0f5af : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5a8650075eb953453996e2dcdb4b0e29c57bc9c3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-04 20:19:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d5283489d75ae77be630be95059d91c18afcc1ec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 20:19:18.000000000","message":"Patch Set 4: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ac11fbf2e52e8807280025130b881b13919ed49e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-04 20:19:45.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"def22a6526cc2b3da24f69f93b855f258cbda0b6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 20:22:40.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ad960d2f9b3c805ec93b6ff6be74e675d8cfccca","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 20:22:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0b49bff375f44ca73758eaa61a427ffcce9d164a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-04 20:23:33.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9c51cb40151aa9f90584525738dcd1d0f236423e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 20:23:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"0b70c430752bd5a5820d7bedd928a4f1acd74b09","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-04 20:23:41.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0c968b4aef5731d4362e01ea61c7eb88a36c6975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:53:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f542c6a0249409d8cd9c5c2369c8807 : SUCCESS in 11m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39b1995875cf4ee38c19c645bc94bd27 : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d8b5aace0904728a6fd8bc95184b8f9 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2117caf9b7c24057a860d53668180c8a : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96bfb4b3ed0743aeaab4ea53da2d0a3b : SUCCESS in 6m 58s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7930d37239e14948a39f998551d14578 : SUCCESS in 1h 14m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5df765df1c344d19995932a42a3829b7 : SUCCESS in 1h 14m 39s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29df7e8ef1cf41e6955dc4bd452538dc : SUCCESS in 1h 15m 14s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9f54959b19b34901a1aaaeb47e5c0c59 : SUCCESS in 1h 19m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb7311190046406eb73d9f1ec0f15663 : SUCCESS in 9m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4d671ae043a24506ae5e8f4adb3f03a3 : SUCCESS in 5m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a001c828300447a3878c16f09f304755 : SUCCESS in 6m 09s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c0caee4f1e24d2399cc32ddc61968d8 : SUCCESS in 57m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fae4c123d86c4ab8bf4b47eab4d1ba17 : SUCCESS in 54m 31s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/aae9e0ccb1ed4d01b852dfb1e9508def : SUCCESS in 1h 09m 03s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/9d9de4ea264b417381470d6a0fb05d9e : SUCCESS in 1h 08m 24s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/676cbb10e54145a685a74b9fbf183315 : SUCCESS in 1h 06m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f384cf54c7c844118fc7ddc2e29974c5 : SUCCESS in 48m 02s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/95bd7a2a055a459285fec707e6c3f1eb : SUCCESS in 47m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b6a4feb8c32451cb6bc806ed214d42d : SUCCESS in 1h 24m 28s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6349839f10e6444e80ea8acd9f691224 : SUCCESS in 1h 25m 34s","accounts_in_message":[],"_revision_number":6},{"id":"3f6835fecdd96d575c4155448ef3550e48898319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 21:53:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a5cc1407bdff7c22d40d9cd68251799b43bae1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:39:46.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3c08bfd359495b8885b151240ab048 : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3374fd2d0134d52b8e76b2ab11bb426 : SUCCESS in 6m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb35a206fd264b3fa6f772cff735fc84 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c603da6a824cfba67a13988d6755c9 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf921166cc72453e9680c54a878505a2 : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/481cd24a22ff495ebb397cb87f6462aa : SUCCESS in 10m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/34380da4201343699eee459794859667 : SUCCESS in 6m 15s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3ed356eec4ab4802aab1e6f435e0a222 : SUCCESS in 6m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4b7302ff263c4fe39c49ee5877ac05bf : SUCCESS in 1h 13m 26s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/694a452fff6445b88a52649953b9d2a8 : SUCCESS in 1h 33m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38dd739bcddd4ce2902f662247ad9386 : SUCCESS in 1h 02m 53s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/91b23c7d3cda488e890e28f87ef19ce9 : SUCCESS in 56m 02s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/47e856df4915454c86be9395960f4b11 : SUCCESS in 1h 12m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7c99c1b574de44c7a96365d7882dc7fb : SUCCESS in 1h 14m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1b2a95ca6ac48a69097950306434149 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/f4946a2e3f074dae9ad16f311adf9961 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a062c72ccafb4e8ca81b17088023a597 : FAILURE in 1h 29m 43s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/4bb20c3fa5934a2ea43586462e2ea5b0 : FAILURE in 1h 23m 05s","accounts_in_message":[],"_revision_number":6},{"id":"044178a8b6c3d5b24c1f87d348b47d9bef005d52","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-06 00:03:47.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ddc9564cec7ac9bbee7fc418cd4ee7e28aa26cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 02:20:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68cc6bac094b4fbfaa14d5e8d70c1cbb : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfebba2a552a45aaa4cd1740a01dfae2 : SUCCESS in 8m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3089c9b0f88140c2b0cd1b621684ec56 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0012e07c23ef44dba64ccb309f6278d6 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca3d9051ce4b4a9994532d6b417c05d4 : SUCCESS in 7m 49s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7ed72f39a66848459745abfdb5dda839 : SUCCESS in 1h 52m 47s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/fe3b28231b4f464492e69a54298a1914 : SUCCESS in 1h 15m 20s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9667487877d48479f2cbc286ad01f55 : FAILURE in 1h 12m 09s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ec23052a5e7345df9d91096614afaf29 : FAILURE in 1h 12m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59420426eb4c41a8a88ffe9aca43a975 : SUCCESS in 8m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/06cc68f6be064fb2ac98792375fecb8f : SUCCESS in 6m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/89d1e63363fc4c0d92d1bb4b0154ad5d : SUCCESS in 6m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53db9771aca741ceb7ebb6ff00e0547e : SUCCESS in 1h 00m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/58d132da498a47f8ad0fda7e4b3e5cfc : SUCCESS in 1h 00m 37s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/1da7e3c67fd4488a9995d0c894335abf : SUCCESS in 1h 05m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/739d79af464447ae98c8dfba2cf479ff : SUCCESS in 1h 10m 22s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/27e3eb1a00ae4aa389b3e9ab694452a7 : SUCCESS in 1h 12m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/001029d8d9ab449c9d7d4983de733a40 : SUCCESS in 48m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/35719837625a4e70a69ef74956967432 : SUCCESS in 50m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c119317fce345649c6323dd8113457d : SUCCESS in 1h 24m 11s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a1a9bc00caae4109870bb046587b78ee : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":6},{"id":"08c0600b98658e88f315d75ea5be5a1f1e91b547","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 02:20:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8c15d8a5247f3d362160ba87d718c56eb0413b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:11:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a103f1e2264149bd961d83e8b3bc9243 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8c2ea6501b447dba5db2b714c55e7a5 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/38f1b3448e674b5ba3bba6e7cb171268 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7c41a7fdd04fa1802940a152a7ad92 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff7e642dba144eeb1077ddda0221fcc : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecf4e7beda2b4fd79d09007d36d81c54 : SUCCESS in 9m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b671a0739a7d4700881eba43ca51b2da : SUCCESS in 5m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1479bec0eef0483488534fc45d16c1b5 : SUCCESS in 6m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e5361d1c9f9842ab816ebeb81da089da : SUCCESS in 1h 16m 58s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1ecf3ec5bebd457586fb0ec705c87e4a : SUCCESS in 1h 17m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b5a4fd2876d4aaebc5b2783496bc535 : SUCCESS in 1h 00m 44s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/79f9c211f4e8463caacfa219ee885747 : SUCCESS in 54m 26s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/9ab6f33f9b574097868d808f8c3c8ebe : SUCCESS in 1h 08m 49s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a34dc26de07e4a7ca4088f28f73762a6 : SUCCESS in 1h 14m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f55bfc8c0ae490c9138f01930070524 : SUCCESS in 47m 04s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/9612e4b657d942ffac1c84992f2d39b5 : SUCCESS in 47m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a6266fa207e42bdbd25ebd106acfdc6 : SUCCESS in 1h 38m 11s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/41826d7bad964e518b89e3ebcf892c39 : SUCCESS in 1h 44m 05s","accounts_in_message":[],"_revision_number":6},{"id":"122c74ab3b5e6f722f3b11580b85fa9f81b3bbb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:11:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a6bc8da7b345b8c345e1ee08fc9e81c957756fb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 04:12:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9b9d052b984cdaa966b02a5a0da364 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b6769c8f5d4f978e2a14f94da77c42 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c8e50b11f90f611c701330b16372a9e8932079ce","revisions":{"aa53b66fda3eb63e5777ec79b765fb9610cb409d":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:17:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/694826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/1"}}},"commit":{"parents":[{"commit":"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae","subject":"Fix Octavia stable/queens pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-18 18:17:59.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa53b66fda3eb63e5777ec79b765fb9610cb409d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa53b66fda3eb63e5777ec79b765fb9610cb409d"}]},"branch":"refs/heads/stable/queens"},"06b13d4bce131ad57d7ddc16b7cff855a85b1c4d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-20 00:13:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/694826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/2"}}},"commit":{"parents":[{"commit":"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae","subject":"Fix Octavia stable/queens pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-20 00:13:47.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nDepends-On: https://review.opendev.org/683537\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06b13d4bce131ad57d7ddc16b7cff855a85b1c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06b13d4bce131ad57d7ddc16b7cff855a85b1c4d"}]},"branch":"refs/heads/stable/queens"},"4a248e4f786e0a6387f4aa25d6d48109e663dc1f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-04 20:19:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/26/694826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/3"}}},"commit":{"parents":[{"commit":"6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae","subject":"Fix Octavia stable/queens pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d3020d92035b2ad3ab0dcdca8e3dc78b8010cae"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-04 20:18:59.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a248e4f786e0a6387f4aa25d6d48109e663dc1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a248e4f786e0a6387f4aa25d6d48109e663dc1f"}]},"branch":"refs/heads/stable/queens"},"b2e6f41f990c8b3dd480f17cac3ef3abb471dab7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2019-12-04 20:19:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/26/694826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/4"}}},"commit":{"parents":[{"commit":"d0653bf8b3aa36d60b30c774ba483826d53a3e12","subject":"Flush the eth1 default route inside the netns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0653bf8b3aa36d60b30c774ba483826d53a3e12"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-04 20:19:18.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nDepends-On: https://review.opendev.org/683537\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2e6f41f990c8b3dd480f17cac3ef3abb471dab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2e6f41f990c8b3dd480f17cac3ef3abb471dab7"}]},"branch":"refs/heads/stable/queens"},"e7589f3ba86d14058072895e390eefe4803141bf":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-12-04 20:22:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/26/694826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/5"}}},"commit":{"parents":[{"commit":"d0653bf8b3aa36d60b30c774ba483826d53a3e12","subject":"Flush the eth1 default route inside the netns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0653bf8b3aa36d60b30c774ba483826d53a3e12"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-04 20:21:25.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7589f3ba86d14058072895e390eefe4803141bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7589f3ba86d14058072895e390eefe4803141bf"}]},"branch":"refs/heads/stable/queens"},"c8e50b11f90f611c701330b16372a9e8932079ce":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-04 20:22:50.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/26/694826/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/694826/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/694826/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/694826/6"}}},"commit":{"parents":[{"commit":"96e5962e05615df37041723ca173873f11d55ea2","subject":"Merge \"loadbalancer vip-network-id IP availability check\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e5962e05615df37041723ca173873f11d55ea2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 21:35:43.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-12-04 20:22:50.000000000","tz":0},"subject":"Fix a potential race condition with certs-ramfs","message":"Fix a potential race condition with certs-ramfs\n\nThere is a potential race condition in the start order for the\namphora-agent service and the certs-ramfs service.\nThis patch configures an explict ordering for the services.\n\nChange-Id: I8e449b19af72d72f6effd52e0a2debb5754a19b3\nStory: 2006823\nTask: 37396\n(cherry picked from commit 1900ee71a51c25d4c0fa1ee79447219cef05b5ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8e50b11f90f611c701330b16372a9e8932079ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8e50b11f90f611c701330b16372a9e8932079ce"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
