)]}'
{"id":"openstack%2Foctavia~694827","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56","subject":"Fix filtering for provider capabilities list API","status":"MERGED","created":"2019-11-18 18:19:59.000000000","updated":"2019-11-27 14:56:16.000000000","submitted":"2019-11-27 14:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694827-1574866472955-c9d8f7e1","meta_rev_id":"2a5e6e98de313ea258bb50da3bd50d98ad8b1518","_number":694827,"virtual_id_number":694827,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 14:54:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 12:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-22 22:26:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-27 12:19:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 22:26:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-27 12:19:19.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 14:54:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc8cd8423e770157054890316f9a14f185a0aef5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-18 18:19:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a8e8831b5a2c563db3529672c3bcd33ef8c7208e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739800f380d84894847e293c8e5aefb4 : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/130340913b1649569375fc9ec10fb910 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e288bad114b49309eead8f69553e836 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9528ae369c094c1f969ea48fe44efcec : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09ebb15d93894bf3a3b1e1c6213b325a : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a64ac6dd96d94ed9be730b1b3d9ac707 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc413395fbf1412eaa9b70050b4d389d : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/943b1fbe71d7405395fe21fed4db9f45 : SUCCESS in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b908e9b68c054f71a80b7c887dd1815e : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fbaee0d5e6a748288b8c539aa18412bc : SUCCESS in 9m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/93d62ffe6cde466385cc45c3678ee34c : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8d0df1141ab47dda3752855d071e197 : SUCCESS in 5m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e7b96b877e7403ab95095b766a7451b : SUCCESS in 46m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0c4d346449364d5da19826cb550da601 : SUCCESS in 38m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54f36dd69c864053b98e7200b4c34794 : SUCCESS in 1h 56m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/20495cee25e741298c0ca1ebadf5330d : SUCCESS in 1h 48m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85f5b0a7e16d4d5db077ab277afd3d54 : SUCCESS in 2h 06m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/55fab9a6c94d427d936153566bdabc58 : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f1fd2de3c66422dafadf3ebc862390b : SUCCESS in 2h 02m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f9f6045917b4e7a83e1f5348256eebd : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36a25c8d9cb7404a8355f9c0034b2d2d : SUCCESS in 1h 01m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8f08c1336f364998842f1bddb0ba08da : SUCCESS in 1h 38m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f817a7c8915f4f74963358a4b26bbf71 : SUCCESS in 1h 14m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a40a842120ca4bca8bb2fb094a1cd4bf : SUCCESS in 1h 04m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d91d008afe2446fd99ec8be6b1f33c66 : FAILURE in 1h 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d01baeb1888212a0725a471a1ef27a95b6242ea","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 22:26:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c9a1ee31852abd3d7463388e34504b5bfb059528","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 12:19:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce8c478584891d6d4dfb3f8a742bff5d024515bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:19:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86f112603c0e8ba15db22e3220a6d59d237df478","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:54:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"afb583cd99cb0212ba8d12621297af228de8ba46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:54:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fdef923025c46c697af840346f29809 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c04a11960c4f668179fee90cef1a86 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2069f6c91ad648fcad6427dd44bafeb3 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e07bfd4d7ac94a7a913db73d8f1b63e4 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50eec2b79408463a9b95930fd2c5395b : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/876835b85ca7417e979294a594d87811 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1eadea1d5a044c0b4a19459b30cc1f5 : SUCCESS in 6m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/637d8952a8634b59871323ef4f3a8d2a : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e25f393aaea542c8a0e31bfcc8b2579e : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4033abc4494f4e838c0ea3580415fdbf : SUCCESS in 41m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1293fa9eb01c44d8a1c55c2bb1138229 : SUCCESS in 48m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5cb095d0340b45008c123cd94d202a2d : SUCCESS in 1h 31m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f0e58b12127c48b5b87d65b612ae62f2 : SUCCESS in 1h 31m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1822294850a24967b71f572136399902 : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2a5e6e98de313ea258bb50da3bd50d98ad8b1518","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:56:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8151c30ba101492b9283537b2b2e6ed6 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40c98d7da0841998d17264cb579d27b : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"735de5e175b8ee85fb87a897c70bba1e3c9f2f2d","revisions":{"735de5e175b8ee85fb87a897c70bba1e3c9f2f2d":{"kind":"REWORK","_number":1,"created":"2019-11-18 18:19:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/694827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/694827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/694827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/694827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/694827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/694827/1"}}},"commit":{"parents":[{"commit":"ce22677502c36245f695b734f09ddd682be67faa","subject":"Merge \"Fix issues with unavailable secrets\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce22677502c36245f695b734f09ddd682be67faa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-12 01:33:03.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-18 18:19:59.000000000","tz":0},"subject":"Fix filtering for provider capabilities list API","message":"Fix filtering for provider capabilities list API\n\nThe \"Show Provider Flavor Capabilities\" would silently ignore\nfiltering parameters passed into the API. This patch corrects this\nsuch that you can filter the providers capabilities list.\n\nChange-Id: I9a2aa4ed5b4ef60545c60a6d5bd06066ef361a56\nStory: 2006817\nTask: 37377\n(cherry picked from commit 28c936ec7101a03496558bcda311edf205ee74b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/735de5e175b8ee85fb87a897c70bba1e3c9f2f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/735de5e175b8ee85fb87a897c70bba1e3c9f2f2d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
