)]}'
{"id":"openstack%2Foctavia~695112","triplet_id":"openstack%2Foctavia~master~I7153ed9a6214ce352b2ac115001f7f6134d95b39","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7153ed9a6214ce352b2ac115001f7f6134d95b39","subject":"Fix filtering with API tags","status":"MERGED","created":"2019-11-19 23:22:49.000000000","updated":"2019-11-27 15:22:06.000000000","submitted":"2019-11-27 02:05:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695112-1574820318227-a4121484","meta_rev_id":"b7cb19ac1c473a0e5f9237dae5d3dab2499591fa","_number":695112,"virtual_id_number":695112,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 02:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-22 18:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-21 08:12:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-11-20 10:22:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2019-11-20 09:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":1,"date":"2019-11-20 21:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-11-26 13:24:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-20 08:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-22 18:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-21 08:12:05.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-11-20 00:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 08:28:43.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2019-11-20 09:37:52.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-11-20 10:22:12.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2019-11-20 21:26:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-21 08:12:05.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-22 18:37:21.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-26 13:24:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-11-27 02:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f438cbfe560166cbbd7da0132a958a971c8eabea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-19 23:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"158aed6e589b758f3a07a7931f3dcbef3dc8c92d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-20 00:51:53.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"ac8d8cc8cc5656983e9cf1f3388c511be0ae80d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:44:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b0e34a9201d446ca81f75ae1d923709 : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/805ae8389c3a4a339c780b49e0ff9f66 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06cc124cdd74c4d9e1a00a42734c4b5 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/192736fa6e3747f497ae913bad593a1e : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9f87941840492fbd2cea00c1f1df6a : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/371d242923ef42c0b87b83f8b7125844 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41015ceda9bb47b2b4b5e46f946de082 : SUCCESS in 11m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dc86f6e8bcc046ad9a31a460d2e64c14 : SUCCESS in 4m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9182449b5fea427e94466e15b192172c : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a704efcac057492dbd26dc36a7aede85 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7859d7bd73164b21870ff8bc7f05ba25 : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b4d3c0786124adb9c90ed877025e146 : SUCCESS in 47m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/478d3810b8944e019013bf890c03d71d : SUCCESS in 43m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f093df445b64406b2fb2a9d45f0f8bf : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cf878c7678284f59a0ad6ee919a11d41 : SUCCESS in 1h 46m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/643494de0f2242b89b15fd6ae789b0c6 : SUCCESS in 2h 19m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff2de7c2cff54cb48da5a8cfd3601e38 : FAILURE in 30m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7a4eba039f14523bb7d2c7211fcd7c2 : FAILURE in 1h 11m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5d0478978f64df58349bc5632f9f2a5 : FAILURE in 50m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6307d12c4a4847c39e0ea452ef482d72 : SUCCESS in 1h 19m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a07f6a77def8463faab83cac501aa7db : SUCCESS in 1h 43m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b62db1fc9484a9bb995e11f5f2653f0 : FAILURE in 1h 14m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/313ecb0607d142a4b30f9a110a316bb9 : SUCCESS in 1h 43m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7bdaa735e26849848d6e3a28e47a0b99 : FAILURE in 1h 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49cd8d15a0a4876498936cfa24bf3169655421aa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-20 02:10:19.000000000","message":"Patch Set 1:\n\nrecheck\ngrenade gate issue","accounts_in_message":[],"_revision_number":1},{"id":"bfbbf801dab0fabbd709e61dd10a240556b16933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 04:25:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b9edccd1084ed0ab014e56bd6ba15f : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e4190f73ca74af2816278f57df876bb : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e1d97ea17b4b3e9c5038497dd9532f : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a93441dba7014ce1a3cd04bff50f521f : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b989f3ee4da4f72afcf296813742735 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4766153c7c7e429495f1f367fc643dd7 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65a8d56668048f0a140cf33ce644292 : SUCCESS in 9m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/241fd2179e9b4362a3ad00a94413cb53 : SUCCESS in 4m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ab6de4ab9bc34ff1af5393c614a565e3 : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3a6adb9fa57349fe8dbadd11f8cbf734 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/182235f0e9be4a88a13123684632796c : SUCCESS in 5m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c611aea7c18846c3b7273eeaca05a276 : SUCCESS in 44m 35s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d8ae099a4e354e9d84fe8f6ad5940330 : SUCCESS in 59m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a354fb8b678460c8d18576f282a6d3a : SUCCESS in 1h 34m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3bdf5ea5e4714f049876f7656c6b8c5c : SUCCESS in 1h 40m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6ba33172be834f6aac9ff34e8663140f : SUCCESS in 2h 10m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8f7c66fca3b4f35abbe46971270d6af : FAILURE in 23m 23s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fede19cad7034ff58cca45358f0f9a1d : SUCCESS in 1h 10m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/31bc6053d4fe4af9b7c8626fc6b2e2dd : SUCCESS in 1h 19m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/921e33144d3f4e42afb538dbac8d7fa9 : SUCCESS in 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dfbd450178c042f5b5aaacd804a238a2 : SUCCESS in 1h 30m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/864454c263dd4f6fa1c3c324fdada71d : SUCCESS in 1h 09m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4bdb8ba6ecd04c39bea62d545e85b4b4 : SUCCESS in 1h 04m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2bbc00ecf38f4f20954ca389320c4f55 : FAILURE in 2h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4021b322dc178b616627b60679cb46b4f65f954b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-20 08:28:43.000000000","message":"Patch Set 1: Code-Review+1\n\nTested locally and works as expected! Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"ee0dbf06bc967c43b505e70ccd5a3e094fa16579","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-20 09:37:52.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c3cdb27ad7f58683ecdf9796bf05baaaeed1262c","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-20 10:22:12.000000000","message":"Patch Set 1: Code-Review+1\n\nAlso tested locally with multiple tags, it\u0027s working fine. Thank you.","accounts_in_message":[],"_revision_number":1},{"id":"a9097c56c459ddf96d8a5375db138ca721188e65","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-20 21:26:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4e77e113d105209f1744d8c7273154c788ee7354","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-21 08:12:05.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ec5ba8155adc97b59b67e850ea8a4ebcdc05fdc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-22 18:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d3822e776f806e9226eda81e6f9c6a555db75bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 18:37:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"983d3223d79364f052188ca99dd87d1940474296","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 19:41:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a86daf0ef9964c4f8dfa9cf40ceafd45 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbfd5c675e73479d989565155b7889c1 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3dc12a39d4b455b8adb7158eb44a377 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b7f42ec4e34437cb8a82c330493fecb : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a22318e97c4c4d4da3757a15cd3d3098 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a16586308e04995857165f0521bb4cc : SUCCESS in 9m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3a0530d1cdb44188b17f2296219b9307 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb719879d20443a4a2561e420b65d33e : SUCCESS in 5m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/877e1c8a3cad448eab6e960f387e309f : SUCCESS in 44m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/00037b34301547969790aca7e4cc135a : SUCCESS in 48m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa8c82cd56744aa1b34a4bc77d7d3219 : CANCELED\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/580505f4aee140688358f43735c10897 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1be01227d6e441c97cf8011c55cea11 : FAILURE in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a63c3eb1b2f92a0d88f1b9e31527d2376ffcd3d0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-25 07:30:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c666c2736d3b79acdeefbe97d1ef6e1f0ac0875d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:35:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57c435a55084c2f83731e5968ef42c1 : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1788f6f1413b4cf3917399544d4e0545 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77b342f04de47fc9146bb7682634d70 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ad6d1eba2494840b210f5daa82d6500 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3b5ce9c5bcf47b0bb8957b712584395 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1178e8fef3042f5beaffb1a13634965 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b4fecc69ae4fa9ab99b55bfd9f53ad : SUCCESS in 10m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/148d84c108764ad2841223ae28d9f249 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0907ab4e36714304bc32b117eb6b4448 : SUCCESS in 6m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04776c84f24e468fa10d40de218b6ded : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e62ad8e0eb4489b8a083c71c9f04745 : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64bb2d5f51aa40968c587847ae1a858e : SUCCESS in 51m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/695431bcbaf64ed194d3c4cafd1e0b9f : SUCCESS in 54m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d51d2e093dc403c887da4901c4e744e : FAILURE in 31m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a52442dbef5345969115e0d021a9f05a : FAILURE in 31m 56s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/486244c9908246bea20bbbde31702d6a : FAILURE in 44m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/287bbe0207274c0596d798d2e001b8bd : FAILURE in 37m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d21824cc31474f289769abfe10f3472b : FAILURE in 37m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e95b8186426f467dadfd0487ba2f9fee : FAILURE in 35m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d340e00c073e482f9dd3c4bb50ea0d2c : FAILURE in 31m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a3fa3e67422c40e893db12f6c41a84bf : FAILURE in 29m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1ab6972b78b40d0a6178372d558c3c5 : FAILURE in 31m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d922f54710de4d8d8019a74c576d19d9 : FAILURE in 31m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f1a4adf77b4242529f1f05ed9b9798c8 : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdf4f6206cf72f9caf120b6b401ef88b83a2cedf","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2019-11-25 08:38:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"169e04fa08cf4a550874666639af11d60a7465b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:30:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2f754def1434b9f9071ce1729e7fe3c : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49a42a69409d42cfac8f2f0cfe0791a0 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d1e47d42d44cc4adfd74eccc732af3 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60deacc861a64fefb9684bbe4052ebb0 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3e056cfefdf481c9528ff9aa54af573 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8e301fc780946658ffe458b29b606fe : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d96e94ab8a46339ab2f03600aab3ef : SUCCESS in 10m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/de4f7f430494433da200dfeed57de0b0 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0bbe951dfe414f56b8547f44941cd4b4 : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/908737708a40495abc681ada3249bc44 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a22ec4745b548dbb672deaef44cdabb : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f27c5c721cc44c2bda75d0495c00598 : SUCCESS in 43m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/649907ec88da456cba9db8b765b7aebf : SUCCESS in 42m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e289b3d3b6e747898edda63b16649965 : FAILURE in 35m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5ecea96c49fb4149b2a857a18e85d989 : FAILURE in 36m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0363068b4ced4ee29ecb8973536f401c : FAILURE in 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ba1d681ad22461c89aa26cfedb83cfc : FAILURE in 37m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a7158889558e4c43ad30bdc576145f4c : FAILURE in 38m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c8e46dff8b74c8287e8a1365e078492 : FAILURE in 43m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05a2c9b0505143c7b0a22d5cbf7570b1 : FAILURE in 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16db0a121da543f0b974651e0ba73b6c : FAILURE in 36m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41fa18f985d7480ba22eb848425e2438 : FAILURE in 42m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/21341cc08bf2433d86512177421754d1 : FAILURE in 36m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d99d85bab4cb446eae37058740391650 : FAILURE in 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4097f53d7c75aab85872e5853347b22e7a35c303","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 13:24:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f0f49a56a65cc0b0889d243395055d15ebc1e102","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2019-11-26 14:05:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74ed7fb71f848a05cd81a8e2c4d6704dcc90c78a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:43:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31b31fe5b7af49a3b070a4055dd25cb6 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06f547043177489385f13fc0a9baa07b : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e29dab020444b59fd6669e31dbd639 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3d91b7007034fd6b7614e0076d20c4e : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79532b73d3184b0e86c2e597947f5363 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/150000f6d0c94180b163fc48f8691566 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e122b5ae2d434880db12e94144f7e8 : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9a770bd5ee494cb89c57a9691fbd8637 : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/be41686cb2eb44b59dd9b0d45af85a85 : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2c12173632ab4d13b299302dfa56dd9f : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/086214fff56d4d81aef52463426f0473 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e894b0288c8414da5c262dec3fcc949 : FAILURE in 8m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/95aa877d3c4e441682c0855fc465ec1c : FAILURE in 7m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96fa8b30b02f43279e9b455dab3e6f3f : FAILURE in 8m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d8ba4852fad7470692b28041cf3e31e4 : FAILURE in 8m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4e2ba2f8c86a4d668e3b9e770fa4c2a7 : SUCCESS in 2h 36m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13bf33190e5342b9836f7bdf67858798 : FAILURE in 56m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/89e1203f3ce64f489142e273fa53ee3f : SUCCESS in 1h 54m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/984795031add42978e23b644074a3acf : FAILURE in 1h 12m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b27cb3ec1d5f420d85df8be42306f615 : SUCCESS in 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df7c5568d85e486587063d17a2584b1f : SUCCESS in 1h 31m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f19907fda32d47f78407341916de44fb : SUCCESS in 1h 06m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1c3bfc18e8aa4cbe9f7c0c54e2b121f9 : SUCCESS in 1h 05m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5c7626aefea24fc6b70bb3b0946fc8ae : SUCCESS in 1h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"200da772df4833bc63de4cc9e6df4d14f64c674d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 22:08:13.000000000","message":"Patch Set 1:\n\nrecheck\nmirrors and gates should be functional again.","accounts_in_message":[],"_revision_number":1},{"id":"77f690c0939ab468b38943cd455b8fe44cc6c7b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 00:24:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7b0fa108c804cd7a70cef0be083db75 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08e4e37ab07944d2a4cf05cc49084e52 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edde9456f56e4bc8a45617f6368d8765 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e65b804776d141be9f3aa9a1ab3345e7 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7733211ef649d1ac8346c544227df6 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f25c1045aa1d4a4184ead0be551c79a9 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62109ab255d14c2baf16ed716bbc40cd : SUCCESS in 10m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ba94a6e096034edf90bd43a3aa02784a : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/609edf5916754e0ca6bb1fb0cbea8500 : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f1d0d65691574c768bf2654b330b4dbf : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/017d519bce854294a007e1402c094229 : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8230dc3540c8444998d964d08542a294 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/74273d7e435c405699c35cf9b40bee5c : SUCCESS in 41m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c88747a9b9744e1ab4a117f5c8e48322 : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0a4905c07acf4e228e52ce1dba91c494 : SUCCESS in 1h 41m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0a27a1ee0de4450699802f17192255b8 : SUCCESS in 2h 11m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/77a99b62be9f4323853d12256a0fa653 : FAILURE in 35m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/365b2f69b11043d898d47f3bd796f532 : SUCCESS in 1h 48m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6fd9bd95baa84f95bfd755bdbc5051e9 : FAILURE in 58m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f710fca6da5e45a1a600464f11ac6979 : SUCCESS in 50m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2732aa3107de44febdc6876ff19817e9 : SUCCESS in 1h 28m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/679067c8bb59459eb12cc99915536961 : SUCCESS in 1h 03m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1d80b79052e340af910d7a234707824a : SUCCESS in 1h 04m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/66e0fe6f79a8455cbcb9b205a705bc85 : FAILURE in 1h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"448d575ba182fdc8ba746aed1b36c9c6fe617330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 00:24:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40bb7251c7d3817f1e3eb02efa24b58c8cb967cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:05:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3a2d18974a94a61a33d90f4309d2c49 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b885cb24266b49eb999d59f3db3d5913 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75bcf3334b7a466c87ead5b27d046d59 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb76a78fd77742719cf296ef17afa6d5 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3263e23445bb40a0be86edbde41e8380 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32667341e66343f4b77b5576c35b37d6 : SUCCESS in 10m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d94800084ff4876baa22858afefdd2c : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/734a240d8e0248009447e830c26bdb7a : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54b5af330a00463e8d21f341fa645bfb : SUCCESS in 43m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4d98569d771d4efd8dbb576b0d709379 : SUCCESS in 50m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f70418cc0d56402c95cb68140920b8bf : SUCCESS in 1h 36m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/def7b0521b5a4d41978d917f24dc5657 : SUCCESS in 1h 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ae25501a9233bea7c5644d03e5f429efc1af2bff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:05:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f53de3d65fc943eff971e06b112be7b4c528c5b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 02:06:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95808bd7c64400390a15f620b6c365e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b7cb19ac1c473a0e5f9237dae5d3dab2499591fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-27 15:22:06.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 19202863e2e459a0c81a872df7dc745aad4d15ea","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a736f98a7fffa8494ea4f00ecfbb929e8ccae860","revisions":{"a736f98a7fffa8494ea4f00ecfbb929e8ccae860":{"kind":"REWORK","_number":1,"created":"2019-11-19 23:22:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/695112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/695112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/695112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/695112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/695112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/695112/1"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-19 23:17:59.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-19 23:22:48.000000000","tz":-480},"subject":"Fix filtering with API tags","message":"Fix filtering with API tags\n\nWhen filtering API calls using tags along with other filter parameters\nthe API could fail with an \"Entity has no property\" errors.\nAlso when filtering with tags that contain a comma separated list,\nthe API would mis-interpret the list of tags as a single string.\nThis patch resolves both of these issues and adds functional test\ncoverage for these use cases.\n\nChange-Id: I7153ed9a6214ce352b2ac115001f7f6134d95b39\nStory: 2006914\nTask: 37559\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a736f98a7fffa8494ea4f00ecfbb929e8ccae860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a736f98a7fffa8494ea4f00ecfbb929e8ccae860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
