)]}'
{"id":"openstack%2Foctavia~695376","triplet_id":"openstack%2Foctavia~stable%2Fstein~I3f776bd018246897c9a889699a2d0ecbbfbb7098","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I3f776bd018246897c9a889699a2d0ecbbfbb7098","subject":"Fix controller worker graceful shutdown","status":"MERGED","created":"2019-11-21 09:25:10.000000000","updated":"2019-11-27 14:12:17.000000000","submitted":"2019-11-27 14:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695376-1574863703376-ff465d15","meta_rev_id":"c62e233fa2deff41f3f6f58742be89d550f75dba","_number":695376,"virtual_id_number":695376,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 14:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 12:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-22 21:20:12.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-11-27 12:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 21:20:12.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-27 12:16:06.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 14:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46719a5593d8c86e9575f09417c03998770a174b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-21 09:25:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f126210cacb9bd95cbff92cc80af40749c1be124","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-21 09:29:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 6fe5df6fd8b80e3022120f2bd74098a8e71152f7","accounts_in_message":[],"_revision_number":1},{"id":"8cbb4b206b50c6fae5038758747c8ef02e20d921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 11:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eee6bbb054af481eace2586dfae50807 : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3ab6e1e6771498fbc96f0ec1068c5b1 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694d5599da7948338686ee0f803967aa : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b534aae37284e59b0c6a6c64e666f70 : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/292a1fd7b55d474da19716b2448c5845 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e6d91923e924f649d53beefd3c5ca52 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b0839ac9b0410f924f422ba405037d : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56d6e90f433d4762828faa51a07e49d7 : SUCCESS in 5m 01s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dffaf40e44ee48cf8cda3036a6f7ad96 : SUCCESS in 1h 15m 46s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5aa526b8c4564c87bf36f99a17d9ce8b : SUCCESS in 1h 18m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/477511a0e8b74876ae04d8c03bce2630 : SUCCESS in 1h 40m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5bbf365a35fc41d38cd7b8c88925056d : SUCCESS in 1h 40m 23s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6fb53f8a76d94d1c82974d62c633fb33 : SUCCESS in 6m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cba2990a48134ee68bacb3178f649f78 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d795f40d67f4489c93543d9dea0b8417 : SUCCESS in 6m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fc1d1ea73d14e2daf2a207d1b425fa8 : SUCCESS in 1h 13m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/26e9979191584c879bc47b878c4a7448 : SUCCESS in 59m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6fd1995fcb249029a4fee7402e26cf8 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cd3686c25daf4796a0ecef02ba2bed44 : SUCCESS in 43m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19137db201854acdb89420d32f8942b9 : SUCCESS in 1h 37m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ac648394e25841de9943716a1bc0bac0 : SUCCESS in 1h 20m 51s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ccb6c69eee5844708c837a5a7cada398 : SUCCESS in 1h 48m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c53874d76f4c4bdb9d3e1d1237d9d272 : SUCCESS in 2h 28m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/447ad89e873a414482cffaddc5cec66f : FAILURE in 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4494baddeb34f92abf563ec67c1aeaf : SUCCESS in 1h 49m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/332b2962a2b34095b98e04520adeda56 : SUCCESS in 1h 49m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fede724c427446ad8f748e5060b8b2de : SUCCESS in 1h 18m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ef7c3265c3b499785fafd52657d3991 : SUCCESS in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77033d8c1c46f266b92930bdd382f884c857ebd6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:20:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"414f15f30cab79ad063ba44b88dce5460ee3068f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 12:16:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bf78dbf270ce2dd63a66306940c807ac5ed636a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 12:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c7cb31de355f5fcc9e25765e7a15e82f3169950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/885a80184f79445aaa5397e99a13c1fb : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87276812c0e04fbc9287064f1c938a94 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/384f32deadde47469dc951ba6d4aa0c9 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a707fac8a7474e8f9d8a7799f253767d : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63a042896b29488dadc22f5f0cc7b091 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/407383a19fa349f6a26328e311b966a0 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4845c9a9d77f4e61a0a63a199ed150ea : SUCCESS in 7m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd8985c3ecdd4349871628ceecd63d3f : SUCCESS in 8m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e7de40b449334e96b3857e2f69eab7d1 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81486c63c6804fbe8128b21bc1280eda : SUCCESS in 7m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c24e954c8984cd6beaf032cda7d53c8 : SUCCESS in 1h 02m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1ed862a786a54bce8b5dc755fa202eef : SUCCESS in 57m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c6f02f4a6aa647e1b1b4a57181b8aea6 : SUCCESS in 1h 13m 58s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d169ae1f07854b409b87ed2b0765f0cf : SUCCESS in 1h 14m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cafffb0cefbb48528500a1425e75c3e2 : SUCCESS in 38m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/30995ce6a98b449581921b5e80097fc8 : SUCCESS in 45m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4556f97ed987464681373d9faa725c16 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/52baf43b354a48dbb366a9d19fcad33e : SUCCESS in 1h 27m 39s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/31d22c47fae341bcb7e27b24e6842205 : SUCCESS in 1h 44m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/122e5b6e44134d22b6cabf86c2734db6 : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"da7712edd06547c17078b219e3e1642b0f5a4ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:08:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c62e233fa2deff41f3f6f58742be89d550f75dba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 14:12:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4bdf54cd0b40ed9f9ded7188b00921 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0188dc68818244febbdc3ef0cd86b4c4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f4b5ca2c28547ffe5436f5af4257faee3f22758","revisions":{"0f4b5ca2c28547ffe5436f5af4257faee3f22758":{"kind":"REWORK","_number":1,"created":"2019-11-21 09:25:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/76/695376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/695376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/695376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/695376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/695376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/695376/1"}}},"commit":{"parents":[{"commit":"29e9e2a3c30a91525d97fcb231898a9289297639","subject":"Merge \"Improve the error message for bad pkcs12 bundles\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/29e9e2a3c30a91525d97fcb231898a9289297639"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-09-23 18:49:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-21 09:24:35.000000000","tz":60},"subject":"Fix controller worker graceful shutdown","message":"Fix controller worker graceful shutdown\n\nFix an issue that prevents graceful shutdown of controller workers.\n\ncotyledon.Service.terminate function is by definition the graceful\ntermination function and doesn\u0027t have any \u0027graceful\u0027 optional boolean\nargument (https://cotyledon.readthedocs.io/en/latest/api.html).\n\nBecause of this error, message_listener.wait() was never called in the\nconsumers\u0027 termination functions, so flows could be interrupted before\ncompletion and could leave resources such as load balancer in a\nPENDING_* provisioning state.\n\nBy default cotyledon.Service terminates the server after a timeout if\nthe worker could not shutdown itself gracefully. The default value\nfor the timeout is 300 seconds (set in devstack plugin) and can be\noverriden using the graceful_shutdown_timeout setting in octavia.conf\nThe default value will be updated to a lower value when work on\npersistant taskflow will be merged.\n\nStory: 2006603\nTask: 36770\n\nChange-Id: I3f776bd018246897c9a889699a2d0ecbbfbb7098\n(cherry picked from commit 215c4c2284209ff46dc62f7cc04432cb2c9bf39e)\n(cherry picked from commit 0fc1ea4789eaf743e04db31d366b64a34172df15)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f4b5ca2c28547ffe5436f5af4257faee3f22758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f4b5ca2c28547ffe5436f5af4257faee3f22758"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
