)]}'
{"id":"openstack%2Foctavia~695589","triplet_id":"openstack%2Foctavia~master~I48315f8f293811e1d99584ea36da05c4211cf275","project":"openstack/octavia","branch":"master","topic":"695589","hashtags":[],"change_id":"I48315f8f293811e1d99584ea36da05c4211cf275","subject":"Fix load balancer update with provider filtered params","status":"MERGED","created":"2019-11-21 21:52:46.000000000","updated":"2020-03-20 08:09:42.000000000","submitted":"2020-03-20 08:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695589-1584691591883-2664e5eb","meta_rev_id":"eab4549198a966cc435544523a2484c915b45471","_number":695589,"virtual_id_number":695589,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 08:06:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-02-27 07:16:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-02-26 17:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-03-20 06:32:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-03-19 19:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-03-20 06:32:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 14:57:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-02-01 05:51:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-26 17:10:20.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-27 07:16:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-03-19 19:07:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-20 06:32:18.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-03-20 08:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85ba56adda3965dc43f78f2e67eaf3b64f661458","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-21 21:52:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"676ec002ba3dda9fe61ceb896df4c1e88278b607","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-21 22:03:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17a868222c52a73e2c1138e0c88e048ed5fd3c57","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-21 22:05:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e83a3738ee6c91e78162ca908d953b91558e2167","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-21 22:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d42c84f44758bef1872b31a792b4d20b47a9eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:37:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c25f5ef08f664be79e63bf037e8c801d : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff7322d178b47248e2f23b70989d124 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2efdc1e36af4d9f8d17c0f8440cd6ef : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf4078fcdd38463eada07b5f1e3379ee : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96e05c8aa40849daa4faa028864ce9a5 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f02af91ed64d4ad09244923c5db596b9 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74c95dd67ea430bbe205e49e2fb7615 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6362a9c962944e8c953c8e82b803cda1 : SUCCESS in 6m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/11bdfb5020b74363ac6bacdd9a6c46de : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05f24c38f5e64c13be8fa64cd7a517b3 : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d76da0a59f5e4ef081fb003a3ee7ea53 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbc2cf37e46a42df93eeff36cdce4d2d : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fcaab4ee51e04643bb1ce8ca03ef5620 : SUCCESS in 45m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/89da4c9b16a34639b90663572edb26e3 : SUCCESS in 44m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/727d1300400c419d9d758c892bf81c80 : SUCCESS in 1h 45m 28s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ad033fcc6a1243e88fda5a608ba3a552 : SUCCESS in 1h 42m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d4e1b042540542abb6781947b2b5fc3d : SUCCESS in 2h 30m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc8fa964d55e4e039fad1d06a8fd22b5 : FAILURE in 32m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f034be45222945128c364cc602fcc6c6 : FAILURE in 2h 08m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65ef5ba4148a4bc6bc619ca877589c15 : FAILURE in 56m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4db484eeb48b4d998435810a711fee4c : SUCCESS in 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c9b9e6d5905443bc9f2885116236e246 : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d76a393db2164a8db18dc6a90dd29ae4 : SUCCESS in 55m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b86018fc71874976b96cb2e07120f8fb : FAILURE in 1h 12m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2288a0d231f440fdb83378fad02088ec : SUCCESS in 1h 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd6a46a5a515c10536a6f9bbc8c74d59f3d98a68","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 11:36:06.000000000","message":"Patch Set 2: Backport-Candidate+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f993dd714cfff7a25abf8185a9f83b5a7e238a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:28:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0e08ba71782461591e5269362b07199 : SUCCESS in 7m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d04e13a3aa24b75bbab8e459118deaf : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad75cc94331d429080aca1dc748faf9b : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2acabfc8c171466a8fd96a3554f2b505 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47eccbc66533452a85600ec7be0df302 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24debadd3c2f44019703c81efa5a063c : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3ee445110e4d6fb95b3c51b2cac94f : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f4cf5340d0a46ad90cb81b0886754c1 : SUCCESS in 7m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6439fc63153442fba74d2442aa95d6c : SUCCESS in 6m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/39a34c41610444a181328a3be3362041 : SUCCESS in 6m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6830242ed93f4c9c9bd349c34b68cbb9 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c74b88c4fb764853a85527503a2dc98d : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23e4dfe00d1c4b018bf445809e0a4985 : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ce4a7b24a8d44c058d4333188f7b84d5 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0691e0c48cf4b2680639f35cab4e443 : SUCCESS in 1h 50m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bb4976ae1cfe407e8bf2691bd93894bb : FAILURE in 1h 30m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/20623da6c48c4446a4b5298703e48602 : SUCCESS in 1h 47m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/39f78b5f5d794e008d9b2c9f8692110f : FAILURE in 33m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c201e3a682fc403fb3e66159f660a9e2 : SUCCESS in 1h 28m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d8fbbc264be040b897d690417c07b39d : FAILURE in 1h 24m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/65e301e1c8a74ec698a8793b2cde3488 : SUCCESS in 49m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fea69e7442144116b54aa61895a8e197 : SUCCESS in 1h 26m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6489dfeb57d43579fe0891377bcc8be : SUCCESS in 1h 03m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7ef2380dfe894c5cb5330e921d521d90 : SUCCESS in 1h 00m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5edcc1a6086f485e8fe3afd787c29af7 : SUCCESS in 1h 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"804d57e75911816489fa4748527587cb27449ee1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-22 18:05:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b271eeee109a2805ec88ad39aed145ecc47be1c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-22 21:01:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ed4974c9213c439f6edd3a3b922fbe7cb7a8d02","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-25 20:50:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40e4d4216a28a31831aba4a7afbae5e194e54c67","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-28 19:19:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa53eefe28e93fdbf1e665b079c780a4835329e9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-28 19:19:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4347b75765ed5ee6e861a833d1343ed75afad23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 21:25:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5361eb5e2e64f1ebae35a04e008ecb7 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19b114de3db04a4bbd793040ba911cea : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d4c94a426f448a87d2f454c7d721c1 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bbb258152fa4c8798d4bc2d6bbcfd1b : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2909f430d67432b9893baaa3a9df58d : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/697c45bca18a449f923ce6856bcc3a05 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47dee1557b134e809b5e895a91b76dac : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78630931ac78488985a33a3bee57ae6c : SUCCESS in 7m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1bfb99219df74b5db01af71cd70ab0cd : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1524fa7caf5f433f8dea213f2c6c21ea : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/88d41a44d0f84c61b980fa1c33d0ade1 : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5acd5cfbdb84c1cb6d01b083584420f : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e32e21fc1c584ec5b14d7766a7b45b56 : SUCCESS in 38m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aeb28efe76b544f0b269e4f05ddb1099 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b12ef78d732043bd869d69bf61220de9 : SUCCESS in 1h 36m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5d43c3e8cd424d30bd25be623c875236 : SUCCESS in 1h 40m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0ce3dd6380224306a1f05e140b1876ff : SUCCESS in 2h 02m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a2d84703f864dbf97dbc1f19af646d2 : FAILURE in 29m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/56ab8d6bc4fc43c08d186a8bb6ede5b1 : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb254052a15e4dd6abc52aa3ebd08aaa : FAILURE in 51m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83376e2548aa404a8d7dfbe9c7878b59 : SUCCESS in 54m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd0529bf73e8463291d9a8ad55727957 : FAILURE in 1h 30m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1aefa604b9e94d5a93f63d04c101b7a4 : SUCCESS in 1h 01m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/37e13548d2734e9da4cfe92408774b46 : SUCCESS in 1h 04m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bd9806daad7844a8b26bb1d7fdff27b5 : FAILURE in 1h 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"311c552670d3e0f125f257aa0ad3d7d891fe1c14","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-13 22:30:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"98a48799420493643c5b41775cee0e8c2ab6a9d2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-22 14:57:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f5969bc2f3a917d6e4e58dbb93efec8a70cba4d4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-23 08:09:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"de3155811c4a65c35bf987ee673efb418144a11d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-01 05:51:32.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"17ee675e83bddf8fd25fbb9141ccee5e139ab9de","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-02 07:09:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5c218e7276ee469bcae368464e0d882f942e9ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 07:09:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4da664b53c80e823ca8e1f0be71d7be0e35effa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 07:16:09.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d89917ad4242eebeaf551f3a628e4a : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc7488f232746f9b58b81d749ec4c12 : CANCELED in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f3b3050cb549b39ade0f79b469131f : CANCELED in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21675c86e4944650870c1dd1345e9ab9 : CANCELED in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7c6cfc10e243d0b4ce9f3d075962c4 : CANCELED in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc09e68de234c59a015381f96d098d0 : CANCELED in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1342e5e55fa46a596aa37116c9fc04c : CANCELED in 3m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c64e00ee89a444c0a4c107e2f08faf11 : CANCELED in 3m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a359dd1a45414d998a231df53973d441 : CANCELED in 5m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec604a73195f4ec8950b04772a84b16f : CANCELED in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"643d047b2ebea5b9f823a40c56caf2b0ec1ad392","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 14:46:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e5098b9716ff32b7197b60ff92fe551ba5c7f4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 20:48:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69ee6b332a7f4a04bbae6fe6944b2ef8 : FAILURE in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3fe6a08f80b4e50b50bf6b3278c95a8 : FAILURE in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d339a506ed4c05b3e48efbd6480676 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63a19ee09a964419bf874635caf0d1d4 : FAILURE in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78658b6357d248538b40ed0d1eac6032 : FAILURE in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11f2fc6be352491884ef962f08f2e89d : FAILURE in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c5b0b4578d4e8cbd7a9533a8acb64c : SUCCESS in 15m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9ec86e270704e4f99dce990bf48ac74 : SUCCESS in 10m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7533418f194044d4b80494aae110958e : FAILURE in 11m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/526980c62e3d446297bd4391ae4ca01a : SUCCESS in 10m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5136f3c241a44aa5b9bc0fdf720a9ff2 : SUCCESS in 10m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee7f8f8157b944fcb6066dc35d2134ec : SUCCESS in 47m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/165240a2313244bfbee8ef054e0209ad : SUCCESS in 1h 36m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3bd578d59bf247a9bd728a0b5955314f : SUCCESS in 1h 09m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5476864329d24e63b0068707e011a8da : SUCCESS in 53m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c147028c00704318ab86f5de5e05d80d : SUCCESS in 1h 01m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0e0de283eaaf49b8a4d4c99d2b4373ca : SUCCESS in 1h 31m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f51622ea9aa3401d96a586a26efe7691 : SUCCESS in 1h 40m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3310cead50dd4b8782369c13a77533e6 : FAILURE in 41m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d43116caf6c049b29209eb0bae0fd904 : FAILURE in 1h 25m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"065abdc5d64ace7d1977d76b9432fd26d3043d69","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-26 03:23:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"db167042a3b70bc96f05585983f5cc98bf672da2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-26 03:23:34.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bdc99016b11b345d2f5f43b9ad63d6faa36a03d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 05:19:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59aa79b888c94b6ea5f2edd3b67422f5 : FAILURE in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04bc3da7514d4605b36bf7c90975430d : FAILURE in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1b81debb084fe181ebec125369ffd0 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75f13402e13478f843263a82d72597e : FAILURE in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56e39858f6034a26bc0c26aea0208fe8 : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41c711e08e0d44d284d39ebb9c0dfa6e : FAILURE in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c66df1b8dd64fbba316e9f7776dff7a : SUCCESS in 16m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c04c241240b344bd9e9b9513307be28c : SUCCESS in 7m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e77457753849482fb6ad6c67244f51a4 : FAILURE in 7m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a0ee4a9464a0481db05874653ab5554d : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea35be1e4b37408086a69c231a12d5d4 : SUCCESS in 8m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/412f7d65d037425e821a3e24d172ef9c : SUCCESS in 48m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5716003ec66e42acba1047bd290673fc : SUCCESS in 1h 46m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58c7081a185d40d6a1b7b006c9e0e736 : SUCCESS in 1h 05m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/69ba29e4bb834a5d9416944032eab034 : SUCCESS in 1h 05m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b9e656472f44c99a6c072facd17052a : SUCCESS in 1h 05m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/322cc2bd913249fba941f53c22ca6cee : SUCCESS in 1h 34m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a4bd68fcc37a444c9665371d2edfd97c : SUCCESS in 1h 45m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42bcae61f27e4884bbcee686dcaeb5e6 : FAILURE in 1h 14m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/674aee861fa24433abaa049e55758779 : SUCCESS in 1h 49m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08149d072de4063f89e1ccd08c96d8b205dfa1d5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-26 08:20:01.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"833d98b22efbd99bf62e86aeb37f0f3cc192490c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-26 17:10:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ab35c4facf21356ec9acfba6af556b4625d87623","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-26 17:10:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"35ecc7c2d406323f895d468ddb70fc436a55381d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:54:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b1d965840b84c88861a5024ecbab89c : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b2c0c79b00e4fc8b6fb1c7f80020766 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cddeaf6281489bb5bec224756cf661 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec6bda8554a4183a307807a97ce5dcb : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccdffc4d0f794105badb247292a0e0e9 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a3de02d54a41c5b828f3baa3a010e0 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5986d3a92ac4075a5d433ce1831853d : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ea754f094d4ddb8bf119ef73d0366b : SUCCESS in 8m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/63aba255bf9e41b08b5fa4cc8b8ae590 : SUCCESS in 7m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5d4aaff374eb4a2985bdb5913c8a5ea0 : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f48f4fc625534766827374a5c6773daa : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3a4de9dccc546b09823d6a68f2c54c6 : SUCCESS in 54m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e5041ed1d8340ada3df38c65730747b : TIMED_OUT in 2h 34m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e8bb8297b4745c5b8026de23835de6c : SUCCESS in 59m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ae9edc2ef1dc499d96c36a92266126f6 : SUCCESS in 1h 01m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/495bf7b09e7646bdaaebd5f50a66c733 : SUCCESS in 1h 11m 21s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d7ca69039414a40b454de5d4e2f48c5 : SUCCESS in 2h 00m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ae0cc8a4e504dc4ab2dfc54d2a52a8c : SUCCESS in 2h 07m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef917d8859894a68a6b8a114c8723ec7 : FAILURE in 1h 07m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f6e9eb6513b40c786decf77e980ff77 : SUCCESS in 2h 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aecb5645b8e5f995842aab7fa64792bf0840b6dc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-27 07:16:47.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b71fc8a321f091d596a63de26a6bf95ce65b65b2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-27 07:16:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cbc79cee4f1e13e0dd416d15b60ad319d05f200a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 13:09:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62693e59ce4a4bc48fecbdc74bf687cb : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab8f8b33147442c5b980764d89c4084a : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bbd4dbd9ae14585ba7f07d9d834504b : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae0dadd722847a890f63f998af4d598 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cbe3e2f9e3a482dbcf968fa5f77be6f : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99a53cba2b99410693ef0b273ebfe259 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f15c547aa04b898b16f027f65896ef : TIMED_OUT in 32m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe34ebc6e5a44f089161e463a65d7ec9 : SUCCESS in 21m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f77c63f0ca344073a5041f046a67ab46 : SUCCESS in 30m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b29a0cb9596e451f8fc29e2f144e364c : SUCCESS in 27m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d500b1a41ce046ac9bb4837b0052e0b6 : SUCCESS in 22m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80278e9dc2134df8a295d68303d1789b : SUCCESS in 1h 21m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0eddcd9a33f4e568d340dd7d6dcd28d : RETRY_LIMIT in 25m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d3607b96dbb41d6a50b29a3423c9e14 : SUCCESS in 57m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7521893a45b843108e3a547a0477701b : SUCCESS in 55m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0e6af7619614cf1ac5a5b6135cc4ee9 : SUCCESS in 1h 03m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5093265a48ae4367ad02776f2d899b14 : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c01f65519604f04b4988ad8c748c64a : SUCCESS in 2h 04m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e11b33dd5bf44ae986b1a6627d4ded29 : FAILURE in 1h 05m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/98e2193d82b2401c880261f71e123c54 : SUCCESS in 1h 43m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92cf9a96b5a29d857a0859cdfd89a88e74f1ef1f","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 04:51:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b3922747fecfbcd18291a38c1b29ee7ba6caf3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:09:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c79531d196d4d3da1818a20617ca731 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d6308c27964dd69f36ebf91d1306db : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8a925ba1cf42b388dc19f051342ff8 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71ae58ce14c8467f8a3e81d9d3c9d131 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2311ebff41734d5c81b7d6c890a3cd64 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60e28d692cf45e9bf85d1c3bb267a40 : FAILURE in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ddae94f04a442687b7dd4e6e500eda : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b5d0cd193844e3a1ba83b701757806 : SUCCESS in 7m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0a70b348db8b449e94d0f060f17a96db : SUCCESS in 6m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4bcc36ced504e3292917f58dbb741ba : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13da887fbfe548139c46bc4586f20630 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4bf8bb1e0fd48b5971b46c934c2afa9 : SUCCESS in 45m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/812b59852255448d92f1624a1be2b31b : SUCCESS in 1h 51m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee301c9777a643ce9eb129b5c100c04c : SUCCESS in 1h 03m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b7f47236e9f0477c8a4d9c16215809d4 : SUCCESS in 56m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a191ac8354d44d0ace5712dbfbfb30d : SUCCESS in 1h 07m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc6b3fd804104472b34d114f55e4a131 : SUCCESS in 2h 01m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f3f0ab9cafd4e92a39487e953fc4926 : SUCCESS in 2h 12m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81fe7713cb324dacbb79bac9275f5286 : FAILURE in 1h 03m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2cab2a7fb2364629ac26dd27303f5097 : SUCCESS in 1h 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c704ef9bf7265166d04691ff227f147fd96cab83","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-19 19:07:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"979d16de5a0aafefc4aedecd39bd2deeeae4a38a","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-03-20 06:32:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bf7685187b836fdaaf7311bcdf8ad24ec5fdaf9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 06:32:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a3a41f747a71dded5489fc683145c23905edf3ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:06:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcaa571cf19f42ff84d37bbb75b12b0a : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da05f926cf694804addf2bb54899734b : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ec96cdeda3d498fafeacff980cd36f3 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4906bdf1760843a0a90d949bfb1a75c6 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24df0f342d274bf190daf97b3a222b41 : SUCCESS in 14m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c90ff14395444e0886518ca01017191f : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28a74c20e0184191b6aa8e9cafac2a97 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c2eeb908332c40ec93c46db2f92d89c3 : SUCCESS in 45m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d4093bdef664e2b9a8d9406937d8de3 : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3faba3b099ce4cbdba5efe9ec8e9c59d : SUCCESS in 1h 01m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/46dc222f0ffb4a7bbd818cf9dc23fafc : SUCCESS in 56m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68898f0206624756854a26eb393a27ef : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":5},{"id":"843a6664716cda8abd04126c674b4c6eb84c94a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:06:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"61c6e2173233e4939fa42bcf531765a588d455e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 08:08:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792b7720b40a4964ac2f137caef9cf5d : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ce5d91e3ba407e94637ef9afb0aa4c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":5},{"id":"dc3d8dd27cf5130f666af64231bd14f2c7f404fd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 08:09:42.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a491285ce4413aff868a6a17151e3c22227e9c2b","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2cfcd71a8659583d91cda3e488c60ded078c9677","revisions":{"5df77fa69cbfa06531cdf08de0668773b45214e2":{"kind":"REWORK","_number":1,"created":"2019-11-21 21:52:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/695589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/695589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/695589/1"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:52:22.000000000","tz":60},"subject":"Fix load balancer update with provider filtere params","message":"Fix load balancer update with provider filtere params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5df77fa69cbfa06531cdf08de0668773b45214e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5df77fa69cbfa06531cdf08de0668773b45214e2"}]},"branch":"refs/heads/master"},"32b6c4c40aec3b2e52e6f71172253a7ae87d3a5d":{"kind":"REWORK","_number":2,"created":"2019-11-21 22:05:10.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/695589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/695589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/695589/2"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 22:05:06.000000000","tz":0},"subject":"Fix load balancer update with provider filtere params","message":"Fix load balancer update with provider filtere params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32b6c4c40aec3b2e52e6f71172253a7ae87d3a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32b6c4c40aec3b2e52e6f71172253a7ae87d3a5d"}]},"branch":"refs/heads/master"},"7d8363e91d25356c216eb566556b3ff08f8e3aca":{"kind":"REWORK","_number":3,"created":"2019-11-28 19:19:38.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/89/695589/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/695589/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/695589/3"}}},"commit":{"parents":[{"commit":"3c05cce89905a5dcc2caac3f6d81e9dc7b82261e","subject":"Merge \"Fix update API when barbican secret is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05cce89905a5dcc2caac3f6d81e9dc7b82261e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-28 19:19:24.000000000","tz":60},"subject":"Fix load balancer update with provider filtered params","message":"Fix load balancer update with provider filtered params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8363e91d25356c216eb566556b3ff08f8e3aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d8363e91d25356c216eb566556b3ff08f8e3aca"}]},"branch":"refs/heads/master"},"4850654df865e12d776fb5455057869fd459ceb7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-26 03:23:27.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/89/695589/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/695589/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/695589/4"}}},"commit":{"parents":[{"commit":"e0ae8a3cac837beda479e23f6015ee6584093e9b","subject":"Merge \"Fix uncaught DB exception when trying to get a spare amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ae8a3cac837beda479e23f6015ee6584093e9b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-26 03:23:27.000000000","tz":0},"subject":"Fix load balancer update with provider filtered params","message":"Fix load balancer update with provider filtered params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4850654df865e12d776fb5455057869fd459ceb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4850654df865e12d776fb5455057869fd459ceb7"}]},"branch":"refs/heads/master"},"2cfcd71a8659583d91cda3e488c60ded078c9677":{"kind":"REWORK","_number":5,"created":"2020-02-26 17:10:01.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/89/695589/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/695589/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/695589/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/695589/5"}}},"commit":{"parents":[{"commit":"e0ae8a3cac837beda479e23f6015ee6584093e9b","subject":"Merge \"Fix uncaught DB exception when trying to get a spare amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0ae8a3cac837beda479e23f6015ee6584093e9b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-11-21 21:31:26.000000000","tz":60},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-26 17:09:31.000000000","tz":-480},"subject":"Fix load balancer update with provider filtered params","message":"Fix load balancer update with provider filtered params\n\nWhen only setting tags to an existing load balancer, the amphora driver\nwould try to apply QoS policy on VRRP ports even if no policy is\ndefined. This raised an NetworkException that led load balancers to go\ninto ERROR.\n\nTask: 37589\nStory: 2006922\n\nChange-Id: I48315f8f293811e1d99584ea36da05c4211cf275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cfcd71a8659583d91cda3e488c60ded078c9677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cfcd71a8659583d91cda3e488c60ded078c9677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
