)]}'
{"id":"openstack%2Foctavia~695941","triplet_id":"openstack%2Foctavia~master~I57dd2151bce0c73d7835078e53961cc3c2f7a7c8","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I57dd2151bce0c73d7835078e53961cc3c2f7a7c8","subject":"[DNM] Testing current state of gates","status":"ABANDONED","created":"2019-11-25 16:14:48.000000000","updated":"2019-11-26 05:47:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8175cebe9315fc992d5787ed28c6a078e80f7ac6","_number":695941,"virtual_id_number":695941,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-25 17:17:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 17:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09eca4fc5621d90753229f1d2d403e72309ab46c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-25 16:14:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34f179ec75032ac94c3cc2dbc58cdd080744d2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:17:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/562b5e15c44d4f77b6b382faed1c49df : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6eceb2f05274357a6551863d28fb55f : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ad41cfbc764af2bb29a30cee504b85 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3b304823c3441d3ada07a7d2e5e0c97 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d04d041d5046cda0a377d63b14a1e4 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adf1cc39a677498284d5ffd301c2ac28 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1d1b8a95bd43e69f09a4ca43650e61 : SUCCESS in 10m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08102ce1dc8947c2ac348bec7f55ec68 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a6a023240d7f4aacab1ed263d49991fb : SUCCESS in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d3c1cd66861f4261ad017c02c33daa5f : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cea48cb71fa847a1a83ae9c42347a1d7 : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/581741ed67704ad8ac5039a43f0b8080 : SUCCESS in 56m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/db6327b1c4a54f168a1cf4257d10ee03 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a52698b400a46f68e3863b7e84a6ceb : FAILURE in 40m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/75481bf52bd841288644c0d73a40240f : FAILURE in 32m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5322c34664d9433f910c34305c08322f : FAILURE in 44m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8134bbbf4a8246d0afdf2f9f9341ff08 : FAILURE in 34m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9609dcfac1d4aa2abdafb74044d58dc : FAILURE in 45m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2dc31f0866c246779340db80079390cf : FAILURE in 39m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/46925539a3e649a5ae051c36d61c3914 : FAILURE in 45m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a815086b69ff44fa8e2bb06fe9644e8c : FAILURE in 33m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88a139cfbc2e4c309f51b4ecb5b74854 : FAILURE in 42m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e87a6452ba2347b78a4eab23b4fe26ca : FAILURE in 33m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d01768ad3628493097e00dee69cc1a17 : FAILURE in 47m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8175cebe9315fc992d5787ed28c6a078e80f7ac6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-26 05:47:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1c905b715e913b5921220512ef5be16b0bf6e43","revisions":{"d1c905b715e913b5921220512ef5be16b0bf6e43":{"kind":"REWORK","_number":1,"created":"2019-11-25 16:14:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/41/695941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/695941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/695941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/695941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/695941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/695941/1"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-25 16:14:11.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-25 16:14:11.000000000","tz":-480},"subject":"[DNM] Testing current state of gates","message":"[DNM] Testing current state of gates\n\nChange-Id: I57dd2151bce0c73d7835078e53961cc3c2f7a7c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1c905b715e913b5921220512ef5be16b0bf6e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1c905b715e913b5921220512ef5be16b0bf6e43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
