)]}'
{"id":"openstack%2Foctavia~695947","triplet_id":"openstack%2Foctavia~master~I25ed1a60c115e24763176af5f96633ba6d140d95","project":"openstack/octavia","branch":"master","topic":"695947","hashtags":[],"change_id":"I25ed1a60c115e24763176af5f96633ba6d140d95","subject":"Fix gate issue with DIB and Migrate grenade to PY3","status":"ABANDONED","created":"2019-11-25 17:11:03.000000000","updated":"2020-05-19 16:33:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d3ceffe514ffeedaf76c283960683cc2ce738b4","_number":695947,"virtual_id_number":695947,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-26 22:52:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-11-26 21:47:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 03:36:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-26 14:44:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-26 22:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6225e854b90465ebb72df6d38899913a7411da95","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 17:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec710b3b573cc9130a7d415df4782070829153c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:11:19.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  expected a dictionary for dictionary value @ data[\u0027check\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n  #    templates:\n  #      - check-requirements\n  #      - periodic-stable-jobs-neutron\n  #      - openstack-cover-jobs\n  #      - openstack-lower-constraints-jobs\n  #      - openstack-python-jobs\n  #      - openstack-python3-ussuri-jobs\n  #      - publish-openstack-docs-pti\n  #      - release-notes-jobs-python3\n  #      - octavia-tox-tips\n      check:\n  #      jobs:\n  #        - openstack-tox-functional:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^releasenotes/.*$\n  #        - openstack-tox-functional-py36:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^etc/.*$\n  #              - ^releasenotes/.*$\n  #        - octavia-v2-dsvm-noop-api\n  #        - octavia-v2-dsvm-noop-py2-api\n  #        - octavia-v2-dsvm-scenario\n  #        - octavia-v2-dsvm-py2-scenario\n  #        - octavia-v2-dsvm-py2-scenario-centos-7:\n  #            voting: false\n  #        - octavia-v2-dsvm-scenario-two-node:\n  #            voting: false\n  #        - octavia-v2-dsvm-py2-scenario-two-node:\n  #            voting: false\n  #        - octavia-grenade:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^octavia/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #              - ^setup.cfg$\n  #              - ^tools/.*$\n  #              - ^(test-|)requirements.txt$\n  #              - ^tox.ini$\n          - octavia-v2-dsvm-tls-barbican:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n              voting: false\n  #        - octavia-v2-dsvm-scenario-amphora-v2:\n  #            irrelevant-files:\n  #              - ^.*\\.rst$\n  #              - ^api-ref/.*$\n  #              - ^doc/.*$\n  #              - ^octavia/tests/unit/.*$\n  #              - ^releasenotes/.*$\n  #            voting: false\n  #        - octavia-v2-act-stdby-iptables-dsvm-scenario:\n  #            voting: false\n  #        - octavia-v2-act-stdby-iptables-dsvm-py2-scenario:\n  #            voting: false\n  #        - octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7:\n  #            voting: false\n      gate:\n        fail-fast: true\n        queue: octavia\n        jobs:\n          - openstack-tox-functional:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py36:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-noop-py2-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-py2-scenario\n          - octavia-grenade:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tools/.*$\n                - ^(test-|)requirements.txt$\n                - ^tox.ini$\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-xenial:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-bionic:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos7:\n              branches: ^(?!stable/.*).*$\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3dd0ab327436bad9754a1ad811b42efbe7f9f001","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 17:12:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08197ce66ea18824c3e4f436212b74d70f1972cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 17:21:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"420f4af4e11e340f6dbe5e4cda91417a0c889e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:02:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7d2f9cc1ce1480580a95c30804889ba : FAILURE in 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fde5eb6fdbb3dd5eb8fc12dc24169064d1da826d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 18:10:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"093d2306fb1e22fa376510552fd46d1ef868187d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 18:18:09.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0b0591dfed224b317700654fdbf9eb54b24c697a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:37:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b635fe0cd1af44d788b92cb26854d132 : SUCCESS in 8m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad27987f114e4550808cd052c02eed39 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c66a3f951f24c5590256530c2937725 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42a4fe82db45402a9ed0addb62b322af : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccab1fa0a3e245a799e25751ca690a70 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73673e38c3394a0089f82fb5c68ebe81 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e13006b8f8b4eb68cec9a51dc88e8e4 : SUCCESS in 10m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5188f9e7b164033a91fcd384ba79632 : SUCCESS in 4m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/53a12a9755f74ceb86258857901aeba3 : SUCCESS in 7m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29d124a9fc764c81a423c642bc37f105 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56809dab595b4a61ade2e8d5f6b497fb : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/691134fe9f194551beaf73bae1b85bac : SUCCESS in 58m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/01beaf725e614484833bdba8ec082e1d : SUCCESS in 44m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/073f7ddc3baa4676affabd692aa34ad8 : SUCCESS in 1h 33m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8213eef49181491bbde8837e3e4043a3 : SUCCESS in 1h 49m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/15ab391c5fea468689404cea8a50dd0b : SUCCESS in 2h 16m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32a79dfa70554693988898d1c1ffc997 : FAILURE in 34m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/33707a7235604b668e22a85087fead79 : SUCCESS in 1h 48m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf31ce223f874ca48f9baa924cc4bf68 : FAILURE in 48m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f5787ea8b38421b9a226816dbc482c6 : SUCCESS in 56m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/27998f67fbdd4b47aefaeed1dd4dd6ec : SUCCESS in 1h 34m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f360b2f61a741f7a529cabc15a752b1 : SUCCESS in 1h 04m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dfefe029b00645e2ac1247940aa401c6 : SUCCESS in 1h 24m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2fac0a26f91e4f70af4549ee2a22a613 : SUCCESS in 1h 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"da4ec11083c43064b85c6c72dccdea0f98e82315","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-25 20:44:50.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 3d28fd56e8857efe47f39490dad8419c4ac0e5ea","accounts_in_message":[],"_revision_number":5},{"id":"8bdc00ac5d7d7a47d5e18d543c6ab482cef82b12","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-26 03:36:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"dac6e075180baf0e0178997d850dd96fe7bc0286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 05:41:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c24aa2a5be794cdd996b4f1d9da5d1c7 : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd634a58c4ae4f5e868168617dfaa577 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1762be07624b46718fbdfd8f03ad0019 : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4ca1528ead14e918e863162120cf4bb : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fc2d4cfd804408e91579d0401915d1f : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/152ea6a6d55e4d6cbd0abf00cc480b55 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277be38be85f4d5192f8159e15951fef : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4e86c030bf7844ceaba4e53c65180df9 : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99a5b2c81fc94b68925cab24ff26726a : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7200575a7407475fb3a5d6984e2d7ef6 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/637717820f174eb9963759467baf7fd6 : SUCCESS in 6m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6613156d97414221ab2ea8c81598f93c : SUCCESS in 52m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/36ee0e1bf0764e5e924a1a06b7d0101a : SUCCESS in 48m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd6437de31d04ab18d7046a51a8f2f77 : SUCCESS in 1h 28m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e3e1cb53569a4441860551bd6e289a1c : SUCCESS in 1h 18m 50s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d820a7e2a9d74f319591a7dc80818e42 : SUCCESS in 2h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a3f03edd96347d8afd2159ddcf23722 : FAILURE in 27m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/306fd2c6fcf34125854c319bd31763a0 : SUCCESS in 1h 06m 52s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77b9004d463f472796638112c2cd6edc : FAILURE in 54m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6d208deb5d5444c9f103272d72a5614 : SUCCESS in 49m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/288f8222d6764df498f4ca21b6e6abe9 : SUCCESS in 1h 27m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78864187f61547caa99593f74f68fb81 : SUCCESS in 1h 04m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/31ad868beb73475d8ce2265f1f468843 : SUCCESS in 1h 00m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e0fac62caecc44e9b6dd613087ac7481 : FAILURE in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"89b68dbef635f9e30e038cfc52d7fe8901da239c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-26 05:51:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cabdffed66d24df2f3d8730c2d0402e271f2b4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 07:53:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/301b93b63996429e8291371b8a268201 : SUCCESS in 7m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c43361ffc8ee4c68b1649f9b89eebae1 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c0074eb1f34c4bb5116482429515b3 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f29616a916104659b48ac98bb30e3a6b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e22330b0d6a4e799240e61088143e4a : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01bf1bf2e8f04013b5b01a9f96a80811 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c63575142fc492ba3db76f7ec5dc9f3 : SUCCESS in 10m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/733884c97fd9436c821f82350a31b9ad : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79c048f817b44240bd53658fe78cfb3f : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7158742de8164551bad63711b408cb38 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b3ad7b5ffd0489a879ec7cf909253ae : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c34528f662fd45b9807edea40da26d8f : SUCCESS in 41m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f7207bfd814f4b10aa225bd5ccd9acd6 : SUCCESS in 45m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e13b1665b5f46c1b4c5675d1cc2de34 : SUCCESS in 1h 36m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b9cf281fd2d24755b32ea3b02c4e0002 : SUCCESS in 1h 55m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6d25e8b49f0f42069d9fb7024ad508b9 : SUCCESS in 2h 01m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3e74be0690d4cb5ba1c486db9fb9f18 : FAILURE in 32m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/aafc3f13ecf84935a04790f699a19f40 : FAILURE in 1h 25m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d124c9dba8049d1a8a541419031c778 : FAILURE in 59m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/892bc1ee65d84dbfa7295b9848d3355d : SUCCESS in 49m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ad5771b906ec44d6a65d5cd872b01721 : SUCCESS in 1h 34m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f488437481804daa83fff0499a351d4c : SUCCESS in 1h 09m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1aeefc6f3d1a43e59745e47f190816c8 : SUCCESS in 1h 05m 51s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/767613d5ac8846ac86897c478f4620a4 : FAILURE in 1h 04m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4a6c8fc3cccc43055901b5dca0b73a0e370acb97","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-26 14:31:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9232e4547db6824a3a7b5ae293ead366c8dc91df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:33:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ae2b7686824fe89f6c5276bccfa1bd : SUCCESS in 8m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a7a6afe1cfc48cbbdebe2dca1651731 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6281a25fc71a428c964a400d6cda54e4 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b61357dd5cf64d148d8702d56f2ecaff : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a8b9d5cd0f40fa9a42d94b0578f801 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cded2fd883cb40c89a843ed1eed8c12c : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f15cc749ecf408cba0206092ad9d5a2 : SUCCESS in 13m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/672b699bff8749ba9146b75ca2c3a4c8 : SUCCESS in 5m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c53091ea947742ee8c82966b602d6ec6 : SUCCESS in 7m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cd7e45438e894eb6a5e99af0d29873ae : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a415ee983524537a2b5fac41de89118 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81d3bea349f84eb49c5cf3b6bde7f839 : SUCCESS in 44m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a01d6fad5ca2431083bfa061fcb3c68d : SUCCESS in 48m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1eec51406e314c248ff00ab9c4cd9466 : SUCCESS in 1h 51m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a5af67cc536a4432a192e45e20bed7ed : FAILURE in 7m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/63d5058b6ed74fc3a3348e26ad5ac168 : SUCCESS in 1h 58m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b068d8d241be4e9ca7b05f4beffdd00c : FAILURE in 38m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/24d9a52c1fdc43419f6c82c54f014311 : SUCCESS in 1h 38m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ab4856039334d94a0f2c09be9cb2c1d : FAILURE in 1h 06m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c834f19c67164d0eb85edc53d2be0f3f : SUCCESS in 53m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9dc35739cc1c4794bf34ea75e149303d : SUCCESS in 1h 54m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fcfcf9ca88948098d89bab6b8633492 : SUCCESS in 1h 12m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c195be921f9c4b64a489e1e834b8b4fc : FAILURE in 7m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3bc74fe482a14ef6a9d6b9616ebfb25f : POST_FAILURE in 1h 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"26e0ad60e08e4b085b49907eada0b471c8195b59","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 16:39:20.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"db1d344ef66975335c4e27c9b11396ceb25eb863","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 17:04:44.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c8f5f6a392558b9661f88bc1a1be21cddddfb898","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 18:33:32.000000000","message":"Abandoned\n\ngrenade is too broken to work around. We are going to make it non-voting.","accounts_in_message":[],"_revision_number":9},{"id":"c4ceaf55e3e8a6ffbfe6d9955c01f43dceae980e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 19:00:03.000000000","message":"Restored","accounts_in_message":[],"_revision_number":9},{"id":"e210ee51d07a38fea7751076b6f179fed2430919","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 19:00:32.000000000","message":"Patch Set 9: Workflow-1\n\nrecheck\nTrying to capture log info","accounts_in_message":[],"_revision_number":9},{"id":"141daad4868f67e78b27d8ecd07dcf335a2b3bb8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 19:04:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e82ff0291e854be0904f6efd6e2de85cbe642233","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 19:05:10.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"f9f9dbd05359c0409e6ee72d3b486d849aea4dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 20:26:08.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fff672698562463fb30ac21654e454e9 : FAILURE in 1h 13m 46s","accounts_in_message":[],"_revision_number":10},{"id":"f8680ee22c6fb181b81fdaf2179288df08c09900","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 21:44:42.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"1917f85597941fc6ef5d5257dca337439ff3b18f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 21:44:54.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"60b887404411a9abf277664bbf635ae5ac626163","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 21:46:26.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"498cf44fa21e1823540ab0399ee71caf9960c12d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 21:47:18.000000000","message":"Patch Set 12: Workflow-1\n\nDo not merge this.","accounts_in_message":[],"_revision_number":12},{"id":"9c1cda2dacfc15af2a9952994cd52e7cf2ee2a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:52:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/39a2a2342d6b46e39f3874a85e0cbef4 : FAILURE in 1h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"397256f79108b6740f8a5ab01e3ab1e941d50d10","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-19 16:33:34.000000000","message":"Abandoned\n\nThis was fixed in other ways.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6ab0d0de309c37c7a5958f13c2ec4f36a09897b8","revisions":{"80bc1f3fcd90f69ff77671813ea3572507611bdc":{"kind":"REWORK","_number":1,"created":"2019-11-25 17:11:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/1"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:11:02.000000000","tz":-480},"subject":"DNM: Debug for distutils","message":"DNM: Debug for distutils\n\nDo not merge, do not pass go, do not collect $100.\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80bc1f3fcd90f69ff77671813ea3572507611bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80bc1f3fcd90f69ff77671813ea3572507611bdc"}]},"branch":"refs/heads/master"},"9018e232901b4ede00125919765bc49c565a1a37":{"kind":"REWORK","_number":2,"created":"2019-11-25 17:12:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/2"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:12:28.000000000","tz":-480},"subject":"DNM: Debug for distutils","message":"DNM: Debug for distutils\n\nDo not merge, do not pass go, do not collect $100.\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9018e232901b4ede00125919765bc49c565a1a37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9018e232901b4ede00125919765bc49c565a1a37"}]},"branch":"refs/heads/master"},"feb39ad0f43de7daf5f0f37d84922811a35a599f":{"kind":"REWORK","_number":3,"created":"2019-11-25 17:21:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/3"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:21:07.000000000","tz":-480},"subject":"DNM: Debug for distutils","message":"DNM: Debug for distutils\n\nDo not merge, do not pass go, do not collect $100.\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feb39ad0f43de7daf5f0f37d84922811a35a599f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/feb39ad0f43de7daf5f0f37d84922811a35a599f"}]},"branch":"refs/heads/master"},"89eefc487484662591b264bb662df3ef89098a47":{"kind":"REWORK","_number":4,"created":"2019-11-25 18:10:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/4"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 18:08:27.000000000","tz":-480},"subject":"Fix gate issue with DIB","message":"Fix gate issue with DIB\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89eefc487484662591b264bb662df3ef89098a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89eefc487484662591b264bb662df3ef89098a47"}]},"branch":"refs/heads/master"},"a7f2f71f01da5b93adc0a9c00bf389b0fe2c17f3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-11-25 18:18:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/5"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 18:18:07.000000000","tz":0},"subject":"Fix gate issue with DIB","message":"Fix gate issue with DIB\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7f2f71f01da5b93adc0a9c00bf389b0fe2c17f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7f2f71f01da5b93adc0a9c00bf389b0fe2c17f3"}]},"branch":"refs/heads/master"},"b956dd337196e414a7d23a7728574252e08f83e1":{"kind":"REWORK","_number":6,"created":"2019-11-26 05:51:09.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/695947/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/6"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-26 05:49:36.000000000","tz":-480},"subject":"Fix gate issue with DIB and Nova console services","message":"Fix gate issue with DIB and Nova console services\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, the nova console services are buggy and causing gate failures.\nThis patch disables them for the grenade job as Octavia has no\nneed for them. Both must be done in the same patch because both are\nfailing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b956dd337196e414a7d23a7728574252e08f83e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b956dd337196e414a7d23a7728574252e08f83e1"}]},"branch":"refs/heads/master"},"6e6b78e68f7f621e5000092db8ba0b7e3859af34":{"kind":"REWORK","_number":7,"created":"2019-11-26 14:31:31.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/47/695947/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/7"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-11-26 14:30:00.000000000","tz":-480},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6b78e68f7f621e5000092db8ba0b7e3859af34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e6b78e68f7f621e5000092db8ba0b7e3859af34"}]},"branch":"refs/heads/master"},"c8943b605eed6bee45c2da11dbd6afe1d29964c1":{"kind":"REWORK","_number":8,"created":"2019-11-26 16:39:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/8"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-26 16:39:17.000000000","tz":0},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8943b605eed6bee45c2da11dbd6afe1d29964c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8943b605eed6bee45c2da11dbd6afe1d29964c1"}]},"branch":"refs/heads/master"},"a0af101030311ffd724ea8290471cbd45df8dff4":{"kind":"REWORK","_number":9,"created":"2019-11-26 17:04:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/9"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-26 17:04:42.000000000","tz":0},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0af101030311ffd724ea8290471cbd45df8dff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a0af101030311ffd724ea8290471cbd45df8dff4"}]},"branch":"refs/heads/master"},"f160aa9cd749984cdd0f59362b4f67c04ab8234e":{"kind":"REWORK","_number":10,"created":"2019-11-26 19:04:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/10"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-26 19:04:47.000000000","tz":-480},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f160aa9cd749984cdd0f59362b4f67c04ab8234e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f160aa9cd749984cdd0f59362b4f67c04ab8234e"}]},"branch":"refs/heads/master"},"918060c13f870c4d2abc80c3e02ad7e446b3aaa9":{"kind":"REWORK","_number":11,"created":"2019-11-26 21:44:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/11"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-26 21:44:39.000000000","tz":0},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/918060c13f870c4d2abc80c3e02ad7e446b3aaa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/918060c13f870c4d2abc80c3e02ad7e446b3aaa9"}]},"branch":"refs/heads/master"},"6ab0d0de309c37c7a5958f13c2ec4f36a09897b8":{"kind":"REWORK","_number":12,"created":"2019-11-26 21:46:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/695947/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/695947/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/695947/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/695947/12"}}},"commit":{"parents":[{"commit":"ce6b2e80e2d8aed52daa60d30494909aec3df391","subject":"Make octavia-grenade job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce6b2e80e2d8aed52daa60d30494909aec3df391"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-25 17:10:26.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-26 21:46:25.000000000","tz":-480},"subject":"Fix gate issue with DIB and Migrate grenade to PY3","message":"Fix gate issue with DIB and Migrate grenade to PY3\n\nThis patch adds a diskimage-builder(DIB) pre-install script that works\naround an issue with pip and pydistutils.cfg files[1].\n\nThis is a temporary fix until DIB can merge and release the full fix.\n\nAlso, migrate the grenade job to python 3. Both must be done in the same\npatch because both are failing simultaneously and must merge together.\n\n[1] https://review.opendev.org/#/c/695823/\n\nChange-Id: I25ed1a60c115e24763176af5f96633ba6d140d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ab0d0de309c37c7a5958f13c2ec4f36a09897b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ab0d0de309c37c7a5958f13c2ec4f36a09897b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
