)]}'
{"id":"openstack%2Foctavia~696037","triplet_id":"openstack%2Foctavia~master~I6d5d39c4cf87ac687355f80ca591c1a611f46f41","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6d5d39c4cf87ac687355f80ca591c1a611f46f41","subject":"Use 2048-bits keys for devstack certificates","status":"MERGED","created":"2019-11-26 09:00:54.000000000","updated":"2019-12-02 23:59:38.000000000","submitted":"2019-12-02 23:58:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696037-1575331084231-7198a097","meta_rev_id":"a2153c4c6d9847d949e368c8cc274ed83fa9d4bb","_number":696037,"virtual_id_number":696037,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 23:58:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-12-02 18:49:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-02 18:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2019-12-02 18:49:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-12-02 18:49:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 18:20:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-02 18:49:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-02 23:58:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14c8fdc9d8fd1546cde756c726a85ee9168ff32f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 09:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be3382910dc51e1d7b6e6b0e6c9fe7d739c93190","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-26 09:07:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca3dc9abbe2e2dc809ca10dd4536fbceec4b88b0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 09:36:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32523f12fa9f146a7f3912acfd9674cec20d8963","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-26 09:36:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63de0b32e3d62a9a7ffac594ccb1bcc0a96868c3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-26 09:38:25.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM. Thank you.","accounts_in_message":[],"_revision_number":2},{"id":"4cc2ea2db0f025ab33c43c79976f9b02c9aa4db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:03:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3d0084171444cd9a5b04e5f4afe3d9d : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42905c6f706c468e99fe054ddeec5576 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03bd40f89ad948399f29dbd1a0183b5d : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0a189e06dfa47b8b0e543f068902e94 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684519959f6e425cb6e2a8d1a66df8b2 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f71597ca83b49cf8a0dddb297bd93dc : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c937eb04d340859f7d2c6bae74436e : SUCCESS in 9m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1972cefd83834a57a784365f35fb0986 : SUCCESS in 6m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/929ae44e51f3484f975373b3f29d05dd : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c2c4450d77e846dabebde6320b5749cc : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8995afc57664d0583158d23584151c7 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b8318b1320d40358cd1e36e72bdcd65 : SUCCESS in 41m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8a9cce9ef6184f44aee3a19399eaffa1 : SUCCESS in 48m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fe0c17b1228484eb4e5ab4a85fc7825 : SUCCESS in 1h 52m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/93620913c09e4ab68309e427d89cf5ac : SUCCESS in 1h 49m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/42eaed5f422b4f2c830ae424d75c563a : SUCCESS in 2h 18m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23bb0de1d91f4eecb468baba4f282ab9 : FAILURE in 36m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3974c2e74fb64d59b1d8e0e96438f550 : SUCCESS in 1h 27m 11s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8eb977bad7ae447b82f669d632cb5ca1 : FAILURE in 1h 13m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0f3ff08fe8d446498cbd50009ea16b1 : SUCCESS in 1h 02m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d45f8fe652bd41bf9e727828d7b7a063 : SUCCESS in 1h 49m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/162fd78798a546a49b8dec082552439e : SUCCESS in 1h 16m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0c465aefbdec48bbb4fe723f41cad3df : SUCCESS in 1h 19m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/50d8843e21bf49b8a7e1c5f50ee13e2c : FAILURE in 1h 15m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e250f6bc7fd1027eb68f1614808231374eb1221c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-26 22:07:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nLGTM and I understand why it is needed. +1 to see if others have stronger feelings about the wording change.","accounts_in_message":[],"_revision_number":2},{"id":"76cf7f02fca543bd5988da00c19fc56051d43f4c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-27 10:50:48.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e42d24fc568900137456e37100ebf76a91e5b5e0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-11-27 10:51:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4dfd17e9cd39c4c0cf081311e248267d859b5b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 13:07:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7eefe4aeca84310b016a1541b390256 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e765f97693740d98f4264820736753e : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c19f59066a46c98f648a1bf4f34e2f : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c697731d4c794f619017a826adeb12ee : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eff2d263f0ea4abcbfb6f6e891e015f5 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00f433bb46334dacae8470fe4172e1ce : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98bdc7f81a9747c78fc2d281926a66e4 : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/84bc3fd3343946c2b0921de7efabb000 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2cb7ad91e7d849babab37e6ffe900f19 : SUCCESS in 7m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/152a7cd164ad4100be2bb4a9f679378c : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9df485a3682b436fa1debdb22c185348 : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e0f1003183bf4e4d9f9898013833ce5a : SUCCESS in 52m 42s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2a8112e776fc44909d79e06cbf84274b : SUCCESS in 44m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f681c533c037489f8592017daa6ee351 : SUCCESS in 1h 34m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/85c8a4e0dc064aed978a68cc9f19375d : SUCCESS in 1h 34m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e6fccbd27d5b44d7bbe17e6f64d75ff1 : SUCCESS in 2h 12m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b70a837cbe67471eb114332454254aba : FAILURE in 33m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c5e7b13e2e24bb984e88200a7a8887b : SUCCESS in 1h 49m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a6fe6d978b8489fab6c4b748bf1f35c : FAILURE in 1h 15m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3eae1aba86a47d29aa75d3d9d8b9492 : SUCCESS in 50m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a40d9549dc04d1eb28e47329ddba383 : SUCCESS in 1h 35m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1977fbba28a04bf2b965066b2160ab6b : SUCCESS in 1h 10m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8df9389fdd0f4a369ed8b7f4c4314eb9 : SUCCESS in 1h 05m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e2d60a44cb3a4037994537e976feda7a : FAILURE in 1h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7947d3e5972db23e26b0060576ae79e3afb0069d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-02 18:20:50.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"5491ccf15e680ded6d4c3da3d39746c3f6839475","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 18:49:16.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1\n\nI guess we would like to backport this to stable/train if we ever want to stack on CentOS 8 in Train.","accounts_in_message":[],"_revision_number":3},{"id":"2a3707552b6a32178f8135657c82011ae11b5013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 18:49:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"50f02c8c10d1306bd9fccb81fa937ab8b8df42c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 20:32:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a37634023f412e9470a4be30ac29d3 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b84ac193dc84e77b79e70a955ca18c3 : SUCCESS in 7m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1226e72f8668489cb7e8db1742657a7a : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41ac6c8a8d0d41d291bbbc039ddac0f4 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fef7d900d3244bda191994ca40061a9 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91aeee76d42471da524e5607f0d70b7 : SUCCESS in 10m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a7a03e56af14bce8e1281c18e3550fd : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d662a5d7eab243758a991a57cd51639a : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/585068f027a34cdbb7146936b2a1667f : SUCCESS in 45m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/722657952b09411d8b99c0803fb8a66e : SUCCESS in 48m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9794a2b677554f2c8bc5f7a7904fc0df : SUCCESS in 1h 30m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a69de49830bb4e8999405011963d96f2 : FAILURE in 1h 41m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5c547cc69c26f3beda11b7829dffb2c0b3ee763b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 20:45:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb16d115784b5202c537524991cd9e6643540378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:25:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc44a8b4943a4684ac71bfce8f37996e : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b9d5eec27d145179a7e363421fd14f3 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ecdb96e5754cb780d48a9929d896c8 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb2f63573404505b151a752836d9e83 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1907cf3203c3422d8842dbb719a5c316 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e6f60c0358465ab211fcb96e657e4b : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f143a4741cff483f84c6fdd10850290c : SUCCESS in 7m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/02d1d681aa7e4f51bd8613082c1bdf98 : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d001141a730d4af8b56d5e71e3452273 : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/482782b9e21e40a2a420fcfd4d1b9236 : SUCCESS in 40m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c356fa05d82b4a1ebcc56adb60e1cb9b : SUCCESS in 1h 17m 12s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d9d18f2c16b453f8de5cf43dafe3719 : FAILURE in 31m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/caf06818670440e7a2cda7ab7b8b7050 : FAILURE in 1h 15m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc8a82cb0db7448eaf4b00da001d6e0f : SUCCESS in 50m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/31dac8680e77480e83cc17198e32c507 : SUCCESS in 1h 32m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/972f3c9c373e4b4c917613ba93ab366d : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4e5d7e89cc827817b773ce856cb686d631a1a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:25:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f39401b0caa402330eefeb0327daf943430a0634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:58:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1550220c00634e769cf0d5d8c0c24648 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1fc0ae140b44b9a78883bc813d087c : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d63356863b49e9b74ad612b73917e7 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed97cdd7acf04584806c2203d6bf3fb4 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143645d80d4d44a2b31574af827b1efe : SUCCESS in 10m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03854517b06f4000829f7cf358266e70 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84c6101ab5f4481fadf3d43e730b0b60 : SUCCESS in 44m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f496f6d17cc94c4bac67269eeba0026c : SUCCESS in 1h 32m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ab8c4ffbe01993c185b3b0f55f5efb890a081115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:58:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a2153c4c6d9847d949e368c8cc274ed83fa9d4bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:59:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0749d4f2a0584926bb3ccba5d85bf8f4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8b6bb936d377f110f7f684d845ec3bd8e76fec82","revisions":{"73943de9f9801f2359ebc961b21a68da7d071ebd":{"kind":"REWORK","_number":1,"created":"2019-11-26 09:00:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/696037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/696037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/696037/1"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-26 08:58:57.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-26 09:00:13.000000000","tz":60},"subject":"Use 2048-bits keys for devstack certificates","message":"Use 2048-bits keys for devstack certificates\n\nRHEL8 and Centos8 refuse to load certificates generated with a 1024-bits\nkey because of new security policies. Increasing key size to 2048-bits\nfixes certificate generation scripts for these distributions.\n\nStory 2006944\nTask 37623\n\nChange-Id: I6d5d39c4cf87ac687355f80ca591c1a611f46f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73943de9f9801f2359ebc961b21a68da7d071ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73943de9f9801f2359ebc961b21a68da7d071ebd"}]},"branch":"refs/heads/master"},"6b43965b95bf7f84d19dd36c5a7a246da7cf881a":{"kind":"REWORK","_number":2,"created":"2019-11-26 09:36:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/696037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/696037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/696037/2"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-26 08:58:57.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-26 09:35:17.000000000","tz":60},"subject":"Use 2048-bits keys for devstack certificates","message":"Use 2048-bits keys for devstack certificates\n\nRHEL8 and Centos8 refuse to load certificates generated with a 1024-bits\nkey because of new security policies. Increasing key size to 2048-bits\nfixes certificate generation scripts for these distributions.\n\nStory 2006944\nTask 37623\n\nChange-Id: I6d5d39c4cf87ac687355f80ca591c1a611f46f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b43965b95bf7f84d19dd36c5a7a246da7cf881a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6b43965b95bf7f84d19dd36c5a7a246da7cf881a"}]},"branch":"refs/heads/master"},"8b6bb936d377f110f7f684d845ec3bd8e76fec82":{"kind":"REWORK","_number":3,"created":"2019-11-27 10:51:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/696037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/696037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/696037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/696037/3"}}},"commit":{"parents":[{"commit":"4a5c24ef6fcd3b5f198a20d780dedd7f7976296d","subject":"Merge \"Stop allowing the deletion of an in-use flavor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a5c24ef6fcd3b5f198a20d780dedd7f7976296d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-26 08:58:57.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-11-27 10:49:41.000000000","tz":60},"subject":"Use 2048-bits keys for devstack certificates","message":"Use 2048-bits keys for devstack certificates\n\nRHEL8 and Centos8 refuse to load certificates generated with a 1024-bits\nkey because of new security policies. Increasing key size to 2048-bits\nfixes certificate generation scripts for these distributions.\n\nStory 2006944\nTask 37623\n\nChange-Id: I6d5d39c4cf87ac687355f80ca591c1a611f46f41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b6bb936d377f110f7f684d845ec3bd8e76fec82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b6bb936d377f110f7f684d845ec3bd8e76fec82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
