)]}'
{"id":"openstack%2Foctavia~696324","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I7153ed9a6214ce352b2ac115001f7f6134d95b39","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I7153ed9a6214ce352b2ac115001f7f6134d95b39","subject":"Fix filtering with API tags","status":"MERGED","created":"2019-11-27 15:22:06.000000000","updated":"2019-11-27 20:11:46.000000000","submitted":"2019-11-27 20:09:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696324-1574885363794-81677534","meta_rev_id":"22015b9da227a37a8f71fb4fcd158eeac6de7d96","_number":696324,"virtual_id_number":696324,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 20:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-11-27 15:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-11-27 15:58:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-11-27 15:58:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 15:51:46.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-11-27 15:58:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-11-27 20:09:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db9afdc317b91730f2e710f813c5ff0838260b5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-27 15:22:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1c753ca42e4c2dbbb51120dbc026a1db98b89157","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-27 15:22:48.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 567a388b95e18eaeff0830259efceeb9b861b2e5","accounts_in_message":[],"_revision_number":1},{"id":"d0f529768ba1d11895bdaf3b473e8aed62739614","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-11-27 15:51:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4508969e272291d01f7c54764310e1e8e570ff1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-11-27 15:58:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f63a529444840a3281e24b65fd4b5aa9d1b15ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:00:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c7cb44dca245c4bf610c0cb8cbc099 : SUCCESS in 7m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79df2065914b4f0c8b249c07a59a1370 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785cc2e404ea40dc9d3bbb17bbf8bd90 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/994835086de64394ae37f36431084044 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03bad530f80444f5b89476c640d961b2 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c80eee13bdf64bd3a0545653f6343402 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6319d79eb8413d923b6ba7f2a5d53e : SUCCESS in 9m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4d99462482164367b5fae4d28b79750b : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/25c201f777f947678fb228312336efb4 : SUCCESS in 8m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/594e7705c65541be9ee3b0e264a6cba5 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a5ac74a21d74260a5513589b5cd0bdd : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5db9a8b74b8949f09c7412d65f8619e8 : SUCCESS in 51m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8d81cae8609544faa187fc326999ba06 : SUCCESS in 44m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/052fce73fad640adb0c85b706469b8e9 : SUCCESS in 1h 50m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ab2e6660fe8849ebaaf74c2c741edda7 : SUCCESS in 1h 43m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/853505f6cb66409f848b243320d32d8c : SUCCESS in 2h 29m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c47e0fe10e941cd9cfd6ecca878bec0 : SUCCESS in 1h 46m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b63fee3d30644931b462289d967fdcb6 : SUCCESS in 1h 59m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b76ef4c55694b63baddb95f17bfc2c8 : SUCCESS in 1h 11m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/302ff67403db4a78a2e3b772ebd7e36c : SUCCESS in 55m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af79497daeb447b3af393e9102f10833 : SUCCESS in 1h 33m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97fde760e15e444489a6643afb74e1e8 : FAILURE in 44m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/64a7d99bcd7c41a480d79ddeeff63c72 : SUCCESS in 1h 00m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85b92e5b1f2e4f4ab968e5fc52dcf2c7 : SUCCESS in 1h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a9c681bc504dc7d6bdbdecd5f95721c3f701568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:00:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1eb1207a2c4413273cae310ef5256e9431c115e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:09:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/303b5a778c8c41159c37115b4714dc42 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ae51a34e6f44448fe55e870eb210bd : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f945e4d6b4c45c196a8ffe26798bb4d : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4e19ed7dcd49bab24bd238434f2c5c : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a18f8a44098f465e8c30b84d561245f6 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bf8cdc074d47ba841f555208510ac4 : SUCCESS in 11m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/78592a2456b740a9860eddadc7dac085 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0dc54feebf314806b0244e672807190f : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80a2c9ee894b41fa8549a391b3f516e8 : SUCCESS in 46m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7d1fb4cf3a2d4cc9875ff965e28d286e : SUCCESS in 58m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/101755fe3dbb452bb3938516eebc0e1e : SUCCESS in 1h 49m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/adf2e68e6e2b422c886850524185064a : SUCCESS in 1h 58m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3145bd9da954ec1a988d960f6b73b40 : SUCCESS in 1h 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5ad2a4ee35031ee0fc4ef16072418d547719bc08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:09:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22015b9da227a37a8f71fb4fcd158eeac6de7d96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433106f6703341b788f7a1f18b696c98 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19202863e2e459a0c81a872df7dc745aad4d15ea","revisions":{"19202863e2e459a0c81a872df7dc745aad4d15ea":{"kind":"REWORK","_number":1,"created":"2019-11-27 15:22:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/24/696324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/696324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/696324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/696324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/696324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/696324/1"}}},"commit":{"parents":[{"commit":"96e6e4af914a919ea49785b50612050e954b5d45","subject":"Merge \"Fix filtering for provider capabilities list API\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96e6e4af914a919ea49785b50612050e954b5d45"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-19 23:17:59.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-27 15:22:06.000000000","tz":0},"subject":"Fix filtering with API tags","message":"Fix filtering with API tags\n\nWhen filtering API calls using tags along with other filter parameters\nthe API could fail with an \"Entity has no property\" errors.\nAlso when filtering with tags that contain a comma separated list,\nthe API would mis-interpret the list of tags as a single string.\nThis patch resolves both of these issues and adds functional test\ncoverage for these use cases.\n\nChange-Id: I7153ed9a6214ce352b2ac115001f7f6134d95b39\nStory: 2006914\nTask: 37559\n(cherry picked from commit a736f98a7fffa8494ea4f00ecfbb929e8ccae860)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19202863e2e459a0c81a872df7dc745aad4d15ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19202863e2e459a0c81a872df7dc745aad4d15ea"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
