)]}'
{"id":"openstack%2Foctavia~696943","triplet_id":"openstack%2Foctavia~master~Ibf30f6368cb811043962844ae5f89e9dfcae2361","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ibf30f6368cb811043962844ae5f89e9dfcae2361","subject":"Add load_balancer_id earlier to amphora info","status":"ABANDONED","created":"2019-12-02 18:56:19.000000000","updated":"2019-12-18 05:33:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0e7f97f6eefd318347eb9596ad878fd34068edc","_number":696943,"virtual_id_number":696943,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-02 22:14:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-12 22:27:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 22:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-12 22:27:04.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"}],"messages":[{"id":"feb1c44823f24cd074645560d5694cd5a12918a4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-02 18:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe0eaaba4948a7fe33ebbc40dc1e4d5cc9ab438a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-02 19:52:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1689cc158f4497a3ccaff473dd3fce16e443ffe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:14:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00de58207bc64f588b587a7cdce9a829 : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b43bed62900549cb9ba7860328212dc5 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6755b9a187447fa379d4e97ca89dbd : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9de9ac51ba44190a6cc034f3bfb3c3f : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ce66b5529144ccbd0fbd3eaacba1d8 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/638fd2a738bc412f814149a9886ed64f : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5dde331f10f4ce28901ecf2790225aa : SUCCESS in 12m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1396b0cc7b4c400f8c9e89a7e28be1f2 : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1cc8840f75984942a1ac0bb587a94634 : SUCCESS in 6m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c31a2a421daf4d9f8adf496ba4eca6f4 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ab0a6ef06e84de38e5f9fcef18a195c : SUCCESS in 5m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d85f5e3d01094f4482deac73fd98788d : SUCCESS in 51m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d4390742e10645f589a693b685147fb5 : SUCCESS in 53m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39425ceb041e44508bd88f7612be13f0 : SUCCESS in 1h 52m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f4a2d7c8156847fc9c2cf90dd82d283b : SUCCESS in 1h 39m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d30d608d6b6e4b14816a5b30d31eeb58 : SUCCESS in 2h 19m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de5bc08855854a9ab4031d508afc4cf7 : FAILURE in 34m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a7d5a29764145a792e0490860431944 : SUCCESS in 1h 50m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7ddc235718c4629be8742245a25e814 : FAILURE in 1h 07m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/127b67193d1b4ecc9e77c5e0bd4827c3 : SUCCESS in 56m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5cf7af1234ab4d958a78ebd69d3af200 : SUCCESS in 1h 37m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95364f3d357e435799b544787ddf9afc : SUCCESS in 1h 19m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a93383aeb96e4f818cb4135fa7d0e2f9 : SUCCESS in 1h 16m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/83e3157c83cc4a569c6ac421d58beb53 : FAILURE in 1h 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0e7f97f6eefd318347eb9596ad878fd34068edc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-18 05:33:06.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/#/c/698082","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78c8c60dd2934196d6a4d1e18d6a5298eb1a3a68","revisions":{"fdc27a6359a7a49131caeabb4833b700611f5c34":{"kind":"REWORK","_number":1,"created":"2019-12-02 18:56:19.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/43/696943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/696943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/696943/1"}}},"commit":{"parents":[{"commit":"3c05cce89905a5dcc2caac3f6d81e9dc7b82261e","subject":"Merge \"Fix update API when barbican secret is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05cce89905a5dcc2caac3f6d81e9dc7b82261e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-02 18:44:24.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-02 18:44:24.000000000","tz":240},"subject":"Add load_balancer_id earlier to amphora info","message":"Add load_balancer_id earlier to amphora info\n\nWhen we create amphora for specific loadbalancer it would be\ngood to get info about mapping pair loadbalancer-amphora as\nsoon as possible.\n\nFor example, if admin need to debug some issues with amphora VMs\n- loadbalancer_id won\u0027t be set until AmphoraComputeConnectivityWait\ntask succeded, so he has to go to worker logs to understand\nwhich amphora is related to currently creating loadbalancer.\n\nTo simplify that - if amphora is creating for loadbalancer\nadd setting loadbalancer_id in UpdateAmphoraInfo task (pass as\nan optional argument).\n\nChange-Id: Ibf30f6368cb811043962844ae5f89e9dfcae2361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdc27a6359a7a49131caeabb4833b700611f5c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fdc27a6359a7a49131caeabb4833b700611f5c34"}]},"branch":"refs/heads/master"},"78c8c60dd2934196d6a4d1e18d6a5298eb1a3a68":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-02 19:52:48.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/43/696943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/696943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/696943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/696943/2"}}},"commit":{"parents":[{"commit":"3c05cce89905a5dcc2caac3f6d81e9dc7b82261e","subject":"Merge \"Fix update API when barbican secret is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05cce89905a5dcc2caac3f6d81e9dc7b82261e"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-02 18:44:24.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2019-12-02 19:52:41.000000000","tz":0},"subject":"Add load_balancer_id earlier to amphora info","message":"Add load_balancer_id earlier to amphora info\n\nWhen we create amphora for specific loadbalancer it would be\ngood to get info about mapping pair loadbalancer-amphora as\nsoon as possible.\n\nFor example, if admin needs to debug connectivity issues with amphora \nVMs - loadbalancer_id won\u0027t be set until AmphoraComputeConnectivityWait\ntask succeded,( which is never for such case) so he has to go to worker \nlogs to understand which amphora is related to currently creating \nloadbalancer.\n\nTo simplify that - if amphora is creating for loadbalancer\nadd setting loadbalancer_id in UpdateAmphoraInfo task (pass as\nan optional argument).\n\nChange-Id: Ibf30f6368cb811043962844ae5f89e9dfcae2361\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78c8c60dd2934196d6a4d1e18d6a5298eb1a3a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78c8c60dd2934196d6a4d1e18d6a5298eb1a3a68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
