)]}'
{"id":"openstack%2Foctavia~696944","triplet_id":"openstack%2Foctavia~master~I5fe263e59393de56fbee383d394f61b07fb598ca","project":"openstack/octavia","branch":"master","topic":"story/2006972","hashtags":[],"change_id":"I5fe263e59393de56fbee383d394f61b07fb598ca","subject":"Fix HM when amphora-agent reports DOWN listeners","status":"NEW","created":"2019-12-02 19:16:23.000000000","updated":"2020-04-20 23:13:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5817205f40fc559b40e6e14c54631340ea2e9022","_number":696944,"virtual_id_number":696944,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"date":"2020-01-24 16:11:47.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-03 00:07:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2019-12-03 00:55:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":-1,"date":"2020-01-13 22:58:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-01-23 07:37:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2019-12-11 16:42:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2019-12-11 16:41:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 00:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 00:55:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-11 16:41:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2019-12-11 16:42:05.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-01-13 22:58:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-23 07:37:06.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"347fbe67c5abc089b136c107b69fab9e982edb42","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 19:16:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2e06ebc371d3ddea08d5f13e213b68614de62aa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 19:20:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3469923222e5173cdb99ada76f56cd3360412fbf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 19:25:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f06369ed7ea71a10394898b4bff0131ac02a820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 21:35:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47c291a3ecf94e6884c1423bac30bda0 : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3ebeca116d439499f729f788648582 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d9f25a90824d5fa540f23e8806ffe7 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/349917202efb4def87d91593e9641d32 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd9ebf36ec24652bf1a88a3330a33a1 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a68e0960a0254da4938ec2807632bd78 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a04dd7fd93f4a909c830efc6d48f76a : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e0c8f9b8be74c70a9ab96ce39a145f2 : SUCCESS in 5m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0cb6da6e50c84286bca13ab8dc443df8 : SUCCESS in 4m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dae7dda965154b6496e547f094ba1800 : SUCCESS in 6m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ada36d2733d4c6dbeaedabe25765d64 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1997ab38fc284f089e9c71fd960e4988 : SUCCESS in 5m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a60a31440174a118156749fbad1f491 : SUCCESS in 41m 17s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4f8f901a4bc9475c879b9aec84da258a : SUCCESS in 47m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad1cc095711c49a78f3b77ab61afd79e : SUCCESS in 1h 42m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/70de937d60924844b74a4dca4a006323 : SUCCESS in 1h 40m 40s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/06a9b51623504e1bbf288c8347e104bc : SUCCESS in 2h 13m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58753f1d30a64410bc61712a2eb76381 : FAILURE in 40m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3fa1470a43ca4a679e914057f8a939a7 : SUCCESS in 1h 52m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/059cd1b8da134473b3f72d73831aeb98 : FAILURE in 1h 02m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a14d8b8932c64d4591c90af732915fa8 : SUCCESS in 54m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3a152566cf2d4e7b9b1075a9f30138be : SUCCESS in 1h 34m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/840698d2c40e4ee380b888843c38b5b2 : SUCCESS in 1h 02m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dcc4c69715a048a0854b1575a6eef949 : SUCCESS in 1h 06m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b3e0d99850f4622826d10690a31d7ed : FAILURE in 1h 08m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"893fad244f3fc69013f196b0e7fef1a7d562baa8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-02 21:44:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"30a036bdc4a16cb556158d5823c333eb869c0006","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 22:07:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7230b9d781fb43b5e1a308f230272d6c91843990","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-02 22:08:10.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9d62d5a82883beb201f246d5e86cd3df651928d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 00:07:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c9572a1f01149568cf14cf78469c143 : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c949d95dd66d4875af9059bdccd9b6fb : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8c754771bf49c8b7cb17d9d391b537 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d56985016b64e649f5a7a9e65403220 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de210085e12843e5a1a26c2c24786fb3 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b67407917f419bb267bd838cfb492c : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d9c8e329fb4fe5a21d027d30b57209 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/777751ef43c64ba8953fca434196b99d : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5f9263cc4bb4cf28de7e5b957e755c7 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a413197ce16d40cfbf743b28bbea6207 : SUCCESS in 6m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e38ba054217c4aa79cb0b60f8e10a458 : SUCCESS in 43m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e2c23d06abd34d35939efd8c2e288712 : SUCCESS in 1h 57m 48s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e691568754cb46b1a5a5f92390a2a253 : FAILURE in 33m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b9a1a2be9124ca2a16760a2f423d392 : FAILURE in 1h 03m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9fed04cd1dc749e0af24e3c4be3a0098 : SUCCESS in 52m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/01a3e0a1a7c643818238bde3c95db802 : SUCCESS in 1h 51m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce383e7152444394be38deda9d29ae53 : SUCCESS in 1h 09m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c00a5ce15ad98e18c91d8c65a185d5b01e3fa2f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-03 00:55:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5d4c0244aaac2abdb35717096224b7610ebbdf10","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 22:58:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3523fc5ea8c606940f32aad77a76d5d3677983ba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-24 16:11:47.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f9cf6c202c473f1ab128a50b379a151ac66e24e2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-20 23:13:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d05fedc41cd9b5d25db7595d41fa48c57a8a6e48","revisions":{"465a6738c574462b6057c67ac03e478ddf3d9fd2":{"kind":"REWORK","_number":1,"created":"2019-12-02 19:16:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/696944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/696944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/696944/1"}}},"commit":{"parents":[{"commit":"3c05cce89905a5dcc2caac3f6d81e9dc7b82261e","subject":"Merge \"Fix update API when barbican secret is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05cce89905a5dcc2caac3f6d81e9dc7b82261e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-12-02 19:07:25.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-12-02 19:15:56.000000000","tz":60},"subject":"Fix HM when amphora-agent reports DOWN listeners","message":"Fix HM when amphora-agent reports DOWN listeners\n\nThe Health Manager was not acting when the amphora-agent reported\nlisteners as DOWN. With this patch, the Health Manager skips updating\nthe amphora health record and marks the listener operating status as\nERROR. Should a second amphora (active-standby topology) report the\nlistener as UP, the Octavia listener operating status is elevated to\nDEGRADED.\n\nTask: 37700\nStory: 2006972\n\nChange-Id: I5fe263e59393de56fbee383d394f61b07fb598ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/465a6738c574462b6057c67ac03e478ddf3d9fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/465a6738c574462b6057c67ac03e478ddf3d9fd2"}]},"branch":"refs/heads/master"},"d05fedc41cd9b5d25db7595d41fa48c57a8a6e48":{"kind":"REWORK","_number":2,"created":"2019-12-02 22:07:59.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/696944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/696944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/696944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/696944/2"}}},"commit":{"parents":[{"commit":"3c05cce89905a5dcc2caac3f6d81e9dc7b82261e","subject":"Merge \"Fix update API when barbican secret is missing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c05cce89905a5dcc2caac3f6d81e9dc7b82261e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-12-02 19:07:25.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-12-02 22:07:43.000000000","tz":60},"subject":"Fix HM when amphora-agent reports DOWN listeners","message":"Fix HM when amphora-agent reports DOWN listeners\n\nThe Health Manager was not acting when the amphora-agent reported\nlisteners as DOWN. With this patch, the Health Manager skips updating\nthe amphora health record and marks the listener operating status as\nERROR. Should a second amphora (active-standby topology) report the\nlistener as UP, the Octavia listener operating status is elevated to\nDEGRADED.\n\nTask: 37700\nStory: 2006972\n\nChange-Id: I5fe263e59393de56fbee383d394f61b07fb598ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d05fedc41cd9b5d25db7595d41fa48c57a8a6e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d05fedc41cd9b5d25db7595d41fa48c57a8a6e48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
