)]}'
{"id":"openstack%2Foctavia~697084","triplet_id":"openstack%2Foctavia~master~Ie8ea3417b1b7fe82d88339ac63ccf0f8b365c085","project":"openstack/octavia","branch":"master","topic":"revert-8b6bb936d3","hashtags":[],"change_id":"Ie8ea3417b1b7fe82d88339ac63ccf0f8b365c085","subject":"DNM: Revert \"Use 2048-bits keys for devstack certificates\"","status":"ABANDONED","created":"2019-12-03 13:25:33.000000000","updated":"2019-12-03 15:18:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2724641eb81628dc689210134308ccc7f0800ef8","_number":697084,"virtual_id_number":697084,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-03 15:13:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 15:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71af068fecd40e4789d9920e25009ec41bb8bf9c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-03 13:25:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7589611b335a9a52807cf1219c2f966df1e3d7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:13:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e2844b7aed044448010079ae9d4fbac : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e97a075d85d24f93be077f22981bb2a9 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99fd63bb3fc140858de4b3ac872b6757 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e2eb8d969d4587b97936ffa2fcf925 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fefb72d5e4842a381bac19c90b78ce9 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51212154c63648e79ed57a93396194af : SUCCESS in 10m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6a00b24102e340d8a1ebe3d9ec53611e : SUCCESS in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/591c426708444e44bb5062876c62c353 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0539237f952f4953910fb5d88c345a6b : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bffb0c9222ec426ab15082cfff1e783f : SUCCESS in 47m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1df74427d721457e9c9275c362e863bd : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f89c64d20134f81b408282215764a42 : FAILURE in 39m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3808e5558b4c4f3eb69ebd89893353f6 : FAILURE in 1h 04m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/691c67e0ac0648a18ef50adf6d8810fb : SUCCESS in 1h 01m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdb09e8747fc485aa43afcde2353172c : SUCCESS in 1h 34m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d40806b375d6482290c9004eb4c72623 : SUCCESS in 1h 08m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29ca8fce7dd7f54064d5f131178df99bb11902dd","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2019-12-03 15:18:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64d6af78df8cd6d1044d2469a14b60c85be22e70","revisions":{"64d6af78df8cd6d1044d2469a14b60c85be22e70":{"kind":"REWORK","_number":1,"created":"2019-12-03 13:25:33.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/84/697084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/697084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/697084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/697084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/697084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/697084/1"}}},"commit":{"parents":[{"commit":"db34110c123a1208a31b5266902d9d128950e60f","subject":"Merge \"Use retry for AmphoraComputeConnectivityWait\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/db34110c123a1208a31b5266902d9d128950e60f"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-12-03 13:24:31.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2019-12-03 13:25:21.000000000","tz":60},"subject":"DNM: Revert \"Use 2048-bits keys for devstack certificates\"","message":"DNM: Revert \"Use 2048-bits keys for devstack certificates\"\n\nThis reverts commit 8b6bb936d377f110f7f684d845ec3bd8e76fec82.\n\nPlease do not merge, I\u0027m just trying to figure out which one causes\nKuryr gates issues.\n\nChange-Id: Ie8ea3417b1b7fe82d88339ac63ccf0f8b365c085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64d6af78df8cd6d1044d2469a14b60c85be22e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64d6af78df8cd6d1044d2469a14b60c85be22e70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
