)]}'
{"id":"openstack%2Foctavia~698117","triplet_id":"openstack%2Foctavia~master~I819d44a37cba3ef952d109a675ae6d55e3eb4aa1","project":"openstack/octavia","branch":"master","topic":"jobboard","hashtags":[],"change_id":"I819d44a37cba3ef952d109a675ae6d55e3eb4aa1","subject":"Update ListenersUpdate for lb/listener dicts","status":"MERGED","created":"2019-12-10 01:00:29.000000000","updated":"2019-12-12 03:14:29.000000000","submitted":"2019-12-12 03:12:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698117-1576120368746-91269b4c","meta_rev_id":"30bcbe773de01b5bc384c2aa95fe6ba9695f408a","_number":698117,"virtual_id_number":698117,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2019-12-11 23:42:15.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 03:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-12-11 16:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-11 21:21:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-12-11 16:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-12-11 21:21:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-12-11 21:21:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 16:10:18.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-12-11 16:36:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-11 21:21:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-12 03:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"210cdbe4d3841056365b0433714f6c804c11aa6d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-10 01:00:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c982a376bd0ff8dc012a2dd33c835252ea57b008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 02:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae79b8949b3741fe9b3ed9911574cc58 : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40a731ae05184a06831000547f02745a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f84055e5cbd54d369299c9f23e91db22 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b05a3e2cfb2f493faef55c22ba00806e : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ddf772d37e24cf285a1e28bdec3da87 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69765d09e43481796cf0039c7b5e073 : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/596c1efda34948029efc89837870da32 : SUCCESS in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5a9be0ed06a843879939c63277c05620 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ff7ab3bef424ce4859625500d9fcc81 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea5403951b1448c183d135895044c0be : SUCCESS in 43m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f164c8c543bc46669bcd6d18550146a6 : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f617938760b49bd8f98908c385c2b4a : FAILURE in 7m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/456df242b9684364af03f8df9fe26192 : FAILURE in 59m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e29a82bc05342d595682e7468d7344a : SUCCESS in 50m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/de3e076ca84d476aac98d9f8a77509fd : SUCCESS in 1h 33m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/547cc74acc944f33b918bd1e7de78fad : SUCCESS in 1h 03m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"223bd900a24a09cb66f6698b3a339c1809276a0c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-12-11 16:10:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b79ae8f72107e18ea88854b63362160d6b83ddfb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-11 16:36:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"954119d1bab524e05907d29338eea19e2dbbdde5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-11 21:21:30.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42a7a18ad93db5f0f69fd5f586e276b8a268fcc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 21:21:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1a2f5f75ddb784161e2626c10e7a2c82491fff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:20:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/122130a5f33a494f8e01ff8f486292a6 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed7245a543ee423b99f5fa64769ab2fc : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6805f4a0e2d48ecb5489ffef3fd1ab3 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8be992418ecf4700bf84808fcfbe3318 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2190c56c6514266849fcd4f1cb86e92 : SUCCESS in 11m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f5b3c26ae084ff2be19f03cff4ae1b3 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e655d3046d584bf3bec81958df9c7959 : SUCCESS in 53m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d3b89781ba647ff94888e51c80f783c : FAILURE in 1h 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c274629465d87aaf8983a7db7ea0ad45b8a707f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 23:42:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"553de08a420dacc570187a175c02eafcc6a06a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:57:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c95522e01d4fe59437f6204728e33b : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd0fcd6af8fc460d82a2bbd787ddd0c5 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f3f864f90a24f5894d70c47c4412a00 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e37c178b4ac41618f21e2fdae5ba2a7 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/026d546662374f6dbce1072947cbab0e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ac0c4b32bf410e86f6a45743ce5db0 : SUCCESS in 10m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0ef9b81bb99b4230a7f8281e95bb1dec : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b2974bb43a1c4c32a3e31a7297d78504 : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ecbebfb71404ba3bcd21e802e247f1a : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ff415a052d54154899dfc30b1b79e3f : SUCCESS in 1h 07m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f9626e4d744965a21685b0a75ff5b3 : SUCCESS in 1h 57m 24s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d97322bb70744a18fb442e17dea3d8a : FAILURE in 11m 19s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e68b89d5d76c4ef2bc063cde84b26633 : FAILURE in 1h 38m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d7d7b03c931e4e50bebeb63a0ff2d836 : FAILURE in 1h 27m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/71edcb5a44664c42b84aa378a82797da : SUCCESS in 2h 10m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc6e91f6318d4a639f99489ea43c391d : FAILURE in 2h 07m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c36168e2c395e16c84bce197ae4fd039df7635f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:57:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"468c69e974bf666a150a67552a139b99b80e23d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:12:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af305d4956454b8fa9516a742f639486 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c58a73ffa04190b8112116620423b8 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa149a9995745fdb273af4201b86b2b : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22910ace74db403cbb26dc1136f352e6 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687a24b17de541269c8e523d40e7f496 : SUCCESS in 10m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/607bc96979664e1299cbe9c3b0c44adf : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c389910258d4ffbb05b1612aded5465 : SUCCESS in 58m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b89f56e78374426b940150cffc072d4c : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"de9afccf21beed05075efaa4a8d3b6209abef598","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:12:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bc4c7967faace62ba98333819b967688fea3fd98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:14:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30e705b1885427fbc4b02ab6efe8255 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc4c634d7c828ed55e4ea1b70049e867fdc27b8a","revisions":{"fc4c634d7c828ed55e4ea1b70049e867fdc27b8a":{"kind":"REWORK","_number":1,"created":"2019-12-10 01:00:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/698117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/698117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/698117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/698117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/698117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/698117/1"}}},"commit":{"parents":[{"commit":"a07de8efb7ec92de4d243cf07d2d3632fe60f076","subject":"Merge \"Fix multi-listener LB client auth/re-encryption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a07de8efb7ec92de4d243cf07d2d3632fe60f076"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-10 00:58:16.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-10 01:00:28.000000000","tz":-480},"subject":"Update ListenersUpdate for lb/listener dicts","message":"Update ListenersUpdate for lb/listener dicts\n\nThis patch updates the ListersUpdate task to accept a load balancer\nID instead of a load balancer database object as part of the job board\nwork.\nIt also removes a duplicate test that is no longer needed.\n\nChange-Id: I819d44a37cba3ef952d109a675ae6d55e3eb4aa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc4c634d7c828ed55e4ea1b70049e867fdc27b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc4c634d7c828ed55e4ea1b70049e867fdc27b8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
