)]}'
{"id":"openstack%2Foctavia~698432","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ic638c056715f8fcea3b095726964e7224c38eae2","project":"openstack/octavia","branch":"stable/stein","topic":"oslo_policy_args-stable/stein","hashtags":[],"change_id":"Ic638c056715f8fcea3b095726964e7224c38eae2","subject":"Accept oslopolicy-policy-generator path arguments","status":"MERGED","created":"2019-12-11 09:35:39.000000000","updated":"2019-12-13 08:40:23.000000000","submitted":"2019-12-13 08:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698432-1576226304019-0c0e24f5","meta_rev_id":"6156c0a82abc32bc66879a3e6612a1b86e8c02ba","_number":698432,"virtual_id_number":698432,"owner":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 08:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-12-11 13:50:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-12-13 07:00:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-12-13 07:00:18.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 13:50:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-13 07:00:18.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2019-12-13 08:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b670f23e46ad3153ea7bb4f85d497a9161303db2","author":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"date":"2019-12-11 09:35:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d675d4e02587e54de8cbe21bf98fc81d04d154ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 11:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1d2c142c30485c8b6a2d1c442705b0 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f561866f47749558a3b2e8f6eb7d640 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f856d56a0c448a6a170528f66c6d826 : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f24bcd9d8d4943f99caea3e7e2560651 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1155cb475f384a23835fea97fc42e77a : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c50bc13b24428ab0f00ee6fedc021f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905c7dedbeda4bf883a087eac0efcbe3 : SUCCESS in 6m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/82ef966cf3f5497c9c42c85e726ad2dc : SUCCESS in 1h 27m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b20bb207b0ff4bc990657f884b7a0afe : SUCCESS in 1h 17m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/008887293ae64599a1a931101989aab7 : SUCCESS in 1h 41m 55s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2387b95040b443dba9faf45b1e6aa0a6 : SUCCESS in 1h 24m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/674d21b4c573487b8c0ab50d3f9ee8c8 : SUCCESS in 8m 29s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ace9aee068d84e11bfa2091b76e41f19 : SUCCESS in 9m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/30e8de899c874ebe8be389d273584fd3 : SUCCESS in 7m 48s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/707444b8d77c44089fb479b1640a976c : SUCCESS in 1h 02m 54s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c74c8dee2a254767a8f39e7d699c8f15 : SUCCESS in 1h 06m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2b13831e44eb4c2d9b81ff06a614ac1a : SUCCESS in 43m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/60e400215d8d4c99b146017cd86bba92 : SUCCESS in 1h 01m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d784e0bdf2941088023627c3e2b1876 : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b18cac5c51354581b7d172f2f468b777 : SUCCESS in 1h 35m 47s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d3882f2323394b92a1535ffa0522114f : SUCCESS in 1h 24m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d17693a2d8c549cea2c0a399e796cd88 : SUCCESS in 2h 00m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da213393bfaf49d3891f856b3aa986cd : FAILURE in 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c747acb8ad7e4926b129a8b343cfaabf : SUCCESS in 1h 47m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0452c424ad844e3a9565da6fcea66a15 : SUCCESS in 1h 45m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b9ccfe588844cd688e8a32a8cfd28b6 : SUCCESS in 1h 17m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21681348277a446bb766cf36b73ac699 : FAILURE in 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdc888b648d3de82b68bd28855256f4ee238704d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-11 13:50:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"826ae2b01836748731a57bb22e9089e145cb24b9","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-12-13 07:00:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"596741ca15aa5a97273d69326c9a9381158df139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 07:00:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bc2e095ce41984ca4a29ddaae5197d3c7db107b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 08:38:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/453d0f02e937445ebc7708979df1203c : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6799346bbff4f6e882693db86272195 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc7e8f073bf4490ca3b2ba693485311b : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b9941f648a004ac4bd6e4bed92ecac8d : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd338da01bce43369c0c40741f865347 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8160be67594c4736a4c52406d1bee741 : SUCCESS in 6m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2bce5be8a95d45ccba8b7be2e6dad5c4 : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1d7b5e74c98b4cd08506458e06a92e65 : SUCCESS in 9m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1207b1dc764647f2bddb242f27f76931 : SUCCESS in 8m 40s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/390708bea3404834bd9d28e74f940f96 : SUCCESS in 1h 03m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7e41ebb1a7034ebeaad997f6f0274789 : SUCCESS in 1h 01m 09s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1e2539585f0a4b868c7423e6cb9211d1 : SUCCESS in 1h 20m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c713db62dfde457dba5899952d77f8ea : SUCCESS in 1h 25m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a08767bbad34cdca358706342822884 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c016ed0fefbb42c09f90d8224f9b8934 : SUCCESS in 42m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f259645935a54ca3ba53c694b5bac5d9 : SUCCESS in 1h 32m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/894ead666f9c493b8bfa2f20ce225f62 : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/06e3e44b48b543e38f6e147e4783f7a1 : SUCCESS in 1h 35m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8fc4f8bddf944db95d7710822378b40 : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d036d3d4b213b896c71733354b0f2b97f4205ff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 08:38:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dae9499a6013fecf369432d2d3ed31ed37684499","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 08:40:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960fcacf5307445f99b6d2cf1e055960 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9904b26a9c40f29f554c56e9e65f6396caa8fea9","revisions":{"9904b26a9c40f29f554c56e9e65f6396caa8fea9":{"kind":"REWORK","_number":1,"created":"2019-12-11 09:35:39.000000000","uploader":{"_account_id":22595,"name":"Stefan Nica","email":"snica@suse.com","username":"stefan.nica"},"ref":"refs/changes/32/698432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/698432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/698432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/698432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/698432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/698432/1"}}},"commit":{"parents":[{"commit":"567a388b95e18eaeff0830259efceeb9b861b2e5","subject":"Fix filtering with API tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/567a388b95e18eaeff0830259efceeb9b861b2e5"}]}],"author":{"name":"Stefan Nica","email":"snica@suse.com","date":"2019-12-04 14:39:27.000000000","tz":60},"committer":{"name":"Stefan Nica","email":"snica@suse.com","date":"2019-12-11 09:35:39.000000000","tz":0},"subject":"Accept oslopolicy-policy-generator path arguments","message":"Accept oslopolicy-policy-generator path arguments\n\nThe Octavia policy enforcer object used by the oslopolicy-policy-generator\nmust pass the --config-file or --config-dir command line arguments to\noslo.config, otherwise those arguments are ignored.\n\nChange-Id: Ic638c056715f8fcea3b095726964e7224c38eae2\nTask: 37743\nStory: 2006983\n(cherry picked from commit 8800e85987a7dd4716fb92466f6a4eff9cb08f3b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9904b26a9c40f29f554c56e9e65f6396caa8fea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9904b26a9c40f29f554c56e9e65f6396caa8fea9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
