)]}'
{"id":"openstack%2Foctavia~698536","triplet_id":"openstack%2Foctavia~master~Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","subject":"Fix house keeping graceful shutdown","status":"MERGED","created":"2019-12-11 16:51:39.000000000","updated":"2020-01-31 14:00:15.000000000","submitted":"2020-01-30 22:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698536-1580423962550-87017a1f","meta_rev_id":"b939cc79e7afc6fe477037d1729b0023ffcf64d4","_number":698536,"virtual_id_number":698536,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 22:39:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2019-12-19 09:38:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-01-30 20:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-19 10:29:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-12-12 16:33:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-01-30 20:55:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-12-19 10:29:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-12-11 18:46:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 16:33:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-19 09:38:16.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2019-12-19 10:29:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-30 20:55:49.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-30 22:39:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4fc53e758d04f0f92796d0eea2ab44b57961c5b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-11 16:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fdb17b440aa1bee60c1e881166b86b99097ff17","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-11 18:08:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b6ff4be20570c419231ded3c28a59b2900b9f20","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-11 18:46:21.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"633727824d957def59bafb03f98d872fce072791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:57:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c637ceab0c68417a8765bd2a97eb6b68 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a42a037272364aa9b7da474c6f55fc90 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7494a5d87774b1a9a93ae4d463d29a3 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76dd688d27f842bd9e78bec8582dc369 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29694010a73146ae992460bd25d42063 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fa3d1a10c94d1ca0412ab47f4d7f1e : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9049d6fa8ef84507928062f274bf45b1 : SUCCESS in 5m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db57c22526304b1bbe161d8607c3df71 : SUCCESS in 5m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a4aadfc94934a0dbc3a7d052c4408a9 : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79461bacf2f14da48eac3d12a679ca27 : SUCCESS in 6m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/33702125d4b1499bb754df6b8bbb6732 : SUCCESS in 50m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e5ce845fa7d42ccbe054c2f6cafa454 : SUCCESS in 1h 34m 01s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad35cae1cd764719a43a04b887696fd1 : FAILURE in 14m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5263f3880c64b49b880f135403308e4 : FAILURE in 1h 06m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e3a209d2735d462da781686d0408b566 : SUCCESS in 59m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/64557c02197d42c8b9903b6a3df14386 : FAILURE in 1h 46m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/618392501bbd47a3b536ebdb56ad84c3 : SUCCESS in 1h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a4aee062f61e8a1c5aeab745a9ff14cc5b9ba40","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-12 16:33:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7c1fa34ab0eff6e3dbd4a2c34afbd775d55af36","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-19 09:38:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe61968e6623f83d8065cfa6f0222cc6f0fec793","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-19 10:29:30.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nThanks, Greg!","accounts_in_message":[],"_revision_number":2},{"id":"3b50543105898ec23cfc20180a07e4a8697b1d68","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-30 20:55:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nNice, thanks.","accounts_in_message":[],"_revision_number":2},{"id":"7294edc4044f188fbb15c153e34270d79df617e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:56:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5fdba8f708700d9380081269241fcdc1da63da35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:39:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/499122fecec1422cbe6f5c2ebd2d25b3 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cf8c0d219d405fa77162e8365f6031 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c90be1a7a004224a14966ec8973c448 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9703ae160d8b42fea8d67c397f440a31 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d0b04d8d7f43119815fd7350d52a9a : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c5334d7b524becb5bae70a0b9aaf3c : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/300f6856f96e464282c66bd90f703bb8 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8f6465cf89a48c7ae75f5b4018d29ff : SUCCESS in 43m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7228f7d61c6b48aeadf434accdf59e65 : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e7c98c829424dbb8f24c65294cd4ddf : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":2},{"id":"af053c23da56fa0db35dd7c217ebae94c7ff6211","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:39:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ce5c40cba5258c4598947a36a41ad6f23c906130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 22:41:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02c12bd4a94446a8e6a936101e203fc : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f71d4b4860224e24b898b1a9ccf310be : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b939cc79e7afc6fe477037d1729b0023ffcf64d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 14:00:15.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 164aed0e676b0fcdbc74a0f5ef26b4c70b70e03e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df0ba06c1e96dc5ad4855d24435970ee926175ca","revisions":{"4165aa9ff2466c6c425b69de2ecf3278bf6cc1bb":{"kind":"REWORK","_number":1,"created":"2019-12-11 16:51:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/698536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/698536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/698536/1"}}},"commit":{"parents":[{"commit":"e7ebd3905a13ee2def9641f2464590f401f758c8","subject":"Merge \"Accept oslopolicy-policy-generator path arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7ebd3905a13ee2def9641f2464590f401f758c8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:16:07.000000000","tz":60},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4165aa9ff2466c6c425b69de2ecf3278bf6cc1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4165aa9ff2466c6c425b69de2ecf3278bf6cc1bb"}]},"branch":"refs/heads/master"},"df0ba06c1e96dc5ad4855d24435970ee926175ca":{"kind":"REWORK","_number":2,"created":"2019-12-11 18:08:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/698536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/698536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/698536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/698536/2"}}},"commit":{"parents":[{"commit":"e7ebd3905a13ee2def9641f2464590f401f758c8","subject":"Merge \"Accept oslopolicy-policy-generator path arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7ebd3905a13ee2def9641f2464590f401f758c8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 17:49:36.000000000","tz":60},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df0ba06c1e96dc5ad4855d24435970ee926175ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df0ba06c1e96dc5ad4855d24435970ee926175ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
