)]}'
{"id":"openstack%2Foctavia~698538","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I96908e6cbdb523f95298aff463a749d15e98e1ad","project":"openstack/octavia","branch":"stable/train","topic":"fix-client-auth-stable/train","hashtags":[],"change_id":"I96908e6cbdb523f95298aff463a749d15e98e1ad","subject":"Fix update API when barbican secret is missing","status":"MERGED","created":"2019-12-11 16:54:48.000000000","updated":"2019-12-11 23:03:44.000000000","submitted":"2019-12-11 23:01:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698538-1576105312671-ad48f38f","meta_rev_id":"dfd4167e751ac37d96b9344602501a65f1960a20","_number":698538,"virtual_id_number":698538,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 23:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2019-12-11 17:54:57.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2019-12-11 21:16:57.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2019-12-11 21:16:57.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 17:54:57.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2019-12-11 21:16:57.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-11 23:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6219e15531ff8b25360549870b4be78f53f11e7d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 16:54:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"82f52d7bcddb0ab86fd10aa416197c550806a1b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 16:55:06.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit bac03aedf0f861dfe81893b5494e60baf1206d0a","accounts_in_message":[],"_revision_number":1},{"id":"ff426090bfb68f9e8d1ee691ae01223419eba836","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2019-12-11 17:54:57.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"97835f750c31d20766104cbdc3fea40c546eb332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:07:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d3eed349144931899d17ab55aa2faf : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f80f5f2a3e74de592e2d7dcfbd302c0 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee924b6cfcc45e3b3e2cf2ce99cdf6c : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c95f2e4b6044be5b564e6b69ba3747d : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b94c5418e20e4fc886655922eab9afb4 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6d954d62f3e43fc93a6afd81db9b627 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48434177685468194a2bce47cbb787d : SUCCESS in 11m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2259507451b0452eaa2f5722ffd3f49f : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/76489edb1dca459183c8bc016a92beb4 : SUCCESS in 6m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b1b5d806caf48bdbdf1f4e53237fd0d : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3214230745b4992841599c4994e22c1 : SUCCESS in 5m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13917200f6e64db7bd90f4595b9ae74b : SUCCESS in 52m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0b5d8bbafd5c4605a3d82b2fafc75f0e : SUCCESS in 50m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59fcbc6b516f43cdb1c9c18833410281 : SUCCESS in 1h 54m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9e699383a0424151b2da5ce7c28a8bb9 : SUCCESS in 1h 26m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/747d91c7b1554620803e950ab89b0cdd : SUCCESS in 2h 05m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf301a1cd345460fbecb1b8bd4f6dff0 : SUCCESS in 1h 47m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5feb52c2c754f4d8c19b5fa059f859c : SUCCESS in 2h 01m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f30ac0a5929456c8754bedf09f83aa1 : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6aa7ae7faa074cfcafba39cc6df31176 : FAILURE in 1h 09m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f6cfd56b88464e668b38fc4aa59dce80 : SUCCESS in 1h 45m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77928980ad8349d0a8a8ea70b9d79bba : SUCCESS in 1h 09m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d149494d62964b14a84db564f4c2513c : SUCCESS in 1h 13m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b76c4933c9cc44b1a9e6f96be6f31cf9 : SUCCESS in 1h 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d210f9be34838396287d273f8bdd60a46ba629d7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-11 21:16:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f470bc45c94c9b0a3f3fbd3f16d14448020ec9f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 21:17:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32059c0ef63f63843bf012848a0498219ad78e21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:01:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26951f74996b4ee6b0287b1c29e71155 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8537770560e648bf9add75605e139d36 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00cf46d86268483fbe7b62a6ebd9b975 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be32809d0dca4ca0aa7cb86850414540 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8b864a14979495fa2ed0c9480a7a5b2 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13090fe5ed194df48ca7979bf337ea1c : SUCCESS in 10m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dbc9268182f64239bcd3e1369da964f2 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd311c1a9cf34dd297e8b65a8f84293d : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9851a70a05a1434da03e549a62f54d73 : SUCCESS in 45m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0e72cb146ce54d86b9563a18c1f32fde : SUCCESS in 56m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4214edd58b324882a803b50f1c8a3b50 : SUCCESS in 1h 32m 30s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d6b26acbf374bddab5c29e8c94e023b : SUCCESS in 1h 38m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72536b56f2fd496c86d8709ef24a3e8c : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e70e8d978e402e9c69b81c45ef82db4029441f40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:01:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"120ca234b3456a2b76bdbd840b259190ffeef875","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:03:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a6e1c69ab348b788c837e47e7bf3ab : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"343e3d8f38dff0310d314a0ca8d7d44d53d506ed","revisions":{"343e3d8f38dff0310d314a0ca8d7d44d53d506ed":{"kind":"REWORK","_number":1,"created":"2019-12-11 16:54:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/38/698538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/698538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/698538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/698538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/698538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/698538/1"}}},"commit":{"parents":[{"commit":"19202863e2e459a0c81a872df7dc745aad4d15ea","subject":"Fix filtering with API tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19202863e2e459a0c81a872df7dc745aad4d15ea"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-29 21:15:13.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-11 16:54:48.000000000","tz":0},"subject":"Fix update API when barbican secret is missing","message":"Fix update API when barbican secret is missing\n\nAPI update calls were blocked if the load balancer had a TLS\nterminated listener that the secret has been removed from barbican.\nThis patch corrects this problem allowing users to update the\ncertificate reference.\n\nChange-Id: I96908e6cbdb523f95298aff463a749d15e98e1ad\nStory: 2006676\nTask: 37322\n(cherry picked from commit 5af97a248b2c20cd12d743b124fec9848b1d965c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/343e3d8f38dff0310d314a0ca8d7d44d53d506ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/343e3d8f38dff0310d314a0ca8d7d44d53d506ed"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
