)]}'
{"id":"openstack%2Foctavia~698568","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I300839fe7cf88763e1e0b8c484029662beb64f0a","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I300839fe7cf88763e1e0b8c484029662beb64f0a","subject":"Fix multi-listener LB with missing certificate","status":"MERGED","created":"2019-12-11 18:28:18.000000000","updated":"2020-01-11 20:20:23.000000000","submitted":"2020-01-11 20:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698568-1578773926394-c94174cc","meta_rev_id":"7eacf26a964d47cc7facea0e408d46331d233b61","_number":698568,"virtual_id_number":698568,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-11 20:18:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-01-10 22:30:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-01-10 22:30:48.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-01-11 18:17:20.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 02:52:45.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-01-11 18:17:20.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-11 20:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a21b20c9155187916e11a4fb55d3538b5d9c96d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 18:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3781f09df80734994a08f491d6f790e58f19a109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 21:08:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75803d949adf4e66a555ca715b0418a2 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c9022734c104f719f16bfd8d5edf6d7 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54827a4c495744538051e2813ee21b13 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41d44e445b764d4fbe94db2ad6cb87cc : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffdd63f5390d4541901b1f855331ed98 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d83d1e2c24a34aba828610b023c127f7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7c97383f3a43d4a8edd723c02f18fd : SUCCESS in 10m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6e2ec10a92ed476abf7286769788a828 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb22705b5eb14f9ebc37f3ebd243d7a9 : SUCCESS in 7m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/63d6bec1b66d4e8186f849cc36706cc9 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a1200d4e7ee42a9817b9bf6cfd1b2bd : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9b93e9382ca4599bd1a792fe76fa324 : SUCCESS in 48m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b75d4dfa74f546d4aed072c78b96b536 : SUCCESS in 55m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/863e83dde820489a9b3d1a6eef0cfca4 : SUCCESS in 1h 52m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/60a324df83674dd980e25bc32a3ad6f2 : SUCCESS in 1h 34m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/27946ba87ac544d99c5bc81ad37fd0ae : TIMED_OUT in 2h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b625ee5316ae4568a841493cd917368e : SUCCESS in 1h 55m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/433816eb46f740168f987a77acec4826 : SUCCESS in 1h 47m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb2d9a49adde44baaa53f44714713922 : SUCCESS in 1h 16m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86f3546865574a86a8e58a51c9e15bf9 : FAILURE in 1h 07m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a1da9dc2360f4b6a816d084ca8a8bbd7 : SUCCESS in 1h 33m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e51a74b175c34b6c9faf01d0cebb95c0 : SUCCESS in 1h 04m 05s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/980d14a000f3457b9f60bcb6772a3dac : FAILURE in 57m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/61150d6896524f10b6d1692f86842f6e : SUCCESS in 1h 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dad1ce02d5fbba9ae2a65b79bc76990053ba9ebd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-11 21:16:19.000000000","message":"Patch Set 1: Code-Review+2\n\nThe barbican job failed on test test_tls_multi_listener_client_auth. Patch https://review.opendev.org/#/c/693586/ needs to be backported. I take it will follow next as one depends on the other.","accounts_in_message":[],"_revision_number":1},{"id":"7f74c805a5fa0b86ff5c01512ade10d8552fc925","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-12-16 02:52:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9748348fedd8e206935f722cb6ff12e954bbb7e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 02:52:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86e3f56130c49f2ca7f6995505a8e1ca1c4e7f8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 02:57:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4a9fb540aa049e29b253dbb3355067d : CANCELED in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13785a3c7ae040e59e7f7bc3bc905ce2 : FAILURE in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eac8b2ad5b824ec896c833c52e2d374c : CANCELED in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a449f663474453ea9de858af5742bef : CANCELED in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae72dd211bd84674acafdc2a90f93476 : CANCELED in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa265cbae074eff92b9bb132ac1469e : CANCELED in 3m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a90f953d15f14858beca63b505bfce83 : CANCELED in 3m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f9e57413d744265b6923541757359d4 : CANCELED in 3m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3947d8352a1d4755a065aafa58371476 : CANCELED in 3m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/80bead2cafe54da48a527401d1224cfb : CANCELED in 3m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/05e21075dc174818ae5f470f190e4d98 : CANCELED in 2m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dd72256227cb47e3ae36f3d3a325870b : CANCELED in 3m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bc6a6bdd8954af49e38a18be65d9a97 : CANCELED in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"855247e538e2b81287356aeb231ee8a9927e2360","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2019-12-19 08:56:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9b1870c331b8455011ce1e87d599ea01258e70f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:27:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dca8099d813547d9bca3a588250a22f2 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f03b30288d2640c58b4731a15f8e397c : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d52bd5e0e2d441d9be389cfa34234f6 : FAILURE in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae65a93b1de24d93b1e8123e28d0641d : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/245b352000304dae93e222d8cc91e9ca : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3796c167cad84c05b9625f43f6165aac : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da20e0b1dec49d0844ea257b0070499 : SUCCESS in 16m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/daff1288c02b4d14893ca453c5efc749 : SUCCESS in 7m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7d4e4f0094e34861ad55aa8c3c4bdeeb : SUCCESS in 8m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0fdccecec60b4bbc9006762e8030e6e9 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc074f339d824c39878eaa2c533d16bf : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18a60fcb7c8b45b3b96417f1f4240173 : SUCCESS in 45m 55s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a8610c8c57ea4b869f27a1d93652f473 : SUCCESS in 49m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7df45c929f449ddab1efcf096c30afa : SUCCESS in 1h 30m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c20cd109963e434092dc02a5da0cf694 : SUCCESS in 2h 29m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/be1ef0e9143644c49c0869d0764e65dd : SUCCESS in 1h 49m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b688259a5999464bbc871d4de86b358e : SUCCESS in 2h 00m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/16d495b1352d47a19129f9b9cb7f23f2 : SUCCESS in 1h 55m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6949b67bdd7e4816b836dedce9c8070d : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6790324c38854bc5adf0fbac5c998266 : FAILURE in 1h 07m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fcef79d4592f48848e52d6e8a3ab524d : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/992f14901d8741d7a4f06b8306c05375 : SUCCESS in 1h 12m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d1712fe4bc344c0fb33069a5a3631255 : SUCCESS in 1h 12m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d2f92b03f2244e3a96baf76030d37e24 : FAILURE in 1h 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09a9eb81f46f0c0641ff58322b9415ad21939708","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-10 22:30:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1b780a071206c02599c1fe260628e8e86263fe2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 00:43:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7f632cb13c8409794ea756031e1435a : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8711bca039a46c9875a94a666887a68 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df57c797185041c78786ace439901599 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebe9157ea77e4677ab942d3320b6d9ea : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/911c76625bc94092990cf734504217a4 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e901352307c041eeb1893c5c3d387add : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43995c5959d421ab0470c59c88324e5 : SUCCESS in 11m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dd27cb17da3b491da71e065339b7875c : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d3ad3b867d37416f9203e807481b425d : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/680fb1cfdb2c4028b1fd42083cd51a51 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aae411cf72c4625b22c2682560e2540 : SUCCESS in 5m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/937d3b8018d1420eb0da1b25b025abe2 : SUCCESS in 41m 44s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/82e2946cc9ad469fbd2537741a9a5a03 : SUCCESS in 45m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcb72e5696454e7389c694a9d324f50c : SUCCESS in 1h 43m 11s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f5acc054a73447068aebc3f3f621c1d2 : SUCCESS in 1h 49m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/02b16a170dc04fd9928e89c0c97963d2 : SUCCESS in 2h 11m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7680bee0bf4a4b4db291f8e0d74ead94 : SUCCESS in 2h 04m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f27520adccf0428fabce373b387e661a : SUCCESS in 1h 47m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4d2a8c0775004f98a79ac52d218a1ad1 : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f9923517723486aba201568da399cf9 : FAILURE in 1h 00m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1dc4360fd394fb090667d005e1386aa : SUCCESS in 1h 35m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ce55598bc984c82b2e757599e4453b1 : SUCCESS in 1h 08m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d66befe4dc141b7803665f9cbf8f728 : FAILURE in 1h 07m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2f64db750c3e44369c0c6941fd0310a3 : SUCCESS in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04a62e2d6ae2fa4e2dd66454e5df95d550a5c68c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-11 18:17:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d47fa16c12e336297045a10843edc701a2b0a9be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 18:17:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d3428f3fb3cc5931d50a9db385e6b6ef530c905","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 20:18:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad9ee6fda7edcb840939c872518312e6c5b2c090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 20:18:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2f87937d96c4610b7bbc192c21a1619 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79594e9d884e436fb5112078d6859bd8 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3270d16466074969a98c8c1ef616836f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a246017a20a46a4803d7b813175c3b2 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b08c84c18b74a5d8ffe1a8b72ffc018 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7215b173b2d4b2696a6f3d10b249074 : SUCCESS in 9m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b7d7faf48904fc4b8cc11368aa2bd31 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b085ee83c6742999fb76573d432ff9d : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0650aebaf12a46b3aeefd8fc4f48ab74 : SUCCESS in 46m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/16fe0b0295624b61a59b32752c240e60 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14b63f0c615b4aa79efa2dabca33b079 : SUCCESS in 1h 36m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8a5e07ca12c548f0963446ececa8fd19 : SUCCESS in 1h 55m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/709501a4345147bcb2a2cc3609771df2 : SUCCESS in 1h 13m 23s","accounts_in_message":[],"_revision_number":2},{"id":"7eacf26a964d47cc7facea0e408d46331d233b61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 20:20:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff1ec86fe2e4fe7a935b674de31e898 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df5db0b000f69dfd92e640a09dc98344ef3be731","revisions":{"d46df21e4594df3b2520026248a87ebbb0bb2e91":{"kind":"REWORK","_number":1,"created":"2019-12-11 18:28:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/698568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/698568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/698568/1"}}},"commit":{"parents":[{"commit":"19202863e2e459a0c81a872df7dc745aad4d15ea","subject":"Fix filtering with API tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19202863e2e459a0c81a872df7dc745aad4d15ea"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-11 18:28:17.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d46df21e4594df3b2520026248a87ebbb0bb2e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d46df21e4594df3b2520026248a87ebbb0bb2e91"}]},"branch":"refs/heads/stable/train"},"df5db0b000f69dfd92e640a09dc98344ef3be731":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-10 22:30:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/68/698568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/698568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/698568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/698568/2"}}},"commit":{"parents":[{"commit":"85f7abbbaf958458ee12415b0de0e938c20fb121","subject":"Cap hacking version to \u003c2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85f7abbbaf958458ee12415b0de0e938c20fb121"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-10 22:30:48.000000000","tz":0},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df5db0b000f69dfd92e640a09dc98344ef3be731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df5db0b000f69dfd92e640a09dc98344ef3be731"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
