)]}'
{"id":"openstack%2Foctavia~698597","triplet_id":"openstack%2Foctavia~stable%2Fstein~I300839fe7cf88763e1e0b8c484029662beb64f0a","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I300839fe7cf88763e1e0b8c484029662beb64f0a","subject":"Fix multi-listener LB with missing certificate","status":"MERGED","created":"2019-12-11 21:48:30.000000000","updated":"2020-03-08 14:10:58.000000000","submitted":"2020-03-08 14:09:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698597-1583676580552-99049c42","meta_rev_id":"66f86960acaa1fc5276cf0d6eff9cd531f3ea925","_number":698597,"virtual_id_number":698597,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-02-04 06:04:56.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-08 14:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-02-03 21:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-03 21:40:56.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-02-10 20:53:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-03-08 12:15:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 08:18:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-10 20:53:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-08 12:15:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-08 14:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf53706b664b5b3522b43aa379c4cd7337382bf6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 21:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdaa1255fbeab90be4d7022cc5a92a5bdee4415c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-11 21:57:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"33d84f68e1373a8adbfc933e0b593e0454acdfc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 23:50:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdb9c79d30fc437cac491ef6ba736834 : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41e3ffada3c04d5483f00283de588056 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bfe66fbeef4d659b9675be0a89ba49 : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3119429a38b74711bc2469b3dfeff4f6 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/828a1040c0284921a1c29c520476a87c : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6f1f36ebbc348809d6e2e289c7fc94e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a018b573b4d04d1396fa70914e654ff9 : SUCCESS in 5m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/3dfc7a6cac614122aa28985d5d0f6f84 : SUCCESS in 1h 14m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/185456fbe7924049a3eabce7921cc6c1 : SUCCESS in 1h 13m 17s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bed36c220a354d90809ed451dfd4ad58 : SUCCESS in 1h 13m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4c053d7c89a24090bb2a6b379d34d2a4 : SUCCESS in 1h 18m 01s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f68fbe7a9af146ee9609a98a14723767 : SUCCESS in 6m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/093039c9f8f6462da79c8016ee5fa1e8 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df58a6c9d5be41aea43f0d88aad7af5f : SUCCESS in 5m 35s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fd16842ac3647159388a766841bdb83 : SUCCESS in 1h 02m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4696c2f6a0ce48b487149edd9fa4ecdb : SUCCESS in 56m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cb479b14daa4eb5b0a29af630f31f7d : SUCCESS in 43m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a1e962fb3656487585e8758709f075f8 : SUCCESS in 41m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ba2898a8bf4457bad6022a5d8726747 : SUCCESS in 1h 21m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/55941d12a0354c429e912f224971bdc8 : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9deabcc655c846eb9e0729d4ad6fbe7a : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/32c6a77125194fb5bb13fbfc9b125467 : SUCCESS in 1h 53m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a176690fe1647889556dad2fb36c659 : FAILURE in 31m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e85efb917f64b6b9926ba94825e19b2 : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/09f5a0cd6bab4e65b032f9aae139f85e : SUCCESS in 1h 55m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49e69177e5f64be0aa2d936a2aa91cf0 : SUCCESS in 1h 11m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13f7687c3aae4349ae9eea35771e72a0 : FAILURE in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0424858284c27ae792af85c3526a89c90765a80e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 19:34:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9c0e3d93beda91ae75670851041bd433aa2a661f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 21:56:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18cd0545064c4c49b8941c538e9fb5d6 : SUCCESS in 11m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79f1cc39b597451487af4564aef77dd3 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b53098c76b94c44b47e4d5d5ea1ded5 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e348f1ecb71d467cb813d065a0fc10d1 : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90c09fd46741409c8098f2aba891f286 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41a3011825094b81a4ae4b92e1dc3880 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fc2cf6effa42569443c440a3adeecb : SUCCESS in 8m 26s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/71fbf8b09199433aad8d8fcf530826bd : SUCCESS in 1h 17m 42s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/84464c777ce7452cab2a4eb3b2a12661 : SUCCESS in 1h 13m 24s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/308a471acad84b159ae0d20a96eaee6f : SUCCESS in 1h 24m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4487450444e34a57907e7776ca3283e8 : FAILURE in 1h 49m 49s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/35d1661fac1f44c3975baa6b3b8234c9 : SUCCESS in 4m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4cbe9ae3ce3e420ba095a4b6841b382b : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a278622efad48b38ed5361548384ea1 : SUCCESS in 7m 28s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1c5b72e91a844c0bae108c005d8a3be : SUCCESS in 1h 20m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/49084ffb0c434625a96a0d0c29a28279 : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c028023a6f040d397d91a25a962a0cf : SUCCESS in 48m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5534338e83a446449f62aa22d75cdcab : SUCCESS in 1h 10m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/600d85f2bd634fe8aef56c86cff21c27 : SUCCESS in 1h 34m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7e20cf43e06e49f0bae5123e72b98a33 : FAILURE in 1h 34m 14s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/34edd6e3bf2c4d37b1708c27cdf0868a : SUCCESS in 1h 47m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/446f2d4e57d5466baed48d7f235da645 : SUCCESS in 2h 17m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/098265480c2c444785f32f029da35f46 : FAILURE in 1h 11m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac8c33c19d7a466ab8e612269d578831 : SUCCESS in 1h 43m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/da5ca3c0e755466ebdc9f510021d1297 : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f50a6bdcb6d458d96e4b8f3f9792caf : SUCCESS in 1h 41m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7f30d1edb1346d78d808fda07b8809a : FAILURE in 1h 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9715700176f22a3a2a2a43f0343ad1f1f173d78","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 22:07:37.000000000","message":"Patch Set 2:\n\nrecheck\nnetworking issue","accounts_in_message":[],"_revision_number":2},{"id":"1264c676f200a7d61651e2a70ddef2d19dd18fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:14:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d580140e7ad4db2a3ac0ef005c539d3 : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/351b2b44ae9343bcb36c6324175bb9b6 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/420d952d1a45408c97e2fd39db2fdd30 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da74c8fd8a2c401b868276991675358e : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04a7f77b95c24d7aba5145e14582792b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc7b9b13c7d4fa0b9fbe8c6d758a191 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2f80b1cde54f969d973045485d4536 : SUCCESS in 7m 21s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/63160ee0e07f496f886615ce1787d69d : SUCCESS in 1h 17m 07s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0601dc9aebc14c44af17caa45363e44e : SUCCESS in 1h 47m 45s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0321ae3254284d4db82abbec637cb114 : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dd69db53fe8645ee8a8208350af4d117 : SUCCESS in 1h 24m 27s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/74368024d79343ae987bd3ab8ddd98a7 : SUCCESS in 6m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c77050b291184e0dbccf9f1bba16e16e : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f6fa38c3cbf4dfebff721936de798cd : SUCCESS in 8m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48a64011f1db40068bfcbfac78fa14d5 : SUCCESS in 1h 00m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2909f64e00944900a6164bab4c0d5ff4 : SUCCESS in 56m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4311e17c02c448df8791261e1fd17875 : SUCCESS in 55m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ae3eb40dcebf441d82d5e477184b88bc : SUCCESS in 1h 00m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/180522106f284cfabb1512fc10dc59cf : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/60352630c8664764b244414d0456950b : SUCCESS in 1h 28m 23s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/7af01ae8d56748e1b73edaceb5261cbe : SUCCESS in 1h 50m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b99517f667d84802993874f00fda9531 : SUCCESS in 1h 58m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b694494d91574394848ec1c2b3d72b1c : FAILURE in 48m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad4e9106a6a04fbeabb3e3998641d1bb : SUCCESS in 1h 37m 32s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe29c288727d446a9a5ee458976aef6d : SUCCESS in 1h 37m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8dd2adc6792c452fb997a951ba827b69 : SUCCESS in 1h 03m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d81290f6b48544ebbe564370fc0d1fa5 : FAILURE in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c910c38de9a2b83bba8bbfd881408070c4ad9eba","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-24 08:18:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cfe5547f8d1e6a370c836dbca84bf0bd9fcf84d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 08:18:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6fe8fd636def61c69e2fff75be84a091daf0fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 16:28:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b178eed558ad405789e85bcd86846b58 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb7bf77c5fd4e509d39c3dd1e77d120 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b721e2a490849ae8fd8522b41b9d1cf : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/84d0dc43cdb543b08178b0421f6607b9 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33210411a74240f6a803cf91c919e073 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76d0e899c2e41d38045f514fe7c481b : SUCCESS in 10m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e18bd2f57e704850bded642bb34c91dd : SUCCESS in 8m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ba5c0d43f7934a6c99f61c4b43cdfdca : SUCCESS in 10m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbadb8ab1b4f4dccb162e941696e9434 : SUCCESS in 9m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f8e7c7da0014879ab1a70066918290c : SUCCESS in 59m 53s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/de7ed18faa9f45faa718761dee6db79b : SUCCESS in 1h 13m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2365269931744717aa70f864ce0b0f41 : FAILURE in 35m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c14262b7b96e4a1dafb5cf6f51320368 : SUCCESS in 1h 21m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3cf09ac74bd48b1b37e98794e4e9374 : SUCCESS in 52m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a3b106639a8f4e0eb26247a22e592618 : SUCCESS in 59m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6342fb5649114429a22acdce48f5aec1 : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9eecb0f0143d44c094ba7bf012dea5bc : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e9acdd36f6be44309a14e8a08f4dc867 : SUCCESS in 1h 29m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b38304c71844313900d9969dae846b0 : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d203d19eec468398764dc6d1dfeaaaded8b2acda","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-25 11:03:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb43f62a8b2be21765976bfc059427aa5132258c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 13:30:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3f67b0852424fccb6f7c0312694adae : SUCCESS in 13m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d13c3db88fc4bd2b150ea1ab34490a4 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e5e394036540b6bb1d8ff6ca39b713 : SUCCESS in 9m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/616db53e14244fdca6af28ef823d63e2 : SUCCESS in 8m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/41da61076dc74dc48dc79873f8657354 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f3cab04f314ca6b80c4ce300362520 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe70d48a9314219adae6a04047ef691 : SUCCESS in 7m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/caa9b985a1054583a4a02d13f09c36af : SUCCESS in 1h 13m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b7d5eb4a4eb5426bb85b817fa42a73f0 : SUCCESS in 1h 23m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ffd17330d3f84e66b49e2298b6135c67 : FAILURE in 1h 01m 23s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a9946afcfa584b5c9b848408e1b9117e : FAILURE in 44m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/35947a8a46d94983b3c016d5c72a80da : SUCCESS in 6m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6f55b45d4b7d40d5861ff2ca26b1b193 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9759949c8a954218b803659adcf751fd : SUCCESS in 10m 43s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f8e35e4e7fe4b18bd4a86609cf7d191 : FAILURE in 34m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/aa9b695d24704a5fb1bad8ced6de845f : FAILURE in 36m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c8b4b07c5a624ff8a95249ff0f50a13a : SUCCESS in 52m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/31d9180e79c644b3a07e216a2d90d4e7 : SUCCESS in 44m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4562089772c44427a95df721ec86a1b0 : FAILURE in 35m 00s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7dd3e92f4f9c4fe8bc6c82be90da9206 : FAILURE in 43m 06s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e52f54728602429da369e6cc9eccccc8 : FAILURE in 16m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/63324a66d8454cd4b919b20e881ee5aa : SUCCESS in 2h 25m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e42ea94b9be4883b65402e34d5e4db3 : FAILURE in 45m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38f8fcde7827465ab1c0509d83cb3b9b : FAILURE in 34m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b3a69d274a846b8be4021aceb00a118 : FAILURE in 36m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cbf6eb53715b4310adfc7c2534b58046 : FAILURE in 39m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b5c7e5586e445f5b7429c7ea64379cb : FAILURE in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57cbcdb589aa1f184cfd9b79511b10c44ec17e80","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-03 21:40:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0b9d2d026f7357e1f73a062b9ab9d04da2aa8747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 23:42:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/936cf3d2e51a47e8a29713c84ee7c2e3 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecc13593fab7473d83f85920fbc9ed45 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62b469b13824398a3237d08018a0642 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7b7a75f124e48cd906b5a52a31a19e0 : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64240ac9e6804a9997d2f877a28afdda : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bee2630721514b1dbf25ebffca7cf0d2 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13016ebf1c264477ab4e5be9287898fc : SUCCESS in 6m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8a84fc6cf8764ed5be5fa9960805632c : SUCCESS in 1h 23m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d0df1c1c428649a49a287e8e9eacf5bb : SUCCESS in 1h 28m 22s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c56246d342c4480a51c928ae1de186b : SUCCESS in 1h 20m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5126563a62eb4f6ea934d3d02ec8ff5b : FAILURE in 1h 12m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d22ddd626fc94c34a6d04b188640b6bc : SUCCESS in 8m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e88ca60716ce419fb5609fbd6fd9f8aa : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ad250ed41a446918d662a681989a1e7 : SUCCESS in 6m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2420be0aa00f43648043868e0cb4b40c : SUCCESS in 1h 10m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ef6ac9db6dc443afb8eac41d3414c86f : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/851d651dbc094127bc9e7be54542e168 : SUCCESS in 53m 07s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/384a7667856141a5857eb4b69b337ec3 : SUCCESS in 44m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cffb4e3fcca4b7391a71df524535272 : SUCCESS in 1h 29m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/28349c53cb7c4f68827035c74acbd5ef : SUCCESS in 1h 37m 07s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/3516c4d67f3447f69792b97aecbb9589 : FAILURE in 31m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ed2723f2f0548f7a907aa98030ce267 : FAILURE in 22m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc461733b65b47fd841712f10af0d0fc : FAILURE in 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a33f4e43cf6408f87ea6870ed5d793b : SUCCESS in 1h 51m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c79f6167ebb4396aa3edad4b34f0a9b : FAILURE in 1h 46m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8abf0bb89a2249c4abe01865f0d1ba02 : SUCCESS in 1h 51m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/648cdfb4f8034b65a7f857cdb8aa157a : FAILURE in 56m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"029f64a60832452afc5eef1598076e95316fe767","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-04 06:04:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"da05c93c415247f6e26e00ce3338922c7b3846e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 08:24:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aada9701c1ff4462abb809950e7a2147 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/249d9e0043094aaeb8f8d5d16b6eb222 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78bad842152344f290c9140fd490f908 : SUCCESS in 7m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/539d347531f44e488bb7d2c75fba2aaa : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f8724e2f9994769bcc0970e4fcdb0bf : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82e858d86eb84c1c8c90bad6c83f518d : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef24eda6c3d74d29bf80e34ed32c4e5e : SUCCESS in 7m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/45a32a8a01af4f6ba2ef0d11f33f1be3 : SUCCESS in 1h 17m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/74cab11b20004045830c0f0bb5f911ff : SUCCESS in 1h 19m 07s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b174c167ff2b416a810f8f1ebb868992 : SUCCESS in 1h 24m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bcea441ead684cce95dd1c51fed99d01 : SUCCESS in 1h 46m 38s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7770fadec9e04fd19b1281f60ec413df : SUCCESS in 7m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fe9ebcd3bf6b40d5ba7192bfc59c26b5 : SUCCESS in 9m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b297b682e5734fbe855d684977dda416 : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec485b7f93e2434a8b2dd9f8df25fd9d : SUCCESS in 1h 01m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1211004b887e469c879f38ec72efc470 : SUCCESS in 1h 01m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e18d8b4634304e9ba4d16f614a769b44 : SUCCESS in 58m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/83d05b5b3dc64bae837d36d7a078a523 : SUCCESS in 48m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e68b5d20349b4355b4b840b4103deb65 : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ef490bb1b1ba46cda65b2f4202649ca4 : SUCCESS in 2h 04m 42s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/510e81c5cacc475c8425a09c2c54a3d2 : FAILURE in 19m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ebae803dce8a484b8196a478e8a16d80 : FAILURE in 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3506670477f471b8a5bbbd42cd404e0 : FAILURE in 54m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59a715a38fc04c8ebea171a836b0964e : SUCCESS in 2h 07m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/41e4e617b99844f39bba8a5cd0054253 : SUCCESS in 1h 40m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b60e20eedac249b88f36b072e7f71409 : SUCCESS in 1h 11m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e32a3163a1f4c9aaaab6ac280691185 : FAILURE in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1dafc041235bde1c38b856c60094c18d0e0b112","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-10 20:53:05.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck\n\nI think the issue with the missing py3.6 interpreter is fixed...","accounts_in_message":[],"_revision_number":3},{"id":"9341cd9df4b85a75bbf71e386c06f8467ba14208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:29:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d11dc16bf374f46b3db028061fef574 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7677c2d5222b46a9aea92bc137cf8d61 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb39f1cf0c9e4f3fa48ac7780c47cf75 : FAILURE in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a2de3a6d945448a8c89ab933a7ef7a6 : SUCCESS in 5m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8e8649ddc5cf4fbfbbb2f497cdd5963e : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc95ab9195d49c9b640d73dc2dd0402 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de6e658566746a4881142b9be075986 : SUCCESS in 8m 40s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a9e1b11c0a3540a3943f5429f2d1a3b1 : SUCCESS in 1h 19m 38s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2c683dbea6fe4f80a040e9b321b7de5d : SUCCESS in 1h 20m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49161178633b4ccf884dab14b0cd02ba : FAILURE in 31m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bdf959e71c8843ddbebfa5055cabbb3f : FAILURE in 35m 35s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2384c26c349244e3b68f696c6e8e2bde : SUCCESS in 10m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f0ddaabecc8f43dc8335703b554e2451 : SUCCESS in 13m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2eb49c3f7264229a84838b1e22fd825 : SUCCESS in 11m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce580f70ed104f22bea5c4f7409c1279 : FAILURE in 29m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3241c3e18b2342ddbd86b6fd77b2ad99 : FAILURE in 32m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e25079b30a3c4bb680c46b4ebbc1dab6 : SUCCESS in 42m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/22c2b349e1184a47a74bc04b2ac13ffb : SUCCESS in 46m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ab81e9cf5874c2da24cc85264861c12 : FAILURE in 33m 52s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/011311c3c80b486ca9fea47fb83a1814 : FAILURE in 32m 59s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a16669de8abb4d0eb912e3c0b07906d9 : FAILURE in 21m 59s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc0d8ac651a14902bec4e0a392c4967f : FAILURE in 31m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fda0790cc844afba87bf381a8d5f5e3 : FAILURE in 47m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a31b2b9807e94dde937205abc43d86e7 : FAILURE in 38m 50s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b7551049cc140cea24407bfc4ffcdcd : FAILURE in 31m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/851ca8a628dc426a9804da89ecabdf8a : FAILURE in 43m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e5dfb3924f6454aa7f234489ba8a0b7 : FAILURE in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f0cbb18961d3c0ad3ea79c9d8cad3f042fb8c56","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-08 10:05:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"270dc076d0b1b416ddc6693b68d26a2785dc19e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 11:59:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dcbcd8482544e759ee3dbe24509f53f : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc7b98389e6d439e87428b776f2d4506 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49e74ee210d42e1904c22f27dea7df8 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51545de6ebf044ff8372917ace79b596 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c96fdb1ac9ce425cad79188825dc8238 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b434ae1238e4c6f8d3f1ac1d071a980 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479644aede6045e7b604ced5899fe92b : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6e7c837c66214229b942464760c85da7 : SUCCESS in 1h 17m 45s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bb46cf7c11b94b97a2bca97ea82afa08 : SUCCESS in 1h 15m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ab89e071b964fd69fa2b0e1632da268 : SUCCESS in 1h 19m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1d64f9822c0f4f5885e4b01290077622 : SUCCESS in 1h 19m 05s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b39558014d3f477fa09e07ca4aa0eb3b : SUCCESS in 6m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a9e6acc01332424ea2a1099f78435821 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61c5bab374d94938a1500a0d81e442e1 : SUCCESS in 6m 13s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/678ee1de272d41578b80c9c4079229a6 : SUCCESS in 58m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b2853887d0d746ffb862023f8ec45061 : SUCCESS in 56m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edec8a207f5d49258b11f7ce3d91fcc3 : SUCCESS in 40m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fbcb9afbf0864c5ba0cf39fe83b57cf8 : SUCCESS in 42m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/373b6552d4574fcc8601214fb79e814b : SUCCESS in 1h 49m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/28277539d5a54c42a0634917096b7de0 : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/68b5b0f03e494d069024e5e14cb078e7 : SUCCESS in 1h 46m 43s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cfa97577d25341969eb2984e02cf2db9 : FAILURE in 19m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4b04a004549442888a5eda50a4e1154 : FAILURE in 31m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36260c07e9b849deb3af89f832e55d5b : FAILURE in 1h 26m 36s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c790fc6073c4471c957f6de2aeeaaeed : FAILURE in 1h 27m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52782d990d954ea69c6deecb2e070da6 : SUCCESS in 1h 08m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb4b6690c3984151b0119ceed84c3e8c : FAILURE in 59m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"64d924c8be71ce28785b5496643b6c34fdc6ca40","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-08 12:15:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bc7785b3da469d291f9abbedbad89e8d4ef97e7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 12:15:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3cfebfe9a4bc9b37d4fe7dd1828d02e36962066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 14:09:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e791accd9fe4d77bb96d47bdb241dd0 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc749d92c8094a539fc022e8154194f5 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9003825d80bb48c8a524effe1d6bc5e5 : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/62c6f5e0d2fd41e6b61c72369e15e0ec : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3d00c666d84e01891be84a8b60c4bf : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca3ecc5203446a2aa67d61fb96b1186 : SUCCESS in 8m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/07e2bc3555f84694b8422a6adc80bee4 : SUCCESS in 6m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1c400bc2aef04d02a71ac98a1bb3fed9 : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39af22ee358b4adf8af972ed8fba8284 : SUCCESS in 5m 52s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63fcc32950b148ad98c5928019aa730b : SUCCESS in 47m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3c63d5af81b441d0a8992803d71a0582 : SUCCESS in 1h 01m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/e92654fad8b549c6940baf9c9bc22575 : SUCCESS in 1h 15m 24s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d226b25fa1d14b5db21e9465c16584a6 : SUCCESS in 1h 19m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e225ce3cbc34ee0ba236d89b4aae566 : SUCCESS in 39m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d6932e87f4454e9d81f44b30c071e505 : SUCCESS in 41m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d8552c75bfd94c1db468424fced95143 : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c46bfb23c4dd47a9b9ce3cb62c37c6b8 : SUCCESS in 1h 30m 07s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/1938bb94dea94d7582b1d05cbd032480 : SUCCESS in 1h 49m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c399b5df3e048ffa4669c5f0f53c33e : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1f4ce685db1db5228128b8afcd5d5884a97d7996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 14:09:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"66f86960acaa1fc5276cf0d6eff9cd531f3ea925","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-08 14:10:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e95b5899bb44f688654e32baff55ea : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ebdce39bc47d368ef3bc096dfaa45727091ede7","revisions":{"705e2ed4568337f13302984ab58fbfd5a70d8c00":{"kind":"REWORK","_number":1,"created":"2019-12-11 21:48:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/698597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/698597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/698597/1"}}},"commit":{"parents":[{"commit":"567a388b95e18eaeff0830259efceeb9b861b2e5","subject":"Fix filtering with API tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/567a388b95e18eaeff0830259efceeb9b861b2e5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-11 21:48:29.000000000","tz":-480},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/705e2ed4568337f13302984ab58fbfd5a70d8c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/705e2ed4568337f13302984ab58fbfd5a70d8c00"}]},"branch":"refs/heads/stable/stein"},"e51b817fcfb36e1a8d9c3ca1a4a386a6fa4c1870":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-13 19:34:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/698597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/698597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/698597/2"}}},"commit":{"parents":[{"commit":"9904b26a9c40f29f554c56e9e65f6396caa8fea9","subject":"Accept oslopolicy-policy-generator path arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9904b26a9c40f29f554c56e9e65f6396caa8fea9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-13 19:34:22.000000000","tz":0},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51b817fcfb36e1a8d9c3ca1a4a386a6fa4c1870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e51b817fcfb36e1a8d9c3ca1a4a386a6fa4c1870"}]},"branch":"refs/heads/stable/stein"},"4ebdce39bc47d368ef3bc096dfaa45727091ede7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-03 21:40:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/698597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/698597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/698597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/698597/3"}}},"commit":{"parents":[{"commit":"c563d24681b814be05e2aa519a777b0247adb995","subject":"Merge \"Add listener and pool protocol validation\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c563d24681b814be05e2aa519a777b0247adb995"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-10-30 18:46:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-03 21:40:56.000000000","tz":0},"subject":"Fix multi-listener LB with missing certificate","message":"Fix multi-listener LB with missing certificate\n\nThis patch allows listeners on a load balancer to continue to\noperate should one listener fail to access secret content in\nbarbican. Previously if one listener failed to access barbican\ncontent, all of the listeners would be impacted.\nThis patch also cleans up some unused code and unnecessary comments.\n\nChange-Id: I300839fe7cf88763e1e0b8c484029662beb64f0a\nStory: 2006676\nTask: 36951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ebdce39bc47d368ef3bc096dfaa45727091ede7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ebdce39bc47d368ef3bc096dfaa45727091ede7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
