)]}'
{"id":"openstack%2Foctavia~698601","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253","project":"openstack/octavia","branch":"stable/train","topic":"multi-list-client-auth","hashtags":[],"change_id":"Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253","subject":"Fix multi-listener LB client auth/re-encryption","status":"MERGED","created":"2019-12-11 22:46:32.000000000","updated":"2020-01-24 17:18:47.000000000","submitted":"2020-01-24 17:16:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698601-1579886186925-9ea6cd4b","meta_rev_id":"9063455d05e3a92cdbfdf00e6f84afb293b446d6","_number":698601,"virtual_id_number":698601,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-01-13 22:08:45.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 17:16:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-01-24 08:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-01-13 19:08:20.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-01-24 08:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 07:38:59.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-24 08:16:08.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-24 17:16:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"999ffbc0ea909c75791247ede4c384ba037b84c4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-12-11 22:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e8cd3397ab39eeef8d47264b71141cca1e54466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 01:01:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/740b9dfee38945ed9a6483cf882d6e6c : SUCCESS in 7m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04a2e042736640069c6f6bd4a951c099 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eec1824fb984d0c9de9d308c18cd8eb : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c78da332ef4f4253822b1f5d92f197df : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e49db8a8db3441e48d58b5528ca7cf65 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb3261911ecf4decbd4951c8176f805e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9c98d53aec47158e341fe822801731 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe0ae55d86546e29b92bc0c81956e0e : SUCCESS in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d8337549482b46c2a87769cca1c40fba : SUCCESS in 4m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/53e8eedf2ab9408fa1313ac9b1b6d816 : SUCCESS in 7m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ec1642db92654326956fc4a6531a36e3 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd3e30c8edde4bc28e4acb887de3cb90 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/619c49baa3584ac88a044c88a6cc019d : SUCCESS in 50m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6bec3d69519041beac89a851c4b584c1 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2cea2a24f3747009671fe578900545e : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/67e446d4401a427da8f9316546e84233 : SUCCESS in 1h 39m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/49fa9d1430764dfd86b3e00ad1fd9b6a : SUCCESS in 2h 09m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60d01ee0d52b4f92b8e842f214fd01f5 : SUCCESS in 1h 57m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5ef93e0bee44371a9dd082222e08f5e : SUCCESS in 2h 01m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/51b43b5488e94063b290b4bc21d34fa0 : SUCCESS in 1h 19m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02b1d8cfdcae43b0a0fcc4d9660403d9 : SUCCESS in 58m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e23a13d2d1154ad0817a0f67013f25ca : SUCCESS in 1h 35m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b84caec7d79466c8864b0d3305992ea : SUCCESS in 1h 09m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1d7e392c412e4f44905aea5f7a349c66 : SUCCESS in 1h 08m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/00530cf6f69b4b05b036a28288029dbf : SUCCESS in 1h 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f2ebcd255b4a2eb5bdf077d0b66f5ea429d482a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-16 07:38:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59b78c6ea8c9b6c67254d43d13a954fdbccfd670","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 19:08:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c3c10bd5d90a26f57c291842bd7fadf40fda94fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 21:39:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e415d7249f046dfbdb2ea15511812c4 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb1bddb49f7445f0997ea6dfca874dd3 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347eb4d1fb7d433d8adf03174a306762 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7862e5cf3e45438591fdb6efe90c411d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92e7d6fba4d14701a21338237ec3235d : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff09e4546df1405eae6444ea6474c5cd : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a037b628d55416a9ef9a9fa8878d8a6 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27dfdcbb16db4f17b28ea81b47ad6bf9 : SUCCESS in 6m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e4ae64f789d843579d916276c4f8f920 : SUCCESS in 7m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb7f96c484bf4aaba8eeb25b2220ea3e : SUCCESS in 9m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d3a87ec7b78466b886ce446aa7a6c27 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72d4993a347f4fa09b79d498f074ba6f : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c17cbd0c939a4040a5eedb58830b0ea4 : SUCCESS in 55m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f9b065c452974697b6720e6e20b1acd8 : SUCCESS in 57m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86b909bf0ecc450cb6aec37bd237f595 : SUCCESS in 1h 46m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/dea67fb7f9c547d786a448b3a8288718 : FAILURE in 1h 47m 23s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/113a6ede445841cdb484d3e5763de6d5 : SUCCESS in 2h 03m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42279f3de0584c799e42ee54727a5308 : SUCCESS in 2h 02m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/94d8a75191244201a684fc70e545f67d : SUCCESS in 1h 57m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de94aac931fe4b209d220dcbcce364bd : SUCCESS in 1h 14m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7fd5d96c7d2b405d94c761cb9fc7938c : SUCCESS in 1h 00m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/83b968b03c6b4409821ac871400e3df8 : SUCCESS in 1h 46m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/965df80706d546bb9f6ee284d713d4cf : SUCCESS in 1h 14m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fb4b9dc6b3c44b58a6f8300e5d2b2745 : SUCCESS in 1h 09m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/163a211a6184472ba816db8cb5d7ef36 : FAILURE in 1h 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9f4630c218504fe12679e69e857dd7867fda1e7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 22:08:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1fa181d2780700a175ef337af29260dccd84d5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 00:49:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/962eb15d351c468c80974334be1825c3 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf1dedec232140cd87179f8a16ba7546 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ff7fddf73a4c03ab0137af08cca7d5 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38501f2ddddd4c30b29e351da369e940 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3729358e784d94b320b8cc663cf231 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20436afa51504adbb91e09f94fe00a2e : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a81d86bd104bd480687c37e70a7577 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb98e298872a483698b4fa4751419349 : SUCCESS in 12m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a06c96783b524962a9577fe72a84a8d0 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cc7ed087e4df40cd86b482174f45d723 : SUCCESS in 11m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0a11a72dcd1e47f7b988f629b0e2010d : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/83a0857b6ad44d4f9db934836cd77a70 : SUCCESS in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5c5090206e94d5a85fce5b32a3fc850 : SUCCESS in 45m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8ea743023e134337b40fc49d2da4e517 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dff417475664cf99059e0e6084943eb : SUCCESS in 1h 32m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d6e4e8d4f0934a9f9635bee7b1eb84af : FAILURE in 2h 04m 19s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d4da2d1656da4692907cdd783a13940e : TIMED_OUT in 2h 36m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02dc76a6fe5b4c80bb103973046e5700 : SUCCESS in 1h 54m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3153b272dee0410bb4f67035f76f4bb9 : FAILURE in 1h 51m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3444db386294077b20cd24b938d6f01 : SUCCESS in 1h 11m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/613983b3310e42dfb67fbbb08e895d65 : SUCCESS in 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c3750808894646449e4128f1b74a0b37 : SUCCESS in 2h 06m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e091d265d2d744da87c21ebf0c89237c : SUCCESS in 1h 05m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5f2f90f5b1ec4bf1aada72153daf3813 : SUCCESS in 1h 07m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/246afefffce34143a3b3a5f0456a598f : FAILURE in 1h 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f819423a22301c6978afb49f1f3a95f88c7ea4cb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-24 08:16:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0042f519f00eec691adfc2e033baa3511d3309d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:39:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7639388264d54c8daa68fe45e97c7401 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2ecca993f04dc4adaabf716ed0d11b : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ca9ec7fb33449082df85ede8f45e4f : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6648a031fa3749ba8569bc735e364609 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e15aa8e4522d458a84467b9fbadcb840 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f9b714cddbf4821a1ea354351e829c5 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91451b546ff446be859435508dab4c81 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2469677ddc6e4ebda6d7ee37864730a1 : SUCCESS in 7m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/38ea18cda76c400c8cb063e69916debe : SUCCESS in 6m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05d7f8d4097947138442bbccae91e137 : SUCCESS in 7m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fc457dc6f301451ea3dd96ee911c6e9e : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0462e7fd82bd4ef4ba3aba38184a35de : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e166b4fd73a7452992d71b8b7194c33a : SUCCESS in 47m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/96e1e01a57a84e9ca47c7610ccf25924 : SUCCESS in 47m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/237be5f0d1dd4eb194160ea3c20eff93 : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3f4103044c784d0e8a0b1662e0c0e792 : SUCCESS in 1h 36m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f972f8524e874311beb6c029ab0d6946 : SUCCESS in 2h 17m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e86f302cc3764b18bf07215b1db27677 : SUCCESS in 2h 17m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e34d8ed1b3c439abcbf1e721201c7fa : SUCCESS in 1h 47m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c50722666364710a6e189988e89c056 : SUCCESS in 1h 19m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b067cd90990411fb01906b2dec90be1 : SUCCESS in 1h 09m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2a07b472070489589061f610f8b4306 : SUCCESS in 1h 37m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0979bf90114e4163b6b5fc0a1447f0c8 : SUCCESS in 1h 08m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2d2d9c2f98c641c9bb585f812a8d53e8 : SUCCESS in 1h 23m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/551cabc5bf9b4d0286b034b389b6f998 : FAILURE in 51m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d7a6c843d7af5c707cf4f2b60f12ac01424b4a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:39:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2da211a5f00faabf20c352bdceaeca65b26bacad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 17:16:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ed86e8ccc73a7c8c091a9a675444b25eece8ccf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 17:16:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4102c878ede4a7bb31d0c07cce213ef : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa11598e1ea41229d51e9e0e5384f9a : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c56732e112047378c2d9193f99468f6 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baac41c172e4402a820ee85c734154e8 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68abc9f0b56446109381d01478b67b0c : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7268efdde9364109bf497fbe87f037d2 : SUCCESS in 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c3786b440b345a39111030019e65d79 : SUCCESS in 8m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e7182f793c7f49f38be1272655d3db30 : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bbf8148a5d94f8a95664b0b949c8cd7 : SUCCESS in 9m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/740766e269a14b8ea6f4de17a413d840 : SUCCESS in 54m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4273b7bfcfb942ef8a41ae86dd10bca6 : SUCCESS in 53m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/557b337a46ed465b8b7ab87412bd0d75 : SUCCESS in 1h 43m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7b1a54d482e848d09314a7ac1e08f237 : SUCCESS in 1h 41m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f78f8367137b40e1b3acaceced6a0ee3 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":2},{"id":"5a8da026f3b1ba5b3301e58583132cb50c5527ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 17:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdccc6c294134fdbadbedb20887f4705 : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/750b499e7ac74c839203eea832544388 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84c10796a77a6cdd06141691e56a2eddf81be940","revisions":{"6bc52af647fda5668eec4a8209cd259712b0be00":{"kind":"REWORK","_number":1,"created":"2019-12-11 22:46:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/01/698601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/698601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/698601/1"}}},"commit":{"parents":[{"commit":"d46df21e4594df3b2520026248a87ebbb0bb2e91","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d46df21e4594df3b2520026248a87ebbb0bb2e91"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-12-11 22:46:19.000000000","tz":-480},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bc52af647fda5668eec4a8209cd259712b0be00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bc52af647fda5668eec4a8209cd259712b0be00"}]},"branch":"refs/heads/stable/train"},"84c10796a77a6cdd06141691e56a2eddf81be940":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-13 19:08:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/01/698601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/698601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/698601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/698601/2"}}},"commit":{"parents":[{"commit":"df5db0b000f69dfd92e640a09dc98344ef3be731","subject":"Fix multi-listener LB with missing certificate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df5db0b000f69dfd92e640a09dc98344ef3be731"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-11-08 19:40:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-13 19:08:19.000000000","tz":-480},"subject":"Fix multi-listener LB client auth/re-encryption","message":"Fix multi-listener LB client auth/re-encryption\n\nThis patch corrects a bug with mutli-listener load balancers that\nare using either TLS client authentication and/or backend\nre-encryption.\n\nChange-Id: Ib7b083e1dfbfd7afcca870ed6f60a871b2e19253\nStory: 2006822\nTask: 37394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84c10796a77a6cdd06141691e56a2eddf81be940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84c10796a77a6cdd06141691e56a2eddf81be940"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
