)]}'
{"id":"openstack%2Foctavia~698885","triplet_id":"openstack%2Foctavia~master~I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17","project":"openstack/octavia","branch":"master","topic":"drop-centos7-ubuntu-xenial-trusty","hashtags":[],"change_id":"I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17","subject":"Use LUKS1 for certfs-ramfs for all distributions","status":"MERGED","created":"2019-12-13 11:49:39.000000000","updated":"2020-01-24 06:31:01.000000000","submitted":"2020-01-24 01:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698885-1579828465240-5aeb02ef","meta_rev_id":"6ba5e0953c5c5b51cb04d48c39051929fc9b7317","_number":698885,"virtual_id_number":698885,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 01:14:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-01-17 12:01:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-01-23 23:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-01-17 11:23:26.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-01-23 23:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-01-17 11:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 11:23:26.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-17 12:01:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-01-23 23:37:12.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-24 01:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87e3055d7e72759928225edbdce8578cba97eabb","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-13 11:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6a0bdc2fd79e8e81b7876bcaa7b70af2581deeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:36:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea0c54f34d64a10936b7110076426ba : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d415f4399aaf413f85d2a3a911d583f7 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffacff4db04e41a6a69ea376e1f68a95 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f04f54a72a479595a465069486c65b : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/946b6f38e079450d9fc88b51e1e2e109 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b07165d16824ad6883419be8b28ab67 : SUCCESS in 10m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7bc9229c8556494eb033c00738fcba35 : SUCCESS in 4m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2c0883d232044999af542282c849cab9 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af2828d73bac4dd6a6dfad474a1bdc6d : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6775386e8a04310867e286cd570783f : SUCCESS in 46m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50bf4c2101cf42dabf269b174d7a8fb7 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ccf405812ec24de98ca8eab05e39bb46 : FAILURE in 13m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/262efb115d1b4a7bb03f12a9097d35ba : FAILURE in 1h 10m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7bc618fd8764f1aaca80ac4e42b266e : SUCCESS in 1h 09m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/057a4b0ea80b4ffea18c471c1ad3dc6b : SUCCESS in 1h 36m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/571fffba41e14b83af8df55966ba4863 : SUCCESS in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdec6e99878a415eee72e9a058806e881bd81477","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2019-12-13 14:44:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"67a706dbca5cb303101540fbe16da137fffd095f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:41:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/419a357d9fd04c978b289c5bb491c452 : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea48bde553fe4ea2a9d9efa8f42e5317 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a4fa39f9a8475e8e3697b8ccd24764 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c660b95a8484d9cbb3edb97ee374434 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cd7b0a3e1e0444eb2de2e7a46ccd96a : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d55181ab9c40baa9b5d2b5cc14c4ef : SUCCESS in 11m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2da05a34a5f24bde9e94c8ce3e8ac361 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2512f557b95b408382f68acda05dc580 : SUCCESS in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/520dba14d7dd4d42bcd762075c7ad8e8 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60ade4b89ec0438089ad2a5761eea431 : SUCCESS in 54m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/280a129e11ec457faafb0ba2dc6c6457 : SUCCESS in 1h 54m 17s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe16f4a1b7a4423eb43f4de8039c1818 : FAILURE in 16m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7af40f8f076a4a60955dcf59bf63196e : FAILURE in 1h 21m 55s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67b36285a9564c6fa4438ea9c8e35378 : SUCCESS in 1h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ab20dfa566ff46daa538843de864267e : SUCCESS in 1h 34m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e932daa731b45f3a0f59d44133db6c8 : SUCCESS in 1h 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"063547f1b5b94f98255a59e68bedcb8c2ca24f87","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-17 09:26:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"989008b01000c66b5b61bfda651be7405150966d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-17 09:31:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88623b6e840f74b7d1db3b37776769b47fb95ed7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-17 09:38:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"869e25905c604aebb67944576661c7749396fa9f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-17 09:39:43.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nthanks Carlos, I removed the luks2 specific commands","accounts_in_message":[],"_revision_number":3},{"id":"fb8f72840bc5f81a4627659d4459095ce9d64c8d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-17 11:23:26.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2\n\nLGTM. We may have to backport this to stable/train as at least one OpenStack vendor ships the OS version with cryptsetup\u003e\u003d2 (LUKS2 default). I\u0027d like to hear from others.\n\n$ cat /etc/centos-release\nCentOS Linux release 8.1.1911 (Core)\n\n$ ls /dev/ram*\n/dev/ram0\n\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: mke2fs 1.44.6 (5-Mar-2019)\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: Creating filesystem with 14336 1k blocks and 3584 inodes\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: Filesystem UUID: 6d7624ab-e4a7-4db2-9f76-2d38d9690824\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: Superblock backups stored on blocks:\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]:         8193\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: [41B blob data]\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: [38B blob data]\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal certfs-ramfs[833]: [75B blob data]\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal kernel: EXT4-fs (dm-0): mounting ext2 file system using the ext4 subsystem\nJan 17 11:15:44 amphora-c81b7024-64d7-4d7d-887a-a217c936e120.novalocal kernel: EXT4-fs (dm-0): mounted filesystem without journal. Opts: (null)\n\n$ sudo cryptsetup luksDump /dev/ram0 | grep Version\nVersion:        1","accounts_in_message":[],"_revision_number":4},{"id":"bc75d080a593cc440b2ecf917590b37a3e07b8d7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-17 12:01:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4eb206693a8a09f08ccd9e568196f1b607285a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 13:02:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9569f3a2164048478f6dd6f83f0525be : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9696c37fc1fc41f5aa1d2bb6a2c28610 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b0c66380e448f4b407308e925408a3 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55dc38569e0447688c745cdba02faef : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c7038a9318143cabd04d6f351b63be9 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc2a8a5998c4cc68b4d95ba66bb9999 : SUCCESS in 16m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ccc45ec45ebb456dbbf7b761f82aa2e5 : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3179ed748714a36bbaceda525bd05b8 : SUCCESS in 12m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe0686c308194448b556f5a093541e1d : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e5887d3e5754fffbccbafaa5df7e501 : SUCCESS in 52m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f3eba14b487450d8f0e69af88e87b7b : SUCCESS in 1h 51m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61391cb9b3d341deae40e7e3cabb512a : SUCCESS in 1h 16m 46s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db3a74579c1f42f2916a387959f166e2 : FAILURE in 12m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9eef66a2e08845e492052cd2f4718045 : FAILURE in 50m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/391ae5ea6bf44f3c9bbb98a0a767c5d3 : SUCCESS in 2h 19m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31f298cbe785478c84c61a09b86a55c3 : SUCCESS in 1h 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"94873b08be732366d896e0d197da2004266ea527","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-23 23:37:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ec5557d43be78deeb5794154126fb9a15e90f903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 23:37:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c2a4544f997cf7b0463b782cc8fb45da833e0249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:14:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8a0df33e2bff436f7282f4e3df43c6402e177795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:14:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e6c64177dc84cb292af395c5f3713ba : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4d21164e5047baa6681c2366412ed7 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22dc25fd12ac45e5995036fac8e4dc1d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37c356732bca4c66ae75b7cc4968389a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f86f454c87469385b50a54a733434d : SUCCESS in 12m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab166a5d05ac4741a38ccbb99eb9fbd8 : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a31b7f075e8c4498923d66d09370a6de : SUCCESS in 49m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88364b77891140fa9eb9c7b10285fd53 : SUCCESS in 1h 34m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0519afd1adcb44bba6b54289245eafb7 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":4},{"id":"f973d98fa832a0d7a8b8f56c0b01e08b32ae82b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:15:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9894856d29a458e92a4b074d83c9846 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4},{"id":"330d47d1b7a6a82f9ee8089bf63e3a22aadde7c7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-24 06:31:01.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a59e7235560ad045b24c58ed722ab62a451174d3","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c2ef7792fd17c5e54e4b82a63ea448d1f8310c31","revisions":{"e488edb56a9e787c32e47c00602e28df59bb6f5f":{"kind":"REWORK","_number":1,"created":"2019-12-13 11:49:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/698885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/698885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/698885/1"}}},"commit":{"parents":[{"commit":"45cb1e4e161239ef93adaf545dd6db2c537c69e3","subject":"Merge \"Update ListenersUpdate for lb/listener dicts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cb1e4e161239ef93adaf545dd6db2c537c69e3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 11:49:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 11:49:20.000000000","tz":60},"subject":"DNM improving amphora boot time on Centos","message":"DNM improving amphora boot time on Centos\n\nChange-Id: I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e488edb56a9e787c32e47c00602e28df59bb6f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e488edb56a9e787c32e47c00602e28df59bb6f5f"}]},"branch":"refs/heads/master"},"72097a9e0cc33ce46689d4940c43068d0e46b40a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-13 14:44:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/698885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/698885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/698885/2"}}},"commit":{"parents":[{"commit":"237b25d3933ba9b2f6475bbc65fb3fce58effaf3","subject":"Stop supporting CentOS 7 controllers and images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/237b25d3933ba9b2f6475bbc65fb3fce58effaf3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 11:49:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 14:43:51.000000000","tz":60},"subject":"DNM improving amphora boot time on Centos","message":"DNM improving amphora boot time on Centos\n\nChange-Id: I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72097a9e0cc33ce46689d4940c43068d0e46b40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72097a9e0cc33ce46689d4940c43068d0e46b40a"}]},"branch":"refs/heads/master"},"7c41c2e5c0e8509534451659ea9a013fba6c3ff7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-01-17 09:26:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/698885/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/698885/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/698885/3"}}},"commit":{"parents":[{"commit":"47e0ef31bcfd34838fac4c619b699edd65e20223","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47e0ef31bcfd34838fac4c619b699edd65e20223"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 11:49:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-17 09:25:24.000000000","tz":60},"subject":"Use LUKS1 for certfs-ramfs for all distributions","message":"Use LUKS1 for certfs-ramfs for all distributions\n\nThis commit forces the use of LUKS1 in cryptsetup.\n\nCentos uses LUKS2 by default in cryptsetup, ubuntu uses LUKS1.\nFormating a LUKS2 block device is way longer than formating a LUKS1\ndevice (15 sec vs 8 sec in my env) and LUKS2 doesn\u0027t provide any\nsignificant features for octavia (it only improves recovery and\nmetadata).\n\nThe commit aslo limits the creation of more than 1 block ram device\n(amphora doesn\u0027t need 16 ram devices), which reduces startup time.\n\nChange-Id: I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c41c2e5c0e8509534451659ea9a013fba6c3ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c41c2e5c0e8509534451659ea9a013fba6c3ff7"}]},"branch":"refs/heads/master"},"c2ef7792fd17c5e54e4b82a63ea448d1f8310c31":{"kind":"REWORK","_number":4,"created":"2020-01-17 09:38:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/698885/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/698885/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/698885/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/698885/4"}}},"commit":{"parents":[{"commit":"47e0ef31bcfd34838fac4c619b699edd65e20223","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47e0ef31bcfd34838fac4c619b699edd65e20223"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-13 11:49:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-17 09:38:45.000000000","tz":60},"subject":"Use LUKS1 for certfs-ramfs for all distributions","message":"Use LUKS1 for certfs-ramfs for all distributions\n\nThis commit forces the use of LUKS1 in cryptsetup.\n\nCentos uses LUKS2 by default in cryptsetup, ubuntu uses LUKS1.\nFormating a LUKS2 block device is way longer than formating a LUKS1\ndevice (15 sec vs 8 sec in my env) and LUKS2 doesn\u0027t provide any\nsignificant features for octavia (it only improves recovery and\nmetadata).\n\nThe commit aslo limits the creation of more than 1 block ram device\n(amphora doesn\u0027t need 16 ram devices), which reduces startup time.\n\nChange-Id: I5cdc0a9ccc01548f195eed80f2ee2848a1a93e17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2ef7792fd17c5e54e4b82a63ea448d1f8310c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2ef7792fd17c5e54e4b82a63ea448d1f8310c31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
