)]}'
{"id":"openstack%2Foctavia~699257","triplet_id":"openstack%2Foctavia~stable%2Frocky~I4c98b3442d2471662488184fa7e91ac64ec33279","project":"openstack/octavia","branch":"stable/rocky","topic":"fix-backport-spare-pool","hashtags":[],"change_id":"I4c98b3442d2471662488184fa7e91ac64ec33279","subject":"Limit spares pool to the spare_amphora_pool_size","status":"ABANDONED","created":"2019-12-16 17:52:43.000000000","updated":"2019-12-18 17:38:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e80b8c2d69087a7b0785aa013669d518f2360f0","_number":699257,"virtual_id_number":699257,"owner":{"_account_id":25290,"name":"Mikhail Ushanov","email":"gm.mephisto@gmail.com","username":"gmmephisto"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-16 19:05:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-12-16 17:56:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-16 17:52:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 17:52:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-16 17:56:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-16 19:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"387f71b730459c2350d170c2bc524502b3511dc1","author":{"_account_id":25290,"name":"Mikhail Ushanov","email":"gm.mephisto@gmail.com","username":"gmmephisto"},"date":"2019-12-16 17:52:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11510ec6cc9fe01b58508b6033c95c1cbbc72211","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-12-16 17:56:14.000000000","message":"Patch Set 1: Code-Review-2\n\nUnfortunately, the patch cannot be backported because it contains a database schema change and Octavia follows the OpenStack stable policy. See https://docs.openstack.org/project-team-guide/stable-branches.html#review-guidelines for more details.","accounts_in_message":[],"_revision_number":1},{"id":"a07a1db98792e63a9bab405009287de34a505a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:05:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae328901beb438eaec8a7af848061c4 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8010a269d0284c638b162691a6a6150a : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ddc192e31a2e41cfb4063239e12afec3 : SUCCESS in 8m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a474f8c1f344d9b919bfac4d90131ab : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36192e022f0441e6bdbf4ae116df79ee : SUCCESS in 7m 08s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/dcc6f2d618fa4d40b2e55ad1a4511289 : FAILURE in 25m 58s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/435d7e2fd7c0464c9c6c0bb3a18fa625 : FAILURE in 27m 24s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75eba269de5d420fa4c722cfe571b075 : FAILURE in 27m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/87217d09188d490baef0889e167ef6f1 : FAILURE in 25m 07s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2ae83e41b7754de293efae3beb3c6383 : SUCCESS in 6m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8136adfe131f4616b36dfb5c1ac4240a : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8286f6b80054f7a93087284c43717c3 : SUCCESS in 6m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b88edcb2b3fc4d84aada043901d25075 : FAILURE in 24m 57s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b409fcbd9883466b9658e3d91fb2c8dc : FAILURE in 28m 57s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/4e49571602164c52926b626c5a2ef350 : FAILURE in 32m 17s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/621831c58c044944b7dd93dc905a4e84 : FAILURE in 31m 06s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b8ecd4fb20cd4ac1b48748939c1f8772 : FAILURE in 27m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5234913e51d41fdafafebe97e23672e : FAILURE in 26m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/86aeb4eb81ba4debab968cdad329c54c : FAILURE in 27m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0d14cd327ac42ed8e9d3f8ca9189acd : FAILURE in 27m 47s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/bdacd6647ce04c5b8d7bc564ef9a2a6d : FAILURE in 26m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0a971e726d1420db513b2e99a37560c : FAILURE in 1h 08m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7ad6e8abfd49708cc1fd64f80ef9c6 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0536e6c7f141cd136e1263bdf746ffbd8ecf120c","author":{"_account_id":25290,"name":"Mikhail Ushanov","email":"gm.mephisto@gmail.com","username":"gmmephisto"},"date":"2019-12-18 17:38:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"327e51b2e22a4090ddb529222a794b3a6164a69f","revisions":{"327e51b2e22a4090ddb529222a794b3a6164a69f":{"kind":"REWORK","_number":1,"created":"2019-12-16 17:52:43.000000000","uploader":{"_account_id":25290,"name":"Mikhail Ushanov","email":"gm.mephisto@gmail.com","username":"gmmephisto"},"ref":"refs/changes/57/699257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/699257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/699257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/699257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/699257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/699257/1"}}},"commit":{"parents":[{"commit":"5c485a69aca552df8c1ed5f43828174898fef590","subject":"Accept oslopolicy-policy-generator path arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c485a69aca552df8c1ed5f43828174898fef590"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-03-12 00:06:15.000000000","tz":-420},"committer":{"name":"Mikhail Ushanov","email":"gm.mephisto@gmail.com","date":"2019-12-16 17:52:11.000000000","tz":180},"subject":"Limit spares pool to the spare_amphora_pool_size","message":"Limit spares pool to the spare_amphora_pool_size\n\nThis patch fixes a bug where the housekeeping controllers may launch\nmore spare amphora than the spare_amphora_pool_size setting.\n\nStory: 2003094\nTask 23186\n\nChange-Id: I4c98b3442d2471662488184fa7e91ac64ec33279\n(cherry picked from commit a205ab3ebe4ae14668f25328c062f9383d6c69cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/327e51b2e22a4090ddb529222a794b3a6164a69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/327e51b2e22a4090ddb529222a794b3a6164a69f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
