)]}'
{"id":"openstack%2Foctavia~699287","triplet_id":"openstack%2Foctavia~master~Id2193612c2829d4fdd02d5b5c5d1dd21b502eae1","project":"openstack/octavia","branch":"master","topic":"call-reset-mock","hashtags":[],"change_id":"Id2193612c2829d4fdd02d5b5c5d1dd21b502eae1","subject":"Fix tests to correctly call reset_mock()","status":"MERGED","created":"2019-12-16 20:45:08.000000000","updated":"2020-01-07 21:46:49.000000000","submitted":"2020-01-07 21:44:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699287-1578433482364-7a9322a4","meta_rev_id":"7f8d27ab975c9818778fa73f876b4bf553780203","_number":699287,"virtual_id_number":699287,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 21:44:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2019-12-30 08:44:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-01-07 20:01:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-01-06 10:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":1,"date":"2020-01-07 13:24:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-01-07 17:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-01-07 20:01:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-30 08:44:55.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-01-06 10:11:10.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2020-01-07 13:24:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-01-07 17:39:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-07 20:01:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-07 21:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fff3ae23a3b8f5ab754833cdf212607ef5a94ad1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-16 20:45:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aac8dfbdbba27e1f86fac83fc7cc83c8a2b5e367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 22:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e28d95341f14e13ad0f9627cf3a9290 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780e6fb42736430da3cb3082728d2008 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01eba70ea793495fbc7431cd76df019c : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e20b83ab8da484e9760ae3d6897718b : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/387be7d60ca0453987428a6c60240d96 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b63d57a9c8749f5a0bb8e612b9bcd04 : SUCCESS in 9m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/23a4d48cdf5444a1b7d03d8ce10c5ddd : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/eb1d0c4340eb42eeae21ee992d13cc3f : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f04c70f072241598cc51be8953c1801 : SUCCESS in 5m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06c202a134e2454091619dad7afe3c25 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3da5dfca22d4e72933a44bddcea3989 : SUCCESS in 1h 51m 29s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7fc818a98b6e447a801598e571371197 : FAILURE in 13m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/125490f5846142b89c9a8e6b460f42d9 : SUCCESS in 1h 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6063a1268de5c0e610ede8e230b5e32aa06a6900","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2019-12-30 08:44:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a66c748992b59dd2d8585d12693558d0f38dbdda","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2020-01-06 10:11:10.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":1},{"id":"5fd60160463e70fa20916cf7bda28c26b4ccfc72","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-07 13:24:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0cd9be1815d116703e13b98a19f43adb5dd251be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-07 17:39:04.000000000","message":"Patch Set 1: Code-Review+2\n\nMy faith in pylint just went down a notch....\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"1635fde472a6f75518d4e49e6375339b0325d9f1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-07 20:01:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4896c3055333a78fc4aa833ac53b106cbaba361d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 20:01:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7b6e0c2f2cde9a7acdbdf697cca15ac084c0e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:44:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bae74e0940024a278b8b0a6143fc1312 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ccb65b368243439665dae1d62acbf1 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53f11c38718e491c96afab68f93c5ddc : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5fa7377c0db45e482dd184ddfa9fdff : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485abb57769f44379b7dbc806753c996 : SUCCESS in 13m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a6c1f68c3c14b2aae913cae9586268b : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/024021b3b8c24f4cb1e55c0b073eab04 : SUCCESS in 45m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2511800fa5564495a4b4a8f4b4271e5c : SUCCESS in 1h 41m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24125b3fe6b041e885d297c3eebe2652 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a4bd93faf66abdb935e718d4dab6278042f1403e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:44:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e7a711dd3e8998c2c886a10ae7d79b64e43fab6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-07 21:46:16.000000000","message":"Patch Set 1:\n\nAh, I don\u0027t think I would have merged this. I believe the correct thing to do here was *remove them* because obviously there was no reason to reset the mock, as they were working with that bogus noop... That\u0027s what I\u0027ve done as I run into them in the past (I have removed about 6 of these).","accounts_in_message":[],"_revision_number":1},{"id":"2c18ef5fd49e9bdc6b90a62077e39a7f419eac7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb360bf4b37b4a1cabb3d0df6b382a8b : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a490119f776ed77c7941e6c335529d414b36ee5","revisions":{"2a490119f776ed77c7941e6c335529d414b36ee5":{"kind":"REWORK","_number":1,"created":"2019-12-16 20:45:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/87/699287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/699287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/699287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/699287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/699287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/699287/1"}}},"commit":{"parents":[{"commit":"45cb1e4e161239ef93adaf545dd6db2c537c69e3","subject":"Merge \"Update ListenersUpdate for lb/listener dicts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45cb1e4e161239ef93adaf545dd6db2c537c69e3"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-12-16 20:33:33.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-12-16 20:40:30.000000000","tz":-300},"subject":"Fix tests to correctly call reset_mock()","message":"Fix tests to correctly call reset_mock()\n\ntest_failover_loadbalancer() has two calls to \u0027reset\u0027\ninstead of reset_mock(), which do nothing, fix them.\nIn both v1 and v2 worker tests.\n\nChange-Id: Id2193612c2829d4fdd02d5b5c5d1dd21b502eae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a490119f776ed77c7941e6c335529d414b36ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a490119f776ed77c7941e6c335529d414b36ee5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
