)]}'
{"id":"openstack%2Foctavia~700091","triplet_id":"openstack%2Foctavia~master~I2b174ef3231956037650eba19aebe5989dffa617","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I2b174ef3231956037650eba19aebe5989dffa617","subject":"Cap hacking version to \u003c2","status":"ABANDONED","created":"2019-12-20 01:55:08.000000000","updated":"2019-12-23 09:31:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"370dd8e466c4fc22d0acf029148f70d8d15acb31","_number":700091,"virtual_id_number":700091,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-20 03:56:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2019-12-20 13:39:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 03:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-20 13:39:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"252e50489f8f54f61f6e4a421b81cd44446de128","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2019-12-20 01:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a380d7b2374cf0b7c23583c14834e7dc75341a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 03:56:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f87b31100f2541fc931d98d5d4ea4db4 : SUCCESS in 5m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/463733c7ed394ba28e38888f16a53fab : FAILURE in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1797c530cdb2441990f46dcd3eea6081 : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9edcb63586ab4530b08cced1205d9f80 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5b335a459b4908a87043b30341f559 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/058344301ac84bde8380c9802b50f3d2 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d6f635ced1f454e95fe84a9c94e11e5 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5a2fed630e498bb62e0ceffd2ac585 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1cad3b3d20c487ead96dc59a4459ac2 : SUCCESS in 5m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9d914280064b4b5999d28d5a6a77c440 : SUCCESS in 4m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/774108c7d13f4859905c49c4cf1404b9 : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/654f0aded4b54231afaf2263a6cfadb4 : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/797ecd8f9e9d4b5192038c0908693671 : SUCCESS in 47m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b172bd2b71634d0ba030200918cde210 : SUCCESS in 1h 56m 05s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b89fe1592d7f4b5b9481cc9682c65188 : FAILURE in 8m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34ab74c930974f12859caba232e5060a : SUCCESS in 58m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62ce403a2c2a45f58c2966902a9a48dd : SUCCESS in 1h 32m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f183603230bf4c1b81e63ca1629f9677 : SUCCESS in 1h 00m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faa1d72a94aa0b896c3b10103666125d49c32c4c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-20 13:39:44.000000000","message":"Patch Set 1: Code-Review-1\n\nI would actually rather fix the issues so we support hacking 2.0.0, I have already created a patch for that:\n\nhttps://review.opendev.org/#/c/699302/","accounts_in_message":[],"_revision_number":1},{"id":"370dd8e466c4fc22d0acf029148f70d8d15acb31","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2019-12-23 09:31:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"332ba3297a58adce420cb637512419d3fa37f4b0","revisions":{"332ba3297a58adce420cb637512419d3fa37f4b0":{"kind":"REWORK","_number":1,"created":"2019-12-20 01:55:08.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/91/700091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/700091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/700091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/700091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/700091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/700091/1"}}},"commit":{"parents":[{"commit":"5972d7b0fac775b2f259153c0f9d41556002c627","subject":"Merge \"Add install guide for Ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5972d7b0fac775b2f259153c0f9d41556002c627"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2019-12-20 01:54:13.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2019-12-20 01:54:13.000000000","tz":480},"subject":"Cap hacking version to \u003c2","message":"Cap hacking version to \u003c2\n\nHacking 2.0.0 and its dependencies like flake8 have incompatible changes\nand bumping to them leads to pep8 check failure.\nIt is good to address pep8 warnings in the master branch, but in case of\nstable branches it is good to continue to use the same version of hacking.\nThis commit is proposed to the master branch so that it can be backported.\n\nChange-Id: I2b174ef3231956037650eba19aebe5989dffa617\n(cherry picked from commit 7a7558b0583fa84f73d5ecb81204cbb76b0ef0e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332ba3297a58adce420cb637512419d3fa37f4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332ba3297a58adce420cb637512419d3fa37f4b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
