)]}'
{"id":"openstack%2Foctavia~701013","triplet_id":"openstack%2Foctavia~master~Ic290d733b2cd927f05a2becef3b3d9984d1b7a73","project":"openstack/octavia","branch":"master","topic":"reproducible","hashtags":[],"change_id":"Ic290d733b2cd927f05a2becef3b3d9984d1b7a73","subject":"Make octavia reproducible","status":"MERGED","created":"2020-01-03 10:48:59.000000000","updated":"2020-01-04 03:46:08.000000000","submitted":"2020-01-04 03:44:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701013-1578109454726-46a14a57","meta_rev_id":"4cb944d30b6ae2d46c009f4fd2337c666fe26bed","_number":701013,"virtual_id_number":701013,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-04 03:44:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":2,"date":"2020-01-04 02:09:06.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-01-03 16:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-01-03 17:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":1,"date":"2020-01-04 02:09:06.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 16:49:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-03 17:56:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-04 02:09:06.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-01-04 03:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36eee540477c77588016309ba7cb9debe8455ba4","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-01-03 10:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea421d817070318fa46ad1ae83a93a1087d3df82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 12:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75b23caba8ed4017b94c4472a05ec4eb : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8079df2ecf9143a3b9f52e06702d330e : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1277c997b94c95a1904ee3130b0966 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8272c30b41c747e79a697367728da882 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b29b42c1ebd045a0a32f4ce3b876c9ee : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f52594365c45d6a9a9702c752098fe : SUCCESS in 9m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e68d4aca127c40049910f1c2d756324f : SUCCESS in 4m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61558de132a44a5b91dbff068f51d99a : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd3387415f1c407b8061ae1b22c00cb0 : SUCCESS in 5m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41fa7a9f9c364d02a918666d363da367 : SUCCESS in 43m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a2c2a381f924885920c6364a39998c5 : SUCCESS in 1h 41m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03c05710ced14bf386c5dbd2cede1426 : SUCCESS in 56m 50s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbb29a8781b44827b5bf3824f3cf71c1 : FAILURE in 12m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7805de85f5364d9daaadfdd873ce8dba : FAILURE in 59m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9d1bdc8262494315ab29465329d5728d : SUCCESS in 1h 39m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8fa78fd4485436182ac0cb2d1df9df8 : SUCCESS in 1h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c871ed2f4c11b89733abbb92b110e345767c32b2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-03 16:49:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c148a3d4b965b40f2a33e465e7ac79f19467774","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-03 17:56:55.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"59bf63211883a7d748c846c94ec830b7dc558690","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-01-04 02:09:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27922d4dc3c8927478cacad32326457ee8a584ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 02:09:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17013597dc925a79e240c0ea33dd3c9c56a5748a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 03:44:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"73205a45b81a2dc584c318b8c57be4ed79bcebb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 03:44:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f06b5f9f979243079de6162528d37f63 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9082c68c252840e3b10cc23d27dae30c : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75a5444abdbd41b88a844c4169ca36c9 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6af3547339e48c5b6086597565d2afd : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d747f42a3b864df2afa78c2931cc9fc7 : SUCCESS in 9m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbcd6ca1474a47f9b462936a97aba3bd : SUCCESS in 5m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/691b1d206ab0406f8a49143f6a07d20d : SUCCESS in 46m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1b0cc0ec5ac4ce698a5930232af8754 : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d53b5ccc82124f799603d3f10aeb1e97 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"83e2116ee180024e4aadb4df835d715d5f6fc969","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 03:46:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f5b9ce329e49e6865c0e82063d8e7a : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","revisions":{"b96107e27a6d6afa755d93e5d0ac1f09affc11ba":{"kind":"REWORK","_number":1,"created":"2020-01-03 10:48:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/701013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/13/701013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/701013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/701013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/13/701013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/13/701013/1"}}},"commit":{"parents":[{"commit":"8f032d884e0f89ac69d5b6e5f5b77d19ee6eb1d7","subject":"Merge \"Gate on octavia-v2-dsvm-tls-barbican\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f032d884e0f89ac69d5b6e5f5b77d19ee6eb1d7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-01-03 10:47:46.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-01-03 10:47:46.000000000","tz":60},"subject":"Make octavia reproducible","message":"Make octavia reproducible\n\nThis is a forward of the patch attached to the Debian bug here:\nhttps://bugs.debian.org/908222\n\nWhilst working on the Reproducible Builds effort [0], we noticed\nthat octavia could not be built reproducibly.\n\nThis is because it embeds the hostname in the config file (which\nalso gets picked up in the documentation and the documentation\nindex files.)\n\n [0] https://reproducible-builds.org/\n\nChange-Id: Ic290d733b2cd927f05a2becef3b3d9984d1b7a73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
