)]}'
{"id":"openstack%2Foctavia~701290","triplet_id":"openstack%2Foctavia~master~I2716ce93691d11100ee951a3a3f491329a4073f0","project":"openstack/octavia","branch":"master","topic":"remove-six","hashtags":[],"change_id":"I2716ce93691d11100ee951a3a3f491329a4073f0","subject":"Remove all usage of six library","status":"MERGED","created":"2020-01-06 22:15:34.000000000","updated":"2020-03-19 14:31:27.000000000","submitted":"2020-03-19 14:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701290-1584628053399-7fe8aeaa","meta_rev_id":"489ac6f7a7bfaf77fab92664da815fa246cc44cb","_number":701290,"virtual_id_number":701290,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 14:27:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-03-19 09:55:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-03-19 12:09:39.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-03-18 21:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-03-19 12:09:39.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-15 09:13:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-17 15:57:01.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-03-18 21:47:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-19 09:55:57.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-03-19 12:09:39.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-03-19 14:27:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64923b8ee13ec6f6a7c1026e7a345f1cf70d56aa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-06 22:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b84a8d8e2adcf4bf4b0b961cadac28442f97f00c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-06 22:16:17.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds a little more work, and there are other things more important to merge first.","accounts_in_message":[],"_revision_number":1},{"id":"318a320a7288644f00ccf595edbd3bc2eb570ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:48:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/708bdc7d87774501810eb3628a8b3edf : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b96adb4d58a4db6b7b00d69a159b773 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37ff998793c049a88b7990985837835c : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b01e6c8b94429fa4f63b2dc2a72add : FAILURE in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e9540936454442c8e098fa619dcaad8 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0519f187726495bb9945b4a731e5138 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52eaba264ec475182545c6f9336519c : SUCCESS in 11m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/efee4abb268f41048868f30607f38f3d : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/72f745f029524a31a9148745eda2a9c7 : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f9184dc52db4a5692312df0a3a59029 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7107c9caa1b1411bb506c17ecf2e0c55 : SUCCESS in 56m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95e2a815046c45528be55005ea11b49b : SUCCESS in 1h 31m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37cf430fff42451cbdc4959e6e20d936 : SUCCESS in 1h 03m 48s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/36a1209c2b6b4deab218bf91a7e5a474 : FAILURE in 7m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89f1f15203ac45ab841b1a15135fcefc : FAILURE in 1h 07m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94ab5373791d46e1bb783e9f5f1ec11f : SUCCESS in 1h 31m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0422dda85d7345c699df94dc5bbe927e : SUCCESS in 1h 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b22d5936dc6bb5d8d10574a28b86b98f8d4ffd84","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-07 21:39:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f5132051ed905bc3ed8003933ab68b43704add1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-07 21:41:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c24a5799dd9b00e8a5f7b4d7e81be3f7b6fa5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 23:24:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/15e7a5e8c2f746708991360451cda652 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30b42668f4c4620a90c6386a39bb053 : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190da386f9834104aeb0110d59ab0297 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7917f4457d1840e79244c8301ccba84d : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1966f7af74d4c3d93256f0615654104 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b812481275b14d2f907107d9a53062a7 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119fb9c715a846aa9161a5ded401e0f7 : SUCCESS in 10m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df904e583d1945fbb8017b7e5bd1ba26 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c11df62253049b19d76d00899cfd468 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d168c5849594227a217a895982a76fd : SUCCESS in 6m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5fa9e61b401d48c4aa9b759e5e9da73a : SUCCESS in 46m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c6caa0641734671b2335ee5244fd8b4 : SUCCESS in 1h 29m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1f98a9623a8453ea19266465960980f : SUCCESS in 1h 00m 41s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c0d2d3330a04fe5b425ebbaa9dd7e12 : FAILURE in 13m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3cfb8f7072848b28ce9eac60ca821b1 : FAILURE in 1h 09m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b371abccb7f4e6393c90cdee63e30de : SUCCESS in 1h 36m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae134cde65e64f1fb3741b9a4792f195 : SUCCESS in 1h 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12de6075670c11811af8b10cb28e008cb7c5c33e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-20 18:02:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ae3339349b31ca40e7babc9845773470c8a98ab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-20 18:03:14.000000000","message":"Patch Set 4:\n\nRemoved hacking checks code and tests missed from previous PS.","accounts_in_message":[],"_revision_number":4},{"id":"15f051d989e4493ac091f977cab0b2226d14785d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 20:02:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/51121e161d464c89952e8b01f7d97c93 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f943e101bae458f99497d306eeeda8c : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a427a9b808742f39d7275befa51f553 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80d08540f2f4685b27a8cf2ac54b51f : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad944ae6232c4cac86b92c374998dad8 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b9578470cd9422d8d43a27f7897d5c3 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821ea17168b44289ac108b55bc8c9555 : SUCCESS in 12m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/860d7ef36e36452d92d710251cad25f7 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0b94bcdb3ef748e4ad1e96045c40f4de : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bb9d57f813244819cc073b841fec5ff : SUCCESS in 9m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38612b47b99346e386971e9520953832 : SUCCESS in 46m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01f5b8c1c10045ebbb4932530890cefd : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b10c6d9842774a1f9cc1437701e6edd0 : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b190a76f5974133a779152e4696b286 : FAILURE in 14m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca9b593318334a859631f376eac8050f : FAILURE in 1h 00m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c617d32625224093aebaa2d2d300519a : SUCCESS in 1h 50m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/535a6992a8a94b4c9c515418db1b9167 : SUCCESS in 1h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3ea9aa0054d16cdf7719b38ab2962f26f8c37d89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-21 14:36:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9b5ceba0b0e02025e22d90cb12e4fd37b61e9bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 16:23:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60303753736649e8979bb5fc53393062 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ecc8e1e15944ada9dadf9b9427ce2ec : SUCCESS in 12m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/359e92abbdce4ebe84598443308e062a : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27d8c65f13f4393a56002c29d435b05 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53fffbeba49643538258b3e5dfdaa836 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fbd2984a39e442a92f7093cd5afd121 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa4863f5cbb411cb483c1d1d6a71e3e : SUCCESS in 13m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3feeb95b86124dbd8007783b3105efff : SUCCESS in 8m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b2cfd22dc8ba4b97baef29650dc24882 : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6796e064f1e9499daed6ca2ede9e0133 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40098d06727340e4a0633f9b3faa55c9 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6d1eacaed8d404eaed151114d359976 : FAILURE in 28m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/726592bb4c93400faa8c2b26a1c24c7c : FAILURE in 29m 14s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1db0645a98284861bf9171a65580d2a2 : FAILURE in 10m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/092300b473af42cfaaab3e3276584a31 : FAILURE in 1h 19m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b43530fa114a4023999cf3e81211dea1 : SUCCESS in 1h 38m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26a60fb1cb2544fa94057d95ec540de3 : SUCCESS in 1h 08m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b4136a2da3459b164ce4fa880023c11f5e7e5db4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-21 19:50:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ee651046a95dcf0143bcb0db6e26865787888216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:48:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f551006594bb44898c10445e128568cd : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aacbf2fe4fc245e2b150613f52de7e59 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/692d83d7601f4e1d910408bc3bee75c9 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1c2eda3b3248bbb027070446bcb64c : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27e84ae5ddc24d028b148e1253e11bb5 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3343b5b3d494f68942ef3ff54d80833 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce185d1b561b4f809ca7038365fe848f : SUCCESS in 9m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/83df7305a07e4ca2aae841f48af8d471 : SUCCESS in 5m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/43ed6319a52743eaad8e88d1eb00aeb2 : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3449477589a4fa89d34c94e57fe18f4 : SUCCESS in 11m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/547c43b107c04c3ba4950cf90cddc3a7 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f563640239634fcc8831c2698d6dc166 : SUCCESS in 1h 50m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70b81105e960466e94487af4f9f9d35d : SUCCESS in 57m 37s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c78ef84c14c436d90bc410d38294538 : FAILURE in 16m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46c9bb952fa34fcdb1f9365bacdb8204 : FAILURE in 1h 02m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c6775fb8e4849638aa946d0099de564 : SUCCESS in 1h 47m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f328b09bad8a4115aa8309b1a3057353 : SUCCESS in 1h 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4462fa122b005b41333751af2f084e65dd637fa8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-04 23:31:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0ecc27c1f7c1bdf94af3b909f8dd4c400edab38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 01:17:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ed244f24f664d9b851e86c142055827 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35782d3fdf1744a1bc42c82afca8d136 : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a01161fd46e44f19256792e262a7121 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742121e3f830471182ab9f6904dbb340 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef88e7e928a14c0ca3ada60e138a7fcd : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d134d0c23d44d0abc2e9cc78e90aa0a : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/464078b0324b4e12a9723746bcd21132 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5167d9b946d04444b4297571883e4454 : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3df9a999516e4f95a9ac6763000e9773 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/63302add5c24431799d73b6cd3b5ea0b : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb8075ffc64a4daba185517f13a32309 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/906fce16f30b4d93bed6d776bab12ead : SUCCESS in 40m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/022c68e656d5496a94563fc2f5e49a94 : SUCCESS in 1h 38m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/20061334f03f4e6db244ff09613f90b6 : SUCCESS in 57m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e8078e354f214c12927c21d11a6fc6ad : SUCCESS in 53m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ccb795258f649468c7f8826c84f2134 : SUCCESS in 59m 54s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b4996d2a71348eaa5b47ee3943df681 : SUCCESS in 1h 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8562845bb9a44a19c174099e07f5fb2 : SUCCESS in 1h 35m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1360f2d747d046589d501857db82da79 : FAILURE in 37m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9b5cc058f8fa4636a30417e4180191f3 : SUCCESS in 1h 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65fa5b85ec3f157218fc5080722cfc7bc6d3dfef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-18 23:09:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"73ebc97062041a2e13ece566efd850d1c0dc22f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 01:09:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/085f5138d24b44db99dcc47b3486c36e : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a1a927178564acf92ee9e2ee77cdc5b : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15540cb4bf18472fb26abe30b6aa49e9 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7744563acbb74bff8fa790d434725e53 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c94eb843a6134645ab4cb1181a7dab8c : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ac33f7afb89499084b8dacd1287186f : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ca0d3e56da641d08b7df3a71f3b57cf : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8de78422cca43bda4e3f2a9d3c23fae : SUCCESS in 14m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dbecdea01e784fc8b1c4321430e11d9b : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/64d86615ce9e4a968b72dd8259b29bdd : SUCCESS in 9m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68f0fb434bf045bcaf60f6cce04706ec : SUCCESS in 9m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6d8d8315d294a29a3d96ef775e6b12a : SUCCESS in 45m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdd61cfd55dc4730af645ae2fdc9130a : SUCCESS in 1h 27m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ebc280443e7460db57ff90ab6090372 : SUCCESS in 59m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d2fd5559dfad46e8ad660ac3ea501e0b : SUCCESS in 1h 06m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f63c9cf53e1f4041a940735216ec693a : SUCCESS in 1h 11m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7941535f73414eb18c09cc29fb87151d : SUCCESS in 1h 55m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ec8f1d2a7ac4a7e8834d1516f21a22d : SUCCESS in 1h 20m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10987498c01547018712acbe13a028c9 : FAILURE in 56m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11700808830b45ceb85b51badb01c3c1 : SUCCESS in 1h 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"775fac5c2e7f32bea135fc28864d67049fb2d799","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-15 09:13:28.000000000","message":"Patch Set 7: Backport-Candidate-2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1394289335e36e59eca316e496265ee2be47c0d9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-16 18:04:00.000000000","message":"Patch Set 7: Code-Review-1\n\n(11 comments)\n\nA lot of cruft is going to be introduced here if we don\u0027t look at a few of these things more closely... What do we want to do about this?","accounts_in_message":[],"_revision_number":7},{"id":"955d6ed41382e66de022bb0a5c329f3a2006d2a8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 01:10:50.000000000","message":"Patch Set 7:\n\n(12 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7a33cb00de1a0819ec5b813d7d525adbcdebf343","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 01:10:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9bc836b4699991dd8ad5e742a024baff41d396f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 01:11:10.000000000","message":"Patch Set 8: Backport-Candidate-1","accounts_in_message":[],"_revision_number":8},{"id":"c48d85d8a26b44971937396b45e4fb194faba089","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 01:32:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"41b512c40f94a1aa04a94ccd00e301e22dcef263","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 01:35:11.000000000","message":"Patch Set 9: Backport-Candidate-1","accounts_in_message":[],"_revision_number":9},{"id":"7543beb530f7eb014a94987c4240fe8fae61b9d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 03:49:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/288bd9bbef7b4597aac56f6bc0e96e2a : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d1ed982478440f3b1472c643d145a76 : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c6bf3be4b94593862a96c577b629d5 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4ea873c7de4b86bb0f9e9544778312 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cc616b1fc25445c985473dfe21bdeb0 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd8213531c6344488f95dde67c9d6a68 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5cee3cf722a448a8f4a2b07f977587c : FAILURE in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3363c21188c84f1fb818a790b929f996 : SUCCESS in 12m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/46f745247b9e41f8b9d32e971b07856b : SUCCESS in 6m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2513a5f629894b85aed06c061b1e15b0 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40a16011bb7c461eb5464836c144c516 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b1c433ca2fc4f57811033184153ac8a : SUCCESS in 47m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d876ae052aa942efab070f7c89105cd8 : SUCCESS in 1h 54m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11ab96ccb428425fa4059c2598bf7234 : SUCCESS in 1h 01m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e5c920f27dc14bddad48161f6e0ffdfc : SUCCESS in 1h 12m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ce9c398f928458eb280e261a15a2780 : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/87ee406512a243de81d2fe2ccd7d34c5 : SUCCESS in 1h 58m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d3067166ef949ea98a52345f35aa979 : SUCCESS in 2h 11m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59218a26f64e4beebcd8b524152817b9 : FAILURE in 1h 02m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fc6511cdfb214ddd96199241321d8e2b : SUCCESS in 1h 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e09095baa25a76dfe03625049c37d66fd6469f1e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-17 04:52:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4be5d354e73ab677b34fb926b322ea0ad429a351","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-17 15:51:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0b8ce2e6e9a80d29946d424c900aeaacd724de6b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-17 15:57:01.000000000","message":"Patch Set 9: Code-Review+2\n\nOK, this looks a lot better. I\u0027m ok with this for now, if we find any stragglers we can clean them up later.","accounts_in_message":[],"_revision_number":9},{"id":"acb5743056f438d7f721492d2cbf4891b4107376","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-17 15:57:09.000000000","message":"Patch Set 9: Backport-Candidate-2","accounts_in_message":[],"_revision_number":9},{"id":"b1306fe73729530645f69a513049ab347ec75874","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-17 21:00:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f89f7462ab123448591771ada1b2c9af18e047e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-18 16:30:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6437ad37632539ef205a4917e4a0f785e23d2934","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-18 19:03:53.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)\n\nJust one concern with the change from raise_from to raise.","accounts_in_message":[],"_revision_number":10},{"id":"c2e04b70782da3d5e3473f2f7a763b759d894876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 19:07:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a374e23730a64b8385396f075deb0075 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f95dbd6f3ed24f8aa5045c13d2f1b5be : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7750fe094bd647e19f23cb0630c44864 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ea226277214c4ab20517f8011d29d7 : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c46527a1b934a54ad8c5fb3dadaf488 : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65327afb8701496692db4f3fbfb91d6a : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d0784d6e74d45cea4eaf0a7741dbb57 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919ddf86c114413b9def184fa4ef6f18 : SUCCESS in 14m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/afbeae80fef44feeaba6d9bf5c623969 : SUCCESS in 9m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c7d2ec2638504c72a72fa0b0712774b3 : SUCCESS in 11m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95552281f44e4a8d8cb575f00bff89cd : SUCCESS in 9m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b53dc32f2d4846b8af6a088105d825d8 : SUCCESS in 57m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1c4cf3a7a734c74b2d322c81994c268 : SUCCESS in 1h 54m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b2e8ffda73e4c9fb9e8cc67a281ea5d : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c8cbeaf1016c4debbedf03be84c1d80d : SUCCESS in 53m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3dc1814136c4489b962d2352caf851c : SUCCESS in 1h 05m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e5d7b1fbfe84167af556f4d8fc8f213 : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ae4fcf0549747948051d948f9660096 : SUCCESS in 1h 32m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5917136613ce4ff2bffbd9c4c4588f1f : FAILURE in 1h 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/708e242f79ee48099a961aa000ff3c41 : FAILURE in 2h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"976fe2e96cc69ee4de8cd00972c8ffc225515066","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-18 21:17:17.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nAlso going to rebase this to master again","accounts_in_message":[],"_revision_number":10},{"id":"fc1f056d7906ce61958f8be4b0df15c148d482ae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-18 21:17:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6dff23a2cf2da05b5c51b4b5c91f8199ec0fa7b2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-18 21:47:38.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"e16b31857fe0aaf6d1a2d29ea174f9a96c8d4ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 23:26:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1d203bd248f4ca8b4008dedf5a6e916 : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d307f7cfe337433c92eefcf217dfb089 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abe250c465b645a993611a4e61129f45 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6debee5c0c547928f1473d0b46e8a2a : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ad68c642adb477c8b5c3f81d640178b : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b49edccb1934dee91d13d0fb9ad35e6 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030edc617e8f4f4583044e04016b1c09 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfc068b9c19456383ff11308a0bd5de : SUCCESS in 10m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9b535f137f4e4eedb6da9113f95586de : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1fe974ae0ef0454498b301bc2ff05fd8 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f170aa486d84255a1359dbb3725de1c : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67542a39db9d4a70a372b708c5357c4e : SUCCESS in 45m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/785eb336141040f99fd9da56535acb91 : SUCCESS in 1h 56m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b5d244b1d4744b808e0f92d1e4df1df1 : SUCCESS in 1h 01m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/37bbf376b0984207ba33dde26d410be5 : SUCCESS in 52m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8436a014dae4f60bcbbc9a95abd8df0 : SUCCESS in 1h 00m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f15820cc4b9e430695de2dd5b1730b14 : SUCCESS in 2h 02m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbf7afb4792f43779ecdbeefc26d1abe : SUCCESS in 2h 02m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd902d33312f4dd9b571bc7279afe709 : FAILURE in 1h 01m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f0cd5d7d4754ad186c8462e3e498a26 : SUCCESS in 1h 51m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f5f9d37e6f2a4789e49eecfdd66a4bc51b538a85","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-19 09:55:57.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"fb4ba41e0b1b10dafcb00958658ddafff7203707","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-03-19 12:09:39.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"9a2d061d9d496c47872c582a45eeedd3545bb9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:09:51.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"bade93b3dc3464ec124eeb6f59580d7c5b546f2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:27:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"6bc89f912e9eccf0d44ea00aeb46945ed65c2f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:27:33.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa3deab0303d487bbeedecb457ea0303 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/020b3720e8974ca6bffff3baa9084f60 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e86a45fe3fa4501a7a381b1a4208f4d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4669eb12f7bc4081801583f7d28afcaf : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21bab68fad754bf8ae289e99aa094738 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b89bfc4aa0414f9ceb26cc60f2e9a2 : SUCCESS in 13m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/604f4b482fff42b594b71d594df98cf4 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/84a546eea8804afd9f7ecc72912af967 : SUCCESS in 47m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/109ff35c62ae4bfe919c83abb9e551db : SUCCESS in 2h 12m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92199025911b4d15aaf66cafaaaf0cfb : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac56ef652206484bab897873c9d9518a : SUCCESS in 55m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b941b0cbee94eec9e00869ce5d0c799 : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":11},{"id":"de8e85fe2d0764305cbef26fa1fe54d088dd1eb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:31:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9390c414e9154c3f8f302db89eaaf3be : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"f6b957e8ee017f0bd5ef6c9412430576c0875e30","revisions":{"f1932f10e2ef6dc16687407943d6e53b28f5ea73":{"kind":"REWORK","_number":1,"created":"2020-01-06 22:15:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/1"}}},"commit":{"parents":[{"commit":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","subject":"Make octavia reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 22:13:30.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1932f10e2ef6dc16687407943d6e53b28f5ea73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f1932f10e2ef6dc16687407943d6e53b28f5ea73"}]},"branch":"refs/heads/master"},"d46653be913e8e7d298817c6554b839686374656":{"kind":"REWORK","_number":2,"created":"2020-01-07 21:39:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/2"}}},"commit":{"parents":[{"commit":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","subject":"Make octavia reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-07 19:50:04.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d46653be913e8e7d298817c6554b839686374656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d46653be913e8e7d298817c6554b839686374656"}]},"branch":"refs/heads/master"},"a6eb3db4053dc23c47f6eb548cb3255ec0aaac10":{"kind":"REWORK","_number":3,"created":"2020-01-07 21:41:39.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/3"}}},"commit":{"parents":[{"commit":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","subject":"Make octavia reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-07 21:41:07.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6eb3db4053dc23c47f6eb548cb3255ec0aaac10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6eb3db4053dc23c47f6eb548cb3255ec0aaac10"}]},"branch":"refs/heads/master"},"54a7296398de50710da620f00ce573b58cd0f248":{"kind":"REWORK","_number":4,"created":"2020-01-20 18:02:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/4"}}},"commit":{"parents":[{"commit":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","subject":"Make octavia reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-20 18:02:03.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54a7296398de50710da620f00ce573b58cd0f248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54a7296398de50710da620f00ce573b58cd0f248"}]},"branch":"refs/heads/master"},"e26951ccf95137599d8faf093d0604a41cc8b3be":{"kind":"REWORK","_number":5,"created":"2020-01-21 14:36:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/5"}}},"commit":{"parents":[{"commit":"b96107e27a6d6afa755d93e5d0ac1f09affc11ba","subject":"Make octavia reproducible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b96107e27a6d6afa755d93e5d0ac1f09affc11ba"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-21 14:26:50.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e26951ccf95137599d8faf093d0604a41cc8b3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e26951ccf95137599d8faf093d0604a41cc8b3be"}]},"branch":"refs/heads/master"},"e01e7256ae44b9eac9a97314da71f11a878a2ebc":{"kind":"REWORK","_number":6,"created":"2020-02-04 23:31:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/6"}}},"commit":{"parents":[{"commit":"fb3700566179a21aaa4065693fe53929f354dedd","subject":"Merge \"Transition l7rule flows to dicts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb3700566179a21aaa4065693fe53929f354dedd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-04 23:29:42.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e01e7256ae44b9eac9a97314da71f11a878a2ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e01e7256ae44b9eac9a97314da71f11a878a2ebc"}]},"branch":"refs/heads/master"},"c7d514c3e0e05c643267e8cc254d4d3618da5b12":{"kind":"REWORK","_number":7,"created":"2020-02-18 23:09:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/7"}}},"commit":{"parents":[{"commit":"49036845c118aec11740f18680b779f99113bcfc","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-02-18 23:07:53.000000000","tz":-300},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7d514c3e0e05c643267e8cc254d4d3618da5b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7d514c3e0e05c643267e8cc254d4d3618da5b12"}]},"branch":"refs/heads/master"},"1a51c52cd8596fdb68d91258f386b9fddf1b8e54":{"kind":"REWORK","_number":8,"created":"2020-03-17 01:10:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/8"}}},"commit":{"parents":[{"commit":"49036845c118aec11740f18680b779f99113bcfc","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-17 01:10:37.000000000","tz":-240},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a51c52cd8596fdb68d91258f386b9fddf1b8e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a51c52cd8596fdb68d91258f386b9fddf1b8e54"}]},"branch":"refs/heads/master"},"844c526367aed3e1a5a177a52b0f93bd3f880d75":{"kind":"REWORK","_number":9,"created":"2020-03-17 01:32:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/9"}}},"commit":{"parents":[{"commit":"49036845c118aec11740f18680b779f99113bcfc","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-17 01:32:41.000000000","tz":-240},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/844c526367aed3e1a5a177a52b0f93bd3f880d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/844c526367aed3e1a5a177a52b0f93bd3f880d75"}]},"branch":"refs/heads/master"},"f48fc12e64a726d32db154a984277054866931cc":{"kind":"REWORK","_number":10,"created":"2020-03-18 16:30:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/10"}}},"commit":{"parents":[{"commit":"49036845c118aec11740f18680b779f99113bcfc","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49036845c118aec11740f18680b779f99113bcfc"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-18 16:30:18.000000000","tz":-240},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f48fc12e64a726d32db154a984277054866931cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f48fc12e64a726d32db154a984277054866931cc"}]},"branch":"refs/heads/master"},"f6b957e8ee017f0bd5ef6c9412430576c0875e30":{"kind":"REWORK","_number":11,"created":"2020-03-18 21:17:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/90/701290/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/701290/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/701290/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/701290/11"}}},"commit":{"parents":[{"commit":"e0d4c98556e3cff232ca89faee005178d640063d","subject":"Merge \"[Amphorav2] Fix noop driver case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0d4c98556e3cff232ca89faee005178d640063d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-06 19:17:55.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-18 21:15:26.000000000","tz":-240},"subject":"Remove all usage of six library","message":"Remove all usage of six library\n\nConvert all code to not require six library and instead\nuse python 3.x logic.\n\nCreated one helper method in common.utils for binary\nrepresentation to limit code changes.\n\nChange-Id: I2716ce93691d11100ee951a3a3f491329a4073f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6b957e8ee017f0bd5ef6c9412430576c0875e30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6b957e8ee017f0bd5ef6c9412430576c0875e30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
