)]}'
{"id":"openstack%2Foctavia~701485","triplet_id":"openstack%2Foctavia~master~I67cf32aac8aa8d8123ec8ca1ab433ff23fb1ffa2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I67cf32aac8aa8d8123ec8ca1ab433ff23fb1ffa2","subject":"Allow the Octavia wsgi to accept argv parameters","status":"MERGED","created":"2020-01-08 01:13:32.000000000","updated":"2020-01-21 08:09:14.000000000","submitted":"2020-01-21 08:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701485-1579594044012-641ca554","meta_rev_id":"dd6922daa2edb6f827249fd2fb14338e8e8ab211","_number":701485,"virtual_id_number":701485,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-21 08:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-01-20 07:10:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-01-13 07:32:32.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-01-21 06:10:21.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":1,"date":"2020-01-13 22:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-01-21 06:10:21.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 07:32:32.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-01-13 22:39:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-20 07:10:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-01-21 06:10:21.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-01-21 08:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90fed08d7e34967d6f33ede781a1435c4190be13","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-08 01:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a92e985635927bc4566cf5ef485ddff157e87b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 03:14:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4ccdbf2fe44f45affe23d3b1ad7468 : FAILURE in 6m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c781dfe421c74aeab78486034d1e11b5 : FAILURE in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f410c7cbcee74280b4c56a697e47b3cb : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98c238c60a5a496cba6c8b15c1f702e1 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5ed048358c8457f84f8f4ad5bb938e3 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820bff7a5b104ce7ba11d9b09751adfd : SUCCESS in 11m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2d858f582d5b45ed9b542cf376f183e5 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/65cf04706a6e4636aa368dd617925090 : FAILURE in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db7a4e3ff08943d89e0408fc17cb671b : FAILURE in 5m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e40535d509b54a7fb965535171754fec : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85318b081bf54ddaa08eede1fb6772c6 : SUCCESS in 1h 54m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6392a0f06aa24fbc8ef05109059b040e : SUCCESS in 1h 16m 40s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5687bba83e7346288cec4b2bb5e88cd0 : FAILURE in 12m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8382e5e2e3c6416d897885e78863633f : FAILURE in 1h 18m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6a443c8500b944c4901c3f9ba88f0c2a : FAILURE in 1h 54m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e46bde499f3488ea50e39740387e5ae : SUCCESS in 1h 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89695a35fc279c68d556765c508500a5adca5e16","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-08 18:10:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8740b6a6bca70f3f84d34300a22361af3eee101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:12:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a6c469e31ca479386b8d347be295bec : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62a19e69be3c40b2ae97851153bf1ae5 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cba83925fcc40bcbbc5a33a522b924d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c02d8bde1eba4d50a2c0c5efe00788b6 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f88b77c64387452881a2f1c01a4c7f5c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f437dfd8b46549b3b666b5ca6a7fbec3 : SUCCESS in 10m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7d55a8ce41340c68de457951a7cfe14 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb622c9363964ce79f3c5d6a4c9db317 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be6cf1bfb4e14b08b2f4532b6afa8ee2 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/163b27db67f448929145726422e1b7c4 : SUCCESS in 1h 00m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f39b6ef33494e209b35889113a1d616 : SUCCESS in 1h 38m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b0637b12de04bcabec2adc31b3e1c75 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f6e9eccb2bb4e7fbc902bf2c54da601 : FAILURE in 13m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2a0f948b5cf4b799b3ca9d52538024e : FAILURE in 1h 17m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1658abdda2846dab0e9714b658bd038 : SUCCESS in 1h 33m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a7bbdf453314a56b0f7a50738738af2 : SUCCESS in 1h 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ccc2cfad6d48764c524e1c8dabd0c5eb8eaa05e2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-13 07:32:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0a2e9224eb260e49593d3c6ed9e581264a5383e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-13 22:39:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1721ea6eb3eb677b26e7432cfcf80dc46bd565ac","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-20 07:10:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7c640544a836cc335b686db4efb65782960daa80","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-01-21 06:10:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c86e2b8c8bcd2ed8b92029ba0b6458f891ba48b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 06:10:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79a9e208441fa921c5cca3dcdd0f25ebafb03760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:07:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9ff00a80f83419a94bd4cc6702f9086 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9197ffbd944c6aa25064e51e1719b5 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a29e24aaab43491b832d1f30546c4b0d : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d2ca9de781543d6bd1fa8fb4b059a61 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a43a28f73d40c88d98382eb05c8f7d : SUCCESS in 15m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b467092527f64010b1371dc8cc7ba6af : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea34df33189e42e2b8e4ffc607a4b770 : SUCCESS in 49m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aabea9f458e04baab879265040c74857 : SUCCESS in 1h 53m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/583e4007c04b45b9a49cf2754f883a27 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":2},{"id":"dda75ad7dbc6bd5b5f6b3424b4f8938c12638e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:07:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd6922daa2edb6f827249fd2fb14338e8e8ab211","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:09:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc5b570377c4907bec5a024b9338085 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c6fbe7efe2977e5b07f5f5661f0306d4bdd306b","revisions":{"9707bf9cd5cc8eec5f22f00efc288169362a5c8c":{"kind":"REWORK","_number":1,"created":"2020-01-08 01:13:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/85/701485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/701485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/701485/1"}}},"commit":{"parents":[{"commit":"bc92de786cc46a7db52cc56f5f4eba97ff20231e","subject":"Merge \"Fix tests to correctly call reset_mock()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc92de786cc46a7db52cc56f5f4eba97ff20231e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-08 01:09:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-08 01:13:31.000000000","tz":-480},"subject":"Allow the Octavia wsgi to accept argv parameters","message":"Allow the Octavia wsgi to accept argv parameters\n\nCurrently we only allow argv parameters to be passed into our wsgi\napp. This means that some wsgi wrappers (such as pbr\u0027s script)\nmay not allow the octavia API to accept parameters such as\n--config-file.\nThis patch attempts to pull in the sys.argv if none was passed into\nthe wsgi setup_app method.\n\nChange-Id: I67cf32aac8aa8d8123ec8ca1ab433ff23fb1ffa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9707bf9cd5cc8eec5f22f00efc288169362a5c8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9707bf9cd5cc8eec5f22f00efc288169362a5c8c"}]},"branch":"refs/heads/master"},"7c6fbe7efe2977e5b07f5f5661f0306d4bdd306b":{"kind":"REWORK","_number":2,"created":"2020-01-08 18:10:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/85/701485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/701485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/701485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/701485/2"}}},"commit":{"parents":[{"commit":"bc92de786cc46a7db52cc56f5f4eba97ff20231e","subject":"Merge \"Fix tests to correctly call reset_mock()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc92de786cc46a7db52cc56f5f4eba97ff20231e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-08 01:09:52.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-01-08 18:10:47.000000000","tz":-480},"subject":"Allow the Octavia wsgi to accept argv parameters","message":"Allow the Octavia wsgi to accept argv parameters\n\nCurrently we only allow argv parameters to be passed into our wsgi\napp. This means that some wsgi wrappers (such as pbr\u0027s script)\nmay not allow the octavia API to accept parameters such as\n--config-file.\nThis patch attempts to pull in the sys.argv if none was passed into\nthe wsgi setup_app method.\n\nChange-Id: I67cf32aac8aa8d8123ec8ca1ab433ff23fb1ffa2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c6fbe7efe2977e5b07f5f5661f0306d4bdd306b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c6fbe7efe2977e5b07f5f5661f0306d4bdd306b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
