)]}'
{"id":"openstack%2Foctavia~701823","triplet_id":"openstack%2Foctavia~master~I5baf7d18105494259361be1f0f411412071f1d36","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5baf7d18105494259361be1f0f411412071f1d36","subject":"Support haproxy development snapshot version parsing","status":"MERGED","created":"2020-01-09 19:52:15.000000000","updated":"2020-02-01 11:39:41.000000000","submitted":"2020-02-01 11:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701823-1580557097104-a46f390a","meta_rev_id":"7173a237417cad7507c03a0684f35db02bebcca3","_number":701823,"virtual_id_number":701823,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-01 11:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-01-29 07:55:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-02-01 10:09:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-01-27 19:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-02-01 06:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-02-01 10:09:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":-1,"date":"2020-02-01 10:09:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 08:24:00.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-27 19:35:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-29 07:55:27.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-02-01 06:16:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-01 10:09:10.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-01 11:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ba35ab539bfe317f2d00594978c2ac4c47faf40","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-09 19:52:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67fb4c56d7a775a1a376d38782e34d3cae558fa2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-09 21:21:57.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2\n\nI guess there\u0027s no need to backport this patch, or is there?","accounts_in_message":[],"_revision_number":1},{"id":"f819fefbf002c73b87183d76372ce349effb80dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:16:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e32b90bf6f8644c0ab5f9987aac00f2c : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a66328508424619adab265d12ec2d27 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9253ec7447243cfbc78e6f8205c1419 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ae25c4db6a140a0999879fc84285267 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23eb3e8d969a4ba0a93f40b60e2d2787 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c05c327d504503ba078c8ce8e69186 : SUCCESS in 13m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f1ea87c377cc4a4eb917399bbe5a9f49 : SUCCESS in 8m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bf5b079d00584a91b0189aaa97a48b0a : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/556d5a4ef24144e8b7da32dae0c6b337 : SUCCESS in 10m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d583cb1ed494dac8418112e806eec3a : SUCCESS in 59m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1301a66a629e464db4e3be7e32ef2cb9 : SUCCESS in 1h 44m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14d5e487dad54146a85abdf1008d8537 : SUCCESS in 1h 18m 35s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96a93d5063ac4d119b16db51e07ebd7b : FAILURE in 13m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc2cf078af0b420b89b1a3756546957d : FAILURE in 1h 01m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8807aca197ec469698bfd34b7e52539f : SUCCESS in 1h 41m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53f975020ed74a78a1e6f02dcc0d2af0 : FAILURE in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d3b392071ae3fc8eb3285da01037f1b0133d132","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-10 21:08:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d8db2c7f036791e026d486f2104633ae2a769db2","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-24 08:24:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nKinda not, but kinda not ;)","accounts_in_message":[],"_revision_number":1},{"id":"f5c203f80c2a6c08829874c88ced741627ffe437","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-27 14:07:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca5595f459e98730d0cbf84dfe78557979454cde","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-27 14:08:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cd0be07f5d9e2bca9752e9913a282a3bca0f869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:08:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d2249fb3624a2eac37e91721f4910d : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e064905b1a574191886e25e61343541a : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eec6d836c8c4517aaa6687c0080fe8f : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d14fb80611e4ccfbdd53863b017ea05 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62f3f7de8cfe4ddf92350fd06dd24732 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfb1accdbd54df1bdf0129d6e2cb399 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b21339e30c04f7e942d07f272fe7cc1 : SUCCESS in 11m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e4b5827c877438f95aeb82567025fa6 : SUCCESS in 8m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e03854328d5c4cb2b9833c829b8cab26 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efbaa9a8781e40bd863012d438db97e8 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58b92b62d1ee4160b78e33e42bbea92b : SUCCESS in 44m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c862992300e740ca82683574eb23fcd3 : SUCCESS in 1h 52m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71745c054cfc4a06b7c0130f69813962 : SUCCESS in 1h 04m 57s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ce899c545124c4ca96710828c28a7e6 : FAILURE in 41m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e69e496d1044943b81d559ef5fe5ddd : FAILURE in 43m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f4377030e994ef3a39fa1e638adbfa2 : SUCCESS in 1h 54m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23e27617c5c343259d09088ee979e6da : SUCCESS in 1h 09m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2698f3e5962f1988a10945c24f2e1519d081e59","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-27 19:35:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1bcc38fe48174f3e310818e81c4207bdaf7b681b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-01-29 07:55:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5547055f1d18a89c5c87a3b73cb54cef97062f57","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-01 06:16:28.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"4c0d113a01b504c1f595ec670014fb07b34fc17b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-01 10:09:10.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1\n\nSoft no backport-candidate as we have not tested dev versions before. If someone needs it, please go ahead and propose backport.","accounts_in_message":[],"_revision_number":2},{"id":"42700229bb2e62b3b27a4abcb59bd6768abe35dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 10:09:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec7d26053b159de87e22f2674ac9ab13e6f63ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 11:38:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d456ce8f4704800b742bff851347f50 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace4de7cdf73442a9ea54d556e33c883 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39e9d20f41074422be667ed92d3606ad : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4720bd265af44af8ec1cdd5279bb68a : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8995560d559c4d86b443296e4566d5d7 : SUCCESS in 10m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7a86dce39df489c8cc41c6441d85349 : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff36859668a3406bb1885606846c18a5 : SUCCESS in 42m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/454eed48d0204ce285f5bd7894c98a00 : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a01f34a45737490abb0baea2b3d93f78 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c37db7a9514bf18077ccc8147ef617ee7f9369e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 11:38:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7173a237417cad7507c03a0684f35db02bebcca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 11:39:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d53e7ae982b4e958327b614992a48f1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7dc54eb9c949dc218ad312a43d56adf6b3969085","revisions":{"79cfdfafa713aa37ce66398f703ef2050b81873a":{"kind":"REWORK","_number":1,"created":"2020-01-09 19:52:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/701823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/701823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/701823/1"}}},"commit":{"parents":[{"commit":"47e0ef31bcfd34838fac4c619b699edd65e20223","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47e0ef31bcfd34838fac4c619b699edd65e20223"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-09 13:41:32.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-09 19:50:58.000000000","tz":60},"subject":"Support haproxy development snapshot version parsing","message":"Support haproxy development snapshot version parsing\n\nFix haproxy version parser in amphora-agent for beta versions (i.e\n2.2-dev0)\n\nChange-Id: I5baf7d18105494259361be1f0f411412071f1d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79cfdfafa713aa37ce66398f703ef2050b81873a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79cfdfafa713aa37ce66398f703ef2050b81873a"}]},"branch":"refs/heads/master"},"7dc54eb9c949dc218ad312a43d56adf6b3969085":{"kind":"REWORK","_number":2,"created":"2020-01-27 14:07:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/701823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/701823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/701823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/701823/2"}}},"commit":{"parents":[{"commit":"47e0ef31bcfd34838fac4c619b699edd65e20223","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47e0ef31bcfd34838fac4c619b699edd65e20223"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-09 13:41:32.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-27 14:07:05.000000000","tz":60},"subject":"Support haproxy development snapshot version parsing","message":"Support haproxy development snapshot version parsing\n\nFix haproxy version parser in amphora-agent for beta versions (i.e\n2.2-dev0)\n\nChange-Id: I5baf7d18105494259361be1f0f411412071f1d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dc54eb9c949dc218ad312a43d56adf6b3969085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7dc54eb9c949dc218ad312a43d56adf6b3969085"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
