)]}'
{"id":"openstack%2Foctavia~701963","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I4706972095d42815a3f3cec0bc5aa82a34922bf3","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I4706972095d42815a3f3cec0bc5aa82a34922bf3","subject":"Cap hacking version to \u003c2","status":"MERGED","created":"2020-01-10 13:51:03.000000000","updated":"2020-01-10 22:01:01.000000000","submitted":"2020-01-10 21:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701963-1578693527387-46a14a57","meta_rev_id":"870159b2d997032b8dc225f676a9e23d7cab9a15","_number":701963,"virtual_id_number":701963,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-01-10 15:46:59.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 21:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-01-10 20:05:37.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-01-10 20:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-01-10 20:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 14:16:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-10 20:05:37.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-01-10 20:20:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-10 21:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09b3440edae5719c9d5acdb822c8105d58b2c674","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-10 13:51:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4910ceb7e81a091c8a580a949f4c7e4558252a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-10 14:16:53.000000000","message":"Patch Set 1:\n\ngunicorn issue too it seems :(","accounts_in_message":[],"_revision_number":1},{"id":"35907c9bf527ab48ad9942da792227031774d06a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-10 15:22:24.000000000","message":"Patch Set 1:\n\nIt also complains about sphinx, twice, and networkx. One part of sphinx needs indeed updating but the other part doesn\u0027t from what I can tell so I don\u0027t get. Same for networkx, it also complains but LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"35e354620558ab631309fd1f3300ff248768d9eb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-10 15:22:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd3e3a09e8222d24e5bfbe25e5d6d25cb186814f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-10 15:46:59.000000000","message":"Patch Set 2:\n\nLooks like requirements-check is happy :)","accounts_in_message":[],"_revision_number":2},{"id":"8162aeb9e67882f3145db0d0125120cc951709b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 17:40:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/512b253a8da144d1bc25271e8df65e8d : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d208a2f409d94e7d931888c793d8393d : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2a6fba79efb4729b1006b4b96e97f9e : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1490f2de074074a59b20407024abdc : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fd52cd80660434b9d9ea1bdd550c127 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ec8f6bfbd044c386ccb6a5ac465f58 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32439a88ccb540a098c45e927739fc7c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e345526596843e1bce8d5ea86ffcced : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c95244e5eb7431d9fb5061128f7d792 : SUCCESS in 6m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf2d3e93db1c44e79afef9f2311aae65 : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0e1cdb0d6d543e78802bc2e7d9bdb46 : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/afffe27ec3a941db83af06a4c6e7fb2e : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6454491752674fee8581a30aae98dda5 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2941c1567c564429bd1841c698da5954 : SUCCESS in 48m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/40acaf114bb049cbbf8000f8bdb0bdac : SUCCESS in 47m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0c1c996ddf240b291d88573887b754f : SUCCESS in 1h 46m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4c73d5a5c2664a5586846358a9e3c185 : SUCCESS in 2h 08m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0bee8e97c7c744609661bfa614260544 : FAILURE in 1h 58m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c9eae318279441e82fa0b5f881f86d1 : SUCCESS in 1h 41m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/130985bc0e5d4849807870e610d1dd74 : FAILURE in 1h 46m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e589b655cdae425fb68de57d0a3e2953 : FAILURE in 1h 07m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfc1c279989d40eeac19d137c854c148 : SUCCESS in 1h 57m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c83582b7d25842e3be16e22d0da34363 : SUCCESS in 1h 10m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ee63917af98a407ab97c7cbebead198e : SUCCESS in 1h 27m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8e7685159e854415a4afc96f099cb74a : FAILURE in 1h 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"879a523a88465f9400f0b01abc0eef5009920d42","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-01-10 20:05:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3b2f65b8b1a234194629973abfb61885de67b1c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-10 20:20:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"81572c63aaad09df67bdcb35fede1d7838cf00f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 20:20:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4cd60d14c057a1b7cc4619943a6ad6e87eb2f38d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 21:58:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8725eda5b00f47ed88e495faa7190821 : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31fe7ac16870409faee5356b6bf6906d : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aff116e648494e9be62fe87609bf22 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff44b1befb51403582680005efeedc1e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1050886ac5b426e80c773a0efac680f : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4257dbff8524c01a9eaab05f4809b3d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d16846c4e12480dbdb45829bc5aa814 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/769c0fd74eb24e14bccda2a895c64b46 : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/af37f2083c9741718dcf2f729fc29f57 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e62a494b5b054ad5aa204164e03fd7d0 : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/980a82d8f8c941418fed7bae1f73ba59 : SUCCESS in 44m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c18f8311f6b04f4bb46e1aef26d4639e : SUCCESS in 43m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a77a0f4613a4dc59701503ca3b4bef7 : SUCCESS in 1h 32m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4d869df73f6447669045e6556669e435 : SUCCESS in 1h 37m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d7a77bad36df47702981df2f86133156f4911f44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 21:58:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"870159b2d997032b8dc225f676a9e23d7cab9a15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 22:01:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6d3d043a3e44a89feb1b0da8db4d6f : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95bc592eb58f416e8f035ee8eb77bdce : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85f7abbbaf958458ee12415b0de0e938c20fb121","revisions":{"6f18953f5d4110dc81213d8a76d3741b4243e118":{"kind":"REWORK","_number":1,"created":"2020-01-10 13:51:03.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/63/701963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/701963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/701963/1"}}},"commit":{"parents":[{"commit":"d603c817f40786ebed0711089ed8d7a166b6c46d","subject":"Merge \"Accept oslopolicy-policy-generator path arguments\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d603c817f40786ebed0711089ed8d7a166b6c46d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-01-10 13:35:52.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-01-10 13:35:52.000000000","tz":60},"subject":"Cap hacking version to \u003c2","message":"Cap hacking version to \u003c2\n\nHacking received a major update to version 2.0.0 that added new checks\nthat fail on stable branches. Instead of addressing those new check\nerrors on stable branches, it is better to cap hacking to \u003c2. This way\nwe\u0027ll not see lint errors again should new checks be added in the\nfuture.\n\nSupport for hacking 2.0.0 in master branch is being worked in\nhttps://review.opendev.org/#/c/699302/\n\nChange-Id: I4706972095d42815a3f3cec0bc5aa82a34922bf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f18953f5d4110dc81213d8a76d3741b4243e118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f18953f5d4110dc81213d8a76d3741b4243e118"}]},"branch":"refs/heads/stable/train"},"85f7abbbaf958458ee12415b0de0e938c20fb121":{"kind":"REWORK","_number":2,"created":"2020-01-10 15:22:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/63/701963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/701963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/701963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/701963/2"}}},"commit":{"parents":[{"commit":"d603c817f40786ebed0711089ed8d7a166b6c46d","subject":"Merge \"Accept oslopolicy-policy-generator path arguments\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d603c817f40786ebed0711089ed8d7a166b6c46d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-01-10 13:35:52.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-01-10 15:21:04.000000000","tz":60},"subject":"Cap hacking version to \u003c2","message":"Cap hacking version to \u003c2\n\nHacking received a major update to version 2.0.0 that added new checks\nthat fail on stable branches. Instead of addressing those new check\nerrors on stable branches, it is better to cap hacking to \u003c2. This way\nwe\u0027ll not see lint errors again should new checks be added in the\nfuture.\n\nSupport for hacking 2.0.0 in master branch is being worked in\nhttps://review.opendev.org/#/c/699302/\n\nSince this patch touches requirements file, the requirements-check CI\njob also runs and does not find matching gunicorn and sphinx\nrequirements. This patch fixes that too.\n\nChange-Id: I4706972095d42815a3f3cec0bc5aa82a34922bf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85f7abbbaf958458ee12415b0de0e938c20fb121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85f7abbbaf958458ee12415b0de0e938c20fb121"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
