)]}'
{"id":"openstack%2Foctavia~702043","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ifed862639d3fc3de23ace4c7ceaea1a4eca62749","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ifed862639d3fc3de23ace4c7ceaea1a4eca62749","subject":"Add listener and pool protocol validation","status":"MERGED","created":"2020-01-10 20:58:54.000000000","updated":"2020-01-24 10:41:06.000000000","submitted":"2020-01-24 10:38:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702043-1579862285527-46a14a57","meta_rev_id":"c5682f87b8674124a1b171ccf8dfd57e47be03b4","_number":702043,"virtual_id_number":702043,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 10:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-01-24 08:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":2,"date":"2020-01-13 06:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-01-21 15:20:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-01-24 08:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-01-10 20:58:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 20:58:54.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"REVIEWER"},{"updated":"2020-01-13 06:33:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-21 15:20:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-01-24 08:14:32.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-01-24 10:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cd0ebfadda9046a2cb2a64fb159634c348a4c7a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-10 20:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13522e3d5bd99106349495f69fc90c48c81ec4fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 23:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0e41643ec2a4d5fa75e419333224df7 : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bc00f89b9494202a3677cc5f51ddbfd : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a4f89f42904c3796dfad90c297a563 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86be513d634049ed89436a0fc412df4a : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1da62981ed14b08807e7ad351f9976f : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50c8fc6b20f24e6daf6ab032b7eee2d3 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43224d1b336a4bae963700d0b6eb6318 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c5f4d3b6614ace94f0c074c762194c : SUCCESS in 6m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8b3fad03954743f2acf9b5d7da49ea43 : SUCCESS in 1h 16m 30s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f705b18663e94827838d2b5fb644644e : SUCCESS in 1h 16m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a496a14c9b46442ab955d79e07b267f0 : FAILURE in 59m 17s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e36f579483ac41808eff9b76117fcce2 : SUCCESS in 1h 28m 02s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0c77b3b3e6a44e238610b87e69032434 : SUCCESS in 6m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cfc3dbea7cab422296e34436573cf6e1 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/107f197a9feb4aab919b9747c760745c : SUCCESS in 7m 04s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c50b3edb06ef4ee5811513cca6a7f671 : SUCCESS in 1h 17m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a610edf5dcbd49c095ace45950369944 : SUCCESS in 1h 13m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a95c9d3223b74876aad2899758577abd : SUCCESS in 45m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3c0dd8ccebc14c77bd11359ad2cb194c : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5566603a133840a98c2aff3f7c5f5aa6 : SUCCESS in 1h 31m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3964bf8f36414ee185c21ed1052878e9 : SUCCESS in 2h 04m 43s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/36d385ea677c445180d54bbf57d88dfc : SUCCESS in 1h 42m 00s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d3fc115ddc30428c8de2f82ab4071143 : SUCCESS in 2h 28m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41ec3b947c7b4f5b8fe082eebf4eb95c : FAILURE in 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56e24f6834894078a5f607dc539fa79f : SUCCESS in 1h 55m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0543228ba5214a3b8f2769ea0fbff9ac : SUCCESS in 1h 46m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dad03da3ec0649838472f2794cd88aed : SUCCESS in 1h 16m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/851d43f841974617b3686a98324f6ee3 : FAILURE in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10921bf9a35920a2f0b9ccf515e427fd6d18f57d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 06:33:02.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"acdec923c3c8dfc411ea42c93803e90ee574675f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-21 15:20:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6fc1206d2117406249ada3c1f28be709243c240e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-01-24 08:14:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b98967dc7a3f36dfb8c6d96d9916a01d5215c2de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 08:14:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c58fcc7cb9736c1c32faf73e2c806f0374606e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:38:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/facabc29e900420f92c1a05da464641d : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a9c1d7c940548c38687958517e6bbfa : SUCCESS in 7m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f83891ca76c4460da732fe4339a01b93 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/653ebb5f758a43069ed9778e9001beb6 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe265d55bd894d259e85c8cad8e52b51 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f275b0c08d045eb8c97fdb7dd1b627e : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/742cbe3a37fe497b847c0cc69957208f : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f29846e75e944da1bc06fa16bd638f7e : SUCCESS in 7m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0d7efc459ad64d7789f50ba585fac6ed : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70de51b74c194aceb2875ae311fb8478 : SUCCESS in 7m 12s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96efefdb0a2b4a8b80e300b5d615b35b : SUCCESS in 1h 05m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/40f6b570700041478f9a6306f80f93bb : SUCCESS in 1h 05m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/db2fd24c4af54ad6a0eafacaa3577b13 : SUCCESS in 1h 42m 30s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f3b93314f3554c07a894b37bb16727ea : SUCCESS in 1h 21m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c26cc2e012da4bd5a016cee1a89239ab : SUCCESS in 1h 10m 02s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/45ea48bf350347909fc6dac38805a73f : SUCCESS in 42m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b44fe9d44be6490c95d47a4b4571e8ac : SUCCESS in 2h 13m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ad3a93ffb7d647c099e2815f5dc1b826 : SUCCESS in 1h 30m 10s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8dee5d88f49e4ab3b7a9df21b720e510 : SUCCESS in 2h 07m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78e1f19726734f29b315a9b791b00d0a : SUCCESS in 1h 46m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9973c8b84a86604ae8a9247098b6e9cc8523275c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:38:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c5682f87b8674124a1b171ccf8dfd57e47be03b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 10:41:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336b01aedf7541b3a31c77ee530590f8 : SUCCESS in 2m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6eee36367df4cfd9d4e2a9293c1bec0 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f37a73eb96f8115abf9da2243dc84a6ca9ade1d","revisions":{"1f37a73eb96f8115abf9da2243dc84a6ca9ade1d":{"kind":"REWORK","_number":1,"created":"2020-01-10 20:58:54.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/702043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/702043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/702043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/702043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/702043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/702043/1"}}},"commit":{"parents":[{"commit":"d0ef7a823150ea84e0fbb8d54711dfbe8fb9e477","subject":"Cap hacking version to \u003c2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0ef7a823150ea84e0fbb8d54711dfbe8fb9e477"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2018-08-21 06:26:14.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-01-10 20:50:40.000000000","tz":60},"subject":"Add listener and pool protocol validation","message":"Add listener and pool protocol validation\n\nThe pool and listener can\u0027t be combined arbitrarily. We need to add\nsome constraints in protocol side.\n\nStory: 2003500\nTasks: 24777\n\nCo-Authored-By: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nChange-Id: Ifed862639d3fc3de23ace4c7ceaea1a4eca62749\n(cherry picked from commit 47e0ef31bcfd34838fac4c619b699edd65e20223)\n(cherry picked from commit 87704d42f42156bbb60138b86e34ea86f84629fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f37a73eb96f8115abf9da2243dc84a6ca9ade1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f37a73eb96f8115abf9da2243dc84a6ca9ade1d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
