)]}'
{"id":"openstack%2Foctavia~704053","triplet_id":"openstack%2Foctavia~master~If5bff05f65c6b231c23bfcffa584da3cc346d77b","project":"openstack/octavia","branch":"master","topic":"stop_configuring_install_command","hashtags":[],"change_id":"If5bff05f65c6b231c23bfcffa584da3cc346d77b","subject":"Stop configuring install_command in tox.","status":"ABANDONED","created":"2020-01-23 19:11:45.000000000","updated":"2022-11-07 18:15:40.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbb5980107bbc6afbedf521f22919277af4006c1","_number":704053,"virtual_id_number":704053,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 18:12:09.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-14 21:17:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 19:44:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-14 21:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b42a6361385f48d43228a9323035c75886a238","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 19:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32da9bb22446dde4c92ab653b87c9a6ea097ec31","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-23 19:44:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4f13fb891e1fbac1669b67550d4fcef52a34a1be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-23 19:46:17.000000000","message":"Patch Set 1:\n\nFYI:\nhttps://packages.ubuntu.com/search?keywords\u003dtox\u0026searchon\u003dnames\u0026suite\u003dall\u0026section\u003dall","accounts_in_message":[],"_revision_number":1},{"id":"3cbb1357a044935c16316ac8d7f612c8d799c57c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 20:12:12.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f63a41abc64100135ac89d99238d37bb5505527e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 20:58:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2326d19ecb444ea6a2a6836878a0e95f : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29fdf2f89a394783822a3d30eb11e9cd : FAILURE in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eac167db2794925ac0c4fc3a7e29bb7 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4f9e5c0e384bc799166e3d86193c23 : FAILURE in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de56ff6b12d473584d29c0ff588485e : FAILURE in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc34c047876e4af99bea5e691407431a : FAILURE in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e9c12252da48e398153d79643f51fc : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/998288eb84c649509f48a70aaf161d54 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b66eee4a2714edcac2869b890dae394 : FAILURE in 4m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/90d29f698a5f49bbb07b973af481e979 : FAILURE in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b97cfef28e064156b1c0573def382195 : FAILURE in 5m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d724e4236c44cc9841f426a5af41dd1 : SUCCESS in 39m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/088b4a52564440028c92e41439c75ad1 : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f28335aa4c1b400d823e207b7ab5c57e : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc6924de81c74a2f8ec275e8aef3e6e1 : FAILURE in 42m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3b63aa66fb8e41d5ac508845e3915790 : SUCCESS in 1h 43m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41e1e1faab1243b1890df74b1b906e31 : SUCCESS in 1h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b0ce6c235ce44db2252e84ec87d621fc7c44139","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 18:11:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5effe46620b4138e59a4fa8f5274d5e1f769ae4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 18:12:09.000000000","message":"Patch Set 2:\n\nFrom the same change in Nova:\n\n  This is not encouraged by infra (particularly mordred and smcginnis)\n  since it\u0027s less obvious and easy to miss.","accounts_in_message":[],"_revision_number":2},{"id":"0e4a885e759a932268726c57a0bbfbfe1fea30df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 21:17:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/95ffd3019cd34dc6abd5e812084074f6 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccbf5ab954234438a8e5b6c9ffa35ae6 : SUCCESS in 11m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4393330b5f9d42fb91995ee0b2e59e24 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1068cddafede4fafab933c99f6d4a7a8 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f50a3bf8bca40359a07f62b8d6669f0 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f51d33b74ef746cab477e1dc67e2c529 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adddcce157b448cc98ff185319dfdd15 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be4bf0e28a0b4aa281e576d679a1ee8e : SUCCESS in 6m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/583c14948c29443a98f400106c18c5bd : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7603b74542b248ef9b6d2f95bd6d62fc : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bbe05903f4cf49e4ba046a8ddf208dda : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c062adcba9d5489aa7099552253b5903 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2c092fec7c0423aa3bc8f334130414f : SUCCESS in 1h 52m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c41f25e00004965bb23608ba11af806 : SUCCESS in 2h 10m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dffb26ff1aac4ad3bc8be77236767980 : SUCCESS in 41m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/906a09620ca241cc9db74934753747da : SUCCESS in 37m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1a4cc0dd06c4e4e8e60afee33c474d3 : SUCCESS in 38m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b9a84de65cff49309d803b6686049b98 : TIMED_OUT in 2h 35m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/702beb28185b45bca272b80050311429 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c36859d0d05141aea6e593e7597c0cd7 : SUCCESS in 2h 11m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72b629a67a4e4dcc904d9d03118ab94d : SUCCESS in 2h 06m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/947e5f0bf169474ca48fc6a4e8b9c44e : SUCCESS in 1h 53m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9ee9e3b9d9424f6492f64cefea0c1cb5 : SUCCESS in 2h 03m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fbb5980107bbc6afbedf521f22919277af4006c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-11-07 18:15:40.000000000","message":"Abandoned\n\nNo cycles for this, un-abandon if necessary.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"272c226cd966cbb57e41d351881c256a839f09da","revisions":{"559d2d754eb02ff8921228a199010f294459b44c":{"kind":"REWORK","_number":1,"created":"2020-01-23 19:11:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/704053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/704053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/704053/1"}}},"commit":{"parents":[{"commit":"b6deb12b5dcac2b2c714d70c195d4ab4420336d4","subject":"Merge \"Allow the Octavia wsgi to accept argv parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6deb12b5dcac2b2c714d70c195d4ab4420336d4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:10:48.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:10:48.000000000","tz":-300},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: If5bff05f65c6b231c23bfcffa584da3cc346d77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/559d2d754eb02ff8921228a199010f294459b44c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/559d2d754eb02ff8921228a199010f294459b44c"}]},"branch":"refs/heads/master"},"272c226cd966cbb57e41d351881c256a839f09da":{"kind":"REWORK","_number":2,"created":"2020-10-14 18:11:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/704053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/704053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/704053/2"}}},"commit":{"parents":[{"commit":"8ea4d40281f4da7e5695128bc5a55994db0a3166","subject":"Merge \"Fix pool delete race on load balancer cascade delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ea4d40281f4da7e5695128bc5a55994db0a3166"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:10:48.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-14 18:11:05.000000000","tz":-240},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: If5bff05f65c6b231c23bfcffa584da3cc346d77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/272c226cd966cbb57e41d351881c256a839f09da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/272c226cd966cbb57e41d351881c256a839f09da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
