)]}'
{"id":"openstack%2Foctavia~704205","triplet_id":"openstack%2Foctavia~master~Ia255c320977deabf0787365cb884b3212164a010","project":"openstack/octavia","branch":"master","topic":"two-node-testing","hashtags":[],"change_id":"Ia255c320977deabf0787365cb884b3212164a010","subject":"DNM: Dependent patch for two-node job debugging","status":"ABANDONED","created":"2020-01-24 23:22:06.000000000","updated":"2021-07-28 20:03:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5e70b335fb1628dba9ed5a88d085b5ad4c68917","_number":704205,"virtual_id_number":704205,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-03 00:49:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 00:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de181b3adb6ed8428c93be92724e419ea4c5f887","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-24 23:22:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8efe6fcc67616f2d8fc0f78e2c2abef327edefd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 01:04:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65a1074e366c4604b39d651f64b92b57 : SUCCESS in 10m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e76e9c10b493479ab8dcee7846faaedc : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30a8582725540d19e4479d367a79e4a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8cbf6d74ee348c485deb6f13233d227 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f1afacb52ef42c98cd40e721f05c360 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e46f861b625941dcbff7599ecd7325f9 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043186c451a34157aa08ddc75a70bed9 : SUCCESS in 10m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/853399f1d8f7489c864d3d64becc00f1 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/011e3eaa3f19496aa81b002db14f65b7 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/013ab261b5b4410388f07164ca810b68 : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c478099bd78443939aeec0aed6d86f11 : SUCCESS in 46m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5289839c6d7c431ab0362ed3ac404b6d : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/efa84413b8a8498185b946b62a229e72 : SUCCESS in 1h 02m 22s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db8b19b9c97f489db4eace2fe9ee6f27 : FAILURE in 34m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/80d8f58488e947f685b6a39b69f8dd5a : FAILURE in 38m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ceda7b407d5e4425bae07a0cedf5df22 : SUCCESS in 1h 37m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9f74f246a2c464a821737a5182112b6 : SUCCESS in 1h 09m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eaa272e4f315da17fdf659b7d2b91b3b1a05903e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-27 19:29:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cdb7f3ae982c376db1ba95d1b408e78dae52a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 22:10:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c32a9a71865a48dca3449784d73d69ec : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fabd551375f8438da4664f3e17512182 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4d558af5c5492194cac1b4bd70a595 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd8a662ac4f454db19fa28828577277 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/414d6488ce644acab77858e6d1282b15 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b764f9fd91a044b4abeca3abaa02bd9c : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796531c492644f0496903b168519ace4 : SUCCESS in 11m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea15af12884346b48008c7088a5f5906 : SUCCESS in 6m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1856016b1c364252a98dcb2e8579bd3a : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f339b804a6a14158951bcce1606635d0 : SUCCESS in 9m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/326a996c579a4447912008c6832cea86 : SUCCESS in 1h 02m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a923dfb57fd493aab317dfdd81cdab7 : TIMED_OUT in 2h 35m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/704de41337384d2f8b9a47ad15ad6b65 : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe049e35848842a8a30041c602f6dc03 : SUCCESS in 1h 34m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69d5e899d9e94251ad7675268d66aaec : FAILURE in 53m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9bd59c8693e64fd8812053cb57b17781 : SUCCESS in 1h 44m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/925a4b808a094d3dbbf80c7339176772 : SUCCESS in 1h 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02cd57503e67e6372bf500450d65cfbf3bf04013","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-27 22:57:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3845ceb07a120d82d230dd8f36d929f4fb5cb107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 00:43:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5673e5d2a494571a590f978d8fe0932 : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5be25c245c548c7b680020c77152a14 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02862b8c5544915a8bc961250565c77 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/890dc008c5324227a7520b37a88e652a : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e38083f27a348d5b6fde026f2b43ac2 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4fdacce01e946999bccae2e2099a6a7 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bf8f2f9461447795b2b5da2b68907a : SUCCESS in 15m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b06712fc65194aa69a22616ce1209f54 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8373bc921594426bb48de3e9d4a68188 : FAILURE in 11m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91fd1282f73444a180db2d90c1c26a12 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7397d994051c4a25bef3d91b61f6ca97 : SUCCESS in 40m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d526ab6468544efc9d31e52b08c86ae4 : SUCCESS in 1h 42m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/532af00516084fa59de26f61cfcd3159 : SUCCESS in 57m 08s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61023052fc4d454387cf2ebf13278ff0 : FAILURE in 1h 43m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1b4ecd72a2b4fea8c062a6431f49095 : FAILURE in 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3fd2b957fa5b4b4995c0dae4d790bfc1 : SUCCESS in 1h 31m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67d83c970e6449abb28ce500981e2fe2 : SUCCESS in 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebf91552d8c664dc69848d5704f80e1953d68730","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 16:29:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7cd3f09effdfd095ff3ade1fcfbf8da76358375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d746f3bd52724c139bcff3646f275d1d : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a1573e0de8b4b29ac2db743a9913740 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec98a77bbd1473897f142de0e2fa935 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e6318048e4d4e18842e56b6c920ec7b : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/143cf3f0d2634a93b28eb7991d07b7dd : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3113ba0ca8441cfbfb8b4d5ef6a5063 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84e3de3e0554ce3adf52065c682f5d4 : SUCCESS in 12m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/472319404bbb433a889743d9e26d9ed6 : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9510339659a24730bae01f6681a3772e : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eaa7e9d5bded47fead3098f1514274ec : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78bf4383783c435a802e17cc10cc4d57 : SUCCESS in 42m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16b042e1135548daacf0f1cdfa757bbd : SUCCESS in 1h 42m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/550d60626d434998b1e0951cbac3a0a8 : SUCCESS in 58m 21s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cdbdcc32918446e29368934ea49dbb81 : FAILURE in 1h 31m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/90ebc2f4c5af42599adf4f8f3abbcce8 : FAILURE in 39m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/413b6bd443c948c48790ca4a305c1e22 : SUCCESS in 2h 11m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/251ed120b33643d08ed50db0d73d12e6 : SUCCESS in 1h 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5277249ea99dab681feef844d737efc949f394a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-11 15:01:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c3f13bba61bffa9bd93198cd599f9747637086a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 21:34:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dff51c84eaee492287c43eed7e4e8d05 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06ed224dd42e42f49f9c6c487bec9eb1 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00009e9a6674b07b40f5521ad61af54 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d5dec2160f645939175d1481a96a3df : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/825bc1b08909497e952c40656ec945a5 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4ed592839274807b51d030b7af6ce67 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2667aee01e4894bb34b856e3b277b5 : SUCCESS in 14m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7e2710a986014bb683998f826dc8a8b6 : SUCCESS in 9m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/60c64471c8fd45d1967110b1f0bc502d : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e804bf03dc441398237058f4777c43f : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8bf7ddd5a8c64486aac2189a61336943 : SUCCESS in 49m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a753afc7aada4243b3bf5fb904f3b783 : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abaa6da33bcf46f3a754bf5d25a78af6 : SUCCESS in 1h 09m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a8fec90e1b9044dfbb51754c68eae998 : FAILURE in 1h 25m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71f7c521e6b041dcb185d97be8bb54fa : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3bb795bcb6e14efea121286b18c8d42a : SUCCESS in 1h 45m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f921a42964d54612bbdfa4cc4b7b5cbc : SUCCESS in 1h 31m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40d9114d179344bf94b1ca27863c95c5 : FAILURE in 1h 13m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9bb2acf4a464fc4860cf4cb9dfade33 : SUCCESS in 1h 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f21496ee3700fe533d5e939dd56aa94f75d07b6f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-11 22:16:18.000000000","message":"Patch Set 1:\n\nrecheck\n\ntwo node passed, spare pool failed","accounts_in_message":[],"_revision_number":1},{"id":"958ecc349ab6ebf99db7531982e33117401d90e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:20:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e164af26068e40fb9e23ece65463b693 : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fef6518a5974c64b76d6a9509080c1a : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43e0fc17d244a17adc4c8e44c4146df : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15fef24634e941c3a5eaa255af033a99 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dac3008bdcd443a0955b5285b9e6186f : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89ae3b652e574314a4c26e2aac974b6e : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bc319d12f34060bbfded0337ceb20b : SUCCESS in 15m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b0765f81e3f4446b36ab38d5e51124e : SUCCESS in 7m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61e097450a6844f3ab71a0145bd62ab8 : SUCCESS in 10m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39807c6542d549c2b7c841b084fdde7b : SUCCESS in 10m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c2fc39bfcdb4bfa9d9cf000375207ea : SUCCESS in 47m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b867fb995f24fdba88ffa2f100325a9 : SUCCESS in 1h 29m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11b27197d10f4f47bf7a341b07c4a882 : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ecf7a1d9dd2c4594b218556e7acd94ce : SUCCESS in 1h 00m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/85bc2b37bb1c44908868f8387f60c41b : SUCCESS in 1h 07m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/33ba9ba09dc84c1498f85c5c4949135c : SUCCESS in 1h 32m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8a1c51856c348cb94e4cbad3174ea04 : SUCCESS in 1h 54m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6871d004e184b7396d9e9c13bde36f5 : FAILURE in 1h 14m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3ae2345e314d4c7a8f89f0d402177536 : SUCCESS in 1h 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aca38ea8318a5b9403ed244c5c371330d13af171","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-02 14:32:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a91ace007d42fb6be584464c51f681da9ab87883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:50:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/666af26724e34c7a8901e6cd41ded74e : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1938fcfdee2450ab6b3424c68057ba8 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4b17b934af49b0b40846f34057d504 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf8edf9922a246f5ad211aa525f9e369 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbe720124856480cb40b682e2b5641b6 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dccd78b684342efb743904122b541f8 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd9cde34ae84ff59d23f0896c7640cf : SUCCESS in 13m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3a544a464d24b119f00d88a880847b9 : SUCCESS in 7m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cdb9df4823b4433a809e401ff36757f5 : FAILURE in 8m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32080d6f395c4eab8e00d05a10bcce02 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56a131cc83124201a596ef1d8f2c2c60 : SUCCESS in 52m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40f3a58de4424b2faa12c872411a273f : SUCCESS in 1h 51m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1250be1cb3e844509156f9f26c9e077d : SUCCESS in 55m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cdb663db64ff4f53a86480bee1d46482 : SUCCESS in 59m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91fb33e65a574605b99991a82b8f05f4 : SUCCESS in 1h 04m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0196b81985d14f3795272b323010704c : SUCCESS in 2h 03m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c210d219ca5649b0b661ea310206f769 : SUCCESS in 1h 59m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c32ad5e34fd141a8bb754b73ab466bff : FAILURE in 1h 04m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f52fb6c042f343ab8c32b90f210f487f : SUCCESS in 1h 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"745a25f2fe16934360b163bb26e956561d760045","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-02 22:28:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"11937a6ec6d58b33ac8ba09278608a676558b494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:49:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3e9c4188cf84c5d9a42d8aaf8d253ef : SUCCESS in 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5895820b6a641328f16676034b3e340 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea172689d954ce2919d238617a60309 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f202c5e7ac0b4acbbf53b736c40b931b : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b33124bad8b41048580ddcbfb6ea53a : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc0756e1ad62473982e4528e161c27a9 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c69576167d7422c83f8ba9392535a58 : SUCCESS in 14m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ffc53664b957492f9f2d0167f652b6db : SUCCESS in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/352db2b9047743dd8b018823ef34c16b : FAILURE in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cecc1388c509491e89f9b26e36dc7ab7 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e67253d17c745b1a7ebad2377987883 : SUCCESS in 48m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3a9d9eb6e9c40ba877b31bbdd4dd976 : SUCCESS in 2h 10m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/321b6acedf8b47c0894025a7d59fa444 : SUCCESS in 56m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/28ae58de01b346e2904837cb513e9cd7 : SUCCESS in 55m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51d18131d17045739256e2307ca11ebb : SUCCESS in 1h 12m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e7904a85ff8141958c278e0548b0c7ab : SUCCESS in 1h 57m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15e558b6bd0b41d4a1071c85916625fd : FAILURE in 2h 02m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4179b13b8f104ebcb3f5836fda3026a8 : FAILURE in 1h 08m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b04b90ed5eb94df4bd81fd0b8ad47921 : SUCCESS in 2h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5e70b335fb1628dba9ed5a88d085b5ad4c68917","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-28 20:03:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"980dd033d609d83c6dd9751e77dc1949b89c834d","revisions":{"d5d8976da1c7eac42d46ff5a1717c6bb5c343e3e":{"kind":"REWORK","_number":1,"created":"2020-01-24 23:22:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/05/704205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/704205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/704205/1"}}},"commit":{"parents":[{"commit":"741a47763d89580953c8c12a4e762c065149985a","subject":"Merge \"Fix the interface filenames for Red Hat amphora images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/741a47763d89580953c8c12a4e762c065149985a"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-24 23:21:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-24 23:21:23.000000000","tz":-300},"subject":"DNM: Dependent patch for two-node job debugging","message":"DNM: Dependent patch for two-node job debugging\n\nDepends-on: https://review.opendev.org/#/c/704202\nChange-Id: Ia255c320977deabf0787365cb884b3212164a010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5d8976da1c7eac42d46ff5a1717c6bb5c343e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d5d8976da1c7eac42d46ff5a1717c6bb5c343e3e"}]},"branch":"refs/heads/master"},"980dd033d609d83c6dd9751e77dc1949b89c834d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-02 22:28:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/05/704205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/704205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/704205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/704205/2"}}},"commit":{"parents":[{"commit":"5cbe0453f6dbe37a30dc4d1fcd7f0cfdbc352be5","subject":"Merge \"Add F5 Networks provider driver by SAP SE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5cbe0453f6dbe37a30dc4d1fcd7f0cfdbc352be5"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-24 23:21:23.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-04-02 22:28:07.000000000","tz":0},"subject":"DNM: Dependent patch for two-node job debugging","message":"DNM: Dependent patch for two-node job debugging\n\nDepends-on: https://review.opendev.org/#/c/704202\nChange-Id: Ia255c320977deabf0787365cb884b3212164a010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/980dd033d609d83c6dd9751e77dc1949b89c834d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/980dd033d609d83c6dd9751e77dc1949b89c834d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
