)]}'
{"id":"openstack%2Foctavia~704650","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I335287da66524d026f0c42086d885b478c568bbd","project":"openstack/octavia","branch":"stable/train","topic":"fix_route_file_rh-stable/train","hashtags":[],"change_id":"I335287da66524d026f0c42086d885b478c568bbd","subject":"Fix the interface filenames for Red Hat amphora images","status":"MERGED","created":"2020-01-28 18:14:32.000000000","updated":"2020-08-14 14:29:10.000000000","submitted":"2020-08-14 14:27:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704650-1597415262424-bd91aa04","meta_rev_id":"11c77f1e353a2fd6d3194695bd7c836ca015df37","_number":704650,"virtual_id_number":704650,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-07-06 17:57:40.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 14:27:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"recommended":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"all":[{"value":1,"date":"2020-02-03 17:49:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":2,"date":"2020-08-14 12:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-07-27 09:11:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-07-24 12:14:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":1,"date":"2020-08-14 12:42:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-07-02 17:51:52.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 17:49:02.000000000","updated_by":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"reviewer":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"state":"REVIEWER"},{"updated":"2020-07-02 17:51:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-24 12:14:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-07-27 09:11:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-08-14 12:42:29.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-08-14 14:27:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0a87138f7fc0154c005a1656828634fc382556a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 18:14:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4ae583be98fc08ed4eaf991d24fc22fc5c01c34d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-28 18:15:51.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport.","accounts_in_message":[],"_revision_number":1},{"id":"d225c7ff9162ea0289151264aff93811a8f28a5e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 19:28:23.000000000","message":"Patch Set 1:\n\nEven the train one fails :(\n\nAttributeError: \u0027IPv4Network\u0027 object has no attribute \u0027replace\u0027","accounts_in_message":[],"_revision_number":1},{"id":"39069d7938820d2c5d108dfed9c73cade67be4a0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 20:03:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bfc247133706991b02d8b74255b60a8a862da0e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 20:03:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"429c1352b7e46ad7e6a86cc3fdbb0dd99365e548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 22:28:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07eb95d6e508457aa57db836e003b06b : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/560620bbf99846b8ab547c83a570c531 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4051a14b244a118c73bd1305dd4bf4 : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/965a15408a8640058b4cde3f00c17077 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47300f2a3fcf4c36b4ef8c829b682392 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/947ff6442e6f4ac282e7e01a4417fc82 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289a69724d3a452e9c767bde5fc48ded : SUCCESS in 14m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2856f819e50f4a0a929b2de6d4ead6bd : SUCCESS in 6m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dcadbe245f574f6ea4521ed7de1b4e6c : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/846e8d8bd34c47e5a72b7855dc8c7523 : SUCCESS in 7m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/65a0a465b5194d11b46b17e52333e721 : SUCCESS in 6m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d03f6efd525548cd8aee55da3c5d1dbc : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8920003bae549d3bd23e8e9a18f194b : SUCCESS in 46m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5e6d5a4374514462bb93c90cec5a04ca : SUCCESS in 51m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/441065d00f1e439aa0c386bda0cfb86d : SUCCESS in 1h 51m 46s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/98bbc90b0c78421aae43c76a69e0805d : SUCCESS in 2h 05m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3d2a342ec55b44fd8cb06afc57206a71 : FAILURE in 20m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/189e2d108a4c4881bee950c98733e4a3 : SUCCESS in 2h 01m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/31f17c24febd40c5befbcf3b08c0ad25 : SUCCESS in 1h 20m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08af43c75a9d40219e825bd63f02c495 : SUCCESS in 1h 17m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b302877f8a8c4280af878546246d65f5 : SUCCESS in 1h 07m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7117c005193640bcb85c1eb06320aa59 : SUCCESS in 1h 36m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79e54559823f4856a5da71eb5db48f3e : SUCCESS in 1h 21m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cfa2bec488524b35baa88cd850b18ab7 : SUCCESS in 1h 07m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8549d4812a5a45e281ce7248f7100c7e : FAILURE in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ddb4a34714e75dc9e4f1e213ce64f173317bc883","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-02-03 17:49:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"20ed625045c43a8a04ef8e16e6692d4c275d5735","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-06 17:57:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96af9a3f29e076d12e1e8bba34fc4642570fe465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 20:03:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fb23253441c4724a3bfe57494b06af6 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38674be2364347e19e6ea01bb1024740 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d307c05c7bfb44e183703ffafc92d223 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48d321f33db1454a9a63cdf56b3dae11 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8506d7432b1842898cb51309031ab231 : SUCCESS in 13m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bc878ac46924a0c9c4e1006344f0e20 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404c8b21fb1b4a86b1a0aebd74b0c948 : SUCCESS in 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/decc76da8ce94229a2ac17ba79f2149e : SUCCESS in 7m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0121bb07704e4d04bc317f5a724dc8df : SUCCESS in 7m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fafd5ebadf0d48e181c72b3ce663b625 : SUCCESS in 8m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ae45261f4b84404a2907a4f742c66ce : SUCCESS in 12m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f249e73c36684a2f826e456fd83a6bff : SUCCESS in 9m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9e94f6d050347c1a8ac5f1005ea588f : SUCCESS in 50m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bc015b7117df4f6bb966969de396f9e3 : SUCCESS in 37m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/431cf0183b5c4c0cab53516710f3f0dc : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d5124e5235ec45729e40ba65550eb842 : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/34db55bfd5194d538f077410868179dc : FAILURE in 19m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42ed793dc81c49929c15e16ee9722b5c : FAILURE in 1h 13m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c22a69a0d41743ba924b5f6b86c77585 : FAILURE in 1h 06m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2588f6739074a178fc15d5675d2cbba : SUCCESS in 1h 21m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b36393787b2d4e999dc86cfc90295e7c : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/01b73924a7484b2aaba51bbc8557807b : SUCCESS in 42m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40b1d7c2c66b4b82ac2dd7be4d408114 : SUCCESS in 1h 01m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7ea1386d665a46fd8b7592d96abd15b9 : SUCCESS in 1h 14m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9384037e2f8f45c8b28a76beee99c000 : FAILURE in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6b7c9a02a2c95362d8006bc5d93a705974d52c9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-07-24 12:14:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"37f12d55f9f0e2aeb243f617c9d586b117d696aa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-27 09:11:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e3b5d19f01d04496f1f987b63f16bf9b29755472","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-08-11 12:34:19.000000000","message":"Patch Set 2:\n\nShouldn\u0027t this getting merged as well since it\u0027s already merged in stable/stein ?","accounts_in_message":[],"_revision_number":2},{"id":"5d9e00f6a7a1ef7b9d2ccab5070308db10289833","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-08-14 12:42:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d546989283f9599eb885efd33f02856461d7b8df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 12:42:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0dfa76ac930d5e8acf55185ff6556409a70d97fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:27:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"deceefa86827bddac6a0bd193f945df9307ff982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:27:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8ec9be1ebf41d0869733fa697b378d : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b827b69a0b441e9751395640817ebf : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/120b36345d28439faec5e026f5265175 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383778907bda4e70bd7934aa27b829b1 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/884b929dae3144098e5b6ac69cd2dce4 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a573629fbc43494bb8d74e610499846c : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed1b0eda629e426b94a0c919ffe48aef : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7ceffe7fa4484ec0aa31abb4f9ea8d65 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a3ba56c3b7e482cb1a814e7ceb9f78e : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8aacb88faa3147fcb41541639575ec4e : SUCCESS in 42m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c42176f5d9874417b25bc18074d89fcd : SUCCESS in 48m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e66fa66863ca441b90b40f547dbe0271 : SUCCESS in 1h 00m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f23ff515ee434af2b0858d26ae9c9bc7 : SUCCESS in 1h 05m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ccbc50d695c841c3801367085aab8842 : SUCCESS in 1h 37m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9e287d30dc64a4c0bc783eeff46352ea107d4d1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 14:29:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b87bf4d66594cd09244c05e792fe17b : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36c96e2523a247b7aa886d8cab86afca : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c760678ccc33fbf26b52c6c3763778b60ed02176","revisions":{"2fc5b4279f96ec021d819538429c1523374996a1":{"kind":"REWORK","_number":1,"created":"2020-01-28 18:14:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/704650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/704650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/704650/1"}}},"commit":{"parents":[{"commit":"f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2","subject":"Merge \"Revert \"Fix gate issue with DIB\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-01-28 18:14:32.000000000","tz":0},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fc5b4279f96ec021d819538429c1523374996a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fc5b4279f96ec021d819538429c1523374996a1"}]},"branch":"refs/heads/stable/train"},"c760678ccc33fbf26b52c6c3763778b60ed02176":{"kind":"REWORK","_number":2,"created":"2020-01-28 20:03:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/704650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/704650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/704650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/704650/2"}}},"commit":{"parents":[{"commit":"f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2","subject":"Merge \"Revert \"Fix gate issue with DIB\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-28 20:01:46.000000000","tz":-300},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c760678ccc33fbf26b52c6c3763778b60ed02176"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c760678ccc33fbf26b52c6c3763778b60ed02176"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
