)]}'
{"id":"openstack%2Foctavia~704653","triplet_id":"openstack%2Foctavia~stable%2Frocky~I335287da66524d026f0c42086d885b478c568bbd","project":"openstack/octavia","branch":"stable/rocky","topic":"fix_route_file_rh-stable/rocky","hashtags":[],"change_id":"I335287da66524d026f0c42086d885b478c568bbd","subject":"Fix the interface filenames for Red Hat amphora images","status":"ABANDONED","created":"2020-01-28 18:20:46.000000000","updated":"2020-07-24 11:41:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd11e34dbb69ed0a0bfb59d04f942010e47b0556","_number":704653,"virtual_id_number":704653,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"date":"2020-03-25 16:18:54.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-19 00:00:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"date":"2020-02-28 10:00:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":-1,"date":"2020-02-28 04:56:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 19:15:20.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-18 08:18:45.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-02-19 00:00:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-28 04:56:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-28 10:00:11.000000000","updated_by":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"reviewer":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"state":"REVIEWER"}],"messages":[{"id":"e74d56f31fdb9e7b6481989e15c938419c263d64","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 18:20:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f170c8005a7c28b326a92dcf2cb5d5e344b7bbeb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-28 18:22:42.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport.","accounts_in_message":[],"_revision_number":1},{"id":"d588a0d5ed35fecd83fe91e9a3ac987727c2abf2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-01-28 19:15:20.000000000","message":"Patch Set 1: -Code-Review\n\nVoted too early.\n\nAttributeError: \u0027IPv4Network\u0027 object has no attribute \u0027replace\u0027","accounts_in_message":[],"_revision_number":1},{"id":"9f06bf19f51c428bd91bac15576fcd4c1d9523a0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 19:18:54.000000000","message":"Patch Set 1:\n\nSigh, that picked cleanly, let me look","accounts_in_message":[],"_revision_number":1},{"id":"f5d2041a87e8be9c70d8d7a0c39c4304dec7c347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 20:02:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83720388043948f7a35c19d0457e9606 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2aea3b774f4dd5bbd2d0de7a74f278 : SUCCESS in 8m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/446e7c8aad1a40c684e226f8ecf447e8 : FAILURE in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6bc1adda6e8b4c7bb003ae501c0467e1 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9721f9bdaa1949de88ee0864f413da17 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/079b9f25a2a644f0bd77feab2eb272f5 : SUCCESS in 6m 21s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5bc96edbb1114ad1bf263314e2969a67 : SUCCESS in 1h 18m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/60f4985f5a624aa9bec6288abbebec05 : SUCCESS in 1h 20m 18s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5db9e1a9a548416da6d71a4cb8c10278 : SUCCESS in 1h 21m 31s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f7ed0c8620da4513823a9392d36b2406 : SUCCESS in 1h 31m 23s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/900bbc67ee304fb1ab5da1333c9187dd : SUCCESS in 9m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8c355440af7842bc8fad6764f472d271 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f24c997fcd1e49cfafb2c5ac637d2f24 : SUCCESS in 9m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ebef98fb8b44522a2dbb03350dcbfcd : SUCCESS in 1h 13m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3cd1b8009205406ca1a46c8ce9f032f8 : SUCCESS in 1h 05m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a3488f56ab1142979bf1d79b32c69bbb : FAILURE in 1h 28m 37s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/29e4bccf46474b6a9554b1d47fc7f921 : FAILURE in 1h 31m 54s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/d0872dc5a29542758fcb5a7e6fe71c87 : SUCCESS in 1h 09m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb49341f01d94ab8a1cf9f7b46a5c167 : SUCCESS in 50m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/161fb47d81e04a73b104485886e156cf : SUCCESS in 49m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c56001c04b9743eb9b54f608ecc5d1e1 : SUCCESS in 1h 28m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/88532a10a3804952aa7e694225f2ddd3 : SUCCESS in 1h 28m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/327fc3d41fed4489bb3d4fb0639e18c2 : FAILURE in 59m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa6536e009b649aab7cb4dc1ec866728 : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fb009a51ce3277d49843bda972651722c653f500","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 21:13:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c4112f389f8b31872e1a4f028ec3982156d63ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:20:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6e03777efe646f1877292a4aa03b42a : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dafa75fcbc1444da5759ba7e8a0b25d : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/987a4edbd6e549d299770d7896ac85ce : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9be81a43b0b948fa969f09af62202737 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b20be83c4a45c79a8751fc49dd74a6 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb49373ae56644ebb1856a9a33778cf3 : SUCCESS in 7m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/301105b05d194c4cba7b989385765978 : SUCCESS in 1h 23m 43s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/2d38120e15db43759b962bf625bbc5b1 : SUCCESS in 1h 23m 49s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfacc595c2c944ba9976f5a785c9c86a : SUCCESS in 1h 28m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e6afa3e366444b4486b0c031b824c5d6 : SUCCESS in 1h 42m 21s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e302f6da8404d59b563cb438147456b : SUCCESS in 9m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f70b9ea3731c4501b9d3363c1551575c : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6efded1f73bb4ef2bb824880d669db9c : SUCCESS in 7m 50s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b0960d323ff4159b421a4390dd346e7 : SUCCESS in 1h 02m 36s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8fb9f1ce2af24549882d6d1607f2898f : SUCCESS in 59m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/909c5c33538b43cd80e8c3febabfaf80 : FAILURE in 1h 25m 00s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/dfd406517bfa4dce83ed8b1adb58b1a0 : FAILURE in 1h 26m 20s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8f0c1c67eba2487482ead227b0c669dd : SUCCESS in 1h 11m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1caf0526e42947f5b72316e616a102ff : SUCCESS in 49m 51s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ac540010032d495898766a554cf9f8ca : SUCCESS in 49m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c346ffe3bcad49e9a3f71088d8b29c2b : SUCCESS in 2h 00m 56s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7ea4c5dcead5412d86cf016b54a69a65 : SUCCESS in 1h 46m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/84eff9fa421946ca8bf05123ce03acb9 : FAILURE in 57m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1826be157c849d4b27affcb462eca9f : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c53c9498fff035cb61270384f12064d135f5115e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-29 15:10:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"00fe0493f953f52522e8cf9c0b20bdc24b02c08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:36:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f52653297d440afba14005cb68922cd : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d1225c9741476a89e259df5cd6bff7 : SUCCESS in 14m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92daaaf90ec940898269109d34051cad : SUCCESS in 7m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ca9bcb4bb174e3bbef2421a1c658d3a : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2193698904e545cda95fcd29c5ed22f1 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f2ddf4353e4e8c94cc95623d9cd5cf : SUCCESS in 16m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b6e4d6b67bb74b3892327241d00da8b0 : SUCCESS in 1h 20m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/697135986a3545889f394438653fe966 : SUCCESS in 1h 25m 39s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22051b44596740018a05751d8f9b3df2 : SUCCESS in 1h 35m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/461fb3f1482f4aaa8c1a03d7bc914ead : SUCCESS in 1h 26m 06s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/14b0d2643b28432d95c9d9bfb93c0fed : SUCCESS in 7m 51s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4cbf6fc358b9452e906c060df7452d05 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28b8b29fb4346019f2641b7e56ec002 : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/562094afd0484748a4fbae1628d08d9c : SUCCESS in 1h 08m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1342f0d3a9be400aa53c3b8e37fbf83b : SUCCESS in 59m 37s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/47838f9538e94eddb725c1452dbd7f69 : FAILURE in 1h 25m 58s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/614d9342724740c380360325e8a12115 : FAILURE in 1h 25m 10s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/73a57956143b4166b283cafe18615f1a : SUCCESS in 1h 20m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1112cfa708840939b86ee193227f6c3 : SUCCESS in 55m 41s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/39b7cccc03de4b109228824c7d9632a2 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a04578b58ca4d61996e18e5c13beef3 : SUCCESS in 1h 33m 26s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/6d16d1e87f2146ac96026f1f8c60a2ef : SUCCESS in 1h 40m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7db19a9b3cf4f89a0c6b034edab4410 : FAILURE in 1h 03m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd7a2ffa16e04bd5941a0318388e502b : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5fcb258c89248edfa2e834f512ab5a52e31440ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-10 17:30:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c8aa2af88e600b7045cf46230d4bba7e1f947b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 19:24:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f866324cb6d74ae4924bcd98aef295bd : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac761744793a4b80a401f207784d564e : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/921e4cab95bc49e8be7025c3b4974351 : SUCCESS in 10m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f26fb59de95b477095ebd7e7db2bb0b4 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c6dc9c77e8418ea94c6a390840b1f4 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8257502ed5634ce0a639af7ce9005732 : SUCCESS in 10m 08s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0923bc73768149a08005cbb17815bee3 : SUCCESS in 1h 20m 53s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8c71ace9e8704011837ae65bc1e412d3 : SUCCESS in 1h 22m 36s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b14ca7356267487698a3cc35213ea663 : FAILURE in 36m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/237080909d624b20b8bb47cb49328b22 : FAILURE in 41m 12s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a342c5c741e349888867e232a5c50ab8 : SUCCESS in 7m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fbedae1338cf4616b4916f412496d195 : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee9959befd144d48836c4acbeb388b15 : SUCCESS in 8m 01s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/720d4dd37f944d77ac95ef540824f09e : FAILURE in 35m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5bc83bb8f374456b901b59fb5d64645f : FAILURE in 40m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/94d7cd919ae943a59f7c4891e9c6b081 : FAILURE in 38m 54s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/88714b4393ea4ccb9a1d59f4b9143f05 : FAILURE in 43m 56s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8406d4ab261545daa34ea3b06da58881 : FAILURE in 37m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fed6f59d02f24984b2b60c1001d9f0e9 : SUCCESS in 50m 01s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bc4be442817549ca99e9bf2ea9742f35 : SUCCESS in 54m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d603f5cab174ab5aac6828fbad0afce : FAILURE in 41m 40s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/d249ae0fec89423495df2ec71b3846b6 : FAILURE in 38m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97a30a983e064cb4a07137314331eb6e : FAILURE in 35m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79b768fee19b4d62a7e7bf001423d94d : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"22740a65c0c6e97fd0c6f38e00edcc4f7baea189","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-02-18 08:18:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"693ad0d3380e257f0aa3c74cf0612bf295a88bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 09:49:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46f6ac064c914bdc8a66fe54cb25de53 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d98fe3346fa4dc99a976dfe303b0577 : FAILURE in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1a7d36e887e410890da650494861114 : SUCCESS in 7m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5345e41c1d14d63a378d3124a9240c3 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac899199dcf49a290c5cb35bbc3de31 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63fe2e949d4949baab7182c6a3fc1e85 : SUCCESS in 7m 58s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/74437075d6f84f8083b30872f181dd40 : SUCCESS in 1h 13m 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/18014c1bd35c47f38e7a2321d77a4c0f : SUCCESS in 1h 16m 50s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0030920b39c54ce4bb90571cf4b01eea : SUCCESS in 1h 15m 45s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7599b13edac14f608c03ce7f9dcea119 : SUCCESS in 1h 22m 15s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3a174415ee814355a071bf2e6ee34b46 : SUCCESS in 7m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/db4244d69fd14439aa67df3d704da959 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c6eef65acee43a3a12be62bfa7e3117 : SUCCESS in 7m 33s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f262489d8484029a95c999414a626aa : SUCCESS in 58m 45s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/72986391ebe24a5f80cb8ad38e27942e : SUCCESS in 59m 40s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ec960375a0a045089512b2047ecb8e2a : FAILURE in 1h 21m 14s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4b83fc1768d74eeab69b47df1cf640fe : FAILURE in 1h 27m 49s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e58127afb8e44043ac6e31600efc9cf1 : SUCCESS in 1h 12m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/799cfa0f23384a0bbc611d9d38d2a4cf : SUCCESS in 45m 52s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bcdd6ff37c2340c793521253a734db09 : SUCCESS in 43m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8ac3c57bafb44429ab5cbed53466763 : SUCCESS in 1h 23m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1bdfaed3f8244a30b88ecb7865e17081 : SUCCESS in 1h 19m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9bade941650438786bcbb5b1fa1e7a1 : SUCCESS in 1h 07m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c2c6238ca14726a9d7cb0c18fbd26d : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5bad457995fe9f561478876f6be65348d0c270de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-18 22:11:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2ee0ff72278abb070a3aef6a9d5d2d2ff126fe9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 00:00:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57e528f9ee294bd8b1f8338809ae2c19 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81589b443fb947aa8114a0efda1e5ecb : SUCCESS in 9m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d70731ae2ee41d6bf9b461d12c3dbf2 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e43153085409403694753c663249c3fe : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78484ee6476348c481dab692bd9008d7 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce2b3cfadb24bde9687563b6ae973ef : SUCCESS in 8m 03s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2903660ddcee4aad8dde18d52e73c219 : SUCCESS in 1h 16m 54s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cc6bedae0aaf4a9da8710d2f88db6c30 : SUCCESS in 1h 30m 52s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed3ffa4be40543babe37618efd403d55 : FAILURE in 45m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c03b67025832446f9522ccb9f84edf11 : SUCCESS in 1h 20m 26s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/86dd2349f98b41d2a31fe7ce13863632 : SUCCESS in 8m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/730ff6f6524e4861955289d6c4851f49 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e5b30d7ce45460ea906c75f10d42d4f : SUCCESS in 7m 40s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d82d3f338bf74b51ab7693128783c22d : SUCCESS in 53m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/58e1a257d93643fbb2ebcf573b6e5848 : SUCCESS in 1h 05m 23s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/445fa84c74cb41bdb13ce712bb4cfa26 : FAILURE in 1h 26m 23s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f8befa4b22214977a29d0bb7caa167da : FAILURE in 1h 31m 38s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/c0bf74586b244c45acdfd56549cdd882 : SUCCESS in 1h 13m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d8d0d949ff1e4a5f9146e1645c97ec11 : SUCCESS in 47m 31s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/90768e7596c5422ca8de4170f1c2387f : SUCCESS in 47m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25f8cf0e30184f14b9c0e7f148ed4eb3 : SUCCESS in 1h 39m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b697d00ebd894e7e91899f75b5c54667 : SUCCESS in 1h 39m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58138a4f72f4445a868b41e11c939077 : SUCCESS in 1h 13m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/281cc91faf47419eb380565273191714 : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b963ef4a13df5c44eece43c0599fd5dcce32c7d8","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-02-26 12:08:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1cbabc5ab0fda97c7ed6f73beef9ead20a9785ee","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 04:56:13.000000000","message":"Patch Set 3: Code-Review+1\n\nSo, I\u0027m not going to minus-anything this, nor am I going to +2 it yet -- my vote is that we wait on backporting this until someone actually raises an issue, as I am not convinced it really solves anything, and proactively backporting it seems just as likely to introduce new issues as resolve them.","accounts_in_message":[],"_revision_number":3},{"id":"070a66605fb5c6af086b0ae7d5dee2042127446c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 04:56:40.000000000","message":"Patch Set 3: Code-Review-1\n\nEh, I guess that opinion actually does merit a -1.","accounts_in_message":[],"_revision_number":3},{"id":"3dfcfb4d461b9ba932910d054b21fe0f2a1defbc","author":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"date":"2020-02-28 10:00:11.000000000","message":"Patch Set 3: -Code-Review\n\n@Adam, we\u0027ve experienced these issues on our Rocky cloud. But we\u0027re upgrading soon so backporting to Stein is also fine with us.","accounts_in_message":[],"_revision_number":3},{"id":"69e1cc24d3a16ae5c9507a972f4da624d7c935f4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-28 14:21:16.000000000","message":"Patch Set 3:\n\nAdam - yes, this was actually seen on a Rocky-based cloud by someone (cough, red hat, cough) :)","accounts_in_message":[],"_revision_number":3},{"id":"8e9a4e638aaa4280c94385975a374a0ac6ab785f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-25 16:18:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2e0ee7fde4bee8a1be1ec2e2f3b731ab962879b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 18:18:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7313da956194d5083cb0392456188b3 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b3b25ac9474f799361dcf71aba66e0 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/391e94f6ab5848eda7e94bb4c9cb2773 : SUCCESS in 6m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d50ad918aa694c60bf1c1a20a3abbc8e : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a249ed2d924a48be66dcf557acf566 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d423035a5ebc4506aeff503b9ae8d2e3 : SUCCESS in 6m 32s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/584a932ca2eb438da2b37d08d03e3475 : SUCCESS in 1h 16m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1a21aa427d5b44cb960f41b542220a72 : SUCCESS in 1h 16m 05s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4374b91419914f7cb1b51b9f643f0c3f : SUCCESS in 1h 23m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e2b5140b79854d4785604a1a5be8a3ad : SUCCESS in 1h 49m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/06dba8816d814d0bb6a95a1000e88546 : SUCCESS in 9m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7c33060fab6d412bb9fa9791b97ad43d : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e61bf048c5554c168d04d78419fd7237 : SUCCESS in 6m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/095edf8c4ec94d299f2b54fe61626b5f : SUCCESS in 57m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c98a41a1fdb0450cbb6986a0a0bd7fa3 : SUCCESS in 1h 04m 39s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7ff57f278403490d80b3ac81008e3606 : FAILURE in 1h 07m 00s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/6472cfe898494e7080b33f9564253ba6 : FAILURE in 1h 32m 09s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/781553fae98345ec8e06dcaaf2884c27 : FAILURE in 39m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38e6c76c319d436c862bcc26ceee9142 : SUCCESS in 47m 02s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d688e33700fe4285b07f9c23cf68f582 : SUCCESS in 47m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63d89e60964d40a9ae40aa3d70c02db7 : SUCCESS in 1h 50m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/3f6ba44e38df4051875f772efdb3a0a5 : SUCCESS in 1h 35m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d4a088174644584b9a895354d252966 : SUCCESS in 1h 06m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f157c87a84184e95b807ef5726c5ce73 : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":3},{"id":"67be0d6696144f22d5215acd2df123a3cf41798f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-24 11:41:50.000000000","message":"Abandoned\n\nRocky Octavia reached End of Life (EOL) state. This branch no longer accepts changes.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"61b95a330d25932ed87f9658e983d2b2bdd65586","revisions":{"32eb3a9ec105ff5ef38a26e898a59cdc7450e6df":{"kind":"REWORK","_number":1,"created":"2020-01-28 18:20:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/704653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/704653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/704653/1"}}},"commit":{"parents":[{"commit":"34545524b6cc194c788172aaf3e565513f2e4d74","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34545524b6cc194c788172aaf3e565513f2e4d74"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-01-28 18:20:46.000000000","tz":0},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32eb3a9ec105ff5ef38a26e898a59cdc7450e6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32eb3a9ec105ff5ef38a26e898a59cdc7450e6df"}]},"branch":"refs/heads/stable/rocky"},"f5f185bddf27b2cbe4351e436226f48c1691901b":{"kind":"REWORK","_number":2,"created":"2020-01-28 21:13:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/704653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/704653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/704653/2"}}},"commit":{"parents":[{"commit":"34545524b6cc194c788172aaf3e565513f2e4d74","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34545524b6cc194c788172aaf3e565513f2e4d74"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-28 21:07:48.000000000","tz":-300},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5f185bddf27b2cbe4351e436226f48c1691901b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5f185bddf27b2cbe4351e436226f48c1691901b"}]},"branch":"refs/heads/stable/rocky"},"61b95a330d25932ed87f9658e983d2b2bdd65586":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-18 22:11:02.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/704653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/704653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/704653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/704653/3"}}},"commit":{"parents":[{"commit":"796363638d1c38caa67ffec19bdc121ec0b0ec1b","subject":"Fix pep8 failures on stable/rocky branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/796363638d1c38caa67ffec19bdc121ec0b0ec1b"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-02-18 22:11:02.000000000","tz":0},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61b95a330d25932ed87f9658e983d2b2bdd65586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61b95a330d25932ed87f9658e983d2b2bdd65586"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
