)]}'
{"id":"openstack%2Foctavia~704664","triplet_id":"openstack%2Foctavia~stable%2Fqueens~I335287da66524d026f0c42086d885b478c568bbd","project":"openstack/octavia","branch":"stable/queens","topic":"task37881","hashtags":[],"change_id":"I335287da66524d026f0c42086d885b478c568bbd","subject":"Fix the interface filenames for Red Hat amphora images","status":"ABANDONED","created":"2020-01-28 19:03:48.000000000","updated":"2020-04-10 20:00:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8774c21dbc016cb7910ed10a2edf081ed9c3df41","_number":704664,"virtual_id_number":704664,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-27 03:29:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 19:03:48.000000000","updated_by":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"reviewer":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"state":"REVIEWER"},{"updated":"2020-03-27 03:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bb98d4294628db8ff199aad63eb0fe23360fc6e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 19:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31cf6497cf540a6ac8ceeedf6a517e53847f413a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:08:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfb23695b644125bb88a17679f4047c : SUCCESS in 9m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc177c9edb4248369ee0f63f5cbe3dc2 : FAILURE in 7m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/717db50105c54589bba99dcaf1f75f1e : FAILURE in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bf67e8afb14a58b8bb88e3bf6ad630 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5793c1109248d08bd5aab733f7d2d2 : SUCCESS in 6m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/714ad36c5d834b32a7faa0859cf0a8b8 : SUCCESS in 1h 20m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a97be369166745fbb21f5809988c8c9a : SUCCESS in 1h 15m 06s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/762b92d225964a1b858f8d1d38909cc0 : FAILURE in 1h 39m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/aded0ed6b74d4df58cc08cdde3eca9b1 : SUCCESS in 1h 56m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/363ca1390a744c319562de8b32de0fd7 : FAILURE in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/74f2d47e3a084e97b155b659065ca3ee : FAILURE in 8m 39s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/bc0e77a0b6d741d592555932a9cdfd8f : FAILURE in 6m 27s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aa9398fce5fc4a92afd082b10ffae280 : SUCCESS in 53m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/318cc52a3e3541e2a5f8719706ac88ed : SUCCESS in 1h 04m 25s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5cac0c1d14cf4ccf8098a3c4624e09cc : SUCCESS in 1h 07m 11s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/45632121510b4307babbcb1c32f87c9d : SUCCESS in 1h 06m 52s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/5dfac6093a324bbf8fb92f59d05d3293 : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5599acc6745d4971810c903afaaea633 : SUCCESS in 44m 05s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/246e805169594fd5878ac7234ed2716f : SUCCESS in 50m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cafb4690d2fc4219a9aa493ed9b9d744 : SUCCESS in 1h 36m 25s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2aa1a03cf0754290b819cff9e2b6750e : SUCCESS in 1h 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ca6363835f58a939262c3a1552f025a83e656c7c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 21:12:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3af317c3ddc3d8ece59f38fe713851a887a9239c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:26:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225bf8f4ebc640a69c80ac248b65e16f : SUCCESS in 9m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f01916c8db8845d8bc8ca426f2fd5bc5 : FAILURE in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9321754269af42e2975aeb061304dffe : FAILURE in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928ecc1bfd7744df916a3cb0593627ac : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8261a869cd984867bffb36218d8bf80d : SUCCESS in 10m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/51e94f0ca5e143abb7d4137958dde06e : SUCCESS in 1h 19m 10s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7827f0e3665c45dbb1cb2f3d79d33b27 : SUCCESS in 1h 16m 37s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaadd46b63c54fc686bbaf07f552af35 : SUCCESS in 1h 14m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8afa8bd8032d42e2921f9a6dd177715f : SUCCESS in 1h 20m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65fdabc8114641d087a4bd3b9c201289 : FAILURE in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/57a88538fce146d7ab898df3ef04883e : FAILURE in 7m 19s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3388c50435224d6b8de138bb065eb16d : FAILURE in 7m 38s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d84cb508ba942179b7497463b7ce1e1 : FAILURE in 55m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ec463319c1e44dc799fa2a30ce3a84ce : SUCCESS in 57m 27s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ce3cfb8964cf45b3b2e820f1cc40ce64 : SUCCESS in 1h 04m 42s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c2fbf310b77344d69544fc2d22414af3 : SUCCESS in 1h 08m 01s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b4377b4dfa7b43d8819006ccb271f30f : SUCCESS in 1h 04m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a38d7b2f47c44cb9a9e845a0a81ec474 : SUCCESS in 53m 00s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/244c6ff95bda45698b59f2f6e4803e63 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/93deeb2b2f764ad6b49b6ec317387535 : SUCCESS in 1h 48m 00s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/2185766a6d9e410f83418129ab2ff74a : FAILURE in 1h 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5830ff483f223d1fe5ccb4e7fde864df2f5c5f88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-18 22:11:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c88827dff4dd0044986a44bbd5f65cc6b56eb705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 23:58:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab2d10174f74c659565b6297ff02db5 : SUCCESS in 10m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eadc153fa66149b5a39396f08626fc1f : FAILURE in 8m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4b5559d6bc0438eaede3ab944083d05 : FAILURE in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c87685a43042dd9377e65d2fd02f79 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b2cdbee235e4d32a3c0dec165eb0f73 : SUCCESS in 8m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b8b2ee20da184459bef8208346719c15 : SUCCESS in 1h 17m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/051c882a2c754407bf94603403a33dda : SUCCESS in 1h 16m 58s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/724c96a43e9f4618898b0d3b1e2df6a7 : SUCCESS in 1h 16m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/37ca769d43ee49bcae648964b4a2062d : SUCCESS in 1h 14m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a463e42575d4a149f0955944ede285e : FAILURE in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e3db61895e14ac298e89d39974b9613 : FAILURE in 8m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/70ea3cc6c4c64961a56f8890e3e49026 : FAILURE in 7m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eebd434413b24319a02640dc0094bd03 : SUCCESS in 1h 08m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/02967947d09f4467bab681bdc0cd8600 : SUCCESS in 1h 07m 56s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e280023d827a4b20941f6ca51177aee8 : SUCCESS in 1h 23m 12s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e882c38686f543c09fe101c9303ede1a : SUCCESS in 1h 23m 38s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/bff8f10f21d34c2a9dc8cece523651de : SUCCESS in 1h 13m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5da4cc6c9cf74731849a3d50648ca543 : SUCCESS in 51m 05s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/8c84e628a71f4af49a9b37dee7c4a4b3 : SUCCESS in 43m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cac97552cef746289ece9f8cfe05ab85 : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f2a3bceab182443f985436487e886249 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e0efe1ed6dbc5131bf0e1d253670d07627c9e8e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 01:38:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"65700fca6b6d8b5946e5a14bad38a838a9dbbc62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 03:29:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d46798552d4e48a218cea23fcc7c54 : SUCCESS in 9m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b9cd5e130e14e9d83b5d071903dad0a : FAILURE in 6m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/33c2d87adfaf49e2b0379fc807722155 : FAILURE in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fa5c31be094ae4ae390b89d26ed9c4 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec03dc2aa68b4efeaa6b9a95a9d45d0e : SUCCESS in 6m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/36481f3ac0ef4df3b38bbc193374045a : SUCCESS in 1h 31m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/eefb644276e14b7a90dc4c1591c18810 : SUCCESS in 1h 16m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50df43d196bd446e9388516ddecc1218 : FAILURE in 5m 22s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/eac8fff1d30a41aa91b6558fd5f236fb : SUCCESS in 1h 05m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/082416de0cad47d6b7162d921bda6192 : FAILURE in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3914e26215034a959d65790013627cf3 : FAILURE in 7m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0956d38ee330400b96bc36646ebb7a54 : FAILURE in 6m 40s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/844e872148634fdfa4c4877885bd37f5 : FAILURE in 6m 51s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/405d2e6ad0d44cb79b4b96fd6e57ef30 : SUCCESS in 56m 48s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/56c39410a67744c09ebeed82cbd9d578 : SUCCESS in 1h 09m 57s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/44888f150c1446b3ad6aedd760f2966b : SUCCESS in 1h 29m 45s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/901cd106d8c54273b1f2c535203489ef : FAILURE in 28m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e852dae79b7b410b9a5694ad20f832b5 : POST_FAILURE in 52m 27s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/c61ef61f28ea4b948798d1c0697067f5 : POST_FAILURE in 48m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbb668d9ec2e4dfb82768c6ccb33ba32 : POST_FAILURE in 1h 26m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/1094f6cfaa9249b2b3e40734f692d15b : FAILURE in 17m 19s","accounts_in_message":[],"_revision_number":4},{"id":"fae67762d9ccf6ac81f0b14470b8e813a3a668c5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-10 20:00:01.000000000","message":"Abandoned\n\nAbandoning in preparation for Queens EOL.\nThis was never going to merge in this branch, as it is of questionable merit even on newer branches, and the risk would not be worth it for such an old stable branch.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5bebe342de5d7bfafd2b556075881d0bebc1fae3","revisions":{"be45e7b1f0bc99e105fc915142682c253484da8d":{"kind":"REWORK","_number":1,"created":"2020-01-28 19:03:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/64/704664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/704664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/704664/1"}}},"commit":{"parents":[{"commit":"ea4ae88378e68b9b4a2c1de78c028112787aad6c","subject":"Cap hacking version to \u003c2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea4ae88378e68b9b4a2c1de78c028112787aad6c"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-28 19:03:10.000000000","tz":-300},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n    octavia/amphorae/backends/agent/api_server/osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be45e7b1f0bc99e105fc915142682c253484da8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be45e7b1f0bc99e105fc915142682c253484da8d"}]},"branch":"refs/heads/stable/queens"},"39cb6000afe3e4f02f7d161c544459e17d906117":{"kind":"REWORK","_number":2,"created":"2020-01-28 21:12:45.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/64/704664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/704664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/704664/2"}}},"commit":{"parents":[{"commit":"ea4ae88378e68b9b4a2c1de78c028112787aad6c","subject":"Cap hacking version to \u003c2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea4ae88378e68b9b4a2c1de78c028112787aad6c"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-28 21:11:15.000000000","tz":-300},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/amphorae/backends/agent/api_server/osutils.py\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39cb6000afe3e4f02f7d161c544459e17d906117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39cb6000afe3e4f02f7d161c544459e17d906117"}]},"branch":"refs/heads/stable/queens"},"37c37c8261e010c638826a90c25ada111e7f074a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-18 22:11:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/64/704664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/704664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/704664/3"}}},"commit":{"parents":[{"commit":"354561580b122fece43caa232485796176e13470","subject":"Use stable upper-constraints.txt in Amphora builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/354561580b122fece43caa232485796176e13470"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-02-18 22:11:28.000000000","tz":0},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/amphorae/backends/agent/api_server/osutils.py\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37c37c8261e010c638826a90c25ada111e7f074a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37c37c8261e010c638826a90c25ada111e7f074a"}]},"branch":"refs/heads/stable/queens"},"5bebe342de5d7bfafd2b556075881d0bebc1fae3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-27 01:38:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/64/704664/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/704664/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/704664/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/704664/4"}}},"commit":{"parents":[{"commit":"5528268e2f360c2eb921d665313a326387c2e1b7","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5528268e2f360c2eb921d665313a326387c2e1b7"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-12-24 13:16:32.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-03-27 01:38:42.000000000","tz":0},"subject":"Fix the interface filenames for Red Hat amphora images","message":"Fix the interface filenames for Red Hat amphora images\n\nCode was not using the correct filenames for the \u0027route\u0027,\n\u0027route6\u0027, \u0027rule\u0027 and \u0027rule6\u0027 files on Red Hat images.\nChanged to use config option \u0027agent_server_network_file\u0027\nif it\u0027s specified, else the file of the correct name, and\nadded unit tests for each.\n\nConflicts:\n  octavia/amphorae/backends/agent/api_server/osutils.py\n  octavia/tests/unit/amphorae/backends/agent/api_server/test_osutils.py\n\nChange-Id: I335287da66524d026f0c42086d885b478c568bbd\nTask: 37881\nStory: 2007051\n(cherry picked from commit 93cd9fc075b3d77490090b89e449f18938710b26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bebe342de5d7bfafd2b556075881d0bebc1fae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5bebe342de5d7bfafd2b556075881d0bebc1fae3"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
