)]}'
{"id":"openstack%2Foctavia~705219","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","subject":"Fix house keeping graceful shutdown","status":"MERGED","created":"2020-01-31 14:00:15.000000000","updated":"2020-02-20 18:39:31.000000000","submitted":"2020-02-20 18:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705219-1582223878490-25e6796c","meta_rev_id":"c8683aa861cf69ceab7dbbce07138724f6a50c2b","_number":705219,"virtual_id_number":705219,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2020-02-20 14:47:32.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-20 18:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-02-19 11:16:37.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-02-19 13:23:53.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-02-19 13:23:53.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 11:16:37.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-19 13:23:53.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-02-20 18:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"653cd00f5dcddbae7a8f70ea854878f7dfc54589","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 14:00:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"17cd8516c7dd4e23edc2cdb1eef6bbee54087ae1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 14:00:28.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c09d72998870a666d6769a3719a5b1309a5fb93c","accounts_in_message":[],"_revision_number":1},{"id":"efe5c02204a4891423907b4d66c3d72812151f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 16:03:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5437476a2bd4258beb693fe5083c3ea : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59889639c2f546e291ea6e7a9fe42953 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f57756fe55c463fb7229eacf87259cf : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7f8b9de30f448d9b623cd0ba8de982f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3259656805a542acacbe8e1f566e8528 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/deec0c878bb942e8a68d756e4d0ab449 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97fafa368e14f7b99003f22c5b2f491 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b25caeeb1ba451bb26f91cf1e4b9c29 : SUCCESS in 6m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e111b43a8ce945bf8fd14d33d5f18903 : SUCCESS in 9m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/99ffee53422e4853b3f8b074f9628179 : SUCCESS in 8m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/551e912e9f31489fbf8c1de33b39bed9 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88c193fdb16c4b60be57ef1d78fa47cb : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bcb6a60579e43dca4e603efa82dac16 : SUCCESS in 44m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/69d890a01d5c43289fe55b11196f8e0c : SUCCESS in 41m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e0f998be4be40369e57dfdccc23341e : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/896a35af11384c71aa73656b1541ba8b : SUCCESS in 1h 48m 37s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eeb83eb05b464abb9be64d3bf555ecd0 : FAILURE in 23m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e5abd22f99e4a06b51abd57d4d9fbbd : SUCCESS in 1h 56m 14s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/08d8b56492db462b968d6a6573b920af : SUCCESS in 1h 51m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e60f6949d68e446c9e1b7954797925bb : FAILURE in 37m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37ebcdf017c44d2e90a6fd69b6ee7a78 : SUCCESS in 1h 08m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8bd48d155fa44499f4fd6729125c75a : SUCCESS in 1h 45m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0527f6577a9f4efabfccfad89fadef43 : SUCCESS in 1h 19m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f223c6321cf345f1bf6b354ad01112b5 : SUCCESS in 1h 16m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0f30302edec84042a642da018fa9b12a : FAILURE in 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21168f7e7ed7ce178228369c8520aeb981e8203d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 17:45:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15a4b4b8d0396176f6d3066f02c0dc0ceb2c0b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f53f971fe39433b96c04b04789250b6 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59c9efc2cac4813ad700d2ec730bd24 : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8774a2724e4844f8b10a0f502eadc82d : SUCCESS in 9m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8365b3d41794461ab2b5f6e22802298 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c7b0b7ce28f461297a8f01a3bce826c : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d01fda7381e545af97b3a53dabc6c9b6 : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7d84a2eda5446fbf3119667211fe3b : SUCCESS in 19m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2517df76648f4b6092fc3e3fa6bb5d86 : SUCCESS in 13m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f19649f1fd54c14864da8fbae8c22f7 : SUCCESS in 11m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d3ebb78c39024b21a5f063314f3f5cc5 : SUCCESS in 13m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c5a747b4110f4247958e53445d22e66c : SUCCESS in 12m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/385e1e4b0068420f9d44efdeab63af16 : SUCCESS in 10m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2f7ca88a581a4be1993f379c1195c0e4 : SUCCESS in 51m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b3fbe38a0d1643c8bb4b6a8dceb4c6bc : SUCCESS in 45m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5932d6e21d734ab189f34f34faad5c26 : SUCCESS in 1h 34m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/760b33a2af6a42fcbeb03ae4925f442e : SUCCESS in 2h 01m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2b60548ba0714775a4972b3d9df13599 : FAILURE in 24m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af5bd61d933146d9a0743044bec3f646 : SUCCESS in 1h 46m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a62575e447046fbaa4859fb6666c3bf : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b4705dd1c32f4efdbc720cd28e94580e : SUCCESS in 1h 07m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/705688da3ef8435891b208a6556692b3 : SUCCESS in 56m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c0ce557acf647af81c262428f7a4634 : SUCCESS in 1h 32m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b515b22e12a462f95b83c97f00e5231 : SUCCESS in 59m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a45bcdac86b44afbab2bec41f503488b : SUCCESS in 1h 07m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/28be9c3725f6453cb11f48b938618e45 : FAILURE in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff15139c3c913ec71daa32d7e4f9dcef8fde116f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-19 11:16:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b63bc152516526c2d5a86aadd85bc829488b43b6","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-19 13:23:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7083001e7e4f91dabb80024386e08072c237e8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:24:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11694e284fbe77c3a9f5b908f32bc0ef58e70249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:33:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84cfdda2d9954dc7b376586f5386718f : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4226d08a3404d4fa4711f4c44ef09bb : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9a6907ab9ac4b4fa2bc9456019af247 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7971c4bc3e8341918abe84834bfcc49a : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72dca69471554ea4aeb6183abc3f5954 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b8bb339aa542129c29b889060ed90c : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/023e1b7a9b0a4103bb6bf40301d324a1 : SUCCESS in 9m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8e919820120048d193f1f547dcb3fe33 : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4716decffdc848919961f6eabcf69a5c : SUCCESS in 11m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7cef1dea1f5a49fbafa724345aaf843f : SUCCESS in 48m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2a67e343b0924953bf5106a369b78233 : SUCCESS in 46m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41e7a14b59354f87bf60a574644880d9 : CANCELED in 1h 14m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/281b206efec648b380affedd5fa4d080 : CANCELED in 1h 15m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6a90962a27c4d2f8495df2c7e561973 : FAILURE in 1h 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b0312766f399207c49f2b13aa111d886500c976c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-20 07:36:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"274bd29eadd92011f946c790664bdca0a4f98c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 09:37:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba191818c1ff4f98a16aeac639a74044 : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e30f1e59a2a44ed98d5bb4706cd4dfec : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5bd32a8bed4fccbfb2d04c88f9e073 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50618fc869ef4f378f07ea95e62887a7 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08cb97ce305467a84c637694b53f506 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9043777b74eb46c3b479c1584cfb2851 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92527480ce154b839f2ba5dec7d36826 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff3575ea944d42c7811cfbc5c631f868 : SUCCESS in 9m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/57f0f4d8fe01449ab3f6c2911c9c7e14 : SUCCESS in 7m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/da83f617b1ae499faf00777475f3d16b : SUCCESS in 10m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/28d7016d755c430ab599d05eed1c7bc9 : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d42be0a314e34526988788251ef92f2c : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff316bab29a7460784e842619c975335 : SUCCESS in 51m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9ff7bd07ea0249618ec37880a00f7b66 : SUCCESS in 46m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95fbdd9233f4493ebc7eda00eaaa2278 : SUCCESS in 1h 34m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a351969781e4402ea75fb0d7c00338e1 : SUCCESS in 1h 45m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8aa1ba213f8049a990a6247d056570fa : POST_FAILURE in 26m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7fafa9d1b194e08980d4ac16406a28d : SUCCESS in 1h 52m 35s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed34160194bf4ace8c09c27e25a98026 : SUCCESS in 1h 23m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6e56ed70bfe4d09a455b069ba60061a : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/954aae5cd4364049a488987ef8a884ec : SUCCESS in 1h 16m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0ca50e31dd34f5ab90fcde2cafa5b2f : SUCCESS in 1h 46m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec3463c1535a4fd18a911cdeb44d3ee0 : SUCCESS in 1h 02m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f17f5932c45a49339330c3601051c641 : SUCCESS in 1h 12m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f988b3c392ca4d1aa0800348d8bba1aa : FAILURE in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce747c0efe3ef95d0b42015b5b1c7cf0ad7a571d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 09:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58808f7ee34e86a44144cac21db65d03848a7d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 10:32:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f718525081341bdba37bf11247ea99a : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b59239393a48268f6c59ef8b722623 : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd3d7f74666f4c06aab6c461bdca722c : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5b121ca89a0401e9dfa49b91629e288 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e9ea71dca2c4874a3a91c436770449b : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ba483ca221474ba0a1f487a57724b1 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7b18533b14d4176ad751173e0e065a2 : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c6613d8e6d7d4ffb9168700230b20e56 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3b3f0f8d2cf4931af7a7add7775296a : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/643c0e1de92c47b5b3692934cca5aecf : SUCCESS in 49m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b26b7f4dfbac4debb1bd00128d9b9df9 : FAILURE in 49m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5ecee73b0944ce99e4bdcf1aaee5fba : CANCELED in 50m 01s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d44008c2c004486b868a757c195ff22e : CANCELED in 50m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2f9bf91d45e4d96be3df87b9469c846 : CANCELED in 49m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3aecbf38f7e0fd670469ca1f1c904ca0fab3be8f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-20 14:47:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"515ae300b06abe9988473628d02219e316f7287d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 16:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012d0ca45031419180d8e5eeed191751 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b105d5a97a3847d792a9a758d73c7403 : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa167ea18884c118a9017616977a232 : SUCCESS in 10m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e7f3a9ba6bf49279a94f268078fb405 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba704bd280e6473eb3765c8e5064664b : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a5b42f79d014be785a3a5706d392d57 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9798353df28744ba94f2b9e0f8bf4fed : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e9712655934205860c7db5435a1519 : SUCCESS in 13m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/084a9cf14cf649dbb46ee0e3c345dc63 : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b263063c24d247e8803775eaee299fc1 : SUCCESS in 13m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2d6fbcc7d421480a9c2d5d7a243b3481 : SUCCESS in 10m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5aadab68da824a7cbac0013e08ace414 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1f2252f422d4746b08c01c804f23266 : SUCCESS in 50m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/447a2d92c2a14cb59194ad8f677df94e : SUCCESS in 50m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91ca98273dfc4864baf6191512414e3f : SUCCESS in 1h 52m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b98d25899fbd48f195bdc7fb37eddec2 : SUCCESS in 1h 45m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/89dbd9b37ecf4dad9599ecba9545aa6e : FAILURE in 28m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b45c3acdf474d85950505218339b16e : SUCCESS in 1h 52m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8de202a797a420da5d9608f760fe8a0 : SUCCESS in 1h 58m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad245386e4d7470cbcb5f89fe36afb75 : SUCCESS in 1h 33m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28aea95457574106a138bbdd0b276b26 : SUCCESS in 1h 06m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b9f16b4dbd534ee68eef6cbf33695ad1 : SUCCESS in 1h 39m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e567f13f5fcc4420a0cf2ca2e72c3a2a : SUCCESS in 1h 19m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5dc5007c640a4a3bb415e9c83f97c2e3 : SUCCESS in 1h 11m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/277db7a99af840fd8c04311cabfb4a37 : FAILURE in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"664cb3c9c379ee350202929503d6a4e327f78001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 16:54:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a9588692bac1a9871f8468f25691db432b9fa59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:37:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9a070fe622f8d2b673cf5766a01ad55f6e18d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:37:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42f772a452484d77bcfcc3ca64e1318b : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33fe186d2854c38879402e89b445735 : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82a052284916407dac90d07b6a5a0706 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/669e24cb8cc94006a6788f84cfea7fc5 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c1ef7d1744d41729cfa1a9ef688acdd : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa8de01e8a944c5944cb8faba4835f0 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab1b013ef55b40c2b01f30c4017a898e : SUCCESS in 9m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4adae801d7b64d98bcd85dd002337574 : SUCCESS in 9m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d75ea62874ef4dd7adc09b09757aadbd : SUCCESS in 9m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/485d2afe4d1b45a097a167ba30b731ae : SUCCESS in 47m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aa451e0dd6804f508b039d0926d90bed : SUCCESS in 44m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab58420b45cf46d3bafaabc9e8b05769 : SUCCESS in 1h 37m 04s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3c2a202f483b49f1a55fbb77c42a3b6c : SUCCESS in 1h 31m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0f8913d637a4877b76a121e5a2f8d54 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c8683aa861cf69ceab7dbbce07138724f6a50c2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:39:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341e2ae7a7864b94adf6d568a2e41308 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb70360348924b10ba9d709d7997a57d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"164aed0e676b0fcdbc74a0f5ef26b4c70b70e03e","revisions":{"164aed0e676b0fcdbc74a0f5ef26b4c70b70e03e":{"kind":"REWORK","_number":1,"created":"2020-01-31 14:00:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/705219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/705219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/705219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/705219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/705219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/705219/1"}}},"commit":{"parents":[{"commit":"f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2","subject":"Merge \"Revert \"Fix gate issue with DIB\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0df1334ddc4e7e22e7271df2b3066eee1e9f2e2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-31 14:00:15.000000000","tz":0},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n(cherry picked from commit df0ba06c1e96dc5ad4855d24435970ee926175ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/164aed0e676b0fcdbc74a0f5ef26b4c70b70e03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/164aed0e676b0fcdbc74a0f5ef26b4c70b70e03e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
