)]}'
{"id":"openstack%2Foctavia~705221","triplet_id":"openstack%2Foctavia~stable%2Frocky~Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","project":"openstack/octavia","branch":"stable/rocky","hashtags":[],"change_id":"Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","subject":"Fix house keeping graceful shutdown","status":"MERGED","created":"2020-01-31 14:00:42.000000000","updated":"2020-02-19 16:36:22.000000000","submitted":"2020-02-19 16:33:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705221-1582130012984-7fccfbb9","meta_rev_id":"85cab9eef5570ea887c3123f9ec0a4ea371ebfeb","_number":705221,"virtual_id_number":705221,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"date":"2020-02-19 09:04:13.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 16:33:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-02-19 11:17:29.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-02-19 13:22:44.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-02-19 13:22:44.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 11:17:29.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-19 13:22:44.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-02-19 16:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"009887c17848a1e8925bb949bbcdf8c3f55851d8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 14:00:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"a75f398bb22c99cdc6f5057e6a9fe29a598abff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:06:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d3dae905dfc47f0a8d9555c861134ff : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083d46f1b7a24fe8a018c1e3c50612e8 : SUCCESS in 7m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93fd591e8d1148f2a90d7faa8d9dcbca : SUCCESS in 6m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15d5c7714cf24c858e75a158094c5cfb : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9ce04aebef48899fe47d0c980aaa53 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907a8154db634927aa5a87241613ead9 : SUCCESS in 5m 56s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a841c62c9659487aa8d6d6e4cd91bd01 : FAILURE in 41m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/05a0c50854134173a225871fc1bbcfa1 : FAILURE in 43m 29s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce1d5cfb652d438c9f1e60bce96c50eb : FAILURE in 37m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3955f30bbe184a0e9a10c4bd659c8ed5 : FAILURE in 39m 08s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6f5cea73ba29472e85e38f6e6a8a5537 : SUCCESS in 6m 59s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b91c84768ff540088677d0ae2e80acce : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8bd8746509114c2c9ce5cdce61feb0aa : SUCCESS in 8m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01190b1550e644dbb59e0259f2512297 : FAILURE in 37m 10s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/365ee099c4d44f648770ea34f38ad28c : FAILURE in 37m 30s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/3161623d746c46cea56aaa3539ee8d12 : FAILURE in 38m 55s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/a6939e8de16e431b9c7b8a3d9567b36e : FAILURE in 40m 25s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3b56756b091c4cf9b6b7aa33d02c336d : FAILURE in 45m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c59b592d75d4fce9f9c3f8ec26fb840 : FAILURE in 33m 11s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/5b6bcea1da9b42c4a97a215706e3b70d : FAILURE in 33m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2c38dea662a46508ed8473a8908ce2b : FAILURE in 36m 26s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dce6ee1043144324b18181f31683dbc4 : FAILURE in 34m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de663503942c4c2782024ec165202570 : FAILURE in 59m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa767a43e8a84496825c2bb983d4685a : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6502a4c297b56edcdf4958d4d9f3040e5b72aa3e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-03 07:03:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18a3264d6eb1d42059b01297006203130ee05578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:26:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eddc3a3ed8e4c6984b8b30f94fd773e : SUCCESS in 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8fdb6e1f7145f88aa6c7f1941d2747 : SUCCESS in 11m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23ce5f7fe10a404cb683b7fec61b168d : SUCCESS in 9m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eb567a189e5846d0b73eaae196ce63dc : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad0de2bbf6b4082885b343c7cf86ff5 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b13bd29d014548dc8410f1dcb69ebfe5 : SUCCESS in 15m 10s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2a5238ccba1a4eabb211a591ddf6d448 : SUCCESS in 1h 18m 00s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cce5b4ad2bd2423997eca7df9fa82f40 : SUCCESS in 1h 55m 03s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a3fce89c895745eda526b0edb1bb771d : SUCCESS in 1h 21m 40s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8ab84af394cc4fcdb8fc872fd1977e6f : FAILURE in 1h 01m 32s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16c48efbb92d4f598a91cd634a6665de : SUCCESS in 6m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5a88af0bdc584770817431073e59d552 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d862ed6c8e6a4173b53e6cfd7c508265 : SUCCESS in 12m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/972e9c9555634c3f85bc849057d329f9 : SUCCESS in 59m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1b4691c278c94172918dc9178d907b08 : SUCCESS in 1h 28m 59s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/a8d8e2e2f0934a0682089aa4500d4b05 : FAILURE in 1h 24m 19s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c698927aa459484d92baff7e275d95fc : FAILURE in 1h 43m 52s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/86a37c83ba6e49c088e8c4ab6e7ebf13 : SUCCESS in 1h 10m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3392ecdb66aa4d939dfc8291993ae1bd : SUCCESS in 54m 53s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d64efd18b9d44deea264e8e6ca6f619f : SUCCESS in 1h 16m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd1d11d222fb4ecb8af0254770e681e6 : SUCCESS in 1h 51m 57s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a46ef71fd81942a09a25ad776b13e099 : SUCCESS in 2h 09m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d95077949a2748f8a0c03c2f1be5edcb : FAILURE in 1h 38m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03543ee5071141cfa9f69b93ac92a6e7 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f6b6d1f41db729bc9f37e03d71f3e1242420da48","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-04 12:40:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a300b9eb3db0cba542ca264126674ea84960138f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:43:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a9388a6a1ff46788addb549a389b8fa : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9222643d4cbf4c40b726eddd361790ed : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/775d9dc926cd44b9982ee96daf4d416d : SUCCESS in 6m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f93ba5e0f89a47518d1523de86b98aff : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2c3c73a3ec48a8a90479cdec3c486b : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a61ec733a28a4e56b586566e7413c9d3 : SUCCESS in 7m 13s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5574b94493ba402182e45b01ebf92555 : FAILURE in 41m 02s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/09e1968af1c7496c9141575bb16e8c0f : FAILURE in 32m 36s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01e0511dd80a4c3f814a246dd1b87d9e : FAILURE in 41m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/438e6af6d56649ad8458afe8d5789353 : FAILURE in 41m 28s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0aa3ba81e0ec427c9382c40c5e89a8a5 : SUCCESS in 7m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/208ab901e2d142b5835a10aaf86d1f69 : SUCCESS in 10m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a57e127f858044fbac97b410e5a4128b : SUCCESS in 6m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0e39f3908a740e2ae71c533b6f3974a : FAILURE in 41m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/69483b6355cf4b15beb6709dc1689313 : FAILURE in 37m 37s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/01399667f70f4c46a9d6da406cf46ac4 : FAILURE in 44m 51s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8d8b55bbbc22476a9919ab80216bd1b8 : FAILURE in 38m 31s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/90c1ca8057404a7a80ecfa2df0f32ef0 : FAILURE in 1h 00m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47e7955b6f3e47d29845a1dd66f6745e : FAILURE in 33m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ae2a6196425d42acaf37dbe349a91dad : FAILURE in 42m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff2ead419b7547ed8f26edec3b884308 : FAILURE in 40m 16s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/f25f560e9a7d4e6fb6b83ddb223cdd42 : FAILURE in 34m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8de18a569f5941c0adcb8ad5e36c4e50 : FAILURE in 22m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b13b6b798354713adf91ca8cff0c654 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7b71a58c9dffc2aeeb73123c749025c945b4d043","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-19 09:04:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f1bccd5a4f03ba5219d0f18fe8f45efec3e46ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 10:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa01530147b4416094048689de081e95 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26f922204324dd5b58f6016df5e801d : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/659c4964394440acac84300d1f4fdfa5 : SUCCESS in 7m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e32615eaf6f44deacfb7516ce7eaacc : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440759fe70814cc785511d1ef982cedd : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faaae37f758a405baab77cc6546e338e : SUCCESS in 7m 23s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5e5e1849303e4936ba5cee0e7f8a2095 : SUCCESS in 1h 16m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/47a80a19c09241e691d19720cb96e6ce : SUCCESS in 1h 16m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a029ade272724b3da57cb81d1b9e2ae3 : SUCCESS in 1h 15m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0158d7c224e247aa896f9f79c4defda0 : SUCCESS in 1h 23m 37s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5edd6f424e454156be1cde13b9a006ca : SUCCESS in 11m 10s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/16afb20c5738486b917f884cc9186f72 : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4e4f2dfbb0047b7803cfc4b3912ce1f : SUCCESS in 6m 57s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d95d810488304649aa7b9dbd022e97ad : SUCCESS in 1h 09m 00s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0ac117fa4b854e41ab5179c05486360b : SUCCESS in 1h 08m 54s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/277b848abd6c44d398d0653f821fa5f2 : FAILURE in 1h 31m 18s (non-voting)\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/1702ed3b663f4466b68f441cbff3f3e3 : FAILURE in 1h 32m 31s (non-voting)\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/99b871cf519342b499292e822c7bca4c : SUCCESS in 1h 05m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8dd0000fa31048538fe780ff64560cf8 : SUCCESS in 54m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/de67aa95a0134f8f8fc2a0f7a7a8c863 : SUCCESS in 50m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a12b9ba41aa74a9d93dc4b4cd5943a4f : SUCCESS in 1h 30m 28s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/84fb9a39cd2e42db9f552704ee0f9b15 : SUCCESS in 1h 25m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ef14f55d1a8942f59da442454cbb2843 : SUCCESS in 1h 32m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fff199e270d431eac0b9750ae4099f7 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"97710d4d5489c90b55cbc5261f747ba47afdc664","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-19 11:17:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cdb5f4f70462b552ac7447b87346443da81f25b","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-19 13:22:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c08a162ef2b56e12ec6fb30fc081e7435e2af48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:22:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbf94e0fd115cbbc00004c0621e4da7f27017ec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:33:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"659985761ab518fdbdd88a1bfd2a49197ac89aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:33:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0125f85719a04465b2e480d5bef5e32d : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4bcccbbb31f46ada881371cd729af17 : SUCCESS in 8m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b65106be4ac4d40a49ec6c1c79a125b : SUCCESS in 12m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c3242287042840e7b52b8e1a97328f6b : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6496879890224512a7ae95d42a870eb0 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/507005a98042447f84aee11386f5b977 : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6a6a9bba4658452ca086d4577bab3f57 : SUCCESS in 9m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2c53454bf9834e6984345fed2daf1cee : SUCCESS in 9m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/814d9239921448da84fd56edac9ef9ec : SUCCESS in 8m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e32cf2a682242c482a77380fd905b7b : SUCCESS in 1h 10m 38s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a70ee855a4234f0285450b55cbdd20e3 : SUCCESS in 1h 15m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/110508fb39e843b99a879b75191a700c : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cd38845e40da43e1a0e5d699db49245e : SUCCESS in 55m 22s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d53e4a5755e84f2fa675abe10590de41 : SUCCESS in 59m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c8bf5bf98b84e908395de9fb359a58d : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/4d642795b2fa462f9d6244003bf3da0c : SUCCESS in 1h 41m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/df9b0f6bbd71494b9f1de841130939db : SUCCESS in 1h 06m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/708ba9b0f96b4ce799e5064d3a23aab9 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"85cab9eef5570ea887c3123f9ec0a4ea371ebfeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:36:22.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c6abc5e34c4464bda4492a3069a186 : SUCCESS in 2m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd135a5a146d4aa0a0cb12c802996f1c : FAILURE in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69d5661c1fa564ab85146e44d9e493248dc01e76","revisions":{"69d5661c1fa564ab85146e44d9e493248dc01e76":{"kind":"REWORK","_number":1,"created":"2020-01-31 14:00:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/21/705221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/21/705221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/705221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/705221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/21/705221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/21/705221/1"}}},"commit":{"parents":[{"commit":"34545524b6cc194c788172aaf3e565513f2e4d74","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34545524b6cc194c788172aaf3e565513f2e4d74"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-31 14:00:42.000000000","tz":0},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n(cherry picked from commit df0ba06c1e96dc5ad4855d24435970ee926175ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69d5661c1fa564ab85146e44d9e493248dc01e76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69d5661c1fa564ab85146e44d9e493248dc01e76"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
