)]}'
{"id":"openstack%2Foctavia~705223","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c","subject":"Fix house keeping graceful shutdown","status":"MERGED","created":"2020-01-31 14:07:04.000000000","updated":"2020-02-23 17:07:05.000000000","submitted":"2020-02-23 17:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705223-1582477520795-354c625b","meta_rev_id":"ecfac5c4cfe2ba0cfb273842fea6b371448936ce","_number":705223,"virtual_id_number":705223,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-23 17:05:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-23 12:35:55.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-02-19 11:18:18.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-02-19 12:44:22.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-02-19 12:44:22.000000000","_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 11:18:18.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-19 12:44:22.000000000","updated_by":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"reviewer":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"state":"REVIEWER"},{"updated":"2020-02-23 17:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"452e8a6d10f8caf41984a0ce81f23947e32a300b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-01-31 14:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3dbf5392cc368b39708ede5f9ba59377a3ba835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a39c2ab432646c1b5a602fb5b223c15 : FAILURE in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cd813f9930647f894991265ef36aa0d : FAILURE in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55ea6c42cb354cfc905f97bf5110adc1 : FAILURE in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848d14258b604d4199f9233c54fa633a : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5124625b0d88494c9dacfdf0ddd045dd : SUCCESS in 10m 06s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bb9788a6ca354706a5b60fb508d503bf : FAILURE in 29m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7d5c38fab0bc494fbb4489e5609ce76e : FAILURE in 20m 32s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44981e21c6264df886a2eab9b73fd7c0 : FAILURE in 46m 07s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0cca7b693fe7479ead82672eb7822b15 : FAILURE in 20m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df58538fd5434c03847c04b785bcb4d2 : FAILURE in 5m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6d6e1cbec91148369c7ec4cad3fda8ba : SUCCESS in 7m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3812450d2fc74a65a743f5effa6035e3 : SUCCESS in 8m 45s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36a55a6494ff400e9739fda4961b0b2f : FAILURE in 35m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/45ca3dc3decc43b58655ad0129aa72a7 : FAILURE in 20m 18s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/3ee47180954c46f19251445bc918b90c : FAILURE in 41m 51s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e6d701ae993d4b28aee9b48aba037db2 : FAILURE in 21m 10s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/9461334e78ae432680690deae79e5c09 : FAILURE in 42m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b258f7d4d8924542b86b466469ee666e : FAILURE in 20m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/48075ad24d534e0db0ac41bc9116f477 : FAILURE in 20m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22790efdb28c45e3ad5aa19c9c5f784a : FAILURE in 20m 39s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/444f1affebad48a7a3f1dc7471aac383 : FAILURE in 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"329771b0d5f0f7b1d6f2f91331d3aa980dfba542","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-03 17:09:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbef8f1ee6df571e91b0f163ef49f70ca8beea82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 19:01:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d220590e1fe4ea6b238047765d5d298 : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/038b9ee01df44b7d87f2adbffc218aee : SUCCESS in 7m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8789281f16349e289b1d713b23303ab : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395a5da450554e2b94d78e1ef9eb72ec : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5371ff2dc8a49b59623c3c1301afca9 : SUCCESS in 7m 33s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5cc4d09dc203421ea8a5797ea29e42b5 : SUCCESS in 1h 23m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/cfd5788ccada437f8d7b3f3e964a190a : SUCCESS in 1h 27m 28s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49026bfe9a4246b6a61d560d847e6615 : SUCCESS in 1h 23m 54s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b63feb4010e14c1dac603fbd816f2bbe : SUCCESS in 1h 42m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e619d9f924494373a8bf0b78691ad16c : SUCCESS in 14m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cd62ad3dd9bf42d08a3415922b967362 : SUCCESS in 10m 37s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/64b3de2995aa474e8bd35ba6011a326b : SUCCESS in 8m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22bdf0db013e486a84d79aacd82b3a9b : SUCCESS in 1h 04m 31s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e58cca8717614c5aa7fbdf8c7f167365 : SUCCESS in 57m 34s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/08321451bb604dc6b9b43f0ea03dbc78 : SUCCESS in 1h 08m 02s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/f4af8572ccfe42c08294894b424b003c : SUCCESS in 1h 12m 15s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/21c79dd50dc44aed8daad88f6ac1db00 : SUCCESS in 1h 11m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9b1a5de9d1c464997f4f0ee7646bafe : SUCCESS in 47m 37s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1d1d1adf89364669bbf3cc3e0e61c4b6 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af683adea85e4f45845e0e80c555c558 : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5fb5e05e07394398993054b229ce0a2e : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c92d4c25871b0e3f5d63708a7488f1029430897a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-19 11:18:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"39895c2aba9069084cfac6f4c8e424c4fc8191f2","author":{"_account_id":2245,"name":"Jacky Hu","email":"hudayou@hotmail.com","username":"hudayou"},"date":"2020-02-19 12:44:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"addab121ca1cb25e28d1019a7260610856561e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 12:44:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c828d532f2befcabe38abc060fedc770387ba78a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 14:30:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56b62172104437eb302fb8f300d652e : SUCCESS in 10m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fceaf8f13e74ea1b1315dbf7c0ea487 : SUCCESS in 7m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca04253534cb4f55809bd26ee3fdd39f : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac5eb07770349b58bce3943b4517810 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c629d6db85443a80fc672aad88f4ef : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/423a22d5314a49a9b43e0a382f3ebcbf : SUCCESS in 11m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e0c06f633494d97afbb7b2a8eea5984 : SUCCESS in 9m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0bd7c53edf1f4ab6a2d459ff9f9fdafe : SUCCESS in 9m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f21a12845ba848edbe05d475f3568fbd : SUCCESS in 1h 19m 44s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/109c6785fb2b4849a54771dd0a078da6 : SUCCESS in 1h 23m 15s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5361d02eec2b41af84b8f81ecaf6358e : SUCCESS in 1h 05m 23s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e48bed0870c840f4a13eb9d5e47396b7 : SUCCESS in 59m 07s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f81bf556fb7049f8ada8deabfa781bfa : SUCCESS in 1h 08m 03s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4bf9f5f69a434e84b69ecffa1dd9279c : FAILURE in 1h 19m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/041a8b3cb6eb4d0696d9673c4c44046d : SUCCESS in 45m 52s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0aa078491ade420998c443d1f3a0162c : SUCCESS in 52m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4a9e4059ac5404abcd67d87cf657e2a : SUCCESS in 1h 38m 23s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ca081d8d0dba4507a05f0aebd2de71d6 : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2e1c726f5953c83cb3db4196d1bd8584b4887d05","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-19 14:35:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b52b38e6114849a4a153db7d923f8e1e04cd1b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 16:15:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b368fab6262048dfb83bebc7c75d22d4 : SUCCESS in 11m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/080628b9cab041eaba18527b2190a654 : SUCCESS in 9m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd294e422dc249cdab019c308242c19e : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04299f85514b40e387dd0abf440138cb : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa2facbc6175424181d28af852b2c28d : SUCCESS in 9m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/06788441b8c3498198141bdbeccf6a76 : SUCCESS in 1h 21m 30s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/af29ea6069cf4f1586da44488a4c6c4d : SUCCESS in 1h 25m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d72b178d04a47faa7a647b1b69c91b6 : SUCCESS in 1h 26m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a9f5f7b25626421bb1fd6642ce7f7a2f : SUCCESS in 1h 13m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c879c2036c48431f8c70e3ab8a3c5b3a : SUCCESS in 13m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3edd6f888b5d402caa8605a224d14e3c : SUCCESS in 9m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f1228cf4aa29400f9650a6344306417d : SUCCESS in 8m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f51ae88e37184cdd835ba530231208f3 : SUCCESS in 1h 04m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7053de309a1b4ac48a9ac792f82ce769 : SUCCESS in 59m 22s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ea4b4e4cc8a346d680b7263d50bfd146 : SUCCESS in 1h 11m 08s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/50456dc4376946309c3dc2b49dc7ba65 : SUCCESS in 1h 16m 36s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/1ea8b5a22147487798882bd32739fad3 : SUCCESS in 1h 09m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5adec3561ac4d6e8c5e3a4accef08ac : SUCCESS in 44m 23s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/23c83a541ad248d19ebb32b2e40ee85f : SUCCESS in 46m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7541473ff8224493b05bda34bb993b11 : FAILURE in 1h 17m 14s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e561609a2ff747549698a991d80d76c8 : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3522d2e671db4bf62315a93c4210305169384ede","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-20 07:38:18.000000000","message":"Patch Set 2:\n\nrecheck\n\nnova.exception.NoValidHost: No valid host was found. There are not enough hosts available","accounts_in_message":[],"_revision_number":2},{"id":"ae21dc6981da4b58758d05c8f69d652a1e825d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 09:06:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d1b0120aad476db81a377d85c046a8 : SUCCESS in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16fe61fced104703bc81503d628718cb : SUCCESS in 7m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1507d07045bd46a190c9da276119b35c : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab94386cdbb4ffead14e8294edb7a40 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2b6035a8709460fbda34d5f0eb99a84 : SUCCESS in 9m 01s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d7720aa54d474f479d84c5ca5ed9798b : SUCCESS in 1h 19m 49s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/05c7dba735984ecc9e258e36c23596ff : SUCCESS in 1h 24m 51s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/183efb3dffa14eb7b0766d959c5526f0 : SUCCESS in 1h 23m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/811830747630405ba34566355ac8649b : SUCCESS in 1h 20m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a7c66eee8a54796b195ee0874992691 : SUCCESS in 10m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d60c226383342f4bceab066e7691d3d : SUCCESS in 10m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cf69212bd80243528b6f009c9922e783 : SUCCESS in 11m 20s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4efd1d2f3a264859aafd0be3c5cd66f5 : FAILURE in 56m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2a564b2ae26b423881284a82d8a7cbed : SUCCESS in 58m 55s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/bf70abd1caa5473087263ffbedc8f27e : SUCCESS in 1h 16m 06s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ffa1fa57195f4e358f940d64bb1da3bd : SUCCESS in 1h 09m 40s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/335b1b6dcb924157a0419d9754401efc : SUCCESS in 1h 11m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8cc4c5af7912479ba1da87dac01686f3 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/dccc3352dc1b4e4ab4b1dacfb9a42da7 : SUCCESS in 49m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4a4bbf8222945b38dfc547fbc65e6e3 : SUCCESS in 1h 17m 25s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7aad9e1384434889bda98ea650bcc55d : FAILURE in 1h 21m 38s","accounts_in_message":[],"_revision_number":2},{"id":"78d0f820c826dafa45b2342288bf2ffbff2601d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-20 13:14:56.000000000","message":"Patch Set 2:\n\nrecheck\n\n\u0027No valid host was found. There are not enough hosts available.\u0027","accounts_in_message":[],"_revision_number":2},{"id":"094de3e51e16eaa08d6dd6bfa693523fd11fbf92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 14:51:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299715bc97a94a0ca7eddad8d506f5d6 : SUCCESS in 11m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6632bd06c70d4ab591dcdcce863a839e : SUCCESS in 6m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/042a0adb5fdc4f28a23e093dffb5fb20 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addd4995dbb2485ab23f979ff5dde9e4 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5d6e47efe6d4b148328c0a3f0087f03 : SUCCESS in 7m 11s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/bf2d305023394df8a76ee5318baa7b97 : SUCCESS in 1h 19m 55s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/be757ab4d529460c9660fe53a62839db : SUCCESS in 1h 17m 21s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b95bed6a4f974616b8cdd4593b4cc780 : SUCCESS in 1h 31m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fa8f0793ca39426dae0f6635cc811065 : SUCCESS in 1h 23m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7b4f49e0b54c6f8209233b0bc7acfc : SUCCESS in 11m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e1b0f432ff2c4722bce95fecb7554ff9 : SUCCESS in 8m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e48f24e84e3647ff9e81b29235fae388 : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdcf1773df0e4055be4e5a3649df2b0a : SUCCESS in 1h 05m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6a0cbf96589f4356997f3ddc688fb813 : SUCCESS in 58m 56s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/bba77cf68bc1484ba03f2b6cd1839cfb : SUCCESS in 1h 13m 57s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/fd907a5560444f158569263611f28206 : FAILURE in 1h 03m 56s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b44ae49725884fe5a5e98a0bca1f6cee : FAILURE in 57m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ef8a8eac27946a58f9e33aea68eb0b5 : SUCCESS in 47m 07s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/6cf0bba7c841415dbf11d1213e45c585 : SUCCESS in 49m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84eec0df83bf47ad841d893cf9ae85fd : SUCCESS in 1h 30m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/511c4824b0494d53a6288162ab1e9e62 : SUCCESS in 1h 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"da9fc6e7a3f3f40aed3416df5b85fca00e25e00f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-21 09:58:41.000000000","message":"Patch Set 2:\n\nrecheck\n\noctavia.common.exceptions.ComputeWaitTimeoutException: Waiting for compute id 9c82b133-9544-4d21-9d29-4169cf589ea2 to go active timeout.","accounts_in_message":[],"_revision_number":2},{"id":"fd6336563108cd6957d9ce68ed65b77d023fd855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 11:43:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef68e6ffaae64e70a59b475cbcc13ea9 : SUCCESS in 11m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad7a371cfb6d45f9a7cda5e0b3de600c : SUCCESS in 7m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f11c7c77746b4ebd87923bf5548577ac : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0128f30d89744b29b8e77d81f1d1dc4b : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5623574816e84e4dbdbcb36dc9449bd6 : SUCCESS in 8m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/4be928ecb7b9472a83cc186e36fcf6b4 : SUCCESS in 1h 17m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d818dfc42ed84e2aa7afd7c1c4ef7b93 : SUCCESS in 1h 19m 39s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9292c7150b9b499abc59825491548688 : SUCCESS in 1h 23m 20s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/65dc3888dfe744d4b5b4e090a11bb94e : SUCCESS in 1h 24m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18add9173024449d80e86734b66f862f : SUCCESS in 11m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3168a532e23f41bea51f2c2b48fffea6 : SUCCESS in 8m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d836e4969d704bd19664e3c1129a2f45 : SUCCESS in 8m 33s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1d81fbfdae84a77af11abe22eefa86a : SUCCESS in 1h 01m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/24b642368b164b1690e03c0da3d4380c : SUCCESS in 1h 04m 34s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/3ee1e7e4880c432ba31c34eee8ddca11 : SUCCESS in 1h 17m 27s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8af2f9c7843448ecbe25ed24ffba3b5d : SUCCESS in 1h 22m 13s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/996fc3a5295c424f86d719cbed48ac92 : SUCCESS in 1h 05m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21a6a2bfb86740cda3fbb9460d08fb1d : SUCCESS in 49m 24s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bd3c69fe65064df1873efae77c3dd35a : SUCCESS in 47m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0221725223a843579843fb01e7e70a0a : SUCCESS in 1h 24m 15s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/bf69835406364084814969a3ca6de52c : FAILURE in 1h 14m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bb572465ee100a95230fb9eb2246e2f149ded4da","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-21 15:32:15.000000000","message":"Patch Set 2:\n\nrecheck\n\nFeb 21 10:46:07.871349 ubuntu-xenial-inap-mtl01-0014729834 octavia-worker[14170]: ERROR octavia.amphorae.drivers.haproxy.exceptions [-] Amphora agent returned unexpected result code 500 with response {\u0027details\u0027: \u0027RTNETLINK answers: File exists\\nFailed to bring up eth1.\\n\u0027, \u0027message\u0027: \u0027Error plugging VIP\u0027}","accounts_in_message":[],"_revision_number":2},{"id":"3d2a199d3fc6721e9985a67f6ba47fb0ad960003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 17:32:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ecd54de76184f52bff07fe049c77f1a : SUCCESS in 13m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfd2f7e2b26b4bd990c8697f677b84b1 : SUCCESS in 8m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/894964cd2f6a4c46a45ac55ddd9b8cbc : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d077784cfa0d4feb80806d527c521bd2 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d59daf05478240be9dafd58c0eecf904 : SUCCESS in 10m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/18a4134dd9884945925fc2824f3779a3 : SUCCESS in 1h 19m 26s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/93ede0c9d4b840abbff881c8d3062efc : SUCCESS in 1h 27m 50s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff792a4a91564b899c0424d1eee53682 : SUCCESS in 1h 22m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/995a7161db7649919018a0afc907432b : SUCCESS in 1h 24m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32e91bee74e847a69c9444e88bb9f2dd : SUCCESS in 12m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/99daab13931b4d55917e395477b61e12 : SUCCESS in 10m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ad3b5356c193462498b9b2fbdcfa7db3 : SUCCESS in 10m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af63aaa778e74686b7178ef6a62cc90c : SUCCESS in 1h 17m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/60724463c5364a949251bd71c6e6581f : SUCCESS in 1h 05m 29s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/02c91e9f728f451db7596372a1c9e569 : SUCCESS in 1h 13m 09s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/14ee3e632c3c4405afbf69312b40142a : SUCCESS in 1h 13m 07s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/f593865bf32646a7b09a97e0700f8b6e : SUCCESS in 1h 16m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5ee99b3297fc4d54843a14ad82e0ac51 : SUCCESS in 51m 20s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/322f46f72a594039aab7719e1b45cbc2 : SUCCESS in 48m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2643f079d5648f882274c8671e02cbb : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/614f6bef2444451fa9468c5d7b27c97b : SUCCESS in 1h 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"347b6739943d124fb14c6d46f04e3a16be8d798b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 17:32:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21614aabbc980d348f4cc9c5c069727e778fdab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 19:17:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b1546605e34c2faa6c75c01e0a1bc9 : SUCCESS in 10m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24a26450ad4d4a56b2494bdd0c368c1a : SUCCESS in 6m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/217dafd6194840f2a612ee7657d3ab14 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e18e0439924484b7e8ccc7f9193f49 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5fa4d7edfd4c279a851597c437ed93 : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a702f25617574dc590819dc76af148e4 : SUCCESS in 11m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/95a8567f8b464489a6e43fe287d1548c : SUCCESS in 8m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fc701af03e384922a079819e41aa149a : SUCCESS in 9m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/59d197d86ec14ad6848707a54115dba3 : SUCCESS in 1h 27m 30s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/70b8ea1921cf40f2aec481c22dc76e3b : SUCCESS in 1h 28m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9abe658e09054e449ad040e087b16be3 : SUCCESS in 1h 07m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d8a109dec37a4048ae3ed2ce996010e0 : SUCCESS in 1h 06m 26s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/fd642f0f9cb345e3876aa22eec2e0054 : SUCCESS in 1h 15m 33s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d26c8846047f4a168e8dd87711d90eaa : SUCCESS in 1h 09m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24e7ca43380949b992b7454b4acf4708 : SUCCESS in 56m 25s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/92b285556d00462b80acfbf6667ce838 : SUCCESS in 46m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/baade470d678478cad06df1e90951634 : FAILURE in 1h 18m 50s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/590ea4c4db7d4f8d952c94477e41f09c : SUCCESS in 1h 42m 21s","accounts_in_message":[],"_revision_number":2},{"id":"73501a03e4269d95491d745f4b25d38ac71e0be7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-22 08:59:11.000000000","message":"Patch Set 2:\n\nrecheck\n\nRTNETLINK answers: File exists\\nFailed to bring up eth1.\\n","accounts_in_message":[],"_revision_number":2},{"id":"c5c97302a0f0063bb111234c8976894a7c0989ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 10:32:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d117ba47240463984d954e012d7fa29 : SUCCESS in 10m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b2a6c28b7824349a2e3d3c96fe86d39 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f55a535bac5b456ebf86e88af1f242ef : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0013e0a7450d4a58bb4bb91952c91213 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b8a1dbb610247d4823de82d3fb4e97c : SUCCESS in 7m 11s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9de8f4db4e6f4782b7b6ac1f22d5a585 : SUCCESS in 1h 18m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0a196779d2cc49a6995a21724b626306 : SUCCESS in 1h 18m 40s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ae5412e9a2d4c469eb76f1346dc9249 : SUCCESS in 1h 19m 19s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/47040d84203444b5a4d72c07c7c22e93 : SUCCESS in 1h 18m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0669c781e9b44fbb576de48175000bf : SUCCESS in 10m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c51432e2668c42da90afa0a9a4d8fdd3 : SUCCESS in 7m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/544eb21b274c4635b5594434590115ad : SUCCESS in 7m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ebb7cca0d584e8dab67e4f09b36ef4b : SUCCESS in 59m 43s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8f97302e6c2e4034b78fe2569d4b42af : SUCCESS in 1h 01m 15s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/e5f3282aa0554491bbd22c2f2842fdce : SUCCESS in 1h 12m 34s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2da3fb407f9a4b2bac46f44e1e09e7e8 : SUCCESS in 1h 07m 50s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8fbf5ac09f2e45c7a56acd7524f411fb : SUCCESS in 1h 07m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/57704fa8a0f44b7882faa55e43bb3c6b : SUCCESS in 56m 29s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/dffe5fecd6c84358b0cf3c40610b93ec : SUCCESS in 46m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33fbb073f8514b8b90d02b6f580a58bd : SUCCESS in 1h 20m 42s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/08b4360cbef84fbe8a6ec66a6908cc84 : FAILURE in 1h 09m 09s","accounts_in_message":[],"_revision_number":2},{"id":"74c37837380700539a0413d49209090c1af84ddc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-23 09:43:59.000000000","message":"Patch Set 2:\n\nrecheck\n\nRTNETLINK answers: File exists\\nFailed to bring up eth1.\\n","accounts_in_message":[],"_revision_number":2},{"id":"6ec4a58b172e475d0632179ac13947f2f2b108dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 11:15:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9da1946d9c465d821fe9e592a23968 : SUCCESS in 9m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a9be5f3683647a6abac06d1b39f71c5 : SUCCESS in 6m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5c1ebc1bd6e49afbab741c29574c010 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13bae10350074173ba6a11392855b9e6 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25b360bee6a8419d8bfd5b3e844fb46f : SUCCESS in 6m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/754e2efa93d74006b2c9496dd678d339 : SUCCESS in 1h 16m 00s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/27734cd89a4644e8b958beeb53073c02 : SUCCESS in 1h 16m 35s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6e7e99479e244788c2e67afbdd54553 : RETRY_LIMIT in 28s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/487eb4c1b0c74a129200d427ac0bef3b : FAILURE in 1h 13m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6e94c9abf63434ab08616824813a522 : SUCCESS in 10m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f2689e0c75024168b7508cfdcce63171 : SUCCESS in 7m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/77662ad44c1240c69985239ee171a81f : SUCCESS in 9m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22c401c213ed46ec8b523130632c3f0c : SUCCESS in 54m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3c9feb5d93cf4aa2b24cd69eb1193184 : SUCCESS in 55m 27s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/92b7aab3dade4b28a66a88d04e08d4ae : FAILURE in 1h 03m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/41d926e0b22443b893cf265a9bc2d38f : FAILURE in 1h 03m 59s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/54747dbff4e44c6ba46f982d32955b63 : FAILURE in 57m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c8242a72674543409d2455e71eb8cd8b : RETRY_LIMIT in 11s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/6b16e864a61d4df29ed10705d848da5f : SUCCESS in 44m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0a70f60e9954001a215e0050ecee1ce : FAILURE in 1h 23m 24s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/4d382ecadc5845a5b6c3e28274fdcc0f : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ab503048fa022e75714ebf1b58ed1e654370ff9a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-02-23 12:35:55.000000000","message":"Patch Set 2:\n\nrecheck\n\n...eth1...","accounts_in_message":[],"_revision_number":2},{"id":"84415684ebbce5711a4da62a67be9cb67d0b5f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 14:05:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d5621734b5416b979ec76b97961c9f : SUCCESS in 10m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f781177bb8403fa070f03b98e22e09 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5648f00c3f3f414eb62f2b4d56ee234c : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aebb43f495740f4b9298236f41a9588 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f47a2e6ee32d47e683e59b6f3e1e3b97 : SUCCESS in 7m 01s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/0fc4b1d8c150435eac3291b7442fc806 : SUCCESS in 1h 14m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/1fdcba15ad0441d9afc3f500b3c669ba : SUCCESS in 1h 14m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c5fc729949247b3ac3649b673959342 : SUCCESS in 1h 09m 41s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2b21d8999ab54ff1b1359dd95283a37e : SUCCESS in 1h 15m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/009af558a90c4efdad5051ecb0ba7d74 : SUCCESS in 9m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6af129fdb5644e32a3f429b42b48cd1b : SUCCESS in 6m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/95b4a21196314ce1b5006d1eaa106ab0 : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ddef7c8d140a4803979c49ab58cd8855 : SUCCESS in 55m 01s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/a739ee0dfc9b46baa2bfae1ecd4d1b37 : SUCCESS in 55m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/0227effa1ac54a8aa48e7a8ad45f8a86 : SUCCESS in 1h 06m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/47cfad2c08e74498923c54f456c62b51 : SUCCESS in 1h 06m 05s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/36c3908c885c44898ab1a1edc8b9b7e4 : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f38bacc258c047659e56510893224703 : SUCCESS in 50m 10s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/205c2bd9cf8b44b898aa5b98c1b07cc2 : SUCCESS in 46m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b94f2ab68a24294b0a32021604fabc4 : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/ee854b464c5d4ff597ed2bb8244828ac : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e10e3dd12f86766c2bf105ca9e1e6f03e445ff9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 14:05:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a87a563a27f83711bfbab1f9f7b0ffebab30e452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 17:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"078d4b89d78569697d44dd8e61b673861c0ef610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 17:05:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203d8722dd194e0db8fee5d564a5c1de : SUCCESS in 10m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20bd8398fa844d48a7c1dfc5cf504dec : SUCCESS in 6m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/49475366a82e4770b5248cd0a7cc4cb6 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd422ab75514c628f92ff3999c34cf4 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f3189b8015f43dcb8c4692a57f6d991 : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c95ee58142641538e71fec2a60472d9 : SUCCESS in 9m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/786ffa4fd9564be6a69b5557456789d8 : SUCCESS in 7m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/088c2ed57f77406284d62379fda0547e : SUCCESS in 8m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/30d51b14023740e08f26a2acc5b16314 : SUCCESS in 1h 17m 35s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0445b8353146427389ddb8075c9542b1 : SUCCESS in 1h 21m 44s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d702f9cf3094933856a404f8dacc0e6 : SUCCESS in 1h 00m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fdc2c772ee3d4c77a81671f96f63405b : SUCCESS in 58m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/5b832df2221e428aa05d194d86ccb3d8 : SUCCESS in 1h 06m 21s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b2d28672c17f4c08be4f5e14fba20158 : SUCCESS in 1h 14m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de55a866d1804b789da1e9fd16c91a91 : SUCCESS in 46m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/479fe2f248194db586b8b3ea9856f24c : SUCCESS in 46m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/00ba2cac0a104f33a23b21fddf0cf1f7 : SUCCESS in 1h 26m 55s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/99d9ea37edc644cfa0bd359acf3f0a48 : SUCCESS in 1h 21m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ecfac5c4cfe2ba0cfb273842fea6b371448936ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 17:07:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa54bfdac3441c8bbc0fe6824cc4f1a : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02751f223b4f4796b363eebdcedf0255 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae69762511506782b88bdef61bcf4c9561dcf740","revisions":{"b8c6c0dd296e345880a642b42c70db23060f8850":{"kind":"REWORK","_number":1,"created":"2020-01-31 14:07:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/705223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/705223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/705223/1"}}},"commit":{"parents":[{"commit":"e3ddec776da4654517a237287c05298a9280e52f","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ddec776da4654517a237287c05298a9280e52f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-01-31 14:06:35.000000000","tz":60},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n(cherry picked from commit df0ba06c1e96dc5ad4855d24435970ee926175ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8c6c0dd296e345880a642b42c70db23060f8850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8c6c0dd296e345880a642b42c70db23060f8850"}]},"branch":"refs/heads/stable/queens"},"ae69762511506782b88bdef61bcf4c9561dcf740":{"kind":"REWORK","_number":2,"created":"2020-02-03 17:09:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/705223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/705223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/705223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/705223/2"}}},"commit":{"parents":[{"commit":"e3ddec776da4654517a237287c05298a9280e52f","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ddec776da4654517a237287c05298a9280e52f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2019-12-11 16:09:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-02-03 17:09:42.000000000","tz":60},"subject":"Fix house keeping graceful shutdown","message":"Fix house keeping graceful shutdown\n\nSIGTERM was not caught by house keeping service, threads were not\ncorrectly terminated and spare amphora creation may have been killed\nwhen restarting the service, leaving amphorae in BOOTING status.\n\nStory: 2007008\nTask: 37788\n\nChange-Id: Ib3d6aa0f2fd550c3a8756c41e0159c5ae8127c7c\n(cherry picked from commit df0ba06c1e96dc5ad4855d24435970ee926175ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae69762511506782b88bdef61bcf4c9561dcf740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae69762511506782b88bdef61bcf4c9561dcf740"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
