)]}'
{"id":"openstack%2Foctavia~705317","triplet_id":"openstack%2Foctavia~master~I04cb2f1f10ec566298834f81df0cf8b100ca916c","project":"openstack/octavia","branch":"master","topic":"failover-refactor","hashtags":[],"change_id":"I04cb2f1f10ec566298834f81df0cf8b100ca916c","subject":"Refactor the failover flows","status":"MERGED","created":"2020-02-01 02:11:51.000000000","updated":"2020-06-19 22:00:52.000000000","submitted":"2020-06-19 21:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":196,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705317-1592603912450-1a856907","meta_rev_id":"22846cd180a6e7fbfc7f8d3d53f8c7e970d56690","_number":705317,"virtual_id_number":705317,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 21:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-06-19 20:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-06-19 08:59:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-06-19 07:19:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-06-19 09:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-06-19 20:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":-2,"date":"2020-06-19 20:30:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-06-19 07:19:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 10:21:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-05-28 22:33:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-10 05:23:15.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"REVIEWER"},{"updated":"2020-06-19 07:19:43.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-06-19 08:59:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-06-19 09:53:11.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-06-19 20:30:18.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-06-19 21:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce5d4db57cfc966379601c08c964a9f206557171","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-01 02:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a06b2a184ac32373f2e6f0a9811bfaac0fe8983c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-01 02:12:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e5f3a2d24385b6a12098842bfc08581016a1446a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 03:44:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd3c44c87ca462282d5ce645adc3117 : FAILURE in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd84a993bfec43c2b307af3098c8152a : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1045ff15f6944b20a4c0ec16d07be04d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8181d51fc8cd4147bf2621d4a9ce0c6d : FAILURE in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fea0b2620c34365b0ab765992d51153 : FAILURE in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/032d6b91857741d99f2e2efdfb874f8a : FAILURE in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e323b7e150342a5a986902541a27c51 : FAILURE in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2d8ff6685d034556a37f44cc6813d807 : FAILURE in 4m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/76606e7d911e492f937fceb874ded698 : FAILURE in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47a69fe90e7f4bedbb24bf8d3a2faa1b : FAILURE in 6m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85cefab524144943a0e1dbb53f15346f : FAILURE in 46m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/10aabceaca1145849f76d91a4ae8efe9 : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a543ce9b9a6450c8682673eea68d775 : SUCCESS in 58m 08s\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/344d828257974da4bc4355de5b614175 : SUCCESS in 1h 10m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/274e795f000c4440ae74205f4305eda2 : FAILURE in 38m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4572fbe9f914af9bd43436f3a22e15f : SUCCESS in 1h 31m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d5c367d8611472ba71d79e5340b609c : FAILURE in 1h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0edd5d6fc91266d72e333eb3e5c32b1a5a494f14","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-08 22:07:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56a6311ec2fa81a1b82a0bb1c445bae760a811ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-08 22:07:41.000000000","message":"Patch Set 2: Workflow-1\n\nStill a bunch of work todo, but amphora failover now works.","accounts_in_message":[],"_revision_number":2},{"id":"2ad074b7bb62f36ec4e488137275d6f18c5825c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 23:51:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22fe4b73f5274b2dac670c67f1bfec32 : FAILURE in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fca3a41d2ace4e6bb2990717d611d2e2 : FAILURE in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa0b3d4c86d4c75940aebd818a2b3d9 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08a42618e06e4748ad369d8e58973f63 : FAILURE in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6818f887e07496bb86198c3e75b72ca : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58da1a15817c41e1af0c9721773c6152 : FAILURE in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9dc3da08334bf6a32f00ae40c752dc : FAILURE in 6m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/447c6075a6c44223a5e737341efd1abe : FAILURE in 4m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb188459904b4fbb8b6254df0e6adf9a : FAILURE in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4ea0169fdfb4f8e934477f5a97f3cdd : FAILURE in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/940f0c6b1fee41728cae8c7007ffdab7 : FAILURE in 57m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e66d90686aaa4bbda1f194a9096b0e2c : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/15cbf13a5f4b4993bab0372e9203c093 : SUCCESS in 1h 05m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45d444f9fbe04c7e91812c04a754895e : SUCCESS in 1h 01m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1fe70e472b44989942fa4592b4b5c90 : SUCCESS in 52m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ebe8bc0d90d84959925c3b7c2d7a225d : FAILURE in 1h 40m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f5a89bf21e448dfb95d80457ed7237f : SUCCESS in 1h 13m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/906f421a3d5448dbaf2564c6d17e190b : FAILURE in 36m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f7ca7ef67a3f47c99ffe27b70598bb54 : SUCCESS in 1h 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a55f1a39a0041a83497d3de0c075a0407c35aa2e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-11 23:17:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cee5b33adb23f7f81913d6eafd1c58128f5ea4e0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-11 23:17:56.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"25531a1316fa067ffffb084300b2dd371a122a2f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-11 23:22:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"50a37a8f0ea7c25f963947582eadb78c47b00aff","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-11 23:23:05.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"969584c00953d6375c21287f39df0774b3731d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 01:28:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b28e4e225121419286daca14eb3e3a07 : FAILURE in 7m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e67a39682040c28fa19245bfc50eed : FAILURE in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925383c14c3a42239297a2885ee0b16d : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963fb4a9999c4accb1ca2497b3c1c3a9 : FAILURE in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed12c661a40e42e59fe18c6374798e02 : FAILURE in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1068eea58f1f48bba3d81dcc26ee93cc : FAILURE in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7252850e61d47098016a7560bc5637a : FAILURE in 8m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c511893da41b4e85bc3798ff6df18d61 : FAILURE in 6m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e002966b4e4947e8884e64bdca329f57 : SUCCESS in 10m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e92cd8cd0d74eb9be90b6c842c191a6 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/247ae244cf3e46a39583aa7b3fefc746 : FAILURE in 44m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7652c29d4e0438ab59c4f7c0b8bb8eb : SUCCESS in 1h 31m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dffcc1b481a948789251f907807b966a : SUCCESS in 58m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/83e23c6d0f0f4104880bb0b7a69a316c : FAILURE in 40m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3dbdae3e287047818684ec0b06270b03 : SUCCESS in 1h 03m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c3ae1302d1a4914ad9520d1eb582e5f : SUCCESS in 1h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ee28262cd054bddac027a642a286ab3 : SUCCESS in 1h 57m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b7c0170f9a745a8ac77318b2e165e93 : FAILURE in 1h 21m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8f0236f9616d451b80ffdac7d5f94621 : SUCCESS in 1h 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e16e1d8140eb34218afd9d530bda4000e991a932","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 05:51:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"65711628dcc42197e3f89dea17b99bc0addb7cd4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 05:51:33.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"123b579782a718c3b7c753cbadedaf379dce8124","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 06:06:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b5408b0f297cc349d4845f3325e2e6639691229a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 06:06:24.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"eb4bec2c482a022120e2128ba661dab3bcf4af83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 08:29:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3890ccfe44d41c981f64f141c4a8844 : FAILURE in 6m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff19765cfd8d4103a28faedf7fe50c1f : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d64342785c4a2bbd41b1988d78ad31 : FAILURE in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d172ee4d0f1b44ed9f63c9b2cd8d3736 : FAILURE in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b20880c5bd64d0d8fdd926f84b5b1a2 : FAILURE in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030c7d1604214a77a34229e9011907fb : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ef3c26ec9740ee8bafe091972484e7 : FAILURE in 8m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1af78763ed324c9ba37dd1de07d90064 : FAILURE in 6m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/836f0af8652a42649a559e4450276c59 : SUCCESS in 11m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e40327cb47043f2b57b5ff33811f644 : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/953f2baf4eef4c058e4aa61530f91d38 : FAILURE in 46m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcb9fd43aa024773a13811a90f901178 : SUCCESS in 1h 52m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/655b82235e124473a0182cbe09161c31 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/30ce3c69e2554925a1d760e2cbe6b9be : FAILURE in 52m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/148edd0f146a41bca52e4447669e37af : FAILURE in 1h 05m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea43707508184b9790086dcee128cb81 : SUCCESS in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec6ab943e1b04b1687c8157b9306cd95 : SUCCESS in 1h 24m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa163875efa04a98905fe032b70a5ba1 : FAILURE in 1h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c1bb9f928fe649efae5813270bf82126 : SUCCESS in 1h 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"045693ca9518c0864952bc65a6969792d59c6aa6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 19:49:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"391da96b5873ba441a2ba60526fd5e4b174e814d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-12 19:50:02.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"fbdd1e7d04398eb4c561de4bd832a866db901896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 21:54:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe84004548b045c688fc5db8e379c61e : FAILURE in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb87ab249ed6473897da1911def29924 : FAILURE in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854aa6239f46430dac9bcf777fae0382 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d445926e3f4785899fd9c6e2a7eee7 : FAILURE in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cae8b411788e4690bea141f4cf6fb275 : FAILURE in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/493f1d30c9174f75ba67c29547dbc5c3 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d06c08cb724be0a9d82ea1982dffb1 : SUCCESS in 18m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/59d3532d017c4d2c8f1dea91ab7165cf : FAILURE in 8m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d381f7ae1e4d44508f8a634db0c98ad5 : SUCCESS in 10m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9772fb8107124a6caa2967f55bc84bb2 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/741c3644d0854f9c8c692136476431a1 : FAILURE in 48m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d0d648c6e404ad99f2552495012cb7b : SUCCESS in 1h 47m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6244d4916a6437b95518595f6d5cd18 : SUCCESS in 1h 05m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d11a00d719a41b182dc8011808dfb87 : SUCCESS in 59m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c8044b78321425499c8cee79ff84538 : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b2e29d042ff84c179957b09b7eb8a76a : SUCCESS in 1h 43m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0af675b52b544f65b68f2668ef9c7a05 : FAILURE in 1h 57m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/33e1950b545449a595318823c8f481b2 : FAILURE in 9m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d3a249b404614b9c875ec8211fa7a4a2 : SUCCESS in 1h 36m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5da0999edeeaade4a9cc2800252921df923a282d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-13 02:06:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9f81aa9619631ab7d7edeabfbc6257dd3f062d68","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-13 02:06:50.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"7524d4e82006fa4480597ff4ad46906d8d8ff251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 03:38:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88b45cf8bc8e414d8c4b9eccea0e66ca : FAILURE in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07093cce0cb44727be525103aa84f560 : RETRY_LIMIT in 2s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c1da48baec4f038e5899851f20a37d : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61f4f23b41df4cfd97d96a224ac19e6a : RETRY_LIMIT in 2s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3764eae5a7d34939b8b02713e261b5dc : RETRY_LIMIT in 2s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b6f1994e7a14a469c3f2a222b88e4b3 : RETRY_LIMIT in 3s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e834291a09e491bb1254dd0946ae13b : RETRY_LIMIT in 1s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ee4a22b6fbbf4a2f9ced2c3c6046ab9f : FAILURE in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/27a0bf3d17ac4943a6da2b231bc9d49a : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43527016eb15413c9ec8755e38b68cb3 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78d0a16220c24103be8b278c3991eaf2 : RETRY_LIMIT in 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1792ea085df4d819b659c2c25cd5856 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5067d8005da64942a96f789c6393e472 : SUCCESS in 55m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/33c58e4fcfd74deb98a0b224fd7032a2 : SUCCESS in 48m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/441159a0c8274d4388a2f9ef38a5fd51 : SUCCESS in 59m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56c86fa4ea644ae881fc879f267517c0 : RETRY_LIMIT in 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c51c4ccf38034b6dad4615425dab5fc7 : SUCCESS in 1h 08m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6105ea6031f44de08c346d621ee5cdd9 : FAILURE in 6m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5317fd94a8954506824e0c9e2d81f0cf : RETRY_LIMIT in 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4489c12dbb1f37ecefda2c884af3d3f61e88ec99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-13 04:25:09.000000000","message":"Patch Set 8:\n\nrecheck\ninfra issue causing RETRY_LIMIT","accounts_in_message":[],"_revision_number":8},{"id":"6d0d6f572ea399f0bebec63d8092a57f814ac115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 07:12:51.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fed1f3a646b47f6959d193136aa0235 : FAILURE in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddd556bebb8646f7a287a50354b1f47f : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6965a6a8f7a24906b129a864aba373cc : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a907e36b2f304fb0adaf4610c81974f7 : FAILURE in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4956620888534c4985ecfff0eca54d61 : FAILURE in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2576d3d975e34482bfe9223715cf2b35 : FAILURE in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdae053c2ce42139ebb163825910430 : SUCCESS in 11m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fc40fdd6207e429cbd4d129b67230fa5 : FAILURE in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/51c2f88c89d042cd9c376e3d6f3894f4 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b69b85d6dd64256b9d4571eb4196023 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcbcc740cbbf4adda7aa064fec1f7fa0 : SUCCESS in 43m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71d1f5eb4fe24a8db18a59ff3d2ebf8d : SUCCESS in 1h 33m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed4222f2a900453a8d995dc1c4fb135f : SUCCESS in 56m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6339606aa9ec4d57b6c2efed86ca97e9 : SUCCESS in 57m 06s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92efa083628a413f93101be91dcfb9b3 : SUCCESS in 53m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/109b700069c64ea49502b42c847a6ae6 : SUCCESS in 1h 43m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d458ffb718aa455ab02dac22bc753435 : SUCCESS in 1h 48m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3baf7bf0c2d4beb96de1d966e382c11 : FAILURE in 57m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11264c9c245d4496b480ab332dc119d2 : SUCCESS in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ec6c31de3d69ebef36f44866a619fddbc439948d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-14 19:26:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"246cf75933d98d8336d04050e115b6e9fe5d710f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-14 19:27:02.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"594de4d0c67eefbd495debc8c7fe6a5d3cacf597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 21:16:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da0fc9ac5e364a7288d7fbaaaad46ca9 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ac5c36264f741d8b5bce86157e0a68a : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dea3e69f6c4bd1a8d0dbfc879502c8 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f054a5ca665491f959676d3f0caa09b : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d94ece339af4b35aa14e8776d78dcac : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b805bca9dd614f769b051a3226a3fe4d : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3c2f0c16aa42e3a4992b61d4072ee3 : SUCCESS in 16m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27f0b309561c4d20ba193b49c0b0eb75 : SUCCESS in 6m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55310aa533c7401c8b2be33a0b58a616 : SUCCESS in 9m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/846d18b794434a49b4b1927f47297049 : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d97ae9c82ea449c84e04e7cf97ed6f0 : SUCCESS in 48m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/205182c450674b41ba39a66249c58068 : SUCCESS in 1h 23m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbe02c0f547145cc9cd879790a44b796 : SUCCESS in 56m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bacb889c993440a3880cf26a0882d545 : SUCCESS in 53m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a2e702e04d146ac89a74aaa631ad2cf : SUCCESS in 58m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7117adadfe5043b590cae8944df41a5f : SUCCESS in 1h 37m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5ee87f8d71843518ea8b7cfaf098a13 : SUCCESS in 1h 11m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8686efeb06547a39b62648d07707a92 : FAILURE in 1h 01m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3b8f1b263d6491fb8652025692e2596 : SUCCESS in 1h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"03bfb2df449759b04532ccdd5abc6c63e2ad0562","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-21 00:51:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8f0cb6e563aa8ddd7c07e9099c384ee2e34abdf1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-21 00:52:08.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"fd69f1837ef6a6f7660bf6625432df867d26d67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 02:38:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c540f29c7da4486b36f041c77348785 : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/244a97426f1c44c1b17c7743225f3bb3 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000828ee003748fc9fd5799597251c52 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cff71ad9c4134b57ae2170a6cfffb6a1 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a6ef81887884e2e8820e537fcf8bc15 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8162583fd2504894a0e3162c61c40a50 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059823fd2f12475895746e999c0efc30 : SUCCESS in 15m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c4187efe588a46b6834f364b89d7b80b : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b359fd14bff24e43bc59ce699e23d1bb : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db7e58e0834a4623994e02336d594c1d : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e648ac49bbf443cbbc317719ba021b2 : SUCCESS in 48m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d704016c40f465f9877871b3c336feb : SUCCESS in 1h 40m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d1671ab2ac6f4a84abf9ceb7e8978f8f : SUCCESS in 56m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4e2423a6f1b84b08adb9852fadde20eb : SUCCESS in 52m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7bff20b4f844b3b92a673b63be930ab : SUCCESS in 59m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d2dd8cb058bc430eb30e2c99d3222530 : SUCCESS in 1h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ebd74a9fac24418a29220cffdeaa000 : SUCCESS in 1h 42m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28826e5f9f4847729e05d4dd4ca6a506 : FAILURE in 1h 02m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/44e2614bfca44d1fb39af13f42c6b3af : SUCCESS in 1h 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c09386741944b362ed768b6e89aec7d49600b504","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-22 00:38:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c0528d3e08d254f0c4b404a279184471b8411525","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-22 00:38:29.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"461eb7c86c84f61bf48f5777fc887768086b8483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 02:36:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5099af0b0710405bbef136d8e5be55e0 : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce4a63ca97b54cb1800e0f7919e60365 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f57c64646a044e79f5bf672c3461447 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de471d91e07e458d8457f0ad5086807c : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55cbdc3d79d043498a025f741d4cde6b : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/931b0a8d1ebd489c99f7c1b5ee712f12 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0abfc29c39c48c198ed700eb796eee9 : SUCCESS in 14m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6436391355d743b2ba67508de7d78b62 : SUCCESS in 5m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cd26060bb66d4aaf9ed1c4b421cd6908 : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/473431de46a2489fbf7886bc253cee13 : SUCCESS in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b35cded972174318ae437597c684a065 : SUCCESS in 42m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/087f23d80f334536af44bfbb19be8ae3 : SUCCESS in 1h 19m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c13b102c62546cca21c2f5c003d3488 : SUCCESS in 54m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7700616f81844cbcbdc53cad6658520d : SUCCESS in 50m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e69353799534dc9b31a551d613ae9e8 : SUCCESS in 1h 08m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1aee1951fe754e7998678200991c8f4e : SUCCESS in 1h 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2dd0d5e3fe8846a5b3ebffb1fe86c4a6 : SUCCESS in 1h 04m 28s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ffac08f8376f4b7d99f8a61d5db246cc : FAILURE in 1h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/001a4563c2ad4d6b9a09dcf23f46d752 : POST_FAILURE in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"63ce412891dc720e53655e39ae37b7f1889b336a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-25 00:05:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1c589274eea1743bf5ab152d2a4746c7b9925474","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-25 00:05:27.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"a431ecc8c5e7457da717549e846c2afb8229a0b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 02:00:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5cd86967afd435e9999b8d650c5f8a9 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3c0e9288a4e42a3bf0b6158c2816568 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b70e29742ec41bda71125f96d593020 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/531b1e0510ce483e827d5218ce4a067d : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cb10d4cb23c4095814f918fec163b10 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f8d11851474a9388192988bb0c5548 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ac75abebcb42cdafcbc4df489c65cd : SUCCESS in 14m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0dc6fb0c84cb42dca5c98c9436f069cc : SUCCESS in 7m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/239f03dc3ac644078f396392d500cb0c : SUCCESS in 11m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6f3e97bb1fd4e6f93d85a641b80c88d : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d60f2b7cdcd04d3ca8ae7f8c10e07a0f : SUCCESS in 43m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af6ac947a2a1407abb7ea3efa0902ff4 : SUCCESS in 1h 27m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bb22b598611454f87b48982248bb4d0 : SUCCESS in 56m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cc515ebfea9046a3b0b98d3360cf2ec9 : SUCCESS in 52m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08893ec097dd4227a184ebb9c9074847 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/546b089770d34fd3988816e0e02d2b5b : SUCCESS in 1h 43m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80ef1b30728745e59d4f71c0c389f0da : SUCCESS in 1h 46m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50971876451547e09950be46459426af : FAILURE in 1h 07m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/43cb7fd50270427bbf9ecc3c758ff455 : SUCCESS in 1h 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c9903c2c6ede8fee492f7531c29acfa6cbd7766a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-27 01:22:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1ff4970b08df87bf58ab3586d4d68d60690ad7b6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-27 01:22:49.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"21a281f5088ff90e7c73f65946cb8d82dc8f425a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 03:25:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbe5ced283854190b13b2c47710f1036 : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d6d5aec66964a28b69c63e842e1f6f2 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d059ec5908401fac39c2b67b0fd5c9 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26cb27ffb4154e35a443dbd5747195d6 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/159bda57dd4e45599f5e196c372486ae : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ecc08b02dde485e9350129db770b1d8 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b9a9b87db84f64a81a64f47ec46e0a : SUCCESS in 13m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1a4b77212b564b4eb679d0501b788f49 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/31e025990a3646aebedaa9e3ff1a1c78 : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/204d04ae97af4a2399930471a148b33f : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4dabcc9014e14a49883da04bcaacd718 : SUCCESS in 42m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdb33767e3fa478aabbd121213afec82 : SUCCESS in 1h 49m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a46a65b1f3964bcaab6783bd3d6c9d6b : SUCCESS in 1h 06m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2769d46b35b1423da6fca04f3ddb4720 : SUCCESS in 50m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64faf13fc0fb400594f0c8b8ff27f24a : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad01f563f1aa486d985a9aaec6eb8636 : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81a3eb7a5b5e423984823eee44d0ba02 : FAILURE in 2h 01m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46a73328ad3440feb2ea36da5c186c61 : FAILURE in 1h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fa0452d3aa694d64a0565c119b350105 : SUCCESS in 1h 37m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9bff6afc8afebcefc61c4a693a6faf58f27e335c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-28 23:31:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a31b0b8a241b10f57d735e45ade75cdf0ab31495","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-28 23:31:51.000000000","message":"Patch Set 14: Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"564e3c37132a37158b8ac8689f6b6041a88df80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 03:42:29.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/202f8a1c41f9403a95e0921a31d429eb : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17418bafeaff45508ea67f32c1933b5d : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4411c6da048421dbd7f6bc56821a168 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04bb57bb3da841ba801d9342a1b9d3bd : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6598f5fae76416bb2e9c4d4e4ef7775 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e538457ddc24e3586177501c5484fc5 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b654823a4d488cafbca4ff534e136a : SUCCESS in 14m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8c84eb21d6ac4f80a765e4a169bc5d9c : SUCCESS in 6m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a6e4a5def68e40018037c7ba652d24c9 : SUCCESS in 8m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19bfee07245b4bc088fbdaa9a26a8b2a : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee266142a0d441b1aee433889a962bb9 : SUCCESS in 47m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec206d3ff43c4655ade6dd434084c855 : SUCCESS in 2h 18m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb39e59cdf3c4d528d1659475e3ef3e5 : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/367b97dfb7e543e595cb700a532131fb : SUCCESS in 1h 33m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7013fc8b7d48410584590019e1a4587f : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f339d6bbbb4a4961bb045e094083c631 : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a968418bbd0a4e8eb9c34164492428eb : SUCCESS in 2h 03m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/595914557ed84eddb747ce9df36a9a81 : FAILURE in 1h 05m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8eebb6f1814d43299e5f31f15f713568 : SUCCESS in 2h 02m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6848aa6ef23552c0e0572987b239faf864609563","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-04 01:03:33.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1b2fac775b9e85ea453a771b716aa6c3ceb6cf01","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-04 01:03:44.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"663208f58a15d5be8e52355acb271497009e3c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 03:24:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83de600376e4ce793ba6591acbaf463 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84dab4ddc3284af0959e4c7338fa00e1 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89a87370dd44a2b94f679d028d643bc : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68bb276a27540ac80bb6cc805bbf2ed : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/910ef1b9bf9f4be899b58e9b54a722eb : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb256f2548d94fe5a2930c010ad0ed82 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccf5f0480a94d1682fad7e3372a8a2d : SUCCESS in 13m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/651b19d3315c4852af94d62ac3254503 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/11b51dfd349e4904805cbbe662f9c8c9 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95f7cf2799bb4c7d97b69c71aa99995e : SUCCESS in 10m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b50b779916dd4cedb364d2cb68b394fc : SUCCESS in 46m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c88b0bd1ebff4abf8fd29faf73f201e3 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad12b31775e745c58a5c6a0688a2c052 : SUCCESS in 59m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b20ce65dee4541a898f037c64997c235 : SUCCESS in 56m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6aadf0c4077a4f318c07e0fd20628cd1 : SUCCESS in 1h 01m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5769d9a0bba94a548d54591eac48aa51 : SUCCESS in 1h 50m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a18d43a14364b93978caf373e2f6f00 : SUCCESS in 2h 07m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad871bb681334bc194a120940d20c34f : FAILURE in 1h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39327d3cc0bd4ce5934026adfc12ad50 : SUCCESS in 2h 07m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"dcd9a65d4f33622008519cf05c4a257bba9733f4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-12 23:02:51.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2363cd92b6f5344d7dcad60652c4f6f597ba1044","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-12 23:14:18.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"e862a02405f828f9ad877cf090ba55bd86649ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:24:57.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a71cdf5e7ab340ea910855809a6ea909 : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/715b50f3609c4e1989c09322bff35500 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b68803bb3bc43fd9340ef3f98d07387 : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08295f5b4695468d9eb7479a2733a5a0 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9cc29d1bc1f480181c962ccfd5508c4 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f970b24c2c4431a95194946ada8d0a7 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e0c91276a84b7d8ed6c02e61bd2f93 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93936cdb5b74c8f9636bc3abc856f1d : SUCCESS in 6m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d7f94f3edcbc40f2a3898aa5ea9a510b : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0ddcffb270cc4957afe2bca543114aa3 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9f2698c98da4c68af21f760c110b31f : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4c5f54434e5403db215c01671f47a04 : SUCCESS in 40m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c88c9dc794b747d381bee6003893ce85 : SUCCESS in 2h 02m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4032095be1ba4600aadab013a500cc82 : SUCCESS in 1h 00m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b1fbe247702543148becfe65976c7090 : SUCCESS in 1h 00m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d4cf28fb925487caf8f8c7f259642e0 : SUCCESS in 1h 14m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36ed3a57c7bd4c7597380246c90bf1be : SUCCESS in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/231cdd8157ee4f4596a5ac72548322a6 : SUCCESS in 1h 56m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/106da99d07cb45a098225e93ce5756f0 : FAILURE in 1h 03m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a95a3a3d4444e67a820a70332e2a144 : SUCCESS in 1h 52m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"30afcd19218161ce10475180c0b4bd614191c0bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-13 16:13:16.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"f22a13f2fc7894567fbd45f7ea9af04c3352df0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:42:36.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b207f0c66730477281a324af15fda462 : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f76c2420a0d497f8c92e3110ac310d4 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f0297a34954559a528eb9a78626218 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b63db8b3ca340faa9e0899dac54f393 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/638afd4e75234f4bb60548cd9cea8d1e : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877b3a3a61e8496a85f83027ce3c1123 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4511ae23c77e40d7bc5bee738b5248d9 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80964418da0d4836a7477caea2afad45 : SUCCESS in 7m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3afd44c1015d4da8b73127cbb29e529c : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4ce8cc2a4fd74b6aa1d2d8fb0746d106 : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3edb4d8384a448a7ba3ad5e1d5a8d90b : SUCCESS in 13m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24788466dcf94b7793834eb34c308e2a : SUCCESS in 44m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e80f949f95ae4f76bb86023c0ae3c28a : SUCCESS in 2h 06m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9dd9de3d17104858a3e227fa3bc635c8 : SUCCESS in 1h 06m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f8d0d015d14c449d930b2ec904a3332a : SUCCESS in 57m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8eb0284d62bc421ab40e05da503a1c66 : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/704004583de44d0eb437379dc8e11210 : SUCCESS in 2h 03m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca87ec8999fa4698b46b45c2375d9b38 : SUCCESS in 2h 06m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dfe416d63c2e4da9a5595f47bc493f13 : FAILURE in 1h 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/19b259a65c524aeba6465b3e9819c8ca : SUCCESS in 2h 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a97946160ac082f3881a4176eab2db2424a10e7f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-23 15:31:04.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"332ecfb8f696028de41258a64521abd6c77f3bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:52:17.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f869b81d54ac4caaada3cb2f5524b8d1 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35c70c67f3a44fc2b3130286f220ae77 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94ee18565ae41e2958475ddf56491cc : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b98970198e314457ae5b0ccdd3385998 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eca97a4fce9a45ab964ff14f0c25d5e7 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33c907fb335e458ba14159b611719c43 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11683817108b47338b92a58414a5452b : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cb1544ec4fc4f1ab2bdd1a2cabaefb4 : SUCCESS in 5m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/afe0995f3c6042b3972d36a0ec3f4950 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62a1685e57864f589a5e89ba271ecc43 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ada3f3cd3d34649932a7c0b916dc5e0 : SUCCESS in 9m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee8698367f1543d49f9843765a621715 : SUCCESS in 46m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82253410d6264bfdae1797f33a0dda5c : SUCCESS in 2h 04m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9224db6051e4ed2b6bccd075ca50acf : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e5361f5b787c4a75abb19723a7ca4fe8 : SUCCESS in 56m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef1280bc637247da89a9cfd505b72df7 : SUCCESS in 56m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61d7905a992c414d9b2ea07fce544098 : SUCCESS in 2h 00m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e5641ac76f54a2fa14b680e0e692a4e : SUCCESS in 2h 08m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6b86457f08e41e6a54ba50f1f46ae58 : FAILURE in 1h 01m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/282607aa5e6e4dedb52353d71d578d53 : SUCCESS in 2h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d94893583f7d117d32e9cb38478973bb630ac067","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-23 21:42:00.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"44411ab77b6305911fb16ab2b54e330dea0f9348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 00:06:49.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eeee9027ba1475eb5c1481c6fd98258 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef3b51ad8a0b497db57fa39e86e08321 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd457a9eda842cdb40b6598700135bf : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb49d6ff83f47a5bd851e017d12c6f6 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32dfbe50bf6e426299d9c34d8eb41911 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c20fc5592524442bdf779048fabbbcd : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413800504d8d43f6afff4121e39b8913 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc07568810c14aeb975087734dc5bb01 : SUCCESS in 5m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/06426b68eaa04039b21184dce28ef6d9 : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/050943bf0386498d9b0a739e27cfc365 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/afa395c962a14a319b033bba121f2d63 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c20e179ea4d44997aa5f26c14de1004d : SUCCESS in 48m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caa3bbf30ae74c7282d090c24e257cfe : SUCCESS in 1h 45m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7fcb48ce4e45496fb35c13ff3a820abb : SUCCESS in 1h 15m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9d78f33fde1f49a180bb2e3c5ba82f2e : SUCCESS in 53m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6b82a1822774209b80c63ff03c1c4b0 : SUCCESS in 1h 05m 58s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8f161aa11ad2400ca66ff12f5c6de471 : SUCCESS in 2h 23m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6009286edb464d359001f1fb71243ff3 : SUCCESS in 1h 58m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d56432f2a618460db81de0a26d92e19a : FAILURE in 1h 07m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1df127bbcf8e4a3fa5b682c1bba7991b : SUCCESS in 1h 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e36a9ab8e393967718da6a04f65265b8ec833627","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-03-26 13:03:47.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"bcf99709565edd2db8550238ddd32a2100ceccdd","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-01 04:18:42.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"dd561b9162e738267113c2f4e003c002a43a486b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-01 16:38:28.000000000","message":"Patch Set 20:\n\nAlso, as we discussed in IRC, it doesn\u0027t deal well with BOOTING amphora that have empty fields. :(","accounts_in_message":[],"_revision_number":20},{"id":"d6c6971c23cd8031c68a021836aa52ddc1b1bb40","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 17:47:06.000000000","message":"Patch Set 20:\n\nYeah, this patch: https://github.com/openstack/octavia/commit/005cd1e6a63f226285486a8287ec523e1d73a7ff\ncaused the \"BOOTING\" issue.","accounts_in_message":[],"_revision_number":20},{"id":"4464acf04320fcd14fa9b9b1d3eda5436be06191","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 18:27:29.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"b0e18d54749a2985699ae46bb569f712619af9fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 18:31:56.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"cf976887dcc79a39f9d70c3bc2bb4bcbe45d52a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 21:03:57.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52b18ef5a97245ab85efc1ce53a16694 : FAILURE in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/110a02c92c4841999301801bf59d0e7a : FAILURE in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66dbce8c733c4936ae8b5b989a7808d5 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3726cfe92f444134ba342b6517d61b8b : FAILURE in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd44a8e8866d4c6c8373cbb2c21d45b7 : FAILURE in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23331ca5e0694553a0793107bb5bdc61 : FAILURE in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5c27529b334d0c81e7f6b2efd69d95 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c01097e98043909574cdd52a4eea3b : SUCCESS in 6m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce01ce6c73724374a28b881725c8683a : FAILURE in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/597590cca0794c03b460df79a790442f : FAILURE in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cb1e8bace4741bb9c3f9ebb67a9be52 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8a6efc6e29541cd8dad9b98d08b70ce : SUCCESS in 42m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50e8b7d66e3542b29f14fab38efad9f8 : SUCCESS in 1h 42m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/61f79788662241a9986fd1e7ff472432 : SUCCESS in 57m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/90d5b839b6d84717a4598762bb260cb6 : SUCCESS in 48m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2fd4f205d7d4fe38e1b42e6cd58921a : SUCCESS in 1h 06m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/32aa59f730024a4d91e0c4203a51a9e8 : SUCCESS in 2h 26m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbbcd8e6522d49a68b1e395072b6d649 : SUCCESS in 2h 04m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb0b894883b34dfc8a956e0e833cc144 : FAILURE in 1h 15m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1fe7753e68b4da792d78fcbba8643eb : SUCCESS in 1h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"e9c184dda86a7fc739c9a5a513c9b1dadb171984","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-01 21:57:50.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"cf8f8d07c6dd5aba74113a40f9780f22776b110d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 23:33:00.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"dbc644a55c7f8dde6d25fcec4de578d3f1de1bb5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-01 23:38:20.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"b39de436ecf2d70c55940186caccc67d7543a42f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 23:43:45.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"6848b29b0cec22110fb9802cec73ecd916c63e26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 23:47:37.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"ac1c8bffab7defda014580ae937a515fea2ae134","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-02 00:30:06.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a33f194f2559a46b9b5967b54cce1b4bd7774dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 02:31:41.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6fb4f6e019f469e84bce0f96d4ac1b5 : SUCCESS in 7m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d973f60b0bb4cc0abb039f2b0b342b1 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b9ad708d3d4177b5b19a05f11595e3 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12a2893cef1f41dba7ddf30ebd9ea47d : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/748eebbe4c0345a7ab79438c15f1d202 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6d74027ec524edc89f818cf82e85b59 : FAILURE in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ca997c52444164b718044acfa8bdaa : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0bc57bf0a62439fa587aea140605eea : SUCCESS in 6m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8fe1463500854affb482b4466e4a6c01 : SUCCESS in 5m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cf1d272dc446476cb42e09a7e2a0b3a4 : FAILURE in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe2156a95d024d219310ba8b27b27348 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0bf86720e06d46dda5d42b06515ebea9 : SUCCESS in 55m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abe8a77bd66447bcb0550967d9343966 : SUCCESS in 1h 56m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5deff009ce343bc8b89269cac2f0754 : SUCCESS in 1h 02m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cedcae960ebd45e89b47a68315f4bc01 : SUCCESS in 53m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ec92c47792e4c789a741b9c29baefd8 : SUCCESS in 1h 02m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/517c21af5f0d464882eec6f80014c933 : SUCCESS in 1h 43m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0a43f850a9e481d8565ccc54ad3e696 : SUCCESS in 1h 18m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8a1fb053fca44cf8158426fc95d3732 : FAILURE in 1h 08m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5589cfa67daf46928b8f67d1d503324f : SUCCESS in 1h 48m 42s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"9e01b033e6b777e4aff0dda7ed3934fb7f45bc65","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-02 05:29:55.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ee1859d538c5478f6be847c82586013ac8df2d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 07:50:28.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d39f701ba45c4f9e8dfd3533e6f24da6 : SUCCESS in 8m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32911b94dfaf4081b70b906bf5a7f6a5 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d794f6ea0040fcb5fb7b054891b717 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6a8f488624c4ca3a8e175b7e6261161 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8645fa5fc0e46598cc6a5ac173458bb : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecea9e417c954653b1fdd36fa0f895ad : FAILURE in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89ca4ae91474b1e8321aa1e9e4820f6 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/159d47b1bd724c85a93b117347be8c7f : SUCCESS in 5m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/50400bce4e7b4d759f0411030b0ebcb4 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/01ab8ab752dc48a191017e99b2dc1f36 : FAILURE in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d27c23f875e44cd4bc3b2155affad209 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24d12ed83ae24a28a0ea1f81e79f20ce : SUCCESS in 49m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19bf041a701749f6bbeeb93209133969 : SUCCESS in 2h 08m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/66598cbea1fc42bf81312a388e63f172 : SUCCESS in 1h 04m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/663bca6c62e24bc1a01fc32d4063a1ad : SUCCESS in 58m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64f4acfc26074360b32e072d27b78199 : SUCCESS in 1h 09m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f88d097f98d1432dbd659650110d8707 : SUCCESS in 2h 08m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8eaf662d5bde4ffeb87c3cbfa204304b : SUCCESS in 2h 12m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/400f8f4df2ab41a7b4687d27cf07f749 : FAILURE in 1h 19m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9591ff9e01f8422eb6aafea840d5c088 : SUCCESS in 2h 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"06c4ef221c2a02cc58285b203928a04cc6a87e8a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 11:22:49.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"73e49210856a26dfa1783cbfb96d95f849a8b8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:49:38.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a4de5ea188d45b08c38507cf8eb7c1e : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf838b16c33d42a6aa4ff544246bc362 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6de6a0fabcf46d8b71fcb29b9ad58c9 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90a61819719e4bde840b76fa62a5cfd0 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3290823378074ecebf724f53bb3057ec : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccc699d0981a4d0dab8a0c106edf2c54 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8aa3fb5ad814251adc16520e5ffd1b3 : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68b8c4af547d41e88dc8c78da3b78af6 : SUCCESS in 7m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c691262eaca044e4a205a554d5fd7ce0 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b4f67800ab7444b9cfbf4f6a2365698 : FAILURE in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1cf33e495aa4b6ab948ffbfe142c833 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42c35a032e8f41a6a0899fe0685ad34d : SUCCESS in 44m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3f87775d9a845228776a40f92236a58 : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba45f76c864f4784b82593d1818db178 : SUCCESS in 57m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/48bc3d2e822e4ca0a05e2956c0a04f99 : SUCCESS in 59m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9343eba4255a4ab6ba0d49f3cc218fd5 : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96d09327219b493f8c3ea9a55211eeea : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba18de703cf14419add0d96a91a4141a : SUCCESS in 1h 21m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74301c8fb82d40a4becfb36b4490f7bd : FAILURE in 1h 08m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2300d71550ef43d0ba911691f6f287c4 : SUCCESS in 2h 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"be762b340106e57b8e7ea75ada3b505172e664bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-03 16:29:14.000000000","message":"Patch Set 25:\n\nrecheck\nUC has merged for octavia-lib","accounts_in_message":[],"_revision_number":25},{"id":"7654b5ff677213de3639fac1d5b78ca5e97464e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:49:13.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06fd998c9f546e2b43299e437de9e57 : FAILURE in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ba7d889247d485b93327dd25ef2495e : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8923ac6fa96423a8bd55edc7ed06899 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1833c77b34254008b31279f868f8485e : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62c3e04f1e0049168ab245a893c4d952 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12822ee8864b45e4aaea36be8677e093 : FAILURE in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776f27bdc8a44299a28c2c8bdeb5ee1a : SUCCESS in 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c0329cceb043f8b1f6364427a0b67c : SUCCESS in 7m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/653ad8724d79404a95460e94417114ec : SUCCESS in 7m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d999720280684532812da9e8661ccfff : FAILURE in 9m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f72c4098004647a889a8db597cfecbfd : FAILURE in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7915d3d07d574e769b549ef5b41e3501 : SUCCESS in 45m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5467e56aa7d14eadbb6a7c7b5b9d0ffb : SUCCESS in 1h 51m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/efcc61719eeb4d77acbfcd8ce0540987 : SUCCESS in 1h 00m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/61b0e087c7e4441d8bb0d11345a2a543 : SUCCESS in 58m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52cabd1eef4144559d676de331dad793 : SUCCESS in 59m 09s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d52d57caa084fb28fb20f63e13f29e0 : SUCCESS in 1h 56m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e09b3bcdd774cd484cc28afb35a5f38 : SUCCESS in 1h 18m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/540186b6635e48189826564475c1c356 : FAILURE in 1h 10m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/99bcc0748b8b40e09e7f221e0889977c : SUCCESS in 2h 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"4ee047a4d7b7c7e7f987c0ece4eddcfdabb3dd88","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-07 21:00:18.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"c756d6b7be59a8766ce4d8dda2f1a055e218d131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:11:32.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fef2d9f33474a568c407f318252e627 : SUCCESS in 12m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c478b526e6b543ac9c667335a4c942c3 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e0c24a355c4c3aa5626e096131a841 : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0b2ed1fa48476ebed78dd8d4ed76a3 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbba551edce047e3b709640e4297f14b : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7d26c5aa9794cf3a9b850f4696f5723 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffb90ce0f0a4b70ab87274870a6ea75 : SUCCESS in 17m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e4840e60e124911a921eb3d666a5f9b : SUCCESS in 10m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1cd18534e7c4452598c12d5e0d4f343a : SUCCESS in 7m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/311b2bf36d3041b8b6b135acc48b449c : SUCCESS in 10m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08a269b495e44e328207cbca16c278a2 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d82610f20184dd8b70a419a0c248b28 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2144276b5631483eaf68ff4990004280 : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b0c6fb49be444a2a6d1a501dc8624c8 : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a7cceb4926d14e59ace27a4a6c8a7a91 : SUCCESS in 1h 11m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ab919a4647e4a718cb104355c53db87 : SUCCESS in 1h 04m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/366907ad3c0a425db9a12b61a2bd7668 : SUCCESS in 1h 57m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ed3474e68eb4cc2b842d16cac50a784 : FAILURE in 2h 06m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64a8371f0e4b48b0abf179981028ef37 : FAILURE in 1h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4173a4e29c6c4d0491833114c9094c18 : SUCCESS in 1h 45m 04s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_base.py\n  Comments left for invalid file octavia/common/context.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_provider_noop_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_lib.py\n  Comments left for invalid file octavia/amphorae/backends/utils/keepalivedlvs_query.py","accounts_in_message":[],"_revision_number":26},{"id":"8feade98725279250cb892601d231cc7db5093fe","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-13 11:39:11.000000000","message":"Patch Set 26: Code-Review-1\n\n(1 comment)\n\nWouldn\u0027t it be possible to keep the existing functionality to simply delete an amphora that\u0027s not associated to an LB? Then fix that in your followup patch to work correctly? Then this doesn\u0027t really change functionality, and is easier to justify merging in the bugfix period of the U cycle... Otherwise we\u0027ll have a U release that can\u0027t delete spare amps with failover, and that won\u0027t be fixed until V. It wouldn\u0027t be THAT difficult, right?","accounts_in_message":[],"_revision_number":26},{"id":"19603595be82a3121eafc36976d297e139ae3cd5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-13 23:50:31.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"9fdc9947d219f0f143b44eb12304e3467b5ccb36","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-13 23:54:54.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"c2000ccb03491767082be6bef2315fde855916b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-14 00:12:23.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"18d2602a37de1cd9c1f199709f66f1515b99200a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 02:51:33.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59c43cb989e641f18a2c8e4c54568e27 : FAILURE in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39796c1967644de5a5a0734bb26c7469 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6fb3d1da7b42cfa2093040a3e2f0d5 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0c91de9a53411fb3eb95242d8eda6e : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ba31e3c6aa74385b861db9f11fa2184 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ebe2af8f28d4158a11284bfb1f66781 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5e1c91e88748748b7330f957450eee : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a150ff10f11c486aa747459e6d853804 : SUCCESS in 5m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71d09466d32b4979b0ba7b90329946b0 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aac1b04f5184486290955bce19940b48 : FAILURE in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ae26576c7bf4844bc1b55c1d7f96bf1 : FAILURE in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0d56f607c044e9eba5b7d0866a38361 : SUCCESS in 52m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7690d59fb1e49349969dbac5dd17ff5 : TIMED_OUT in 2h 37m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/431a7e8da5184601ac327e668784c013 : SUCCESS in 50m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5d690ac390114a769e7b36a7a88ccbf8 : SUCCESS in 1h 13m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90cebbd748a74125a04dc6aa12fb0ca7 : SUCCESS in 1h 25m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/074a08d288284c46baf698b1f45ac784 : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b00a7ad003b64f1fa09942e09087a2c5 : TIMED_OUT in 2h 35m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2eeb2fe5aed94020847cf4fd32ebf725 : FAILURE in 1h 07m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c32d0f5f4b9c41ccaf086e63957d677d : SUCCESS in 1h 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"27223bfad694809898f09cf957f3a89cd6da6891","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-15 06:53:15.000000000","message":"Patch Set 28: Code-Review-1\n\n(2 comments)\n\nCouple of major breaking issues.","accounts_in_message":[],"_revision_number":28},{"id":"3253b53f413c2ed429b4d18205e23c5f12f042ec","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-15 10:50:54.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"cdb4f447d9ddfcb13d6b4b9939395544d80f6239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:24:38.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c11863ee1b94eb8bce8e0f4dd89104f : FAILURE in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0301f0533568436ca4a860ec592f538c : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0ba9bdea504c4f847a9ba7626e4194 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f28c9ab825e4919955903df5c0fc554 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14f512f622d4477d9dbb77374c1f0d7d : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a737f68c9c74a7b95465f4be61747a0 : FAILURE in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6896bfbb964cd196a62307bfc45e0f : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdc9532dea0545a5832b38167c21c7c0 : SUCCESS in 7m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7fead5792ce445b296a56717377bc1cb : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/02cfd8f779454f9aa4dc47df5bd6c1a6 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddd260413aba4fcb95986bf9e8e0bad4 : FAILURE in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b3eac82fa49429285e13d5cd1df65c0 : SUCCESS in 45m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3861a98f0c74cd4b8e77f8f1212290c : SUCCESS in 1h 52m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a75fa9529bd4633a8f2d9daebc40caa : SUCCESS in 53m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ac7990391134f7689b4aad8e28448cf : SUCCESS in 50m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fa3f8d96ed24652ae8552c08d47cc8c : SUCCESS in 59m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/266b6e8e383d4096968b9b430260de5f : SUCCESS in 2h 28m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d8084d46e77148e9834bed362e4e05a1 : SUCCESS in 2h 10m 17s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8288dd3b37c74af7a54f70303bb1bc72 : FAILURE in 1h 04m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/79cc84b63986476287db86b417027b0f : SUCCESS in 1h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2f42c85bb609ab0a11094291360ac5dfbb2a5716","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-15 15:35:54.000000000","message":"Patch Set 29: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"7027087bb5ee215aede9656ba88005efab131bbc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-16 06:03:50.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"adb030e51b9377462a3fdf10aff52b462f1473ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-16 06:04:21.000000000","message":"Patch Set 30: Workflow-1","accounts_in_message":[],"_revision_number":30},{"id":"7da5f4ad3aa43824b026d5ef5c24dca0455c9bae","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-16 06:18:50.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"e8944ea0c8a61861f40c747e48f3d03741389d49","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-16 06:23:15.000000000","message":"Patch Set 30: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"bb899267aa5371bffd1b66af251f680e231f4962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 08:26:28.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f61f0ac5caf948ff9403a62642878397 : FAILURE in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/279ccfb26ccf426e9c10917c8a32859d : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f9744c5b844c5ca4e4133e592f2ad4 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c380a4b3aa54931bf82c3f6ca349524 : FAILURE in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18fa03a9bbf3417cbfbf931568483e85 : FAILURE in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f92ea6f6074f3e970a67b5d49f0411 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ba9bcf957b4b03a2372828a92e95be : SUCCESS in 14m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc73fc9f248443b9f8c9c9f3df32a7d : SUCCESS in 8m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b011dca550c2460daeb54fe0635ecef0 : FAILURE in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/92c17b2b681f40f88e909c8b3fa2adbe : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1eb4801682914be88f0217bcc706c88c : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01a037d43cfe4a08be5dcb393962af89 : SUCCESS in 44m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/914c2fc8dc0a4453b707c01f479659bb : SUCCESS in 1h 57m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f804160a324a4965ab77177b08484856 : SUCCESS in 1h 02m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e1bd9985a3aa4310957bf1b2a45d8d44 : SUCCESS in 54m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43a4cc8f4bfe478cb4fcc65bbf84de6c : SUCCESS in 1h 08m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e9f01aafa204f1f96a989744d84a6bd : SUCCESS in 2h 03m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/26bac101c2d0414f8ee0b16891988453 : FAILURE in 1h 27m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1c4bfc87013643f5b2be018ee977a4f0 : FAILURE in 1h 17m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/585b0f8b3e804ddb9fb2de2cb8fcdf3b : SUCCESS in 1h 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"169b57a1d6527165321c057070866a2beb58e4b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-16 19:26:11.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"7b57700c47644c9de8c4d243117d5795cbccb53a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-16 19:26:23.000000000","message":"Patch Set 31: Workflow-1","accounts_in_message":[],"_revision_number":31},{"id":"491d78564d04c9bab72c1074271834e2a04c2e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:41:26.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ad8321b078d41e0972a7008fff16e5b : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e84499ec292f43e6a84aea7ae362b17f : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1628b91c56084100ab0827ce19b51312 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0701d9d710d644059c01edf6dec76a35 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06ca83ad60c549f185c520742a21f2e3 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af8ff7f7f3fe4c36b1f2335e8949b6f5 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58098e4ae26428eb740c27cecbbd714 : SUCCESS in 16m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf42de72e1c42fea544f317d5a0abd3 : SUCCESS in 6m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/313799a57b2447cd8eaa9a61480af4d8 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2cc38f47ef204c00a30f956cf37d6f3e : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca43cc324ace46618610037cc99e7062 : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fd60e4a442f47c9ab51b5c4c616e3a0 : SUCCESS in 51m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8efaf2bca0de4accabf95f08e022a9ed : SUCCESS in 2h 08m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/155ae9e04ea04c979382d1f08b982ced : SUCCESS in 1h 00m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/983ae07f0d37442982ba5ea6ef65815d : SUCCESS in 1h 12m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac95b05ea69942e4969653f36557c9fd : SUCCESS in 1h 04m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/390eb4cb8ffc45feb61eb5c3f4858d62 : SUCCESS in 2h 11m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/134f924617f149f593aa34acaf3ed232 : SUCCESS in 1h 28m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/406366acf429493aa478a74d9f18b6d4 : FAILURE in 1h 20m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/91e7d8c49e4442829370b312e2097649 : SUCCESS in 2h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"9a66bcd40d0b4eac84fd50bdb375064d3792187b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-16 22:34:58.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"9c12a8d70167b1cddb6848a6bb6f19b30ebb7ba6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-17 00:42:38.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"106c55d88e4d0ff05405d7f7c4132852b343625f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-17 00:42:49.000000000","message":"Patch Set 34: Patch Set 33 was rebased","accounts_in_message":[],"_revision_number":34},{"id":"d488a813b25145e05eba277ed9f9d6223d45a1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 03:13:09.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69ea357c8c4f43fa979114b77f71158f : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c924a74adc2346db9fc1f17e4797dcac : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42cfb01525a43c3bb8b858638bf7f74 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a18af7fbec604b92b00c89d905999392 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a24b631159c4709874978daebc9ca25 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1bfb8a10f784fb7a423f52833d15d52 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3868e1d9c4c446c875d0c185aa4a887 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d2dc8b74eec4f749d38a78a88f81c75 : SUCCESS in 6m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/979fc79d6be349c580c92cf863bfacbb : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d35b9504039942939c1ce743ab136c2a : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40a8207f90854ca2bf0c25eb777306c7 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/442276d6b574447291c43223bb6f3f08 : SUCCESS in 44m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f717112a421b42c5bea8d247bc943989 : SUCCESS in 1h 50m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/964a0183943d44c1a59ed76948b0460f : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d40238108e5e4d0882504a554066357a : SUCCESS in 50m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38356b01475a438cbe4e0b0d9aca6a59 : SUCCESS in 1h 11m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d4ec0e9c26a492ea06af926a1a3fab2 : SUCCESS in 2h 04m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f15ebdb11726413fb2a7af40de0246dc : SUCCESS in 2h 20m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c191a20e8e704e0182ad00cd7975f8a7 : FAILURE in 1h 00m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e51e327067eb4af1af4bc7e5ecf32e76 : SUCCESS in 2h 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"15f88acc2077f9c96936f386f89147616d9c3e14","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-24 20:53:58.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"bab72f332dd7f7e8736fa4fc1645caaefbcbeb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:22:26.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b0e8bbe0e0341d29da82480f14ef0f9 : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ea857b9d28443d1b13cfcc082969c68 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e318bdd4ba74ecebe0fbe60161e6230 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b7ecebba264541bb00935c29cca1df : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45b433cc52714634896c5dbe105ba05a : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872e4805568d4b42b9ed1bedc50a71ba : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31f89435615b485697d3ceb1ed9484f9 : SUCCESS in 8m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/701be9f600ec4e70b6ff401c3a0fd46c : SUCCESS in 7m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/130346a822bb4a2bb032b9a0d1807cd1 : SUCCESS in 9m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ddaf39204e04ca996943a76744ba028 : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2882f40b7d1347d2baf8c4d50e50db05 : SUCCESS in 54m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ec12de39c16437c854cc626f73037c6 : SUCCESS in 2h 01m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/471cbf4b024f48fd8efe52f7bc37e65e : SUCCESS in 1h 05m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fabee4a8e1344d8a96c408b6b934a1b0 : SUCCESS in 1h 07m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/056f673d76444b7e870637eb6c4b39ff : SUCCESS in 1h 08m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b93f7617a4294c21825124ed32ae0574 : SUCCESS in 1h 52m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0118800dc30b4f38ba1b232cd51eddb4 : SUCCESS in 2h 17m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6288a6d0a9c54048a56afed0dc45f99c : FAILURE in 1h 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f99473a7aaa34f1896d043af774ece9b : SUCCESS in 2h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"f193c5150cd2e95fde4aa175a26ec3d92d758f6f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-30 14:20:54.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"b35e2a3dce0381017ba8ad0dc0f007595b8dcf6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 16:22:37.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84fa8ac36c5b460f8e15a735a258228e : SUCCESS in 14m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb342ff8df91425eb04ab32a52aded16 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3947697305b949c2bfdb6474ef04694f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf23a576e18540148311ade2aff36edd : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b3d103f01d24114873dd3f5e9cc21c1 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4615bac14641178a1c231378dfc181 : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d863df3eebd646d983df877db79fc836 : SUCCESS in 7m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9fe5411655df4090b34d745b99963bef : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d97e40ab9a524a0bb4a35412cacf4798 : SUCCESS in 11m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed8d24a0ad5946d2b486051d65e99898 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55f039b3a1c14478880112a4a824b48c : SUCCESS in 53m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a26df63e9eb4cf785caa1cd1f31a7f3 : SUCCESS in 1h 16m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ccfad77516a946e29edef19cc9940213 : SUCCESS in 1h 00m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/559e563982cb4934a0f0f7c343ce3b7e : SUCCESS in 57m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84492188c59648ccae638968dd2c117b : SUCCESS in 58m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/589c197e84c64edc8d4c2c5fccc3e2c7 : SUCCESS in 1h 44m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8aaeff7fdd3d4357a9fad5df9f1c9e34 : SUCCESS in 1h 57m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b96b8dfd9809495da265a0a33bb501f5 : SUCCESS in 1h 16m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2a3eb7bc3a69482c9c8ee922ec8affc0 : SUCCESS in 1h 55m 37s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"09fef7f973893f229a63405576a10f7c62fe629b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-04 22:37:13.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"4d734ed1186e2e25378411e32e7794724dba8375","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-05 00:07:51.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"a988048ff9cf70e73dc6cd64da67f1869126f59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 02:22:48.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ccd3be5fbd44a26b768ae3922eb385e : SUCCESS in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dab20ed743a4984ab1252f465e005ec : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04154094f2114523b2c2cec040deb849 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bda5fc8fb0a41e2b5cfe1efe02fced3 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/207321e9a8bc49199f39ec5a70fa3f2f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334e724fa4054895be7a2ed72a97e43f : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/936bd82331814171b391c63b436d7a27 : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85b17e86450049b0b1bb7ea97f926eb7 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/04493711d08f4c61aadcecec91616dcf : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa873e7a2a634ffdb601b58f5719f382 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8518a9447384d28ae9b8149ff02d9ad : SUCCESS in 43m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b87069814bdf4df3b52eeec21c408081 : SUCCESS in 1h 59m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71f63e329f4f4abfaeeb5338b0cadd03 : SUCCESS in 59m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/778e7d7d1ad943119b1a73904f42306f : SUCCESS in 53m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/277ae270aebe4acf95a5b7be149dff51 : SUCCESS in 1h 03m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/58f34506e1d24af3a94e7b67c1efbc3e : SUCCESS in 1h 50m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96fd4353e45c4dc09bb295052d563bce : SUCCESS in 2h 09m 36s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f37f5f0fa3b4e8a94d065506a1bfebd : SUCCESS in 1h 11m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e609dc181c144b0b9d423fcf43fcffa : SUCCESS in 1h 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"332f3ef785514bbb5e6bbf74d3ea470457456d9a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-11 21:25:25.000000000","message":"Patch Set 38: Code-Review+2","accounts_in_message":[],"_revision_number":38},{"id":"7308b99f52fdbdcc6c791468f53b33a730b2dd3d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-14 05:11:47.000000000","message":"Patch Set 39: Patch Set 38 was rebased","accounts_in_message":[],"_revision_number":39},{"id":"34ff44fb3e57402c28aaf5c820cc706feabebaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 07:16:17.000000000","message":"Patch Set 39: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33e85208ed6049a09bebb761c163219f : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9be3f0c06e21472ab4bc166d4aaa6764 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6619ac5c6841639d878f7354e612d8 : FAILURE in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66540bfef07745b68bb3906a37b30481 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3ed25567dc4406b80ae9fa22b29a2f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31667060d2ad442896ad67ec1ce4c5d0 : FAILURE in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d46342432a414a0eb7e185d8b7b3b917 : SUCCESS in 6m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9112064054f64854bd3063d951b5ece5 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b79fd72648e847bd8ff20aed02969192 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b1bc179d8cf40d8b87f73b09e1d89ae : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/747987ebf65542d1b9b53468d292811d : SUCCESS in 40m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9916f94f70224e10941d03d4bd51c5f2 : SUCCESS in 1h 52m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53e656abdda848deb30899f98517bac8 : FAILURE in 52m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/33a880dd3fc64fcf8a670db5976c3e4f : SUCCESS in 1h 01m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fef616d1f4e43069fea22dc328e9174 : SUCCESS in 1h 02m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b4767e55f2e446c91249b8ab3c67c2b : SUCCESS in 1h 29m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e39d9a63e4644f0cae364e82748e02cc : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b94f2ae88c44b67a8d099480d16e6cb : SUCCESS in 2h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/362862827a114982bf41298488a06c68 : FAILURE in 1h 38m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_load_balancer.py\n  Comments left for invalid file octavia/api/v2/controllers/pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_quotas.py\n  Comments left for invalid file octavia/api/v2/controllers/member.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_amphora.py\n  Comments left for invalid file octavia/db/models.py","accounts_in_message":[],"_revision_number":39},{"id":"bca74f4d46a35d5ba74169ee411c5ab5ac0c7582","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-05-18 08:00:28.000000000","message":"Patch Set 39: Code-Review+1","accounts_in_message":[],"_revision_number":39},{"id":"8ea9642776a4560d249c46685dcf6fb16e8fd309","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-19 09:51:18.000000000","message":"Patch Set 40: Patch Set 39 was rebased","accounts_in_message":[],"_revision_number":40},{"id":"d7c3eec4af9b3b3eca9412abd0bc8b88e1c52f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 11:58:03.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af47c51fe81449ec82fd5a521fe0f41e : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1b07c1311ee4b65b0d4da0bb623ae20 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bedf4852b2440f591de5aeaca0878f1 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c31ff74f99274b06871baa07158a63c2 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a3c715fd6c4417af89bd83dc4c95d5 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81cb11cd98f14fbdb1f87b5aa01d227f : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3df33ab4f48846ae9221f0f64fc3f4ef : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/80a391313d3e4dfb878deb3c13158f1c : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/997145a539084c008aca6831644578b3 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29764425e5bd41c88424f92a2937f8f9 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c89760f5ffbf4d15a9abfbf95c558167 : SUCCESS in 44m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b4db5873a4e4a6893fe27ea9abe2133 : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33d7173baa0a474f8dcefb6ddad926d9 : SUCCESS in 1h 01m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/97a8b87b47ee408e936df866a9693e8a : SUCCESS in 51m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e05c41fe1c5e4c02ad48d53fb561c9a7 : SUCCESS in 1h 03m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/313152487e974b5e841831d5e437474d : SUCCESS in 1h 35m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f93d48d74814af0a328b6c413065558 : SUCCESS in 2h 01m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5767fa8199ce46e3a266c924e3751d4d : SUCCESS in 1h 52m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8e871d4d91034d868b064d1fa1bd571d : SUCCESS in 1h 56m 43s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"6ad160df2eadd5bff618995220276215f37039f2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-27 20:37:31.000000000","message":"Patch Set 40: Code-Review-1\n\n(22 comments)\n\nOne thing strange is I couldn\u0027t review any ip_advertisement files, kept getting server not available just for them :-/","accounts_in_message":[],"_revision_number":40},{"id":"6bd471d84e53b9f19602d537b7ac7d6c75c7de77","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-28 15:14:35.000000000","message":"Patch Set 40:\n\n(5 comments)","accounts_in_message":[],"_revision_number":40},{"id":"b457f018d015df865d0ccf73cb61dcb4b80f9d8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-28 21:49:47.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"2e6c72cff325426a019b60c1cbb509716ac6e278","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-05-28 21:49:56.000000000","message":"Patch Set 40:\n\n(27 comments)","accounts_in_message":[],"_revision_number":40},{"id":"b7453b0c19d1c85dc84541a7bdee72cfb4bba78b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-28 22:28:51.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"394c4cba986c4231f684ce7d58696a4c63a9d2e1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-28 22:33:14.000000000","message":"Patch Set 41: Code-Review+1","accounts_in_message":[],"_revision_number":41},{"id":"fdaf164291149557bd76d58fd7b5120a2ff6981c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-28 22:40:36.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"8598e775ecd9cdf2b3b34ea02f446938ca343e9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 00:01:00.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5995333c07d64d63aa5f5ecae83332dc : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6891331c1ea247aea633395f7c30506d : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dcbc7d247d4e2da435cb97bb29f8e2 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b719523e92a4f1d92b1f11309d720f8 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffab22ee483849ec92c038594a7a6fb3 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031e3985f8ab4ebba8aaad30613bc431 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86f6b5cbb3014bdab2c93c210ebf0936 : SUCCESS in 4m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d84b66a42cad4d00a4b90c03362a17ad : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6c3ccf55a9fe40c4957cee7e4c1a8e34 : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddb0c1cad2554ceb889dbe3077cc00f2 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5f44ca6597c4346be5147243baa57bf : SUCCESS in 40m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b9ba76d58ce4b1faf55604ed1274dbe : SUCCESS in 1h 39m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0cdc6a25ce82499e9adc68e086a2f3aa : SUCCESS in 55m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/00c54628aa8d49dcab211a513e663db4 : SUCCESS in 49m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bff8592eebb74ab9a68e8bb8f419dee3 : SUCCESS in 54m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fb69e5214d64303be979b3b916f9275 : SUCCESS in 1h 27m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b90c8fda4d504728a1449b9d0f825bef : SUCCESS in 2h 03m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eef8157909e4424b850c6c88ba68b663 : SUCCESS in 1h 46m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10f9d2d564ba4abb8eddd82c00d94750 : SUCCESS in 1h 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"9ee9f6115b30c917f540c78ba929e05af662bb1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-01 18:22:11.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"77668e7ce3e724c7f3a28e186341b89858d83e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 20:35:15.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b31e3824a784cd19e3da9e4b9cf1690 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fb984df108f49a68ae74862956fa90c : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cccaa7197e3f4216aee87e2cb216494e : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17e79455f584e299f30254abd8784a9 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb807280f1af4fc587dc53d7f2d7c958 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed20ae286f0349a0bf1043599585e34c : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2f0d9d62d0433f857d2e332e16c588 : SUCCESS in 5m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d727ac7ed3144c1caa1eb37f00abf307 : SUCCESS in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/405186eef3d74a20b9c2e130dd292bc5 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71c02b821b0a45db994d8844430084bb : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b38984af277640ac9e1bd684ed1d933f : SUCCESS in 43m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/263e69b9e5d6441da50a4f042a2f811b : SUCCESS in 2h 06m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b60d96f561ca4e5593c24e52cabd4430 : SUCCESS in 58m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e1a2ba115c82495faa2facb71b84db41 : SUCCESS in 49m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5a73c8086bb407486c1321b4c202275 : SUCCESS in 1h 03m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ec060751b36f4446a7d414de924033b8 : SUCCESS in 1h 29m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/888d4f33819f43e2b63fe272ad373dd1 : SUCCESS in 1h 45m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c50fca718cf944bc87a8948c3c908bcc : SUCCESS in 2h 01m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fbaa226f0c4c45d894ec6abdddfff763 : SUCCESS in 1h 54m 08s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"0c4ff16df87c592ce6a518fd70db31b0c12991fe","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-06-06 06:13:15.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"7914659698a952588df18be680f42a395ac9b417","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 15:59:21.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"e2da92c9c36ca4dd6bb6292047ce1f9b45a6d16b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-08 16:02:51.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"ea9cc77a32db7d6888753ff6e36613ac3931bbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:17:31.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cff78f2e3ee4d8284f2073b22a620d3 : SUCCESS in 10m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2136d59b1ab74c34bd683a4bd913e7db : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c79d6d815924e54941790b44ad227f3 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6401baab6664ce1b99b52ae0aaa91f7 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c15a087d3d042bd807c8b7f7516e723 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4dda9d064e4f94ad3f53f2e9f95dda : SUCCESS in 14m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/139edebe2f3f46ecafd72665e4a2dda4 : SUCCESS in 8m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/81e192d9ca864c1fa278e042e61ac71e : SUCCESS in 8m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d8e0cc3664a7403eb2a7cea5a25dd560 : SUCCESS in 10m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/193d951130cb455c9d75ce9d074868b2 : SUCCESS in 8m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c23c6c19c8f540fda0ff550c44e2fa90 : SUCCESS in 54m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f66417a62cd414ab138973c72027b4c : SUCCESS in 1h 08m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1039206b04ac4b1e9d4d92e739d5b930 : SUCCESS in 1h 02m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c46e3cd198d04a429b4414ce9a353eec : SUCCESS in 54m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a9b820a1da9413081091d705a368a86 : SUCCESS in 56m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d08b0509f7834e04a3e83cf97b022c3e : SUCCESS in 1h 41m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3e2377b9b7ae453b85982ab676e5473b : SUCCESS in 2h 09m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/750b24a9b4f845c69f345c4a83bca6fd : SUCCESS in 1h 38m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/97b04c61f67d4eafb5092fd82d940c22 : FAILURE in 1h 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"2843d898acdb157b068c6bda81c447776a17daff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-09 16:33:32.000000000","message":"Patch Set 43:\n\n(16 comments)\n\nNote: uncomplete review. I have draft comments on older patch sets, sorry.","accounts_in_message":[],"_revision_number":43},{"id":"4a92c0f76c78ee016af66366bf5b1c70446ac628","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-09 21:14:14.000000000","message":"Patch Set 40:\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"39299afc0509890f0da81b7dd3c5b549c408d986","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-09 23:15:09.000000000","message":"Patch Set 41:\n\n(12 comments)","accounts_in_message":[],"_revision_number":41},{"id":"4a9c0694b97571c04b7ce8dae34e9014f773838b","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2020-06-10 05:23:15.000000000","message":"Patch Set 43:\n\n(6 comments)\n\nGenerally speaking look good, but, maybe only me, have some confused places.","accounts_in_message":[],"_revision_number":43},{"id":"57224c55e53349037b5d050b696587b5b1105b23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-10 19:00:51.000000000","message":"Patch Set 43:\n\n(6 comments)","accounts_in_message":[],"_revision_number":43},{"id":"320019e42c3939ebc52ce4813c4fb884956a93d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 01:52:47.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"5721fa5b9c634921c62e0567aa154e6f47ac87a5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 01:53:31.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"3cdbb6f1501da0dc6d5c359890b051027b30723f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 03:56:39.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7b16172f8c9478697db811897376d0d : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebe395d7f1db416ca0856db3149e4cbd : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8687fbc76d674ce8ac3c61a18a92edf7 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/313639134d8f48d5a17f3eb3f085b437 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46de4fb32dce4beaacf6a16eaf784d43 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469335d8ef274ef5879a50dce6300065 : FAILURE in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5cbd96aa2d14eb4bc79d6a766f8e721 : SUCCESS in 8m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b7dd45863caf48cfb011871d8344f103 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/546fd53e85d74d3da4e12d076e856d5f : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5d1f731324b4ff09d1b582a1cb4a36a : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ec78169e5764ba688bd382fdb767766 : SUCCESS in 44m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7950504e4e24ea7b26eef65866901d2 : SUCCESS in 1h 57m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9aed7277a9544418c019642a91e0273 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64c83bacc33a48c5ba4a0de4b96ad660 : SUCCESS in 50m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2df309de4644b119cf2f05761ce8e00 : SUCCESS in 1h 04m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40eeff327c5540a493a1f775c4eca654 : SUCCESS in 1h 41m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0879cc2c3c3e49b48613698016456756 : SUCCESS in 1h 43m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c11e40a9e8a429e89bf2905d746cd27 : SUCCESS in 1h 42m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfcabcba94ef475f8ca7cbf521da9cae : SUCCESS in 1h 06m 50s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"bede2929e0e7314df05a0a8c0dbb63bb505a462e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-11 16:44:18.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"63acd944edf9a3fe692c84650e6112364f2037bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:49:56.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00834d3e6ac1443e984f51a1e794c1ca : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4c9c429ea044ff48f21c72034de320a : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea5a0874500464980e4543db2531c4a : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d20e2853e6c74707928f3c74c68a6758 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee446c56981445a79de80d2ea153f95b : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba92227a7a243b7ae6e06240d0c6096 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a4c8b8ce84b4a29bb66e1f2e8982dcc : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/daac0abc0dc74d70a4ebafe5f06063c6 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/47cd8eddde964e5c83d5e45ef4b60d09 : SUCCESS in 9m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b2f9f50ebcc4f9a94fe48b60feba2c5 : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92d178d818dc41ae8b30e15feaf4bd72 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f1ff31a162448ebbf383f7f2b4c7c1d : SUCCESS in 1h 56m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/407daf5be0324be09581e54482a72541 : SUCCESS in 1h 05m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf4fd4b3d26749d18a38ba43ad23be8e : SUCCESS in 56m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53f7dbc308f747c6aaecacb75c4ba0e8 : SUCCESS in 1h 07m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5a6c3af9f9a4500bf787c3123f10577 : SUCCESS in 1h 50m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/daf0af9fe62546d4a92c7bbd45a400fb : SUCCESS in 1h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2dbd5f2af6e74b9296b5577de95c6c9c : SUCCESS in 1h 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/23715446e5ff4b6190e12f78c550e9cf : SUCCESS in 1h 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"e5c37ee04b0b3b1d35560c4884a1f455a07593f1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-15 18:38:12.000000000","message":"Patch Set 45: Code-Review-1\n\n(19 comments)\n\nReview part 2/N.","accounts_in_message":[],"_revision_number":45},{"id":"dfb9f867e415ee996763f052e02ac7a758c08cb5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-16 00:14:59.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"36a7b80b36d641699c2eb988d9e9e9a642fcc291","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-16 00:15:08.000000000","message":"Patch Set 45:\n\n(12 comments)","accounts_in_message":[],"_revision_number":45},{"id":"64a9bebc9e5a92fb419cdd2257a19ddb4caf9838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 01:00:15.000000000","message":"Patch Set 46: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4823d4736bb498f8b343514b12d23fc : FAILURE in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59fd1a2c57f48f1a5f01738044ac231 : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36fccb916e24e4f949a953d8fa47e63 : FAILURE in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705b0f55b97c4e24a297bd1127495164 : FAILURE in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd65a6f2b075480caf8aaa83c2d16d48 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e4e83f2d714d0da272328220f759df : FAILURE in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd5e25b9fd9c499d84a9ad8640d45bb5 : SUCCESS in 5m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d3fa3e2981b74e15a2e117b3a294945a : FAILURE in 7m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/832c1d3f285648c29f7ff7c83392572e : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e34d9736d54c4830b9259b95cdaa0ec0 : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02c523c77c4648bbbed96948f81f7290 : FAILURE in 18m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a211af404a8c4cb0bc5d9e9705a79207 : FAILURE in 27m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e11cc97e94b845a2911191dd03adca31 : FAILURE in 35m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6e1190dc18fe4df1a3aca860235b361f : FAILURE in 32m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/147fd2d01abf4d9caa5c69895d959636 : FAILURE in 22m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb25238149bf40849127223fe3673e36 : FAILURE in 15m 05s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af18c7a34eb34b9e9712b9a1aae3fab5 : FAILURE in 39m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef64739cbd044d21b6117ef9336a9366 : FAILURE in 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/538a481171264974867fa9f402f98e6d : FAILURE in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"99fa43f9450a38a4a5052973d825f06c5f8b248d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-18 14:14:53.000000000","message":"Patch Set 46: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":46},{"id":"54ac0fd0db862e848a53222c998681e466e952f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-18 23:13:14.000000000","message":"Patch Set 46:\n\n(12 comments)","accounts_in_message":[],"_revision_number":46},{"id":"32f659e35ab73363b561a818b9f742bd77b5eb26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-18 23:25:24.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"ff0501913dc343fbfb5a2dd67bdb7c6b53d6e61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:38:57.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c03c5b209c864fbaad376564be53a79f : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0ec36aaee3d40018a02cb8eb8d0d1fb : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4d46f2fd334eb99b254534f77140f2 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf980e9b6ab48218ee2f817a92f83c2 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f85f0d4109234726b73aa83c96fd584e : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856b8ea09da743ddbdf886d3fa062a19 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ce4e08628a48d48f287bf36d78f1e9 : SUCCESS in 6m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9881217ddc914e3f9cff8f0e9d63c34a : SUCCESS in 8m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ca293ebc696949db91946c076ffa6e67 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dcbbb52c0c3340a1abc244892d5cb4b1 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/163b448514bc45db930d0eb722ad0d12 : SUCCESS in 49m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11255bd95a034a0e991729e3ed5aa2f5 : SUCCESS in 1h 42m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b070955f6454583a52eaaa187609391 : SUCCESS in 1h 30m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/eab7e16f1ecd4b5da31987fcaacda173 : SUCCESS in 1h 23m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28bbb24b697b4f519baf7e06a2d922d3 : SUCCESS in 1h 28m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7d10d15f4c541cdbf3c9235562a77b5 : SUCCESS in 2h 11m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3adc0553edf64ba5808857fa388fd3af : SUCCESS in 1h 55m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d646f2805ef442e6a1a51372b8dad559 : SUCCESS in 1h 57m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8655a62bd20141b0bf37b5b52db4c184 : SUCCESS in 1h 47m 39s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"83abe3cc4839e931192ee9a0709b2dc8426ae1ee","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-19 07:19:43.000000000","message":"Patch Set 47: Backport-Candidate+2 Code-Review+2\n\nLGTM. Massive thank you!","accounts_in_message":[],"_revision_number":47},{"id":"8906280c83c6018225396d713917ceafee489559","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-06-19 08:59:48.000000000","message":"Patch Set 47: Code-Review+1","accounts_in_message":[],"_revision_number":47},{"id":"e70ccf79aa1fe76790910ef90b5edab5bed57431","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-19 09:53:11.000000000","message":"Patch Set 47: Code-Review+1","accounts_in_message":[],"_revision_number":47},{"id":"3aeb6bf48d7c0da4215a8c8104bfe79c5cd520f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-06-19 20:30:18.000000000","message":"Patch Set 47: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":47},{"id":"ec12a7404b1966bb77e7e72719438e8575d4c035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 20:30:33.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"573f7aca2de41d56a013448a85b9bcb3173c3bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:58:31.000000000","message":"Patch Set 47: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5a68758e0e043f4b1b8fd2a38ac66f7 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4ffe17b1274a85bd03bdeca6e37207 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92bd0e149ad6473a80e6944132118f0a : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d0922c086cd4bbc8a42c932b13ce671 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40bf40b552ef42438b35a17ae10a5bbc : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6bd0141d90f49739246cc8ce9eaa489 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79467c1505794349bca6f74401cacde3 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87549721ea804c3c8b38738a45e88171 : SUCCESS in 52m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d444658f2cb41e3bef7a7a1b97a0815 : SUCCESS in 1h 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/458f1512b296480a91a67b73d969464d : SUCCESS in 56m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a2db9aca2f1e43359cf4358a6fbc72b8 : SUCCESS in 56m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fadd482633dc4b5198f05420e4e7536f : SUCCESS in 32m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8226f654496841c49e27671674f0c1cc : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":47},{"id":"68a2de9b8d576d33e2e21cc55c792b2bd46640a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:58:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":47},{"id":"0b779fa1274a6fe70ea8c2b7a2edf235b579c0b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 22:00:52.000000000","message":"Patch Set 47:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c833719ac94ee3bf86c9130d1aeea0 : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6425cd423b2747e4ac191f62bac10eaf : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":47}],"current_revision_number":47,"current_revision":"955bb8840616d96ed74de3086f8959ad4190a472","revisions":{"498d382674622c0d354b4c55ab1034e99279e857":{"kind":"REWORK","_number":1,"created":"2020-02-01 02:11:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/1"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 02:11:49.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail, amphora failover\nisn\u0027t done, coverage is horrible.\nDragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/498d382674622c0d354b4c55ab1034e99279e857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/498d382674622c0d354b4c55ab1034e99279e857"}]},"branch":"refs/heads/master"},"3ba1f07619973c5266997e692739903224b5a7b4":{"kind":"REWORK","_number":2,"created":"2020-02-08 22:07:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/2"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-08 22:06:08.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ba1f07619973c5266997e692739903224b5a7b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ba1f07619973c5266997e692739903224b5a7b4"}]},"branch":"refs/heads/master"},"e91361ca494ad1f7e9b6aa8ea52d903de96793ea":{"kind":"REWORK","_number":3,"created":"2020-02-11 23:17:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/3"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-11 23:17:34.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e91361ca494ad1f7e9b6aa8ea52d903de96793ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e91361ca494ad1f7e9b6aa8ea52d903de96793ea"}]},"branch":"refs/heads/master"},"8fd960146dbcdca6cee4a06e969ca19bb39e1f9d":{"kind":"REWORK","_number":4,"created":"2020-02-11 23:22:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/4"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-11 23:22:49.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fd960146dbcdca6cee4a06e969ca19bb39e1f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fd960146dbcdca6cee4a06e969ca19bb39e1f9d"}]},"branch":"refs/heads/master"},"bc7a76cc3fe643f3eb39b57d109fbfb845b44107":{"kind":"REWORK","_number":5,"created":"2020-02-12 05:51:01.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/5"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-12 05:50:51.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc7a76cc3fe643f3eb39b57d109fbfb845b44107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc7a76cc3fe643f3eb39b57d109fbfb845b44107"}]},"branch":"refs/heads/master"},"7b83a3acefe3d062ff383b377761f23b1083c43d":{"kind":"REWORK","_number":6,"created":"2020-02-12 06:06:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/6"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-12 06:06:05.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b83a3acefe3d062ff383b377761f23b1083c43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7b83a3acefe3d062ff383b377761f23b1083c43d"}]},"branch":"refs/heads/master"},"8461f7847fa1702229b5a86cb86f7079fa166066":{"kind":"REWORK","_number":7,"created":"2020-02-12 19:49:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/7"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-12 19:48:51.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8461f7847fa1702229b5a86cb86f7079fa166066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8461f7847fa1702229b5a86cb86f7079fa166066"}]},"branch":"refs/heads/master"},"0ef89ac4687355468edbc14b7ce62ebda1d27569":{"kind":"REWORK","_number":8,"created":"2020-02-13 02:06:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/8"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-13 02:06:24.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ef89ac4687355468edbc14b7ce62ebda1d27569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ef89ac4687355468edbc14b7ce62ebda1d27569"}]},"branch":"refs/heads/master"},"033f1dc0deec76111cf1f3effb1ffe38bb401450":{"kind":"REWORK","_number":9,"created":"2020-02-14 19:26:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/9"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-14 19:26:44.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/033f1dc0deec76111cf1f3effb1ffe38bb401450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/033f1dc0deec76111cf1f3effb1ffe38bb401450"}]},"branch":"refs/heads/master"},"d432c6896ecfdb3b03b3df598c46e0fc364ec3af":{"kind":"REWORK","_number":10,"created":"2020-02-21 00:51:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/10"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-21 00:50:16.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d432c6896ecfdb3b03b3df598c46e0fc364ec3af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d432c6896ecfdb3b03b3df598c46e0fc364ec3af"}]},"branch":"refs/heads/master"},"d4518903296629ebf77ae6154b2166e8bff64f10":{"kind":"REWORK","_number":11,"created":"2020-02-22 00:38:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/11"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-22 00:37:38.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4518903296629ebf77ae6154b2166e8bff64f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4518903296629ebf77ae6154b2166e8bff64f10"}]},"branch":"refs/heads/master"},"d45851744d1768589509abe4c2ebb71cf49c5287":{"kind":"REWORK","_number":12,"created":"2020-02-25 00:05:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/12"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-25 00:03:50.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d45851744d1768589509abe4c2ebb71cf49c5287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d45851744d1768589509abe4c2ebb71cf49c5287"}]},"branch":"refs/heads/master"},"852f817ab3aad0f12064c2c7fabf91d2f5ad7652":{"kind":"REWORK","_number":13,"created":"2020-02-27 01:22:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/13"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-27 01:21:13.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/852f817ab3aad0f12064c2c7fabf91d2f5ad7652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/852f817ab3aad0f12064c2c7fabf91d2f5ad7652"}]},"branch":"refs/heads/master"},"cad351a4c6b583a8557179b9c8c558fc8bda5672":{"kind":"REWORK","_number":14,"created":"2020-02-28 23:31:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/14"}}},"commit":{"parents":[{"commit":"4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8","subject":"Merge \"Fix house keeping graceful shutdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4bd9c65e4ac61a34e5c6348fd66ae333e6fef4e8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-28 23:30:58.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad351a4c6b583a8557179b9c8c558fc8bda5672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad351a4c6b583a8557179b9c8c558fc8bda5672"}]},"branch":"refs/heads/master"},"a9bff6bccd3c809b2d26187006faa7f056b9dfad":{"kind":"REWORK","_number":15,"created":"2020-03-04 01:03:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/15"}}},"commit":{"parents":[{"commit":"2c76209003d5ca09a302881a40a35fd14762db55","subject":"Select the right lb_network_ip interface using AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c76209003d5ca09a302881a40a35fd14762db55"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-04 01:03:08.000000000","tz":-480},"subject":"WIP - Refactor the failover flows","message":"WIP - Refactor the failover flows\n\nThis is far from done, a lot of tests will fail,\ncoverage is horrible.  Dragons, fire, and brimstone in here.\n\nIgnore me please. Just checkpointing this work in progress.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9bff6bccd3c809b2d26187006faa7f056b9dfad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9bff6bccd3c809b2d26187006faa7f056b9dfad"}]},"branch":"refs/heads/master"},"e9013688407fb35bf82c6645826f2a473eb1a54a":{"kind":"REWORK","_number":16,"created":"2020-03-12 23:02:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/16"}}},"commit":{"parents":[{"commit":"2c76209003d5ca09a302881a40a35fd14762db55","subject":"Select the right lb_network_ip interface using AZ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c76209003d5ca09a302881a40a35fd14762db55"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-12 22:57:10.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9013688407fb35bf82c6645826f2a473eb1a54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9013688407fb35bf82c6645826f2a473eb1a54a"}]},"branch":"refs/heads/master"},"81ca3a8db0b3cc9a4ba00baed12a3f1d50584aa8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-03-12 23:14:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/17"}}},"commit":{"parents":[{"commit":"801b99ed0069f438646c106389bb9b84f37fb619","subject":"Update the available provider drivers list.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/801b99ed0069f438646c106389bb9b84f37fb619"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-12 23:14:18.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81ca3a8db0b3cc9a4ba00baed12a3f1d50584aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81ca3a8db0b3cc9a4ba00baed12a3f1d50584aa8"}]},"branch":"refs/heads/master"},"7745c4a4e4fdcdbd4e908967e359c84501e1c80e":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-03-13 16:13:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/18"}}},"commit":{"parents":[{"commit":"978d57f3a41395688d3d0ff88db34f9eb9bde3b1","subject":"Merge \"Use sqlalchemy isnot(None)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/978d57f3a41395688d3d0ff88db34f9eb9bde3b1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-13 16:12:35.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7745c4a4e4fdcdbd4e908967e359c84501e1c80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7745c4a4e4fdcdbd4e908967e359c84501e1c80e"}]},"branch":"refs/heads/master"},"88208f0cc7300fdd6cb7a9766708fe927ca0fed4":{"kind":"REWORK","_number":19,"created":"2020-03-23 15:31:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/19"}}},"commit":{"parents":[{"commit":"2cf00d8c85692c571e23b221da7885dc0f179d14","subject":"Merge \"Fix padding logic for UDP health daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cf00d8c85692c571e23b221da7885dc0f179d14"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-23 15:31:02.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88208f0cc7300fdd6cb7a9766708fe927ca0fed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88208f0cc7300fdd6cb7a9766708fe927ca0fed4"}]},"branch":"refs/heads/master"},"2aed6f33749e5dc464a48b0bba91abc90ec1e6fd":{"kind":"REWORK","_number":20,"created":"2020-03-23 21:42:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/17/705317/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/20"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-23 21:41:34.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aed6f33749e5dc464a48b0bba91abc90ec1e6fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2aed6f33749e5dc464a48b0bba91abc90ec1e6fd"}]},"branch":"refs/heads/master"},"be87e2eb362fc1d40cd2b4cfb4d9f685e52ec4de":{"kind":"REWORK","_number":21,"created":"2020-04-01 18:31:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/21"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-01 18:31:55.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be87e2eb362fc1d40cd2b4cfb4d9f685e52ec4de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be87e2eb362fc1d40cd2b4cfb4d9f685e52ec4de"}]},"branch":"refs/heads/master"},"f39933bb897a2462cd337cff1e0eaead8f5cd03d":{"kind":"REWORK","_number":22,"created":"2020-04-01 23:33:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/22"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-01 23:32:45.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f39933bb897a2462cd337cff1e0eaead8f5cd03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f39933bb897a2462cd337cff1e0eaead8f5cd03d"}]},"branch":"refs/heads/master"},"704cb2755c5311688c9aaf88726efd2193d6c853":{"kind":"REWORK","_number":23,"created":"2020-04-01 23:47:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/23"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-01 23:47:30.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/704cb2755c5311688c9aaf88726efd2193d6c853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/704cb2755c5311688c9aaf88726efd2193d6c853"}]},"branch":"refs/heads/master"},"7f7840be64a01ca2ebf321715d5b75ab41c882ee":{"kind":"REWORK","_number":24,"created":"2020-04-02 00:30:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/24"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-02 00:29:56.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7840be64a01ca2ebf321715d5b75ab41c882ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f7840be64a01ca2ebf321715d5b75ab41c882ee"}]},"branch":"refs/heads/master"},"79422f813932d23470e0f8e3d0532fc640b16f9e":{"kind":"REWORK","_number":25,"created":"2020-04-02 05:29:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/25"}}},"commit":{"parents":[{"commit":"894fdcafa4ca8b5e606d457c19e59b84e68d9dc3","subject":"Merge \"Allow AZ to override valid_vip_networks config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894fdcafa4ca8b5e606d457c19e59b84e68d9dc3"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-02 05:28:57.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79422f813932d23470e0f8e3d0532fc640b16f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79422f813932d23470e0f8e3d0532fc640b16f9e"}]},"branch":"refs/heads/master"},"728c4eac24390516847c43b250d0b5bc35cf05bc":{"kind":"REWORK","_number":26,"created":"2020-04-07 21:00:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/26"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-07 21:00:17.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/728c4eac24390516847c43b250d0b5bc35cf05bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/728c4eac24390516847c43b250d0b5bc35cf05bc"}]},"branch":"refs/heads/master"},"893f2bea4a40ab4ce9a0b21902bdd8e5061e305a":{"kind":"REWORK","_number":27,"created":"2020-04-13 23:54:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/27"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-13 23:54:52.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/893f2bea4a40ab4ce9a0b21902bdd8e5061e305a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/893f2bea4a40ab4ce9a0b21902bdd8e5061e305a"}]},"branch":"refs/heads/master"},"5b8d64abdbcc34ecc0a81bc50e67fcc888ec0dc5":{"kind":"REWORK","_number":28,"created":"2020-04-14 00:12:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/28"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-14 00:12:17.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8d64abdbcc34ecc0a81bc50e67fcc888ec0dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8d64abdbcc34ecc0a81bc50e67fcc888ec0dc5"}]},"branch":"refs/heads/master"},"59fd8c90e2ac7ca5103ae2987544bc827a833c14":{"kind":"REWORK","_number":29,"created":"2020-04-15 10:50:54.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/17/705317/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/29"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-15 10:50:42.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59fd8c90e2ac7ca5103ae2987544bc827a833c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59fd8c90e2ac7ca5103ae2987544bc827a833c14"}]},"branch":"refs/heads/master"},"2db50e02506e73523f133b0d658bbe4441041246":{"kind":"REWORK","_number":30,"created":"2020-04-16 06:03:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/30"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-16 06:03:48.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2db50e02506e73523f133b0d658bbe4441041246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2db50e02506e73523f133b0d658bbe4441041246"}]},"branch":"refs/heads/master"},"8c90cb00e2511b44d6a0aa590339658ed95b76db":{"kind":"REWORK","_number":31,"created":"2020-04-16 19:26:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/31"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-16 19:25:51.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c90cb00e2511b44d6a0aa590339658ed95b76db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c90cb00e2511b44d6a0aa590339658ed95b76db"}]},"branch":"refs/heads/master"},"9a5906db18924057bd2fdd975a5d44009bed15af":{"kind":"REWORK","_number":32,"created":"2020-04-16 22:34:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/32"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-16 22:34:07.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a5906db18924057bd2fdd975a5d44009bed15af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a5906db18924057bd2fdd975a5d44009bed15af"}]},"branch":"refs/heads/master"},"49dd4be53f5e28e7d5a05db5e7f25aafef19ca2b":{"kind":"REWORK","_number":33,"created":"2020-04-17 00:42:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/33"}}},"commit":{"parents":[{"commit":"78863cf7ee27b8d4932da162c5f87c96ed3005b8","subject":"Merge \"Don\u0027t register cli opts on import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78863cf7ee27b8d4932da162c5f87c96ed3005b8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-17 00:42:31.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49dd4be53f5e28e7d5a05db5e7f25aafef19ca2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49dd4be53f5e28e7d5a05db5e7f25aafef19ca2b"}]},"branch":"refs/heads/master"},"dbffa953ffef1d69551474dff23287a31c7a1dae":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2020-04-17 00:42:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/34"}}},"commit":{"parents":[{"commit":"0ccb997793035588e9289c23fd3f2d0ec7fd578d","subject":"Merge \"Run taskflow jobboard conductor conditionally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0ccb997793035588e9289c23fd3f2d0ec7fd578d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-17 00:42:49.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbffa953ffef1d69551474dff23287a31c7a1dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbffa953ffef1d69551474dff23287a31c7a1dae"}]},"branch":"refs/heads/master"},"7484a736cd4a593ea2020e740875a6dbf0aee77e":{"kind":"REWORK","_number":35,"created":"2020-04-24 20:53:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/35"}}},"commit":{"parents":[{"commit":"c037a9b374223176248e0cbd7ce2ca382da21049","subject":"Merge \"Add Python3 master unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c037a9b374223176248e0cbd7ce2ca382da21049"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-24 20:53:57.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7484a736cd4a593ea2020e740875a6dbf0aee77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7484a736cd4a593ea2020e740875a6dbf0aee77e"}]},"branch":"refs/heads/master"},"0cf2fa72d98ff2dfb4f8e31e6c4d96758b062d2f":{"kind":"REWORK","_number":36,"created":"2020-04-30 14:20:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/36"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-04-30 14:20:53.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0cf2fa72d98ff2dfb4f8e31e6c4d96758b062d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0cf2fa72d98ff2dfb4f8e31e6c4d96758b062d2f"}]},"branch":"refs/heads/master"},"1f090bd0829be2eeac4829ddd84d67836b331f26":{"kind":"REWORK","_number":37,"created":"2020-05-04 22:37:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/37"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-04 22:37:05.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f090bd0829be2eeac4829ddd84d67836b331f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f090bd0829be2eeac4829ddd84d67836b331f26"}]},"branch":"refs/heads/master"},"bb9f1aaf838028f64a64adfbd52d95f9e5283565":{"kind":"REWORK","_number":38,"created":"2020-05-05 00:07:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/38"}}},"commit":{"parents":[{"commit":"07a93de755c0ec461f51b94e3465708700ff1549","subject":"Merge \"Fix healthmanager not update amphora health when LB disable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07a93de755c0ec461f51b94e3465708700ff1549"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-05 00:07:45.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb9f1aaf838028f64a64adfbd52d95f9e5283565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb9f1aaf838028f64a64adfbd52d95f9e5283565"}]},"branch":"refs/heads/master"},"72f38d87f92a1e8bc33f136dfc9d27f552c9a819":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2020-05-14 05:11:47.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/17/705317/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/39"}}},"commit":{"parents":[{"commit":"5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd","subject":"Merge \"Migrate grenade job to native Zuul v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ec5fb73f8cbf3b4d3b437a117159f50411ae4bd"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-05-14 05:11:47.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72f38d87f92a1e8bc33f136dfc9d27f552c9a819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72f38d87f92a1e8bc33f136dfc9d27f552c9a819"}]},"branch":"refs/heads/master"},"f8a680a96445b4610e2795d2d0c254e94bced276":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2020-05-19 09:51:18.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/17/705317/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/40"}}},"commit":{"parents":[{"commit":"9a1d6d3585070da570bcf6313e237b486bdac81b","subject":"Fix E741 pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a1d6d3585070da570bcf6313e237b486bdac81b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-05-19 09:51:18.000000000","tz":0},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8a680a96445b4610e2795d2d0c254e94bced276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8a680a96445b4610e2795d2d0c254e94bced276"}]},"branch":"refs/heads/master"},"d502d039a248ceee5c9b8686e82920f729f6e5c2":{"kind":"REWORK","_number":41,"created":"2020-05-28 21:49:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/41"}}},"commit":{"parents":[{"commit":"9a1d6d3585070da570bcf6313e237b486bdac81b","subject":"Fix E741 pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a1d6d3585070da570bcf6313e237b486bdac81b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-05-28 21:49:45.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d502d039a248ceee5c9b8686e82920f729f6e5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d502d039a248ceee5c9b8686e82920f729f6e5c2"}]},"branch":"refs/heads/master"},"c01bc8d6bc08d741e416f796cf9eae8d56a8ae2a":{"kind":"REWORK","_number":42,"created":"2020-06-01 18:22:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/42"}}},"commit":{"parents":[{"commit":"2c7c7747b74e866e5ab41139e2f3f01d9dce15a5","subject":"Merge \"Workaround peer name starting with hyphen\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2c7c7747b74e866e5ab41139e2f3f01d9dce15a5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-01 18:22:09.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c01bc8d6bc08d741e416f796cf9eae8d56a8ae2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c01bc8d6bc08d741e416f796cf9eae8d56a8ae2a"}]},"branch":"refs/heads/master"},"a1368d3ad1ce5d6c1a90645292b923b92b4ee4e4":{"kind":"REWORK","_number":43,"created":"2020-06-08 16:02:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/43"}}},"commit":{"parents":[{"commit":"c9e9fd93356f8d6839dd5aa0299f14e5bb316a9f","subject":"Merge \"Change default tox envs from python37 to python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9e9fd93356f8d6839dd5aa0299f14e5bb316a9f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-08 16:02:50.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1368d3ad1ce5d6c1a90645292b923b92b4ee4e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1368d3ad1ce5d6c1a90645292b923b92b4ee4e4"}]},"branch":"refs/heads/master"},"c683d24fea37ce82acd4fc78aee88346a67e09e3":{"kind":"REWORK","_number":44,"created":"2020-06-11 01:53:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/44"}}},"commit":{"parents":[{"commit":"be2acaeb3648f36572b5e79339bffe75c96328c1","subject":"Merge \"Fix test_barbican_legacy for Python 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be2acaeb3648f36572b5e79339bffe75c96328c1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 01:53:30.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c683d24fea37ce82acd4fc78aee88346a67e09e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c683d24fea37ce82acd4fc78aee88346a67e09e3"}]},"branch":"refs/heads/master"},"c4e15190b14bcbf5ee5cc2d5976a929427f126fe":{"kind":"REWORK","_number":45,"created":"2020-06-11 16:44:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/45"}}},"commit":{"parents":[{"commit":"be2acaeb3648f36572b5e79339bffe75c96328c1","subject":"Merge \"Fix test_barbican_legacy for Python 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be2acaeb3648f36572b5e79339bffe75c96328c1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-11 16:44:10.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4e15190b14bcbf5ee5cc2d5976a929427f126fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4e15190b14bcbf5ee5cc2d5976a929427f126fe"}]},"branch":"refs/heads/master"},"a75a2786e06f61d044323c319d1cecc6af0aaeb0":{"kind":"REWORK","_number":46,"created":"2020-06-16 00:14:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/46"}}},"commit":{"parents":[{"commit":"be2acaeb3648f36572b5e79339bffe75c96328c1","subject":"Merge \"Fix test_barbican_legacy for Python 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/be2acaeb3648f36572b5e79339bffe75c96328c1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-16 00:14:45.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a75a2786e06f61d044323c319d1cecc6af0aaeb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a75a2786e06f61d044323c319d1cecc6af0aaeb0"}]},"branch":"refs/heads/master"},"955bb8840616d96ed74de3086f8959ad4190a472":{"kind":"REWORK","_number":47,"created":"2020-06-18 23:25:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/705317/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/705317/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/705317/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/705317/47"}}},"commit":{"parents":[{"commit":"f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf","subject":"add the verify for the session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f26ab8b97be7a478fdd391c9e31a5c1a86a2a3cf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-01 01:15:50.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-18 23:25:21.000000000","tz":-420},"subject":"Refactor the failover flows","message":"Refactor the failover flows\n\nThis patch refactors the failover flows to improve the performance\nand reliability of failovers in Octavia.\n\nSpecific improvements are:\n* More tasks and flows will retry when other OpenStack services are\n  failing.\n* Failover can now succeed even when all of the amphora are missing\n  for a given load balancer.\n* It will check and repair the load balancer VIP should the VIP\n  port(s) become corrupted in neutron.\n* It will cleanup extra resources that may be associated with a\n  load balancer in the event of a cloud service failure.\n\nThis patch also removes some dead code.\n\nChange-Id: I04cb2f1f10ec566298834f81df0cf8b100ca916c\nStory: 2003084\nTask: 23166\nStory: 2004440\nTask: 28108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/955bb8840616d96ed74de3086f8959ad4190a472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/955bb8840616d96ed74de3086f8959ad4190a472"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
