)]}'
{"id":"openstack%2Foctavia~705927","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ifacbd1953c3d8719539a9f4012eb34669bb164e2","project":"openstack/octavia","branch":"stable/train","topic":"use-stable-requirements","hashtags":[],"change_id":"Ifacbd1953c3d8719539a9f4012eb34669bb164e2","subject":"Use stable upper-constraints.txt in Amphora builds","status":"MERGED","created":"2020-02-05 11:20:29.000000000","updated":"2020-02-13 13:00:13.000000000","submitted":"2020-02-13 12:58:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705927-1581598704253-7c03db43","meta_rev_id":"2311a3b68698599d606837dc320c875dc9b8456c","_number":705927,"virtual_id_number":705927,"owner":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 12:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-13 08:55:09.000000000","_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-02-05 18:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-13 11:08:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-02-05 17:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-02-13 11:08:44.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 17:06:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-05 18:56:20.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-02-13 08:55:09.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-02-13 11:08:44.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-13 12:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1ce39d82f747824f051b42070f0e7ba7214c5e7","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-05 11:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f98bcae57d44d224df0ca3dd7a7fd0421544352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:42:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d817e79643a46a7ab5b24890e6276f6 : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41239de9c61a4ace8eb5b30dcb68145f : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6526632a474d2d97353606cc25774a : SUCCESS in 8m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3469fad662b7483fa1ebe935022cb1ae : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/690f0c5c525c45bf9126cd893f3d84e8 : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3b0d888f4e441c88a928ffab650c506 : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1d4a78f37d4ec084f0698deccbf79b : SUCCESS in 16m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3875ecef71ef4250987a5a1e78c0015f : SUCCESS in 10m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/139c03c0255d4388ab3e28a361c63aba : SUCCESS in 12m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b85fc5eef83247ae9d24e3487455e671 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b93c847281c74185925fca94473e63df : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da8a958a28ff41678e3d57633072717a : SUCCESS in 58m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d77afc6495e24b01ad013e0b857ab6bd : SUCCESS in 52m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e37fc22084934502bd1aac5bd3047cf1 : SUCCESS in 2h 19m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/011cfb3c624d497a94bf71a03324b7a8 : SUCCESS in 2h 02m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8f2477adb8cf4ab1bcb264016882183a : FAILURE in 22m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17a65c527dc44785b169fbecf02cd697 : SUCCESS in 1h 24m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/56703ed87037461abb5b897a06bfb397 : SUCCESS in 2h 11m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cab86b338d0e41dfaa925af6c27df4e6 : FAILURE in 43m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70038132dd454841a33d8f93f664bce4 : SUCCESS in 1h 03m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6def038c31a84d03a3881bfeb3adb5b9 : SUCCESS in 1h 30m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e241d9a959d42809026ae102805e5db : SUCCESS in 1h 04m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7172d3d17b2640429b23ca67907daa41 : SUCCESS in 1h 16m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f5e819b41f664be8a7f4818dfd0f498a : FAILURE in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e6bf6a21c8ddddd59409e3e569fdfcb15c90c8d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-05 14:37:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62caf4d0d591b0821b78e8ebd193c39b7a54d391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:42:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6be0b1ef191c43998e03590e8d5c7adc : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0e98f54923941f195949d58d0778246 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72f2d2210c0438eb551eafc70fe5e49 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75a3fa941234446fb5b4559e081bb176 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dda597aeb0fc484e88e8573d49be7b10 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1df2158451e840578101a1805fde3903 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfc5f3188374fbcac66fc43cea0c313 : SUCCESS in 11m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a469078dab2945e2a2e55311dedb0896 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f61610bdfcb74a738867e2a66263426e : SUCCESS in 12m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f79e4fa94da44d62899e36b31fdca197 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d3d1b8f4da64d7abfd53a19822a9126 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2e3be3f025e48af90c52fd01dab9318 : SUCCESS in 54m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d3000ea400534148b90fa759c06a1831 : SUCCESS in 55m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/beeb5998bb50448ca07b79cfad42226c : SUCCESS in 1h 39m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0b10b632343745368bc358ebb9ece95e : SUCCESS in 1h 46m 57s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/17b200e2856c4510a400af95690fc121 : FAILURE in 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d74b326d571c4ac0a614ac8dbfbebc20 : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/91a626105d8b4d08896f67af5245e0c4 : SUCCESS in 1h 57m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8301a413f07d4f9eaf375b967e8de30f : FAILURE in 41m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49b7bcf2f9b14e3f8bd92bbade651475 : SUCCESS in 1h 03m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/349a9d12c2b146778d324d6bbc3ca168 : SUCCESS in 1h 39m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f477ed6b45064f0783fa0becf9c89d4c : SUCCESS in 1h 19m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/507ddac10313402b96dbe7b37eb39f9c : SUCCESS in 1h 11m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/84f49e53390c4219a35157b92005c674 : FAILURE in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbce54053d85cc93c20e9bd77b2ca02a2c959a42","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 17:03:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f95ff52a76ab1b73147f2202bcfc6e0e2f0daae5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-05 17:06:47.000000000","message":"Patch Set 2: Code-Review+2\n\nYep, that should do it.","accounts_in_message":[],"_revision_number":2},{"id":"00892f3c19f80ec6e8126fb40194301fe22dde50","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-05 18:56:20.000000000","message":"Patch Set 2: Code-Review+2\n\nSeems good? But looks like grenade is failing, we have another issue possibly too (or is grenade still flaky)?","accounts_in_message":[],"_revision_number":2},{"id":"f6c91547444ec933968174eb2a7f36a6b865d9e3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 19:12:42.000000000","message":"Patch Set 2:\n\nGrenade deploys a stable/stein cloud first and then upgrades to stable/train (branch of this patch). Because stable/stein is not fixed yet, DIB still pulls master u-c.txt. We could start merging backwards: queens first, rocky and stein next and last train.\n\nhttps://36d0834365e8b1dc2848-76fc92e5e7c4e5a091c792a95503ad1d.ssl.cf1.rackcdn.com/705927/2/check/octavia-grenade/4dbbf0d/logs/dib-build/amphora-x64-haproxy.qcow2.log","accounts_in_message":[],"_revision_number":2},{"id":"77ac42bae7dc0f1187e62304ae9a9093412fe9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 19:37:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a73ea53e72884fed98c06aa22afcb2fe : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95dea0eb90d34756b96c2ae8152bc2e7 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0016441a9424dbcbba0f1da0e1b2a02 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afcb4c38e3e54d51b72858d96a5848ae : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b26a868d020a45a98d0664a7b68704d1 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fed83c4b8d914473ba55b9cab57dd084 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106c53783a104578a7753899a597d1f2 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dcc7f9263fb4fb2be7a3e8016546e00 : SUCCESS in 7m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d10449f6afec4307ac106630f13a1a3a : SUCCESS in 7m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d3d38cf295264e1fac5a8a377e3b8293 : SUCCESS in 9m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/843c9c5d2b3546bf8207da1cf59400a6 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79d61ee2f842478cb63ff4dcb6d4c7f4 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f52cf07a4eb347c79ddd7f846d0688b6 : SUCCESS in 42m 50s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ad9a08a3ba554d828dbdd5e05926635c : SUCCESS in 49m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71734df9a4ad4b35baec92e351dc9c0c : SUCCESS in 1h 31m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/92229876c58041a08b5f63b22c990d60 : SUCCESS in 2h 30m 33s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/295ef8008a5b43afa0eb6ecff6bad96e : FAILURE in 25m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa0f544606cc4d19ad40db7d4b75eaad : SUCCESS in 2h 03m 01s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/369d594c433e4088879c56f2bfd33004 : SUCCESS in 1h 46m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4dbbf0dd73c8497b827df121fa93d7eb : FAILURE in 38m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/109802544fa743b496e6111c30997af3 : SUCCESS in 1h 02m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b7f8423a4e240e99827dc0d45b6caab : SUCCESS in 1h 44m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/051e5c0ae1064711be1574182f463ea4 : SUCCESS in 1h 03m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1d14653263c3470a8288e174970987bf : SUCCESS in 1h 05m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b5332a1921da463fb0d0adf97ab8eb12 : FAILURE in 22m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed3ccca2671a124363a79c05fe6d4d0e7d4381a7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-02-13 08:55:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9f568e37b5fd43c14702d7ff5757b158fb299f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:58:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a11c64d935489c8e67f4f58c6f7a94 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42ef332bbc644f5191de2f7140e20b73 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e52d7130f84f50b6dd2f9b46149fe9 : SUCCESS in 8m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86820da4b6be4ebfbb6782bd40180f2a : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5baeadb47904f63ba9e7a15fb6c731f : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25184b8d9e49421795954ae99ed5b802 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a63c6362254f9986f14af4d8646500 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f012538283014ae2852964d45523eece : SUCCESS in 7m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5b1c97ab70e048cab0bcd502e4a7b105 : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/77574583b66041059625ef87be5760ae : SUCCESS in 8m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a9aea64cc01647b1991223c9168e6c4d : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c5ceddc5796497c9c104afdfe46ba65 : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d11b28ef3d94456819cdfa667a07409 : SUCCESS in 47m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d0d26466a19e4830b68e4bc0bf8099eb : SUCCESS in 48m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6992ff9758e046edb8faa45eafe71916 : SUCCESS in 1h 28m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0771d84dbf0f49b5aed3ed37314e6c94 : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f417b4f98f9f4a7bb21957826621cf90 : FAILURE in 22m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/750055a41c8d4be89501ca021de703e1 : SUCCESS in 1h 55m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/88e6de18dfcc464f9eca7b22e29687fa : SUCCESS in 1h 51m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/821f136c9b6b41599e6796e612ec900d : SUCCESS in 1h 11m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a216c8f163f4074b38d6cc8a1ec79e3 : SUCCESS in 1h 05m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f01d30bed5e34b0f99e7ca5daa446f81 : SUCCESS in 1h 39m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7d52659a88d4f43be66954f7e458d32 : SUCCESS in 1h 01m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/873069e3e37b4120904e59c331c3dfe7 : SUCCESS in 1h 23m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1ef7de6c0be4445c8fb0bede23b0f7bd : FAILURE in 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"791132841d4bd7a0cb8b0786558b3857c99619aa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 11:08:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nShocking, the grenade job started working. I won\u0027t complain.... :D","accounts_in_message":[],"_revision_number":2},{"id":"940b7d0648c6f226548e594f4fc75476bba25fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 11:08:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"073e9aa5549436cb0fc78e952867ca41bc422772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:58:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f6341d29aeb485e820d74a3e6302f7e : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6d5fb738774351b2a1e55c29200750 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2af3c27570cb4ba2b3d8539e82d934f5 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b98e00e1b14e6a88b4930fceefe4eb : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3dfa0eb1e244483ca0b0737dcfef5142 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2865a89746b14e24b8c456934fd261ba : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d09b2c35b0e74d3992ceca555afd85cf : SUCCESS in 8m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5f8a7ed349604ffa83bd5aca73412846 : SUCCESS in 11m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77f9a78bdafd4076b466df7cb68fd228 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59252a7666f74bbfa95fcb15025c884b : SUCCESS in 48m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ed6310a545834714b3b29b0acb65fa11 : SUCCESS in 47m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6e99d221e8b481ca47b9384cf7bda0d : SUCCESS in 1h 36m 37s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f0aaa000d1b04f1bb9406ee17b408a84 : SUCCESS in 1h 46m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7c5405ae5df4e59bd55060c88fdc58a : SUCCESS in 1h 24m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ce1dc90295fd99307d2c55b69f2c48f7b1f4ce65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:58:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7f8c916553754c1771bb02f1f9b971e37e405374","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:00:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e524969fc2804f1780fe494197b5ba33 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff84ba482a2246d1a8bf0f004e851fae : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b033485985f5c3888a1cca996af64bb304bfe14f","revisions":{"66f6280f45f9c117b0cedb5a8d19c952db1386e5":{"kind":"REWORK","_number":1,"created":"2020-02-05 11:20:29.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/27/705927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/705927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/705927/1"}}},"commit":{"parents":[{"commit":"4eb2fdacec0420081b43f6b0f2dd18eda47d35b2","subject":"Merge \"Use LUKS1 for certfs-ramfs for all distributions\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4eb2fdacec0420081b43f6b0f2dd18eda47d35b2"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/train branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f6280f45f9c117b0cedb5a8d19c952db1386e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66f6280f45f9c117b0cedb5a8d19c952db1386e5"}]},"branch":"refs/heads/stable/train"},"b033485985f5c3888a1cca996af64bb304bfe14f":{"kind":"REWORK","_number":2,"created":"2020-02-05 17:03:22.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/27/705927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/705927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/705927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/705927/2"}}},"commit":{"parents":[{"commit":"4eb2fdacec0420081b43f6b0f2dd18eda47d35b2","subject":"Merge \"Use LUKS1 for certfs-ramfs for all distributions\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4eb2fdacec0420081b43f6b0f2dd18eda47d35b2"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-02-05 17:02:47.000000000","tz":60},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/train branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b033485985f5c3888a1cca996af64bb304bfe14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b033485985f5c3888a1cca996af64bb304bfe14f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
