)]}'
{"id":"openstack%2Foctavia~706050","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ifacbd1953c3d8719539a9f4012eb34669bb164e2","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ifacbd1953c3d8719539a9f4012eb34669bb164e2","subject":"Use stable upper-constraints.txt in Amphora builds","status":"MERGED","created":"2020-02-05 17:16:32.000000000","updated":"2020-02-18 20:07:06.000000000","submitted":"2020-02-18 20:05:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706050-1582056314308-798fc2db","meta_rev_id":"8301235d6e61986a527c08475bef0e855cf6419f","_number":706050,"virtual_id_number":706050,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-02-14 23:03:41.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 20:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-02-18 18:32:35.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-18 15:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-02-18 18:32:35.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 17:16:32.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-02-18 15:28:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-18 18:32:35.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-02-18 20:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b56f6895d1dea62fda65e9966c6825dc0675f64f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 17:16:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d97a1319a115b13de2a8a3bd8d60928819c02b60","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 17:17:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"43ffbe054ff8f6a62e951fbabc9e5fd27ad9d79d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 17:18:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18eacdd4043758826428a7ae1a2f957945f64a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 19:32:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea028a21aa245039fa180d5444edf11 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9d6f661494f48f7963bd6710447e464 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bf8214f8eba423fbc96471d0130347a : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4394332c345e4f27af9f32eb459d650c : SUCCESS in 6m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cd37f9c044b408f80123f1e48d3e045 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c93628593644e0ea53298ffa4318629 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610a2218aaa64d8f988860f214864d2b : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba5140765cc04bab821699f8fec9b759 : SUCCESS in 6m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/f585684ca03045619937c64f57785ee0 : SUCCESS in 1h 18m 21s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b6853e1244664ddca63a9dcb78787ba7 : SUCCESS in 1h 18m 14s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae398e1a0cc04185b81b94f6330673e9 : SUCCESS in 1h 23m 29s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c71cb4f831c345cdbb1b072e00927835 : SUCCESS in 1h 27m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fbe33f841d8145098723eeae2d7ff3d5 : SUCCESS in 7m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/982bd04982094de58f0e755fce261c75 : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d876da68ced4f1e85dfb1f3073d53dc : SUCCESS in 8m 46s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69deefa768ae41ab859023ed4aacbdc0 : FAILURE in 1h 02m 02s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4ec7870cb5ad437eb8d3b739baea0f04 : SUCCESS in 1h 06m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfc38f92dfba416eaf112b0140054d45 : SUCCESS in 43m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b2a304a3058d4b79a41492890142427a : SUCCESS in 51m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a11e6fa3627a4fb29a7c0f903778519e : SUCCESS in 1h 35m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a99e565cf10542b8bbd15aaa9c7d82d5 : SUCCESS in 1h 37m 01s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/75ef662e55dc48cebd84af259e90738a : FAILURE in 20m 30s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ce3d0a1ea4245c6bae0b51e472422f5 : FAILURE in 28m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b18bfa87cb07423c8f9ea05796db1b8f : FAILURE in 37m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6bf2bd2b74443a0b1eaa17e1312d0c0 : SUCCESS in 1h 55m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa566fc8a5524c61817e592001bfbb36 : FAILURE in 2h 03m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d942c662a2844141af1121ad1d6c243a : FAILURE in 47m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57fb8f901b0941868cc0fb3c3da22969 : FAILURE in 1h 06m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99d46d0293420d41f5eb6001fa03cdea04450ef7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 11:09:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"019a72e6c1ed3789d573cb45cbe20f941b197356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:13:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/418329e85a0c49f0b82c0c6ab523e589 : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b66450f3283f476c9a328ab633a02cc8 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add5f790da2040b39147d66e86fd2232 : FAILURE in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/736590768840484eac71e0e07d47a228 : SUCCESS in 6m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca98dc4bfa1a445ba2f9324999a74487 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/426fdf7dc91441a2b880237e4a690d45 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2af63da41c0479c94256bf0ab871c4b : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/272e3f94f2bd4523a84bec1c623bffe9 : SUCCESS in 7m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c1b677df048e436a88f2486a6a614727 : SUCCESS in 1h 16m 50s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/092a30b59d7c4fe7bd57050af75395af : SUCCESS in 1h 18m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c534a615fe6647cd9968e76741ebb25d : SUCCESS in 1h 14m 13s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e4b52c773af24889b1698843cb3df3c6 : SUCCESS in 1h 18m 17s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d0500ea697ec4bca803ff11595f1810a : SUCCESS in 7m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fb491de3232743a78b6b26888ce34d42 : SUCCESS in 9m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b70666bf811a42dfb899a5b86ba8989f : SUCCESS in 8m 53s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b24fbab398124da592ad2011fbf1f5ea : SUCCESS in 1h 06m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0ad61873818f45bfa0d5bc6ebf814b42 : SUCCESS in 58m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/780825eac0fb44618db592ae40289a64 : SUCCESS in 46m 34s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7b746521573c454e818c71853dd12387 : SUCCESS in 43m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac2fd15748b5450f82a47ac7140c0276 : SUCCESS in 1h 33m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/307c4bb5669741c0890d664a4e5c2e82 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0967bfce0fa5471691f679a8d2516c0a : POST_FAILURE in 1h 57m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/91a6573dbc2141f39db9db6cfb56ff34 : FAILURE in 19m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56acc056694c4609912e779f270a4f8b : FAILURE in 45m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/831c852459794f76a83f8bbf46d0966b : SUCCESS in 1h 23m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2d5068749b04defa69d689380da2b79 : SUCCESS in 1h 37m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95fbab699e6d42d48fc2942a1336f6c0 : SUCCESS in 1h 10m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09b0506320b548e09b62dccab9b7f2cc : FAILURE in 53m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc6e669f2e12433cb8e98c0f7a0a9156fd103704","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 17:19:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6d4dd20d4d364769ba25aa16bdab248b4626379b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:30:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9d3ac8263a345118f3be82d77c13b85 : SUCCESS in 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0434fa0526504ed591782fb8e67bc0d1 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f40b30975646988f6cea1ac56fe001 : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66ff33e8486a407780d3dbce589478ef : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c87cf33b21bb4630b458a7fd6725c896 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21ec2372d6fe46c4a35789f4bf28eb0f : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3762bbbe835f433dab063b51fe6fdbea : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7584e2e23d9a47a49d153705e79cbefb : SUCCESS in 11m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/a6062d49633d4ae79c4062f828ca2a1e : SUCCESS in 1h 45m 40s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f06008739ed34304b9c9f46bc95202ca : SUCCESS in 1h 21m 33s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5f262f3bed64e818ef2e58ccff0f2cc : SUCCESS in 1h 27m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f2028b8976f5466db3c39aa4a1fa22bd : FAILURE in 12m 18s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/42f7708c7f9c48ca945d430a03703bd6 : SUCCESS in 10m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8a25077b80f84d2ab2de364477d3e352 : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5aab04ecc814b50ba09ad413e4acaeb : SUCCESS in 10m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71ea2ec3ad474764bcf681a50695f35b : SUCCESS in 58m 43s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/84143834e6d64808a923288704aaa98a : FAILURE in 57m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2f7da58cf694edf8881469e60d870f8 : SUCCESS in 1h 12m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/de3bc92b5c124a85aa350215d9c63dc1 : SUCCESS in 1h 03m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfdece8a1c434a95bdf1882b2fdfefd8 : SUCCESS in 1h 33m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/83bea171c8fb44b2b016966d0fa2ca0d : SUCCESS in 1h 28m 52s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/28639533f03c43559d48914592c1dc9d : POST_FAILURE in 1h 44m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/663afc8b33ea4213a5283179572a8899 : FAILURE in 24m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1ffda2a071141d0a4aaa6503cfb4f8d : FAILURE in 1h 00m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c59a93693ce54f9e851afcc5e2f97784 : SUCCESS in 2h 02m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d516413b20848dfba744b624e733537 : FAILURE in 1h 41m 01s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1dee4d7d2be458eab844604ea8c17be : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3fc6259076a4b3b9c9cd63b844e3a6b : FAILURE in 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4743739829853338992df07bca8483c16044ca0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 20:21:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2aae6d247024335a87634b7841c7b32ddcad48fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:14:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070380638c4e4ad08ca667ed76d56f41 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ef2ac600acd4d29b031c6418405f550 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d7cf02eb1f489bb1140acd83f4c641 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/106cc730f51c4251b79d15223a0f725c : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/111486f48dea4093841d39ff99753b58 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18ef172f92d6434a830701261a22b5c5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c2f9b13841471cb0c641100d7998b2 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b26b50178d047cba9074098fe152d67 : SUCCESS in 7m 02s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6ce82b7d8a91409aa2684a32acb129dd : SUCCESS in 1h 19m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8436299bcd874af08d5478e5886f8720 : SUCCESS in 1h 17m 05s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ae30e9f93d84030b48ee4b0751801e0 : FAILURE in 1h 14m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bb188fffd039423b94cd646404f3f387 : SUCCESS in 1h 41m 20s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dec499d1daf04e14b1a4558d3e12f147 : SUCCESS in 7m 03s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c25fe605968e4c13ac07f842eae29c07 : SUCCESS in 10m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46c420a710304e0c93625072d48b3704 : SUCCESS in 7m 23s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/093c3d3e79fb4134910a57c650e43a61 : SUCCESS in 1h 15m 27s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c1998f71c6034a77a41e1cda5b03c68b : FAILURE in 55m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/011a5fa7e8c9457eaeb00e9eae5d24ca : SUCCESS in 43m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4d2dfafc334d4cf98d45fccc253e6a9a : SUCCESS in 41m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f44fd2d4e6774f93ab01806d1f83fe48 : SUCCESS in 1h 19m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2f6e599ac7624517b9f8dba441171ae1 : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/baf0805528ab403a9396054a5db694a0 : SUCCESS in 1h 34m 36s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f6423cd5099b45cd91898c18cce18397 : FAILURE in 18m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b42adeddb74216827feebb78e476ee : FAILURE in 45m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42ee1a1e5fcb48d9887dea2f7b190bf5 : SUCCESS in 1h 12m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/758bd5f49e9647829cd82d85f2206cdb : FAILURE in 1h 37m 02s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c22826c00274417a815aaff028336e8 : SUCCESS in 1h 12m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c4fed518fa84f0d9d204fb98db8879c : FAILURE in 1h 08m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ddf6eef7c304c58079fd930d488dce075df8f0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 23:03:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a6a1c03f305adbed9d26df933547ba9af9bc6745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 00:48:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb209ffe626d467d90d5e81890828d7d : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c15afd1c4143399e4194d3c011275a : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c403e62f72e241c9a173b9a748804581 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d0c6d25cc4640359f5660a84ad5844e : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66cbef08e533402cb184c4dea7850539 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f08d4b363747039394d97bfe1e4b1f : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b839ebd4ae304e95ac6e827256cf365b : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a01a64721e4842dfaea4317a338606d5 : SUCCESS in 7m 11s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/65c796ce84d74f85a1cd269983338dce : SUCCESS in 1h 18m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/7c22f0ecc2434d32b176cd684f7981a9 : SUCCESS in 1h 30m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f70ba62ff6e4a119c4b69bee5ced787 : FAILURE in 1h 33m 24s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b4ceb1136b764708998c564105ed6a54 : SUCCESS in 1h 34m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e234b2c1d5d44283a4e773356fdc284c : SUCCESS in 9m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1af12f22a088462ea97af733a97f1924 : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b61590118bc645ba91a02266569b0429 : SUCCESS in 8m 54s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46a7093ea98c495a933999fd9ae03ed5 : SUCCESS in 1h 01m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e2070a4451864a2988625dfce7fe69d3 : SUCCESS in 1h 02m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/090380e6106a4710b695f7e2f0a3d77d : SUCCESS in 49m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e5a0bdad424644e3bd6943339aa95b71 : SUCCESS in 46m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fb76064c91a439ba095aef677fbd651 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1899ed89ba104514bbdc46d3edf9bfe5 : SUCCESS in 1h 35m 00s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/02ddfcca931345e289aac28359d592cc : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/13091bc8b281498ba455b3482c073a20 : FAILURE in 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82b2ac0b431b473daef758617e726a9e : FAILURE in 35m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bdba5842df4c421e9181fcb4d0040c11 : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/478312b953d04eb1b56cc72e137ffe83 : SUCCESS in 1h 32m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ed10d21ab25457793473dddbeb77d16 : SUCCESS in 1h 07m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2d5789d9c8c4362bba83d655aec83d0 : FAILURE in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9dbbc7a2ceab1771c1cb381e95e706811c50b771","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-18 15:28:50.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"62563522d4598dc390de5881e39662805a308197","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-02-18 18:32:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4e1e184920ef23c1415983eee357606956faa768","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 18:32:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c0775277cfd4a2b86d0f58787f03d74c253f4c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:05:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9a9d40e85f219166dea653aca0cc82ed76143e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:05:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04aabb43b25442d0a6d8ca7521b7822b : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae88e04a8ca14dbbadcdff418dd6e91a : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e41ac9f7b9b4e04bbd8103c704beafa : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/151851b754c94b60988b4b67327ed4eb : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6bd83d49b434215b9c59f7c6a4aef42 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d8b4ee03724bb18198b7b4af169a53 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265e8b94098b415faa9962892aefba98 : SUCCESS in 9m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/77e1e95f76754a3194ca20a3792efacb : SUCCESS in 9m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1371617f2ed346ec9f62795d92ad4af2 : SUCCESS in 11m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70aa3d21fa544cd6ae9b418a0bbb42a3 : SUCCESS in 8m 18s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9939b0d21bd74160b51e40a79d1465b8 : SUCCESS in 54m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7240f929682b4e5fbb3f286479513442 : SUCCESS in 1h 08m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/25d35ba4ddcc4ea3baa2faa060335212 : SUCCESS in 1h 19m 57s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bd62a086a9244717aa3d5426922f5dae : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07c4c339503945d2bfa46127052b8a28 : SUCCESS in 48m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/af0c22e9fd0f4df79ba9334cb20f62cb : SUCCESS in 48m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b60300125cf54c16a8a04b7c60da85a4 : SUCCESS in 1h 25m 54s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c07da0bff3b54f81aa1b9922d67a31d4 : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-scenario-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5179d3cc31d541958f1c9c6c1ae3d3b9 : SUCCESS in 1h 27m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9f0486dd1144439db60c768f8c397280 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":3},{"id":"8301235d6e61986a527c08475bef0e855cf6419f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:07:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41546d485a4649e4b4ee9fc5d5db178a : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7ff0c85eb4647feaa92c5533f60f99a : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"24858a5af2da05bb4dd90d3b5e186a7586f730f4","revisions":{"61c5f762ca6ab7da71bae1a47b3f99232efebfb5":{"kind":"REWORK","_number":1,"created":"2020-02-05 17:16:32.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/706050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/706050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/706050/1"}}},"commit":{"parents":[{"commit":"c563d24681b814be05e2aa519a777b0247adb995","subject":"Merge \"Add listener and pool protocol validation\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c563d24681b814be05e2aa519a777b0247adb995"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-02-05 17:15:58.000000000","tz":60},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/stein branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61c5f762ca6ab7da71bae1a47b3f99232efebfb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61c5f762ca6ab7da71bae1a47b3f99232efebfb5"}]},"branch":"refs/heads/stable/stein"},"196f78387141253f5300bfd8ee809606576fc67b":{"kind":"REWORK","_number":2,"created":"2020-02-05 17:18:16.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/706050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/706050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/706050/2"}}},"commit":{"parents":[{"commit":"c563d24681b814be05e2aa519a777b0247adb995","subject":"Merge \"Add listener and pool protocol validation\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c563d24681b814be05e2aa519a777b0247adb995"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-02-05 17:18:01.000000000","tz":60},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/stein branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/196f78387141253f5300bfd8ee809606576fc67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/196f78387141253f5300bfd8ee809606576fc67b"}]},"branch":"refs/heads/stable/stein"},"24858a5af2da05bb4dd90d3b5e186a7586f730f4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-13 17:19:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/50/706050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/706050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/706050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/706050/3"}}},"commit":{"parents":[{"commit":"84020ae658ec99b5505036f0e991e42a7fbc6916","subject":"Fix pep8 failures on stable/stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84020ae658ec99b5505036f0e991e42a7fbc6916"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-02-13 17:19:39.000000000","tz":0},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/stein branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24858a5af2da05bb4dd90d3b5e186a7586f730f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24858a5af2da05bb4dd90d3b5e186a7586f730f4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
