)]}'
{"id":"openstack%2Foctavia~706052","triplet_id":"openstack%2Foctavia~stable%2Fqueens~Ifacbd1953c3d8719539a9f4012eb34669bb164e2","project":"openstack/octavia","branch":"stable/queens","hashtags":[],"change_id":"Ifacbd1953c3d8719539a9f4012eb34669bb164e2","subject":"Use stable upper-constraints.txt in Amphora builds","status":"MERGED","created":"2020-02-05 17:20:19.000000000","updated":"2020-02-14 17:48:41.000000000","submitted":"2020-02-14 17:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706052-1581702423278-18a4250f","meta_rev_id":"ebfad07c8792490b63047e1c4e1f2575025ca351","_number":706052,"virtual_id_number":706052,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-02-14 14:12:09.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-02-13 20:19:58.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 17:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2020-02-10 16:58:18.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-02-10 16:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2020-02-10 16:58:18.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-02-13 09:07:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-02-06 15:42:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 15:42:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-10 16:49:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-10 16:58:18.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-02-13 09:07:46.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-02-13 20:19:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-14 17:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93ecd97aab4d6a4daa2478ff729b70923f85cecd","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-05 17:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1a429ba5bbf410c38d8935fc0bd6d1502b8d0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:34:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71097d3b77948aaa29bbda1d6ec5799 : SUCCESS in 9m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3aaf1b8006c4c14b8a12806017208bc : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f89b7ce82354cfc8bf33235a932da66 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b06ed59af3b465eb4825de8e250ea50 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb1cb181ec274f2d9a47f3cd98b8df7d : SUCCESS in 9m 57s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6b73ee380eb04b6087db1b44ad1a0250 : FAILURE in 29m 02s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/fe79110ebced45dba2fb933f9ceb0aa6 : FAILURE in 22m 47s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a16852aef5a4bbbb65acf42fbf0c87b : FAILURE in 20m 01s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4af5ae5291514853abded558701e2ec5 : FAILURE in 24m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ac1d48342dc4f67bd2b947f4bbe33ca : SUCCESS in 10m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bc08b8b65e324e2cb73678632e66c7b5 : SUCCESS in 9m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5ae2d91e0d8842b28f41c4f7a208e311 : SUCCESS in 9m 10s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38963576de044690a8b51793893ebbc7 : FAILURE in 43m 31s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fd432d3808754b84bd676c6859f6bc72 : FAILURE in 23m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/adf8837cb7594b38932d5a6bb97c2426 : FAILURE in 24m 11s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/e7b1609c6915459d906a7b711cfad674 : FAILURE in 22m 47s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4f5ea68a7c944285883f4f8ba3d648ff : FAILURE in 23m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3eb218c1d8bc4e419d5d9c8d320d2fb1 : FAILURE in 20m 41s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/781bce389ed7465c9bd9bfe71dd41472 : FAILURE in 20m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef0ced3f0fe34ba08f184c97ba1eb36f : FAILURE in 21m 34s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/66acbf6bd2214d94912c942562d532c7 : FAILURE in 20m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d77c7179f1d525d7b5a68685e80e6f176838c36e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-08 21:11:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbc3b860798bf8690da7fc36bc1a7c7de2fbba67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 22:56:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed5e14c0b194636afc8d93d7e86923e : SUCCESS in 9m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb5f04b7175449a9b4dab988e011f1e3 : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37398053fd634570903be04ee55aa76f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a076d499eb4c4105a2a4ffd7d0cfd032 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/430476cc9fe84fd6a734adfcc1557d21 : RETRY_LIMIT in 3m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/fd92b911400c46678a4e2ac8cc83030a : SUCCESS in 1h 16m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/90eb55e6a5c546bf9b5e1d110e125b5a : SUCCESS in 1h 21m 42s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3173b3c45fe64058ab66185b7a0bc551 : SUCCESS in 1h 17m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/52bfc373fdd449f0a159ca50ba6bbd35 : SUCCESS in 1h 16m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d29cb6f2490c4603acd11c1cd7e32d89 : SUCCESS in 8m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9b7ed77ba33e4d219b50229ffabf66f6 : SUCCESS in 6m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0b7f316058304db6ac35477a39f60b3f : FAILURE in 4m 53s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8241890813774ec6a594390257dc6e7a : SUCCESS in 48m 04s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2239efb93e914070af2c9851654d6554 : SUCCESS in 52m 24s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/c4a476500b3c42d7b3e8bebba684c055 : SUCCESS in 1h 06m 27s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/71512fbcabdf4f819a3d789a23175911 : FAILURE in 1h 03m 21s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/e1be860db6cf42c78edbbee2c13c71d5 : SUCCESS in 1h 03m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b32702e96cf7482da308a024b3246549 : SUCCESS in 43m 25s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/370449dba1b1426595558e1ccd8fb285 : SUCCESS in 41m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73d5cf1e93554bff821620b6da7524da : SUCCESS in 1h 34m 24s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/c2c7cf12cdec4397926a230000f52d6c : SUCCESS in 1h 39m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5606a2caeb7327acb930ec505f231c5c168cf534","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-10 08:20:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f850784c1ec043ab0e3c97b4cddc653da1ae75ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:06:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75c1b782bf949d9a56ed0f7f1106d56 : SUCCESS in 13m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1ab6d551d5e4558906b1ef27606334e : SUCCESS in 7m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0591b9ae86eb4b51a4f7a4087b1f8df8 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae4452f5d194405af7cf73d62656e27 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c51149715a44ac78457c51e51f0e097 : SUCCESS in 8m 51s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/626855f400324221916051fd51201ec7 : SUCCESS in 1h 16m 51s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/f9c1e145b5d0466992782ea7206ad2b7 : SUCCESS in 1h 18m 44s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e40c525949614589a49621dbaba616cd : SUCCESS in 1h 33m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/10842e2b83504c27ae9cedcf42d6c64a : SUCCESS in 1h 27m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b942900f78ff400d948a776134217f39 : SUCCESS in 10m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4e0ec7926be44668889569937c9225ef : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/219e5d8aee5a4390a254a50190552e4a : SUCCESS in 8m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1edfc3bdf4d049f4ab0db42fa01a0bf6 : SUCCESS in 57m 20s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/63109a0ec1e24da197f1cc8a1fd13419 : SUCCESS in 56m 51s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cb0d0ef8675d476884d08734a563d7eb : SUCCESS in 1h 07m 20s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/2dc1c73eb70940dd80b475c0d7c8cabc : SUCCESS in 1h 04m 33s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/7b26da333df34040b573797adf0bfc3d : SUCCESS in 1h 04m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b59608ac1d5f41e183e964af2cda91ed : SUCCESS in 43m 36s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/cc6d0f713ac041d1b2d4ad3fd9eb0bed : SUCCESS in 49m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1244c86c82fa4ac68cd7857cdf972cc8 : SUCCESS in 1h 17m 52s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a364fbb9f9ae4fdb92bf6cb403ad2606 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7cb9a2206f104e11391f0d50cd45fecd47ad6a98","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-10 16:49:25.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"55b701b3dd071f558e41273fe3a0bf39baedf9db","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-02-10 16:58:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"305506c1af5e21862581ea5bcd5d8d1d1875664e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:58:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b366bdf06eee4f61508dd89953ea5ab14433bf7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:31:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1fb4c7385b34d64ade70bc8ddbbad2f : SUCCESS in 10m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d137bf218a943d6ad3967114b26a15e : SUCCESS in 7m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66f15508e7fe4826bed385e9a2f95f05 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6275f7eb8324e689d41a08c146038a3 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01be0606948747138be143d2b9d23b14 : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87fbec9d104749f9a42803a928dd344c : SUCCESS in 13m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dcea9294d082406da7bc9e507d6a6ef3 : SUCCESS in 9m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ae85794369e64bb189fd2bbc31b8f32a : SUCCESS in 9m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/12662bd1818c4bd4a3f9568debf46c35 : SUCCESS in 1h 16m 36s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/79e511e74c7a4a8193108ddf22ca3bf8 : SUCCESS in 1h 22m 27s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51c723a888eb4957a417ae5367d445f1 : SUCCESS in 1h 01m 12s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2e520f787d3f4e06a4f35249bb9b15b1 : SUCCESS in 1h 03m 55s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/d34cba17caee43b29d33b96d33e396b7 : SUCCESS in 1h 17m 29s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8507ba45b95a4f0bbcaea066f7e10662 : SUCCESS in 1h 16m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/950c06c76f58468abf7607d1d967d2c6 : SUCCESS in 58m 13s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/fd681c39e5774773a7ca25a388a471a1 : SUCCESS in 50m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bf602a980d34b8a85dd73cca7a0666d : SUCCESS in 1h 24m 35s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/23379860a5484cac85baf2d9ad203858 : FAILURE in 1h 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"efe2ceb17db178cfa455fcaa21938c835a132d9c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-10 18:46:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c213edddeb82febbb8a407c585073916a1a6f1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:47:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1035c073040644beb671f85dc7c6fb95 : SUCCESS in 9m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/adec684a357b42d9b40a4c0f98b5db7f : SUCCESS in 6m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3a4f01497dfb45938e460e9cdd402bae : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1885887ad774f7ca7f0231dae812956 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0729795337cc43c4adf2d5daa33f238f : SUCCESS in 10m 25s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7ad5c4d767bc426a972413cae25d13b5 : SUCCESS in 1h 14m 47s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0465d9a9d3654cd98b3f69830d2ce2d8 : SUCCESS in 1h 26m 56s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5541762625d4f798c4062a0035a8d7c : SUCCESS in 1h 20m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/60cf5cbc126e416195ddc902ff2e45d2 : SUCCESS in 1h 44m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a4255cd8e62442ea5de0f78d610fe6e : SUCCESS in 11m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/abd96b5f4cfd454fb50b44baa47b58ea : SUCCESS in 11m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/486cfa4ad78d49efadc683d59109771c : SUCCESS in 10m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3d8fb22f2514c708c9d9c5e0695a634 : SUCCESS in 1h 21m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e35e3271dbaf4b07a15e6354819c7b63 : SUCCESS in 1h 18m 15s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f45a054158cf4460934c23bc9bcb670f : FAILURE in 50m 41s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b044666c16864213acb044bab3fee4f2 : SUCCESS in 1h 11m 33s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/b39d194653ea4a13a0fa0b6f61491f8d : SUCCESS in 1h 13m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af22c8a2c39d4c2bb1c4488355a642c3 : SUCCESS in 50m 51s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e357b2324ff84730a13e8ea5b491bfff : SUCCESS in 49m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/255f7c020df4410b892e52146bd4ef4b : SUCCESS in 1h 27m 55s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/978bc1bc569e441b976657712b3d4767 : FAILURE in 1h 41m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bbb9f78f9c1dfcfcabc21f2f3d89235f1d790583","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-10 20:49:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5aa9fea9f133d554d267a36057bb561b0de1a2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:36:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d89e557511bf4c81a452b97cf6d0ab25 : SUCCESS in 11m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a46edab14304a94a6eba36017e6497c : SUCCESS in 6m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7582b35d83e48f6b7a7309c805631e0 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce8b6efd95947acb715afd022ce0e4f : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a30760a7c844e56931c894bcb5bc974 : SUCCESS in 7m 55s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b99b5a78e90e43a180e5288dd1bf0d1f : SUCCESS in 1h 21m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/04b1e4896b2f439f8938c03ab3041f81 : SUCCESS in 1h 24m 10s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/681a04a7eb0b467488450204fca1dd0f : SUCCESS in 1h 26m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ff01610102774dc8ba5992810a980eb6 : SUCCESS in 1h 14m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45fba794b57f40469318b9d1a0db27c9 : SUCCESS in 10m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d898d96d91564e8aa506baf288f4a44d : SUCCESS in 8m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0337e0331b274bd48c8fa8766d3a6e1c : SUCCESS in 11m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ee693a5a11c4c62bc794b42421aef63 : SUCCESS in 1h 02m 10s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d923a44a1a2f461da1aeb410a396e0dd : SUCCESS in 1h 01m 26s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/2cb3e54317a94603a523306df6cecca3 : SUCCESS in 1h 05m 40s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7c023ca53d8343b5b136759836e9f033 : SUCCESS in 1h 12m 58s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/06fed0955e6c48c98ce4fde4aacbe47b : SUCCESS in 1h 13m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/90306969751840528cf13d9b927ebd72 : SUCCESS in 1h 00m 09s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/6582f69dd30e4a3fa00ee2b1d87d161a : SUCCESS in 53m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be19950620a24a77929ab17512642865 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/66d9e6c498c34d31b289964c9681cb8a : SUCCESS in 1h 25m 18s","accounts_in_message":[],"_revision_number":1},{"id":"36c68fa8ba53d188f70653f83e3aa2fcc4c44df4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 22:36:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36f5cff44097da9e135788d8806c00f3965c8a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:21:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfbaf6a40a941d5a013cd9b2376d85e : SUCCESS in 10m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a46d006d3f134862bd2c0af7530856d8 : SUCCESS in 7m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/732aff0eab114e6baebb02ff31496148 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98bd292c79c4420789fffcedbedd99c6 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e42ca4f7ca45462b87879df3583abd47 : SUCCESS in 7m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eda3d3e50c5493d884f1f7a11c2ba3a : SUCCESS in 10m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8943cdcb9b5a41b3a33166f9578a9c82 : SUCCESS in 10m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/327b6ff503b04634ae8b357711dad952 : SUCCESS in 6m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ecbc7b3e72e242368f1fb17b9d0d83c4 : SUCCESS in 1h 33m 43s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/49c4abcf6024487ab1ac00506beb22ec : SUCCESS in 1h 22m 06s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6901d4921dd24ceaa3a94d496dc9c3ee : SUCCESS in 1h 06m 17s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b87c29b1cc8e459cb6b2a559eef6a4ec : SUCCESS in 1h 15m 51s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/28f3f4f08e85421888912a9c5eb05696 : SUCCESS in 1h 23m 55s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/79311348feaa423585765715e541620e : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad6587d97344410ab56c29ef13c1ddd1 : SUCCESS in 51m 10s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/2963c1d1cacd4ab3886a0c1c829416f6 : SUCCESS in 43m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd49dce02af949cda3a3f673b03c1332 : FAILURE in 1h 20m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/98c01ab34dca4b1c9d201388fc5838b7 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8fcf57710a83e97ff53ddb68c69aa3bcbcf41a50","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-11 00:43:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"617b0d8b39ccf5ee6fa1aaa4341a0682fd32cc70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 02:22:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39720c1642ea4fbba535f14bc32e211d : SUCCESS in 11m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/513e258c4590440f91dd1d63a009ea01 : SUCCESS in 7m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af5ffa2bb7594f8a822fa1af7f72fcb6 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e65dcbb1574a2ab84eab4cc61ba658 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b53107c8468498a8f9316fa9b69cd9f : SUCCESS in 9m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7acd1f561d0549c58d61277ea460725f : SUCCESS in 1h 21m 17s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ba661d289d514263a3ebe458fc02e670 : SUCCESS in 1h 20m 27s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6643ee6d48254626b9ca19c4ca259759 : SUCCESS in 1h 15m 30s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/198b868da83b4bd590d0cd5e4d5a0ede : SUCCESS in 1h 25m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7847db13b6fc4ffcb7995e62d254de7d : SUCCESS in 11m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7a5207975b92445ebe130aff13802c9c : SUCCESS in 9m 22s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cbd745a78e9944bb8d282f896f6cc6d5 : SUCCESS in 8m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52bcff18667f4882a25922d9e9113a61 : SUCCESS in 1h 02m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/31a5e26496184489acdf2df3613a8c1d : SUCCESS in 58m 17s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cb9b0fc1ce154475b32bb3203ae4cdb8 : SUCCESS in 1h 10m 42s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/58c61a2c28f347d28798ceb69682cad5 : SUCCESS in 1h 20m 02s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/40279531b0964ad1ae2391d9418267e5 : SUCCESS in 1h 10m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/372c367ff2b54d97af1b8ca58f5bd719 : SUCCESS in 43m 12s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/1471ed33bb9b46eda17773816cf4aab4 : SUCCESS in 51m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1e401f148034febb1b10a8b93a51a8a : SUCCESS in 1h 29m 09s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/463802ed01f542378c4554c8774022a1 : SUCCESS in 1h 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7cae2bb145c9c47edd93ef374225232f2656fe6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 02:22:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b5ba2c659a799f0522f05bf4d482faf98e5ee3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 04:01:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35afc8956424c17bc77c3e36d4096e5 : SUCCESS in 10m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62c0843ffe444aa59696f7e13cd25091 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0af3f93c6ecd4dcda9b34b98d6cc5482 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c2780cf8734edfa968947c778b2340 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7ea6b8005e04ad3ba6007db7d2a5f35 : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1bb459b7b4e4887873287ae875b10d2 : SUCCESS in 10m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/95408500432f4911aa560299993a62ee : SUCCESS in 6m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ca1e124cb3684915bfc23bcedfa01ee2 : SUCCESS in 9m 05s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2360a89730dd4743a9a1f77f7b8a1045 : SUCCESS in 1h 19m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/73990a4a77fb41e08c57a6afa1d5b487 : SUCCESS in 1h 20m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c38daf7d38942a09a9243d3c3729686 : SUCCESS in 55m 14s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/605dcb583e49435087c89830b8d7ff50 : SUCCESS in 1h 16m 20s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/7124f8f8b0a34670a27a1e72aa1b8103 : SUCCESS in 1h 14m 01s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/93f65a811e924f808358e8bbb2d08fec : SUCCESS in 1h 13m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6e64ff7406cb4ff5a2cb4e3389e5c9cb : SUCCESS in 46m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/626dd9c0d528491fb0af153ab5e8b8a6 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c333e7127e4494ca0f6bc3e08bfd594 : SUCCESS in 1h 11m 41s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b9945f08488a4b3b8e0c10d87aa24dc0 : FAILURE in 1h 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"acca499a07d082e856ee6fed698ac56f7808c53b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 07:04:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74f2c0d1ed50158b6ca87d75b82d9207773ac521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 08:48:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bb132a3e8a4d9db66f304bc2d255c6 : SUCCESS in 9m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5399c8370e04e529f7f7ecc2fb1b72e : SUCCESS in 6m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/418588b521cf4257bd6e0eaf7ad87eaa : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd01796c1bd4f1a807779fe38217046 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b7f846d2fde4129a204818725dbb8eb : SUCCESS in 6m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6556e30733bd432eb8f8c7eb0a76f8cd : SUCCESS in 1h 16m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/96a5ae23a5f0403e8c0bc5a4ee947cc7 : SUCCESS in 1h 18m 05s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a32bd91e51124fefa5007153e034351a : SUCCESS in 1h 29m 46s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/81316a7be81f4f9ebafd5de2d1ad049f : SUCCESS in 1h 22m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89fd2230a43a41abb49e061319bc29c4 : SUCCESS in 8m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb2cb0e0d1e447189918ad845b3396fb : SUCCESS in 6m 41s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b88b504561fd4b62a3a2ee7199590152 : SUCCESS in 6m 53s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29b6da5bc1d743349608fbcdc48a2479 : SUCCESS in 58m 24s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/21247c610c9e43c1babbc29562345f9b : SUCCESS in 59m 01s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/73a3fd954a274fcea7073fd06528c4b3 : SUCCESS in 1h 08m 21s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/8538e99cf1384c3598600348d2671a20 : SUCCESS in 1h 09m 50s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/80a0b109899840448563c9f0f26aed04 : FAILURE in 10m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e880998fe0a43379b1a0045f2397118 : SUCCESS in 43m 34s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/171372b025574f9c9219e29f84e21376 : SUCCESS in 43m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1a1e6a9701a492e90251ec37f7ecdee : FAILURE in 1h 23m 44s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/71185cbb00dc420eafc4ec62f5efc779 : SUCCESS in 1h 27m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3da796c715f9a3a8b92908345dbd302ce59ac0db","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 08:51:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d438db68d6eaeb99c26483bf33fb01dc6c01ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 10:25:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df53ad9a03e94dea99f3c666a34cf09e : SUCCESS in 9m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8c2154bc3f4402f9010c70295403330 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad8448b3555b47aa89d709b3af6db8d6 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33247eb21a924ffab437dc9a38569f2e : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c3955029e454f51a4e3d357ddbbc9fd : SUCCESS in 5m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1cc9d2d3df704b8ebbdcee7f7f0a44d7 : FAILURE in 14m 24s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9f8e37a7fe434c839d8bc9424f0ea11a : SUCCESS in 1h 21m 20s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca6e199375c94df59e52eb99def3431c : FAILURE in 13m 38s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/59b9d881f35847708dcb43cfcf2a18ac : SUCCESS in 1h 21m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d9c0663c01547f08623523ecf0202ec : SUCCESS in 15m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b864e5859c7549a68815693235f9e589 : SUCCESS in 5m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/236bd5998b02447c8cc80df10a5cccd4 : SUCCESS in 9m 19s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9249f45057ce4798a6dcd5af1de5bac0 : FAILURE in 11m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0261b7e568f048c8a3ad1235cadb4e0f : SUCCESS in 58m 35s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f558e3814e154b87b026f561c3ec329a : FAILURE in 12m 48s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d891af541f064b5190a00d816e3f8a9e : SUCCESS in 1h 07m 48s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/5c2342f70fdf435498e071ea6962b5b2 : FAILURE in 8m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a32fb30152ab405d806c6ac3fc26979e : SUCCESS in 50m 30s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ac31f7009f2d42cda48f7233349ecd33 : SUCCESS in 41m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7b5fb5601c9481288919b10d5bbc498 : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0941d26785e843eda1f3dc73879628db : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e397a9e624aceb7a5b4f17f62ce1b569b1b71a09","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 10:27:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"90eb1727738e34ed17d74beb1dc97066deb22dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 11:59:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa6d2792bc443e98518391f0c609c0a : SUCCESS in 9m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb0402b940524234a67c5c786513d4ef : SUCCESS in 7m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/128fbf3603434cd1b1930bf2f27d200a : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2057928d8dcb44fb96ef348ac22e4a16 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ea2e54d48c49348181a0725fd9ddea : SUCCESS in 6m 33s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/aff8cbd6c38b4a78b781ca2381f06802 : FAILURE in 13m 56s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/3174a0b28a154dcea651cb3786373a2e : SUCCESS in 1h 17m 57s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6f7cf52e7234169ae94619248de452d : FAILURE in 14m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/139132a0550c48cd9cc7cf0d6259c3f0 : SUCCESS in 1h 24m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81e3f8a42a4e4ade9482a08f26064eed : SUCCESS in 9m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dfecb3e0a0ce47bba74da201d7e3a83f : SUCCESS in 8m 55s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/89617c6b97e14947ba571b2f141cfe85 : SUCCESS in 8m 02s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd5215107e4743c29fad18531afc1916 : FAILURE in 12m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/28336a4275024323afa5a5af95f4d742 : SUCCESS in 58m 50s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cb869614041347c4aeb97ce82574a399 : FAILURE in 20m 00s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/304ac78ac89444b78dbfd45ac5524f1f : SUCCESS in 1h 16m 22s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/ce7a3b5f0710428486e1d0f894e2f425 : FAILURE in 13m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da8fa4df5db74fc399996c5cdcaf6348 : SUCCESS in 42m 03s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/cfb3cbadf1864dc5b24ef96ecd177553 : SUCCESS in 42m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01a504f5e260433b96ee0d95d9554f6b : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9b7f3f08dc2247489f44e55b51f64cac : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b537ff2870e68a71c1f275498123fc242f2083d9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 12:44:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56245afbc1ab2e5a6ba85e7c7f0eeea9ecb72b86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 14:11:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d78baa46df4dacabcc8520d6f19fa4 : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/234b921a40024bfa8da992cf3f8df97f : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/80569a3a4d1949d0af762c7e57395b14 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304074e6079f455786be152db6076ead : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c6303ee4bb42d7957b36a4d0a0c83c : SUCCESS in 6m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/17361be03094484687142b527554be65 : FAILURE in 13m 05s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/e9a252dc68ba47c4a51cbb0a88f06e57 : SUCCESS in 1h 17m 37s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/236bdeb079ca444daaeedba161161807 : FAILURE in 10m 51s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/03901f473cbf473c805d89c9ed653545 : SUCCESS in 1h 21m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c8de68593744f02921637b86ba8d27c : SUCCESS in 9m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2bcc4dcdae3149e5824ff87de5db9d18 : SUCCESS in 7m 24s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c2de1e498637407cae5b2782ef9322f3 : SUCCESS in 7m 25s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/394b9ef66a034b22acc1ae28a1c73932 : FAILURE in 11m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/37c1812c55d043fab36278678a587b0a : SUCCESS in 56m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/87de2941fc264bdbb192ad10c83c729e : FAILURE in 15m 07s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5342efad6ecd47a69a40820ea6e465ca : SUCCESS in 1h 13m 23s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/fc5643d0299e457e82a8f0325d581665 : FAILURE in 11m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3349f113d5bc4a0a8198ddaaa9a0948f : SUCCESS in 48m 22s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/e44b42cb07424adc9fdb2d42887dbbef : SUCCESS in 45m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bab1186aabc9413b88e9768c9a88a88d : SUCCESS in 1h 10m 03s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/8ffda5ec08494319b9ecd31340e30218 : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1e5c5725717f269efeb6271d8d2a3b7a591be943","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-11 14:49:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"193b19536645a872ca76215713c812a04da24c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 21:09:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc0a26b1952748e29b4c81e659db8209 : SUCCESS in 11m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e97dd37a5124e20bdeadc1128cfedf8 : SUCCESS in 8m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/888ae6e4c09444268c23a4ffc052524e : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658f2b391cf24b55a3c0ec9dc3e20946 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2540f2e52db448908f3a430fa6b6e50b : SUCCESS in 6m 48s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c8361f8a698d4deb96df3d38c8b28972 : FAILURE in 16m 17s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/80ff11c8d2ef4efcad01e89d69918330 : SUCCESS in 1h 19m 44s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/196269b812e34649b79cf114ad5f083a : FAILURE in 11m 59s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ca66525734824cc9b5f1a7181a3e2494 : SUCCESS in 1h 39m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac05e1342d32496a8f3f02c8667cab5c : SUCCESS in 12m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1dc7d81d3a3949af9e9090e5ab133d72 : SUCCESS in 10m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/555c9c4945e74a94b4132c881a67c71e : SUCCESS in 7m 42s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1f0db4e679a4f4da37eb73fe253b1d3 : FAILURE in 14m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d4056b454c294aaf9ca191d47e6b7b33 : SUCCESS in 57m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/8b457c78e8bb494e9dec03e3b3cb6062 : FAILURE in 15m 30s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c671cbd45e5f4211b38135afc2a64272 : SUCCESS in 1h 18m 50s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/43a48c68e3344fb1bc1d73d1c2a318a6 : FAILURE in 13m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7db623f3b773438295ec32918a6b5598 : SUCCESS in 56m 58s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/60b111a965f643f4bccc9e5f31e8f1cd : SUCCESS in 49m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed7649aaf22c43a094fa907cf4626b96 : FAILURE in 1h 33m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0b925263009e4cfb83856f1672e9eb78 : SUCCESS in 1h 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"83b99483f168ecde9460c7d612ec0e71363dcd11","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-12 05:55:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8666743e79a20dca84e00aa66d82b1d77d215b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 07:19:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4274520ac684ab19db5908a43418aef : SUCCESS in 10m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2351a8f37cbd47639631b20f67dd3ea6 : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/309b66e609d04ad88a171103eaaa7375 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be23c5cef25c4fe69ae023018c2d905e : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d544f25a39d4638b6bd2be18fd726d1 : SUCCESS in 7m 14s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c9613f1547d14e2d95e1353f08f177a6 : SUCCESS in 1h 15m 50s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c43bcd28ca2a4788b85f9121ada311c7 : SUCCESS in 1h 17m 51s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91829a24035649b3881a4d714e8009d3 : SUCCESS in 1h 19m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0886755830e3454ea10dc6bdcb488ec2 : FAILURE in 42m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef27fe32fc446a89b4431aba37dc041 : SUCCESS in 24m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/005289ecc29f4557863728a9140d14d3 : SUCCESS in 8m 02s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c2a6e12e78e847129639b1971c96080f : SUCCESS in 7m 26s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66ba1a4443754a66ada83cf75fbc9b78 : SUCCESS in 57m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f1d141d31574497e8cbee8ace7c5bd23 : SUCCESS in 58m 10s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/19819567981e4318b1b1376fb138793b : SUCCESS in 1h 09m 21s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/dcc2116f5def49799e97d34854eadcbb : FAILURE in 50m 14s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/a9d038d8679249c1ac25d651499a7ae2 : SUCCESS in 1h 04m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a4a0fb487564d6d930a9f094f6a3482 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0069648aee9f49da99d59f9543eecd83 : SUCCESS in 40m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59f4fe7485c6462885627dc86c92a82a : FAILURE in 1h 16m 39s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/0f3221551c4d43e5abf8be2c266123c0 : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5bb6303a53c9060317eec006824cf394bea58d32","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-12 08:52:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"827066df70fd081df6eae5093b098e39639c62fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 11:08:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8009266abac046f3977729d0c22d3dce : SUCCESS in 11m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f2ddad711a740bfa12a215d561e0813 : SUCCESS in 8m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aced3842e95e4e82bbe0bfab7780ab52 : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d64966257ac47b7a08ef20d4a941d99 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52521209c6f745ba99c91beae64b3948 : SUCCESS in 8m 20s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/680ee75090a8419a8745ca420e5d5a47 : SUCCESS in 1h 51m 54s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0f1d171b8847435485b6f01ad0b9583c : SUCCESS in 1h 17m 48s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea475fec4a774a119aa386b4c754a181 : SUCCESS in 1h 16m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/74a5f3e46db94c44b1ab34fdb1616197 : FAILURE in 2h 07m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6711e5512f9046aba5a205a4d88a14f5 : SUCCESS in 11m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/15febccc56194e2f91c9648bfcf079bf : SUCCESS in 8m 39s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/300e41e987c44d7cbe6ac763335dacae : SUCCESS in 9m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7df64c9ed3c46c2b9d32b6315b48cb4 : SUCCESS in 1h 01m 18s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b20d83231b024bd48f184ad4c9d79d52 : FAILURE in 39m 57s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/fcec6e1e05d847498bb9b4db3f856411 : SUCCESS in 1h 49m 59s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/647e0822c5b946078aa1c5bc3e5b2039 : SUCCESS in 1h 14m 20s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/13be296e4cd04ee5a6e8baa37abc3a36 : SUCCESS in 1h 09m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87694a90886f4d90b5f327aa04dc1573 : SUCCESS in 46m 44s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/a4e8357ca53d4fb88a71c5f14447c45b : SUCCESS in 50m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20e6969f8e544e0ea886209cf984ec2c : SUCCESS in 2h 00m 15s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5700631ba8894a39bf70e92d746f4a94 : SUCCESS in 2h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0ddbcad3a4c3dc953196984f0b29cb61c209bee8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-12 14:57:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89c5e51ebbb2b0efcd17673a3d5408f48751a4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:41:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220ae8ef4c4147bdbe65ffde8a2b5d78 : SUCCESS in 12m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/719ecf4d77da48c98fed7fdd25bf21d1 : SUCCESS in 7m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/816dc4061c9b4d6e82c0fc23560dcfd9 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f7fa9b79b64f0997dc8ba9c59726b8 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f188e9daa0f84cc98df0f666146d3f86 : SUCCESS in 8m 46s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/689065cfb0124dfe9bd4e420db0ef5f0 : SUCCESS in 1h 20m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b7836bbc639f45e9a2d5fc8bafff2881 : SUCCESS in 1h 19m 41s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/881aa7e280b942d983d9f8fdbbfe013a : SUCCESS in 1h 32m 27s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f1d05c1200ec43658f7203ee70c94e01 : SUCCESS in 1h 22m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e44e5e67572d45118c6f33ea64292894 : SUCCESS in 14m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6299df923a5845398db9ac5c89c52047 : SUCCESS in 10m 16s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1327103bbf2c4b22bd216006d1726c4a : SUCCESS in 8m 26s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/231e647de73a417ea9e1bc9a9cd18a1e : SUCCESS in 57m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/279bab81a36e4c3b9210f0c4fa1c0513 : SUCCESS in 1h 07m 57s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/ef0b1c7d582345fe84eaa1f7232b99a6 : SUCCESS in 1h 14m 56s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/69c2f141f01e48f5b23dff0eef546d76 : FAILURE in 51m 29s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/44de1f5247274b4596473a95e4817525 : SUCCESS in 1h 10m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f94efc5f62ba41dab1127c2372e21be5 : SUCCESS in 45m 51s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/0197b494548e42fa85c62f0396ce0a5c : SUCCESS in 45m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d0d2452c30e4010bc81e8dfeb9b25e4 : FAILURE in 1h 36m 38s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/b6ec87218bfd423894859433949bd137 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"eb3d742a885d2d4c854d36dc50f1987ac7cd8b57","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-12 16:48:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5d6699f60dad04951c5ff86ee389eda0942149a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:39:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af05f84016b54053b49b96cf87b96c96 : SUCCESS in 11m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e882dc5761814b27be438f9f99644119 : SUCCESS in 6m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b5958bde4c944dea4bb8c03d08bd38c : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7e018aca9047d2b479a71032562bf6 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08727deb3140434982a148f2e56efa5b : SUCCESS in 8m 30s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/588bca1eda9342ffa33e0f07d291c3ff : FAILURE in 9m 11s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bec3a1f732d74a53983a0f4934f99d3a : FAILURE in 7m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7adeda959c684f7dba5d9b42c58aca61 : FAILURE in 4m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9e51066c8d614dd5b280c89ed6451afb : FAILURE in 6m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/992253f65e0a4473a1fc98089271daf8 : SUCCESS in 10m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b04fca8621844c6780e8cd0318c49a44 : SUCCESS in 8m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/819416ca98a74a24a528523d45361d25 : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afea4aa0d2ec4f42a79b8fdbe51c712e : FAILURE in 7m 34s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/79970733e1f8425e94b8321915abb7d5 : FAILURE in 7m 56s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/77dbc1dee1454084b31a889dd963d3be : FAILURE in 8m 10s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/5a0f3475b0bf488b81894495364d3143 : FAILURE in 11m 34s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/3cb251a3d5ed4abfbcaffd0a35315b34 : FAILURE in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3578a32a96b24599aac7086be8be8441 : POST_FAILURE in 41m 14s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bb1808f30f5c4363adec894bbfa4c4cf : POST_FAILURE in 49m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52b539cb413941fd9db572cb9a6d5dc3 : POST_FAILURE in 1h 43m 51s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/89b23de9ec084bcd96af6eee89e81b2e : POST_FAILURE in 1h 36m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a31a9602f2de71436dbbe4caba352b725cdd5c84","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 06:53:31.000000000","message":"Patch Set 1:\n\nrecheck\n\nhttps://review.opendev.org/#/c/707441/ merged","accounts_in_message":[],"_revision_number":1},{"id":"bae0d4dc788bf97e8cfc56cd92aec468832a6fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 08:28:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535dbb787e564cd7b1c542419649f4da : SUCCESS in 8m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bcaa92e9e7149c897518004b7a287b8 : SUCCESS in 8m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0c0576dd4e694909898663a59bd97a47 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6d9e0b999a4aa7868ad24899619e2e : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c8bd88a7694edfa3d348557140c9a1 : SUCCESS in 8m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9759d99d51c64f038e8b506a0c05923c : SUCCESS in 1h 29m 27s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/be455fba2990490a9ed093f9a61db37a : SUCCESS in 1h 20m 38s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42dceff739b84fe7bfe8a9cd456cfcb9 : FAILURE in 34m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/108110b143164c1280cfcc2734707ae9 : SUCCESS in 1h 14m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a42397af44f4b7d954ba0874156b92b : SUCCESS in 10m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2fd2a977396448dd9c85ab2db3697635 : SUCCESS in 9m 09s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/830f3fecf944441585329ae813dda689 : SUCCESS in 8m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5d46676ccdd4f1eaacafc8b14cef3b2 : FAILURE in 37m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/51a933a7c729422db4421165dee76abe : SUCCESS in 1h 03m 41s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/b8b7c1ab991441039639863b8ff35291 : FAILURE in 40m 33s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/c4a30149f2f74431acbc2f03e8acc721 : SUCCESS in 1h 07m 51s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/045591c2d6e14021b014217205114346 : FAILURE in 38m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5eae58a0d8ab436cad4d183267d7fbc7 : POST_FAILURE in 49m 07s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/80301a95087944399ea31893accc8aba : POST_FAILURE in 52m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9ae9fb643164165a044a3b63815123e : POST_FAILURE in 1h 10m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/5ab261dcc6664395999c34798f818eed : FAILURE in 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d9d32064b58b78ebe49fe6cd3b7d8e49c610a693","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-02-13 09:07:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a27febf658afa7db35fc672dd85fdafa7ef7c947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 11:10:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e3697158beb4db4af4ab73007aea42b : SUCCESS in 10m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f75ea2375d4849b419b9e611edb140 : SUCCESS in 7m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7aa4f8e58fd9421bace47c8721ccc8a9 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d01b1c5c35248f38c598406cb717cd6 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae90d81568f945ff819261a7a88eccf7 : SUCCESS in 7m 39s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/ae3a2a2c77fe46509c691db1f014fb5f : SUCCESS in 1h 17m 59s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6954a20a81414c41acce705751b215b2 : SUCCESS in 1h 15m 54s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3acdff9fdfa74ecd883f46382b62feea : SUCCESS in 1h 11m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8381453d2d46449698eec331092c8ff3 : SUCCESS in 1h 26m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f914153fb34b1e8b0ae488c777f61d : SUCCESS in 11m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a22dc3c196294a8b926ce343010b4352 : SUCCESS in 10m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ab8cf1f45d28457a979fbea7f32f3958 : SUCCESS in 7m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83a6e4c36e77480db232e969167d1c4a : SUCCESS in 1h 06m 28s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/eee2d793a2144234a564783ad9354114 : SUCCESS in 1h 03m 18s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/f2734ac694a84f5d816f4607126d28a3 : SUCCESS in 1h 15m 52s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/4c75ff0a946f402baaf670c3f48b267f : SUCCESS in 1h 11m 45s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/95d6389a6393419d9b8aa707950d3fd4 : SUCCESS in 1h 08m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3ea9dedf3f84680a11422f792fc4dc2 : POST_FAILURE in 54m 09s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/ee25b2a430f54540ba1590cf1f662f9d : POST_FAILURE in 45m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a003cb3fdec2437b9fe7246e44c32f6d : POST_FAILURE in 1h 39m 46s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/116533f24de24b24bb76a035dd81d341 : POST_FAILURE in 1h 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"945ed344bee154785ae961d2a0a1df1d916f76d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-13 15:02:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de341f5bf4729026c9d389c7ccbf869078faf05c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-13 15:02:37.000000000","message":"Patch Set 1:\n\nstackviz should be fine now.","accounts_in_message":[],"_revision_number":1},{"id":"ccd8cfd0b4fea90b5d5896432acc5bb9f6bfc73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:51:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cc0734b872424484940289f62ce031 : SUCCESS in 10m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cfe26c2f5194ac2a039ebb66becaeca : SUCCESS in 8m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad4077c911b94f6bb3d3cd9715dd76fd : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb90f594e544cd99f05ed5d238a7d6c : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60aa38127b740d09d414b848dbd2e87 : SUCCESS in 6m 37s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/27670ea68f294c1dab74b193ae737107 : SUCCESS in 1h 16m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9d049c2d114847aa9fd8a57f4526f328 : SUCCESS in 1h 18m 22s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ed6853ef4f84a8f980b0445b273a950 : SUCCESS in 1h 16m 49s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/4586f0253e0847cd8ab0157a0d6c8141 : SUCCESS in 1h 18m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b26c3dc9d5741ffa433369e66ea15e9 : SUCCESS in 10m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/737822db18d04cd7812de9699b763b5b : SUCCESS in 9m 53s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3d7826c32c4d41b2ac0ff7c3a989e9f0 : SUCCESS in 7m 24s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5ba674dad1544858e94056388a46764 : SUCCESS in 1h 03m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ba16d8dee9ed47f78ec86cd78f11835b : FAILURE in 55m 29s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/cd5cd70d70834c89aa97043537a0c711 : SUCCESS in 1h 05m 38s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/35c9bb93d20247748eecce87db0ece3f : SUCCESS in 1h 10m 13s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/a469c6b968174a17b94f7c8cbe455f99 : SUCCESS in 1h 05m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5b19a12859549939adfdb31dbf2f097 : POST_FAILURE in 48m 15s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/bd5c30fa45d946dcbc7d49995de742ad : POST_FAILURE in 51m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b8d097759b24edc92f98edd478a39bf : POST_FAILURE in 1h 34m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/e2e3d250097843bc84261a33e85759de : POST_FAILURE in 1h 43m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f6911d250cfd9fcbddb020967b793f0ff4ef25c7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 17:17:37.000000000","message":"Patch Set 1:\n\nrecheck\n\nmerge timing issue?","accounts_in_message":[],"_revision_number":1},{"id":"ab7a06ce94ff8d251971c5cb020ae78fd6a0dd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:20:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43e77185f4e4877baccaba2c2868a95 : SUCCESS in 10m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/feca28ea178345bcadd7d90bd1edb8c4 : SUCCESS in 7m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66eb6bae21324dbca751eb676a44ca12 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0342b275017141b29621596b207d4fd1 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0274da0373a4c69a04ac7f6b465826c : SUCCESS in 6m 59s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/92cbc37fe15549e0b1c9ae0b701f689e : SUCCESS in 1h 18m 06s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a40abbe4454b463aaceea4f5569caaf3 : SUCCESS in 1h 18m 11s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de8adf80af2748b7aebe60b6229c5742 : SUCCESS in 1h 27m 03s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ea7ddbc003a24eb7b821de14c8d8b6d4 : SUCCESS in 1h 58m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df2e0b89a2be47e59c0475fcd9e546b0 : SUCCESS in 12m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e849cf77987e4ce6966e1cd2ae3ad374 : SUCCESS in 9m 34s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/be49132be9de4b2c85a1f6ebe22cc9a8 : SUCCESS in 9m 13s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f62dfeb03954481da5173c85b1cd6a18 : SUCCESS in 1h 02m 43s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/13d9c7b05afe400889a80e9419e8f278 : SUCCESS in 56m 17s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/200934305596486ea74ae7db53897ed9 : SUCCESS in 1h 09m 38s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b4acf7fd68b24e22b8f53062e3287e78 : SUCCESS in 1h 29m 50s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/4c1a099e87f641f5bcdb5fb80aea5e07 : SUCCESS in 1h 27m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d62cbb4a840b44289d0be65f11cf538a : POST_FAILURE in 49m 52s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/37389c5e595048c49127a33934c6c9a5 : POST_FAILURE in 46m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f29aabc63c68485fb23d15c580078441 : POST_FAILURE in 1h 24m 39s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9b837817335b46d9985e32ec173cd1bb : POST_FAILURE in 1h 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2eda26c584da24865a2a351e2e92846bddd5a666","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-13 20:19:49.000000000","message":"Patch Set 1:\n\nlatest stackviz is uploaded now - https://tarballs.openstack.org/stackviz/dist/","accounts_in_message":[],"_revision_number":1},{"id":"92a99e5efff97edc6198bdcdf549a9524ba2a163","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-13 20:19:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9537891ca783aba3f430ede7ab4b204b8553acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 22:00:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb4451fdf964bb38096eed131a7dda2 : SUCCESS in 10m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69abbf46356f400a9f74e0e53b7b9613 : SUCCESS in 7m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dbab7fda086444a195e5ab7e040f81a8 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c1df1bccf64705845dad3ac7eac790 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feb4efa2ab6340c4ae42ed44302a1f67 : SUCCESS in 8m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d14d03eb21f7419fbdcf6f4128390b31 : SUCCESS in 1h 19m 31s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/6b1b1bd8e70641d59cbbf909724ee2a6 : SUCCESS in 1h 19m 53s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/45a763256e83435999dc5390018d6bcb : SUCCESS in 1h 14m 05s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/c1a9d661eb90496c842b520688e1906c : SUCCESS in 1h 23m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a55efd3470c4e65b5647c7f951b618b : SUCCESS in 11m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a531e1e7f97448fab79e30de25af922a : SUCCESS in 10m 13s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fa33de5fbcf74f1ca86b1b69840bf260 : SUCCESS in 8m 32s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/817375c85afa4d2681bf37fc3d65927e : SUCCESS in 56m 49s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/143202e24c9a468c8ab94b50d75fa955 : SUCCESS in 56m 21s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/acf6280fbac74e7abc3f01abc4fc3e0a : SUCCESS in 1h 13m 19s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/ace89ea0c2824cd1b1b55a4f58885143 : SUCCESS in 1h 10m 00s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cb80bb651c1b4a669f80f3ba308eddc9 : SUCCESS in 1h 12m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd5b451410d749aaa6c9ffbec38d46a4 : POST_FAILURE in 48m 16s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/d5a6b9833ed74659965633e2c7522fe0 : POST_FAILURE in 51m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3d9df39f11a84da280dcef704c631a9d : POST_FAILURE in 1h 24m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a36341775515427cb351190159be112e : POST_FAILURE in 1h 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"77f9a3924148611f977311150a28ac9b150aeaa5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 05:46:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c498d2c23073e60e79064ef0e20f2bbc0c968bd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 07:20:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d2d3554e1b45799f5339931edfd2e9 : SUCCESS in 9m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d0d09c4bfa946e4b8028ff6fb3cdb61 : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90aabfb44814475fa395ab5bea7a39a7 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ca8bb285fc49cca5be0a33074355a4 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba73ddad64dc4f879a75ef25ca407129 : SUCCESS in 6m 40s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/053b01445f954e97b1255a0bdf86595d : SUCCESS in 1h 16m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5ea62970a9f0469fb34a2773399e5e90 : SUCCESS in 1h 24m 19s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dcd959710a240e0ac6e47a80c991cdf : SUCCESS in 1h 18m 42s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/974d67fb52f24a2fa90c43a2a6a58d43 : SUCCESS in 1h 26m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdcc4dcdb59b4f9c9fdd8601ced8f3d8 : SUCCESS in 10m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ed7332b5c9c3492b8b52e321a18e0be5 : SUCCESS in 7m 49s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a5e3072fdb514a009079731a08ae0f38 : SUCCESS in 6m 56s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/197225a08ebf4283aa488c7b15cc17d8 : SUCCESS in 1h 05m 19s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2a9f051f3a4a4f3bbe6f43f9deb6f470 : SUCCESS in 1h 01m 45s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/958505e077a747c38d73361f7f05aee6 : SUCCESS in 1h 19m 14s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/65ac84e61a594f37b63ef4997c96d060 : SUCCESS in 1h 10m 10s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cbe06b243f524a2a85483d0a4e316061 : SUCCESS in 1h 11m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/15c07296f88643fc89a697b8b277789c : SUCCESS in 46m 02s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/278bf845ed804f6db0d3be0f9605d316 : SUCCESS in 50m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d81059db59c54ab8a425df889575316e : SUCCESS in 1h 25m 17s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/7a4e582a44e84fc7ae639e22b46527ce : FAILURE in 1h 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c98d28bc18e83fc13f231efbe608546f02f0782f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 10:19:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e79a1ece332c59eaf7849589ecc0542fa3ffdbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:03:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c445a7539014837a4a55efba2a86e73 : SUCCESS in 12m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/546463f9edda473cbfbc4bd70865cb84 : SUCCESS in 7m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fcee5450915848188430a54abdd5c3f1 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd103ee5fcaf48cd951c07e48a5a18b8 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03df9810130f4419946c5fa6279a9b6a : SUCCESS in 8m 34s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/08ff41f104d14ec0b330811e46557713 : FAILURE in 39m 01s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8eba42df1e8846ba903169ab45a12503 : SUCCESS in 1h 21m 01s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ae9473885714d43a0b425e34184a60a : SUCCESS in 1h 36m 44s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2cf6aedad29b495390d5ecaf06925eea : SUCCESS in 1h 27m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3af7233b1fe4bfba4ac3b30fe435bf6 : SUCCESS in 11m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/11babe76254b44a693075563c27a9f49 : SUCCESS in 7m 28s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8e5a642aa618400b907d3e337573676e : SUCCESS in 8m 33s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dacea80e4bd44dabbef5b0fb9d89e35a : SUCCESS in 55m 10s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/6a1f48d3ba5c400e9e2954022ee4cfc6 : SUCCESS in 1h 04m 21s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/25305a9fb2774e4c96ea306a8de6547b : SUCCESS in 1h 12m 00s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/b6a0cba724ba40bdb9169a4547743afa : SUCCESS in 1h 16m 12s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/8101e434cb444d3bb9dc1246177b3cbd : SUCCESS in 1h 02m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a2b752cc2bc4734ad38af7f3f928b0e : SUCCESS in 46m 07s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/cd46734630504de68e20abede70861ae : SUCCESS in 41m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86ad6fed893f440782bce924d2d0a07e : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/fc9c096222584fcf890728d2a9baadbb : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":1},{"id":"93a29a3d452b4075e3c5a46eb5ed01eadbf77f13","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 12:11:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"112ed06ce58114afe217dbeda5c855ebf99bdf36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 13:49:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0db8946914f4bf880a00f046b808d9e : SUCCESS in 10m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/99d680205647452abc75528f8416d871 : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f61ea4109f0846e484dbfed255688b9d : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c43134cd77d4fcda6924477f6ddbb84 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f181d305ccf48bc8ecd1a1d35fed60d : SUCCESS in 8m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/efa3724974944113b709701129a2cbe4 : SUCCESS in 1h 33m 38s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/eda58f59afbc4a10897361a6207b6a96 : SUCCESS in 1h 28m 59s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d41d886a44a54cee961e94ac9fd027ca : SUCCESS in 1h 18m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/72cd7785dc5240b293fa92b35ada9746 : SUCCESS in 1h 24m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/938959c3420a4da989b498bdb55904cf : SUCCESS in 9m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cdd0f32bb058440fa6cfbfc576423539 : SUCCESS in 6m 48s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5f9d67188198428fa17edf9fa45d398f : SUCCESS in 7m 20s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e428dd2ad5a44bdeaf7678509e245d7f : FAILURE in 40m 03s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/2dc9384aa07149b699bc0da7fd06cb7d : SUCCESS in 1h 01m 47s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/410262e143ce457c9eee2c4d8c4fcd6c : SUCCESS in 1h 13m 22s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/d417ab859e354875a86272c219a78073 : SUCCESS in 1h 14m 48s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/cb9d100fee4f4814babfc8aa1ace5658 : SUCCESS in 1h 07m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dba7b3a5aad041e98ae8124d2042a5ed : SUCCESS in 48m 42s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/43a9cd5045f248efa66054e273591027 : SUCCESS in 45m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6ab005c646a4d3e8a73c0d0512b13a0 : FAILURE in 1h 02m 08s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/dd3346717bcb482aa731c985f8ca4499 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"724f5e9219b345755274a633df6c755ff33e544e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-14 14:12:09.000000000","message":"Patch Set 1:\n\nrecheck\n\ndie!","accounts_in_message":[],"_revision_number":1},{"id":"685bd5353ed0a77063ac42e87a73af1263b29438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 15:52:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71b90cb19a44831b7eb56390703ad2b : SUCCESS in 12m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fea57132e2b42ab85344394178964a6 : SUCCESS in 9m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/174c73b575154463840e70d29ead389f : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080d869f547a49a6aac9e321cfe08be6 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/185cf9e3ec6e449bbd8fcd30fe18eb68 : SUCCESS in 8m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/1abf0f5a8da84098a5108f32d4ad8851 : SUCCESS in 1h 21m 49s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/181a08d032904bd4a683dcfb7f6d4a2a : SUCCESS in 1h 18m 25s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b0dfac1d03841bcb2ee13d36b218b2d : SUCCESS in 1h 27m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bca782bf87554fa78cf2eb6b1a1e7533 : SUCCESS in 1h 17m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e5df0c416674ee89bb1800f5fe01004 : SUCCESS in 10m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1dd61cb047fc45bcbe6bd21eba268ad1 : SUCCESS in 11m 32s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b986327e563c4b1ebcf8fc2bd2177c3a : SUCCESS in 9m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4eee2951eefd43709d4eeada0edc72ef : SUCCESS in 1h 07m 55s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/57b47d7b6d0c4890af37f6d4279d21da : SUCCESS in 1h 09m 54s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/56f55b95be2a4843ae251a9df7503f7a : SUCCESS in 1h 08m 23s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/7a61fc78b3814428bc9ae71cd7716d29 : SUCCESS in 1h 24m 03s\n- octavia-v1-dsvm-scenario-kvm-centos.7 https://zuul.opendev.org/t/openstack/build/222793a9e58a43878f3e872e4f9711ca : SUCCESS in 1h 18m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e70984f1551e4778be84594a64572888 : SUCCESS in 47m 08s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/32c99268d7b34c29a89eab1b05af1065 : SUCCESS in 54m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edb0f71cd92c42e2a398eda8753d6e21 : SUCCESS in 1h 29m 58s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/a2dcdda763c74a15aeed41a80dcff518 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"dc166223d152de4a825a77d7528d3573bdae7dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 15:52:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91e669c3bc653dacc39ff25c9083fefaf65e7753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d371a4a0b440aaaab065e64bf3326c : SUCCESS in 9m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/770d4ee44757498f89a0483c1b7aef58 : SUCCESS in 7m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/57ff0057c1ad4630af2d25f73f57acd1 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6ba652473a4829a4ffb2e72688983f : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97b2225a74fa4a0d8e99d835c1e42ac5 : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d1daa24158f434ea8fe219d9c7ef0a7 : SUCCESS in 10m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0953b061551240b79ff561e96489c9a6 : SUCCESS in 8m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b39b08aa0f1740c591852726909cfae0 : SUCCESS in 8m 38s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b1d6896cf61542dd86124f2e6c964369 : SUCCESS in 1h 20m 20s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ba8726dc9f5f4166b195ec91b06db05c : SUCCESS in 1h 33m 10s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de637cb1602349d4abcf89cc2c17d4d6 : SUCCESS in 45m 06s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/923fa330c9794b7e92a46abb64a0d516 : SUCCESS in 58m 17s\n- octavia-v1-dsvm-scenario-multinode https://zuul.opendev.org/t/openstack/build/edc795065e1d4ef28e9525403ff8964b : SUCCESS in 1h 18m 31s\n- octavia-v1-dsvm-py3x-scenario-multinode https://zuul.opendev.org/t/openstack/build/41a7925aaba440da9ee0c0720b8202f7 : SUCCESS in 1h 21m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f29e5fab6160419397c271e7c5940b5c : SUCCESS in 48m 59s\n- octavia-v2-dsvm-noop-py35-api https://zuul.opendev.org/t/openstack/build/7e089e15eb7d4c748a006e0ee923d2cb : SUCCESS in 49m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d17d273603c94f848d94e611002d80db : SUCCESS in 1h 31m 27s\n- octavia-v2-dsvm-py35-scenario https://zuul.opendev.org/t/openstack/build/9a7ab86d6b1049f6be4e0a7b23292787 : SUCCESS in 1h 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"29e6caaa2dd8c3a9d1ccce0ddd729045788fa728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ebfad07c8792490b63047e1c4e1f2575025ca351","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13742c00e6784af0a9745953d1af0dff : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9673aad3db3e4b89bc3d806aaab526ea : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"354561580b122fece43caa232485796176e13470","revisions":{"354561580b122fece43caa232485796176e13470":{"kind":"REWORK","_number":1,"created":"2020-02-05 17:20:19.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/52/706052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/706052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/706052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/706052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/706052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/706052/1"}}},"commit":{"parents":[{"commit":"e3ddec776da4654517a237287c05298a9280e52f","subject":"Add listener and pool protocol validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ddec776da4654517a237287c05298a9280e52f"}]}],"author":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2020-02-05 11:18:29.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-02-05 17:20:04.000000000","tz":60},"subject":"Use stable upper-constraints.txt in Amphora builds","message":"Use stable upper-constraints.txt in Amphora builds\n\nAmphora builds are now using upper-constraints.txt file from master.\nThis is incorrect and makes the build fail as some newer versions of\nlibs are not available for Python 3. This commit makes sure the file is\ntaken from stable/queens branch.\n\nChange-Id: Ifacbd1953c3d8719539a9f4012eb34669bb164e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/354561580b122fece43caa232485796176e13470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/354561580b122fece43caa232485796176e13470"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
