)]}'
{"id":"openstack%2Foctavia~706153","triplet_id":"openstack%2Foctavia~master~I32ff97ff97022a59731358e60b2604c118a506a9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I32ff97ff97022a59731358e60b2604c118a506a9","subject":"Use routed network filter if it exists","status":"NEW","created":"2020-02-05 23:20:51.000000000","updated":"2020-09-15 03:33:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd9ff59bc4dbd3f803f183ca963907ea9481a2bc","_number":706153,"virtual_id_number":706153,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-15 03:33:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 23:01:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-15 03:33:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"951df521b61a39acdbf76db2084ca8ccb11905aa","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-05 23:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7efe8d961680da997e3044eecb4c06d2b628dbb5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-05 23:21:57.000000000","message":"Patch Set 1: Workflow-1\n\nThis won\u0027t work unless you have a custom nova scheduling filter installed to handle \"routed_segments\".","accounts_in_message":[],"_revision_number":1},{"id":"3804f9dcfb7337b075b55dff0abda7705e9556ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:44:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b5044f0eff942069576d6e706175a46 : SUCCESS in 13m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccdd1c4dd56e4149afae5f92a3bc8645 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eb57baf8f8b4077920f02ac146fb501 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064fdf707c434d508338d9d15e1b8d5f : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35a28ec253684dbe8bdfdff9b00a6ef1 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d1551ad3433494aa9cd4f73193ab8f4 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40173f9ff70849e3be0b8faa84b657b0 : SUCCESS in 11m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5cae4a5085e44113bc23842210943abc : SUCCESS in 6m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4fcac859e5b448938dd335012cdec408 : SUCCESS in 10m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f67a6d07202e4f629a7a734fb5947ea9 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c792d2e17cc4e76850dd63f4a612b13 : FAILURE in 38m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f67132d247c04a5eaf622faec1b0f782 : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/831c349bda41443fabdbea894eabaab8 : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e488756216df4154b3066e4da9e20a1b : SUCCESS in 56m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a92314abc4443f4bdd61df64cd0d49a : SUCCESS in 1h 05m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/814360490f7e4464bbe6c7f64a7a678e : SUCCESS in 2h 17m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8b2da815744242e397dcfc0ad9cfb7de : SUCCESS in 1h 45m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2557717d19747e4ad0864cd36d5f43e : FAILURE in 42m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67d069a140c7440e973503dc5f1d5a18 : SUCCESS in 1h 43m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"187183dfd7890a34d8780a531c6327abb0e7e029","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-06 03:54:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6690b7e5a22b52f50a1410bbce244d1820da9080","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-06 04:11:58.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"7b53c601b027e4748f355421634537b2217d1f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 05:49:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c317e22e9931404a83ac3b9fee365f9d : SUCCESS in 7m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/248b3a4c8da24a69be34c2bc3211650c : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4c2a919d334f02ab249a15915b3666 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7135e6d3a8af4e6bb77290e48c3c09e5 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74df394e2aae48589ea16183b1a8dd1c : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61c9a63fee544d8a9c6e9ab207d0ffad : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f792723f06b9494093a48399ff5cce68 : SUCCESS in 9m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0646247b612f4560828fca6b3eae936e : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b5c5fd4f1b244efa892b311d21efc15e : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ca8cec2bccd4d4ba14b4d5a2698302e : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e50fadab9d64478eb6f567d12c18798b : SUCCESS in 44m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf739e9ab11747d3afec8c178fef4741 : SUCCESS in 1h 30m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/996089e1a461420a9721302700fdea76 : SUCCESS in 1h 10m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ac51e605644c4b6dbfe811991779a025 : SUCCESS in 54m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d0ff963d9946bc9a9d57eec55e4891 : SUCCESS in 1h 12m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c09f618d5d041cbaf38ed792150fb63 : SUCCESS in 1h 49m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f604dd054afc422a8511fa5bcea0b3d3 : FAILURE in 1h 53m 51s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98c32d64059b4e1dbdeabc7155c4e32a : FAILURE in 38m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a82d55719eeb4566b87c5e78583c3676 : SUCCESS in 1h 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe28a13e30cbae1a3e1c5e6d7f6de32a4434fd13","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-22 07:22:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1338525055bac1758127803e8c4466751510c66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-22 08:56:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9095dcdcf1c4b55be845ec8e137d4ce : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab2462d6fd8242d7b6d1182dba6d0fa7 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f35e4c57ae4421b577b6f1544f69e8 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4de2b710cc8c44cc9047197d3c1f11f5 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9d35ab974954773a36af23af2670044 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8083efd5b11d45f39f5b0208ea109769 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04da4078178a458382b2da93737d5f64 : SUCCESS in 16m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d2a64e9451a466295cc5ed8aa1392ce : SUCCESS in 5m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/112c324ba8e94f6b99dc830f2b0d8003 : SUCCESS in 8m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/656155e2e52e4996a33c3cc96df8a1f9 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ba5f231616e459cbf3c009eb70b6907 : SUCCESS in 42m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18c8cda187d0451abaef19c735d445ac : SUCCESS in 1h 25m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fbc4a4735d35467683282a57e2d5b9e5 : SUCCESS in 51m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b4ef8168b87b4de699a7a3097815b03d : SUCCESS in 50m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f7d9d648e5a469183a9218a2edbce02 : SUCCESS in 1h 01m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d26d0320bcc446ddad3f5fc43736ae8f : SUCCESS in 1h 31m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/284eff98f10a4d26bf8e2eb8a53e2934 : SUCCESS in 1h 31m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/04230baa74024686bcf372c359274a48 : FAILURE in 1h 08m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b208bc58acc64e97a79e115b306019f7 : SUCCESS in 1h 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a19850ed13c934752a89664bcdce83b9d57a36f4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-02-28 05:06:36.000000000","message":"Patch Set 3: Code-Review-1\n\nIt seems that nova is unwilling to merge the filter that this depends on, but recommended that we can bundle it ourselves if we want to use it. I don\u0027t know if that\u0027s an acceptable approach or not.\n\nThey are working on an alternative that would also supposedly solve this for us with a different hint, so maybe we wait for that?\n\nDoes anyone else actually need support for this? I am ok running it as a downstream patch for now, and fixing this to use the new pre-filter method when they\u0027ve got it working.","accounts_in_message":[],"_revision_number":3},{"id":"888315f6ff29c2ffad303df2262d53dd65744556","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-21 00:28:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c80c1e7e1f3ce057951c3dfa9b3fde5b895051fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 03:03:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d14ea174c06420ab7c68310d1f36f62 : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc8131c5073746349e5a49f1f73605f7 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2b0708b5604ef4b26cf0d57477eced : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9181e2f84e8345e89fbe9e7deeb1133b : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8095481fd5f5445d95a51b84c2c5cd32 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4b289ac90f94622be2e58a67d6a391d : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e6eb705308412a8c778d36e8172156 : SUCCESS in 18m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/43684044c0654b838109f250c7064369 : SUCCESS in 10m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9c7638f382cf44f589f9b24e13d030ce : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cce411f711f4bc398202c26ddbeaa60 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/acf632d8a5c641b894b372c42c63af23 : SUCCESS in 45m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/399e499fbed441f7a859f8975a198319 : SUCCESS in 1h 52m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42154d15f43346bca5a120d0db79ad48 : SUCCESS in 57m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e5ff379158b54481a9ea14dd85bcedea : SUCCESS in 1h 00m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a342ce8405ce476cb356944ae5da1c17 : SUCCESS in 1h 16m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b55b2bb94f974f87945f69215d96d94e : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2991de2324484bf69157a3df17043fa5 : SUCCESS in 1h 11m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfdd7fe185e94c6d825572a6920ca508 : FAILURE in 1h 26m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7368799b7bd0487b88bc6882dab48298 : TIMED_OUT in 2h 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"23e41434bfe32de07d8d605cb6cbba72dba17a95","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-23 23:01:50.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3660f0190f230097f85a5e5a5258485afe4fc7fb","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-13 11:23:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b04fa1852399bf35f35007981ffb30a38daa03e3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-13 11:25:59.000000000","message":"Topic failover-refactor removed","accounts_in_message":[],"_revision_number":5},{"id":"2c408470ef93bbf443803e0542ea2f4de17bed90","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-13 11:26:22.000000000","message":"Patch Set 5:\n\nRebased on top of failover flows as there was an annoying conflict to deal with.","accounts_in_message":[],"_revision_number":5},{"id":"57ca981c1e53beaaa521fd9fd6d1b0e6f66f6220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 13:45:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59cedf30b9db4cffa9ddade92b2387bb : FAILURE in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c718d0ef1fd34ada9104113d41d988c4 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d38d7d1e9b5473e8f25afa169644946 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7fb89040cd7420a8b8d5b010b4f421e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fa27870af3c49ab86e25d8ca0dfcafa : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2e19f8861884d0ca47e4f46e94ef3a0 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed8968e21684b8ca2882b00b4e35aa3 : SUCCESS in 11m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cbd6a6efe8914e58befe8fe91bd6705e : SUCCESS in 4m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6141bb8a3cb745a6b9e50d9e8d091dad : FAILURE in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4d0acaaac354033857d1b5ed3d64bbd : FAILURE in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5eeb230398a44502900e15e74cd92b86 : SUCCESS in 42m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2f66821c49cf44aca1a2c179c1cc1626 : SUCCESS in 2h 15m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/228b5ed9d9f14bf68e62291459510ab4 : SUCCESS in 55m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e446829a969847779a52b36fa6fc09bf : SUCCESS in 50m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e82f4b7d65e44d4886aa7cb5e99f51d9 : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f1f8d0e1e9246ef8b4d92592f319970 : SUCCESS in 2h 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/41e0fdf0521e4ed28111a14317b0fbb0 : SUCCESS in 2h 06m 44s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8041bccdd55e45d7890a8d885d921c71 : FAILURE in 1h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dcdaf41771764d878a77727780e3485d : SUCCESS in 1h 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f64373f35f2ceefdf5e3a04cb02d40a535ebf815","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-15 10:54:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6902c819c909f8a535ef1d4579ccdd8c63ae2915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:18:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de2ca5c92a7441c8a2890381526ea079 : FAILURE in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93b1338d00ec4bba99113ae28d2458b2 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6649f59bed4221b6ec2bea81303664 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db3c7f2d6f941cf9f1001ed9692c925 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55f245db5cb04bffa7d7868b3f6c2c6e : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f7041fa6ce94b22be940fcb73ed7ed6 : FAILURE in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e5d645498041a7b788d94a4b1d5a16 : SUCCESS in 13m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/15deebec3a50429987a19ad69e26eb56 : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aadc87bf644643bdb9fa9a23a4427d87 : FAILURE in 10m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c069779f42be47a5ad96e169781d6550 : FAILURE in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e605c8f877f4863aeb20379c789570a : SUCCESS in 42m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e13cccad744457c91a28c42210dbf63 : SUCCESS in 2h 13m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02a68242a1e14d5d834ca1714cb5abfa : SUCCESS in 1h 12m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/95be386d879e436faff622b93b7fe344 : SUCCESS in 59m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb53d2c577f4407989fbb86ef5748901 : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c996ec882f254dd6a5b496faf73948da : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a39b3d59c244599ab13f93eceb893f9 : SUCCESS in 2h 04m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bdb4d6dca36048bf99766ea7e4d5fd97 : FAILURE in 1h 21m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1dbe25f1826d4e0cae2d143538b67787 : SUCCESS in 2h 02m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"303ed1f3bc56c308312b8e277cb26df56c6eb759","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-24 20:55:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"2972cd2d3bec9647bbc8bb8688068985badc1632","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-04-24 23:44:45.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ba9c3ae57acc07c93653b3751e3aaa913fbaa743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 02:06:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc91280b32674e61bce20b78ed30fa7d : SUCCESS in 14m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73db8c8a60134778abeee80c998e6102 : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad403bf1b9e04f6da26752ee743e91e0 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee71a0cc88248ccb67c2027f8c1d9d5 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92db76b32eeb4b4ebb02dddf3eb58d6a : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa874eb51614b83a40da59e6b5a5e53 : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3ed869624005438482430e62cf91a806 : SUCCESS in 8m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0b05b9d41e4c4b14a9f0f32e74d3cfbb : SUCCESS in 10m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e07e8ec24f244c3894fcd08f8352aecd : SUCCESS in 9m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e41905300454940a1eebf4456cefdb2 : SUCCESS in 45m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54a69fc0fd944798be1ae0ded89b3817 : SUCCESS in 1h 45m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29287a78625b4bd79af3fc113488a0fb : SUCCESS in 55m 49s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a4c7e0fb47b4446a9f055f125b7785f7 : SUCCESS in 50m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8ec39fc3b094ef5a0d9a27c4dec937d : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ec831acd4e5497984a2efcd99d47f6c : SUCCESS in 1h 48m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d4187ed4ecac46cc9d57f2b545151b38 : SUCCESS in 2h 07m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c01345a13e54f32ab7c94a239cf3558 : FAILURE in 1h 05m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/24c54badd94e453b80374e1ff87ac54c : SUCCESS in 1h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1b84ed59d4fd4717d891202c17aa41dc5f6e9649","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 07:40:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"89e7b7077f33db89c0140c7e182f6499587aea6a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-07-16 07:47:08.000000000","message":"Topic failover-refactor removed","accounts_in_message":[],"_revision_number":8},{"id":"0362143a55bb1d5fef39fbcb6424618abb0a56d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 09:13:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b552dc5f2264ca29eb765102b10f435 : SUCCESS in 10m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed9e1c316b9e48fd94b8de84a531989d : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e9c827c4bc4f7dbc2d553ec253873f : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3247330eef704f24a733180bc5d9215c : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a3e2a82538b439a949a6ea51871908c : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a4607dab02343a78b0d9db0711accd7 : SUCCESS in 12m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32e9cefd449f4d42b7cfd13a498fd526 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0bf69c67309e4a28959b76e809774165 : SUCCESS in 8m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88b321373d36491390446593fadf0782 : SUCCESS in 8m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da34e890a41642ab8fd5f156e13b5f58 : SUCCESS in 49m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd15a548d44147f3914cafcd5051e5f8 : SUCCESS in 43m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a04294f3425a4880b1c4bf52d3c683e6 : SUCCESS in 33m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d1738f6fc7b64b4880379bb384e36e41 : SUCCESS in 31m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54d5ae1060fe41db9d819fccdcdc7ae0 : SUCCESS in 33m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8e49fe9063d44f0c96eba6d659a00dd5 : SUCCESS in 1h 28m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/318e023be02845269315fcbedaacdb90 : SUCCESS in 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2a1b922ab9f4bd6bee4bc4f64d6c84f : SUCCESS in 1h 16m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/711448493b064878a3a2b3a47abf8b57 : SUCCESS in 58m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f54add18b6041349db539677f6fce9c : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7979973c4e1f8e705746a43b9f424d003c6db2dc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-11 19:03:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"876c3517384b2530f4465ab80cc55b5d77d31bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:03:22.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"8f2f94ba6d07cd2a1bf9af09de2ff1642b258dc4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-14 21:10:43.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"441400bff632f8ebc72f98adbcbdc77ad8d7054d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 21:10:59.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"e4a6d72272dbbfa7c58ac200ccd0d7547b8789cc","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-14 21:13:39.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"133ed8b56dcd98530d57bb2b20003f3a6bb47ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 23:47:00.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0adcfe79ab69478a9b356812673ee185 : FAILURE in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f45654e4d444af19debfde4d711d92f : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a21dd0603f4e818552e1234b484dee : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3129ae22e14a4b7da92a8e7f1d7f85b5 : FAILURE in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86cd4b78b0746958449109420b16ea9 : FAILURE in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93f378a026446a7ba2cb9ac81d08723 : FAILURE in 20m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36ff9bd8c98b4390937a52f3004640a2 : FAILURE in 4m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/724104a15dea44b688f3d2cbd83fd3ac : FAILURE in 3m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba8a581d37b749109f2d9c3a8c1cae25 : FAILURE in 2m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d423de2b4ddf467f9adebf7e1f7d0d8c : FAILURE in 14m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6764eaab10ac4bd2af4bcfc8fae8beb8 : FAILURE in 2h 12m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbacb65d1c004c6faf9c5e1cbebed587 : FAILURE in 12m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ecec555778a422cbc3408caa29c4827 : FAILURE in 11m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/326ad57b076d4c5dab50561bb81e9351 : SUCCESS in 37m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb30738d79b1442d9989d5d989a9bdee : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/715643779e73460e87a006c7dfac2652 : SUCCESS in 1h 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a84097ee71f44c64836a5f1dfd9f91d6 : SUCCESS in 2h 27m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f87ca05b19d6449fa1e90f5367336a3c : SUCCESS in 1h 33m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6afe47a0a5c641dca7e1ed65d5e21eda : SUCCESS in 1h 47m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7e05e4598f194d3f90817ce3c0c6e246 : SUCCESS in 2h 01m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bd30fac723b240b88f34b0384103d6f4 : SUCCESS in 2h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a41e771ca863a650640e4ad5314111944b7eb96c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-15 01:04:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bd9ff59bc4dbd3f803f183ca963907ea9481a2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 03:33:32.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c70fd05ad4417fa2cd353e12d9faf3 : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f24d056e95004cb7b0ba00438c98549b : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2fd6960b6f5476d938f663ec42297a7 : FAILURE in 2m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7697a526fb8410dbb046576abb48e5c : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81f2c976ae484afc92f63bc76b2d6c7a : FAILURE in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52d32a07c0f411eb6f4f54e651fdc6c : FAILURE in 11m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ad131218134145fa80ecf457d5f88b4b : SUCCESS in 4m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dea677eb0ce84238ad6bf309722f3497 : FAILURE in 2m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79f560b62c4e4dd58e1ba792eded941d : FAILURE in 2m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb802c13a44245199f88360361408d8b : SUCCESS in 44m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c95fb51add8346338c502cd0a5fccbed : SUCCESS in 1h 37m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44500a17e03e4be9bcb62d68d2211d65 : FAILURE in 11m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0fe597bd41284077b8743123122a5167 : SUCCESS in 42m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/009c5eaeaeea47de9df0b3dc59f1835d : FAILURE in 10m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb2351b4e50341fb85a57e54ac01e391 : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07b8bccf18014fa991261b7412d8ef2e : FAILURE in 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/288d83a525db4bb6a173fff49e7f8b03 : SUCCESS in 2h 26m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d83b9e0afb494b11af7736374e91f01f : FAILURE in 10m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7978d5e5dbaf4b0fbc7b6472e217fd07 : FAILURE in 10m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/adbc81bb7d68458799cedf0ae00a068a : FAILURE in 11m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe7c2d81ba39429a88670db7954ae154 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"cfdf218ca2f1c42d15ac4b91d362278d6667ba19","revisions":{"b1155df4e9aff185ca5322170176e71ec5f32af2":{"kind":"REWORK","_number":1,"created":"2020-02-05 23:20:51.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/1"}}},"commit":{"parents":[{"commit":"005cd1e6a63f226285486a8287ec523e1d73a7ff","subject":"Update the lb_id on an amp earlier if we know it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/005cd1e6a63f226285486a8287ec523e1d73a7ff"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:20:26.000000000","tz":-480},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1155df4e9aff185ca5322170176e71ec5f32af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1155df4e9aff185ca5322170176e71ec5f32af2"}]},"branch":"refs/heads/master"},"a9657c3b77ddb221428df519a8c4462e52381cc5":{"kind":"REWORK","_number":2,"created":"2020-02-06 03:54:23.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/2"}}},"commit":{"parents":[{"commit":"005cd1e6a63f226285486a8287ec523e1d73a7ff","subject":"Update the lb_id on an amp earlier if we know it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/005cd1e6a63f226285486a8287ec523e1d73a7ff"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-06 01:20:41.000000000","tz":-480},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9657c3b77ddb221428df519a8c4462e52381cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9657c3b77ddb221428df519a8c4462e52381cc5"}]},"branch":"refs/heads/master"},"ff7de4fb026d2365a923039edcea83673d7a2701":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-22 07:22:48.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/3"}}},"commit":{"parents":[{"commit":"005cd1e6a63f226285486a8287ec523e1d73a7ff","subject":"Update the lb_id on an amp earlier if we know it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/005cd1e6a63f226285486a8287ec523e1d73a7ff"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-22 07:22:42.000000000","tz":0},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff7de4fb026d2365a923039edcea83673d7a2701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff7de4fb026d2365a923039edcea83673d7a2701"}]},"branch":"refs/heads/master"},"f9d99b6340d2b5b26e254a42c36b48a85a7c3c5e":{"kind":"REWORK","_number":4,"created":"2020-03-21 00:28:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/4"}}},"commit":{"parents":[{"commit":"d2b432b07e1cccfc38deac9be681411921416071","subject":"Merge \"Fix load balancer update with provider filtered params\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2b432b07e1cccfc38deac9be681411921416071"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-03-21 00:27:43.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d99b6340d2b5b26e254a42c36b48a85a7c3c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9d99b6340d2b5b26e254a42c36b48a85a7c3c5e"}]},"branch":"refs/heads/master"},"2791d362782e5b01bbdd6b5dccb73e523d835481":{"kind":"REWORK","_number":5,"created":"2020-04-13 11:23:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/5"}}},"commit":{"parents":[{"commit":"728c4eac24390516847c43b250d0b5bc35cf05bc","subject":"Refactor the failover flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/728c4eac24390516847c43b250d0b5bc35cf05bc"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-13 11:22:43.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2791d362782e5b01bbdd6b5dccb73e523d835481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2791d362782e5b01bbdd6b5dccb73e523d835481"}]},"branch":"refs/heads/master"},"6c5663fd4fc10ef06e8e0750c824bd3c58c9eed7":{"kind":"REWORK","_number":6,"created":"2020-04-15 10:54:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/6"}}},"commit":{"parents":[{"commit":"59fd8c90e2ac7ca5103ae2987544bc827a833c14","subject":"Refactor the failover flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/59fd8c90e2ac7ca5103ae2987544bc827a833c14"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-15 10:54:12.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c5663fd4fc10ef06e8e0750c824bd3c58c9eed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c5663fd4fc10ef06e8e0750c824bd3c58c9eed7"}]},"branch":"refs/heads/master"},"42eb360d09e527e3e6dc2b20041aa3afa64a6938":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-24 20:55:40.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/7"}}},"commit":{"parents":[{"commit":"7484a736cd4a593ea2020e740875a6dbf0aee77e","subject":"Refactor the failover flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7484a736cd4a593ea2020e740875a6dbf0aee77e"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-04-24 20:55:40.000000000","tz":0},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42eb360d09e527e3e6dc2b20041aa3afa64a6938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42eb360d09e527e3e6dc2b20041aa3afa64a6938"}]},"branch":"refs/heads/master"},"66c20380f15234804b0ea96bd6c21773673d1a6d":{"kind":"REWORK","_number":8,"created":"2020-07-16 07:40:28.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/8"}}},"commit":{"parents":[{"commit":"8ae82a47e1928f99dd32f08785f876b3d1d9c3cb","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ae82a47e1928f99dd32f08785f876b3d1d9c3cb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-07-16 07:40:17.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66c20380f15234804b0ea96bd6c21773673d1a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/66c20380f15234804b0ea96bd6c21773673d1a6d"}]},"branch":"refs/heads/master"},"988fb65efe0137817d2fc39a8c1365c7df4c5783":{"kind":"REWORK","_number":9,"created":"2020-09-11 19:03:04.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/9"}}},"commit":{"parents":[{"commit":"a446c04f779faffa8262d2f1d2c9ce7e506d6525","subject":"Revert pyroute2 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a446c04f779faffa8262d2f1d2c9ce7e506d6525"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-11 19:02:50.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/988fb65efe0137817d2fc39a8c1365c7df4c5783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/988fb65efe0137817d2fc39a8c1365c7df4c5783"}]},"branch":"refs/heads/master"},"c4641c612ce20480fdcf1b54ee0508f91b8b6cf0":{"kind":"REWORK","_number":10,"created":"2020-09-14 21:10:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/10"}}},"commit":{"parents":[{"commit":"a28bc2a55a1e61a45c15b0f9bfcd057369423bbb","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28bc2a55a1e61a45c15b0f9bfcd057369423bbb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-14 21:10:21.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4641c612ce20480fdcf1b54ee0508f91b8b6cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4641c612ce20480fdcf1b54ee0508f91b8b6cf0"}]},"branch":"refs/heads/master"},"cfdf218ca2f1c42d15ac4b91d362278d6667ba19":{"kind":"REWORK","_number":11,"created":"2020-09-15 01:04:46.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/53/706153/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/706153/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/706153/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/706153/11"}}},"commit":{"parents":[{"commit":"a28bc2a55a1e61a45c15b0f9bfcd057369423bbb","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28bc2a55a1e61a45c15b0f9bfcd057369423bbb"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-02-05 23:19:47.000000000","tz":-480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-15 01:04:31.000000000","tz":-420},"subject":"Use routed network filter if it exists","message":"Use routed network filter if it exists\n\nDepends-On: https://review.opendev.org/#/c/709280/\n\nChange-Id: I32ff97ff97022a59731358e60b2604c118a506a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfdf218ca2f1c42d15ac4b91d362278d6667ba19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfdf218ca2f1c42d15ac4b91d362278d6667ba19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
