)]}'
{"id":"openstack%2Foctavia~706393","triplet_id":"openstack%2Foctavia~master~I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","subject":"Add a periodic image build job","status":"MERGED","created":"2020-02-06 21:52:25.000000000","updated":"2020-03-20 22:01:21.000000000","submitted":"2020-03-20 21:58:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706393-1584741491145-7a5aba1f","meta_rev_id":"e27e142992a0b02f76f6aa1ae4223a27813b63d3","_number":706393,"virtual_id_number":706393,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 21:58:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-03-20 19:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-02-07 16:55:21.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-02-07 18:28:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-03-20 19:56:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 16:55:21.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-07 18:28:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-20 19:56:35.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-03-20 21:58:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4f5de616f3bbe3f1bcaaeea3ca5b50ad0543353","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-06 21:52:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e545de4942d8475864ec9e378ab17785caa258c9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-06 21:53:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8508aa632b7631499be32f00973a0753ef5b95b8","accounts_in_message":[],"_revision_number":1},{"id":"a2e58e7b896a169fdf2d3c4dcde416b181865574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 23:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d633ae2fe97245a59cb6a078b5852546 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37fddd30783a4baa9dd71e99a009aaa0 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd30cee3e40a4225872f5708c176cbf7 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a3c74f385a546acafc01f9ad30d0ea5 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8caf9b2bdd644bb59b0447698ce6b5d8 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/106c42cb32234dd6aa6dbe5ebb08fbe3 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26e742a02d9468aa725cffd76167de6 : SUCCESS in 10m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5a94aa9c1c3e4fb3831997fe6788d383 : SUCCESS in 7m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2661ece84cf548cb80cd514b727e7950 : SUCCESS in 11m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ea85eb989694d7bb3be6a53b442200b : SUCCESS in 9m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c17ec6e5715e4b1faa8792a4463e9227 : SUCCESS in 42m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9f9e9260e2d4e679b5cf764b8b5644e : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7efe940c4a84461d8f795f59e9537664 : SUCCESS in 59m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d623eb0626c7409b8cf7da230451b989 : SUCCESS in 51m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd32a0b310b1427bbec3b2af40dd7258 : SUCCESS in 58m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4437902c938b4fcca4f738749201597f : SUCCESS in 1h 39m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ff5a6e12f0f4a8f97f668a182b5edac : SUCCESS in 1h 37m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e8acb25c2e740dabd6e12363951ff71 : FAILURE in 37m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf3b481e444f4e17a45a0544fd372278 : SUCCESS in 1h 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2d848e94ce08f6c0c7ecc362136516c927e7281","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-07 16:55:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"529e63fd3f6835c6afe98195cdea1952eb1c9454","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-07 18:28:15.000000000","message":"Patch Set 1: Code-Review+1\n\nWe can update the failure dashboard once this merges","accounts_in_message":[],"_revision_number":1},{"id":"b5b23d3993067b704180015661fbf94be549c800","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-03-20 19:56:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"509591f1a93165c382fa37dd580399ce58c1d54b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 19:56:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f3ad9cc1f519f91b4105dc4d87b51d7f850695e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 21:58:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f82e77eb4ca44028d9765a18ce024d1 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84a4510eb4445e4948d4eb3d62eff95 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b7b35949654bba94d1cebd578ea09b : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c81ce491446c4c089885868ccef1adc8 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101a4a0e2df546eaaf5d1832bc052cfc : SUCCESS in 11m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e870c93c078d42ee84923d2e6aa19783 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c4f569d071240519591b547b2892a2e : SUCCESS in 44m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd45278f7d42469e9860d59f71a281fd : SUCCESS in 1h 56m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/520d71192fd548e595590c22ee89a9d3 : SUCCESS in 58m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d72fba37de6b4e8bbd5ce372194101c8 : SUCCESS in 53m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1c1caf74fb94d88901c8376ba8d1d2b : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"090da2b3f0fce367970a248fbfe0df067691d7b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 21:58:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e27e142992a0b02f76f6aa1ae4223a27813b63d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:01:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94eb5cf1ce594699ad9dc4f5245b2c98 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f9915037ad68356231691607b461c30b28df3b7","revisions":{"1f9915037ad68356231691607b461c30b28df3b7":{"kind":"REWORK","_number":1,"created":"2020-02-06 21:52:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/706393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/706393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/706393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/706393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/706393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/706393/1"}}},"commit":{"parents":[{"commit":"af9edc30d7f2f93fec2e0fd2a430e31f6846541e","subject":"Merge \"Complete dropping py27 support goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af9edc30d7f2f93fec2e0fd2a430e31f6846541e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:52:24.000000000","tz":-480},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f9915037ad68356231691607b461c30b28df3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f9915037ad68356231691607b461c30b28df3b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
