)]}'
{"id":"openstack%2Foctavia~706394","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313","subject":"Add a periodic image build job","status":"MERGED","created":"2020-02-06 21:53:33.000000000","updated":"2020-05-06 13:46:20.000000000","submitted":"2020-05-06 13:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706394-1588772537975-b9db8533","meta_rev_id":"b1a7149520f4dbee02b87a687521458984bb913f","_number":706394,"virtual_id_number":706394,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 13:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-05-06 11:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-05-04 07:50:06.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-05-06 11:47:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 07:50:06.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-05-06 11:47:55.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-05-06 13:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69af6ec468226cdf86a79097cb6135daf58cc1fc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-06 21:53:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f2a5faa9b4186363145030edfc85013bd484107b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 00:08:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afa78555594146b3a14692c81d6d5fbe : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32084e81df1944aa94b48ff26c88a9fe : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d04ec9468f044e6a50291626dc1e60e : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/993dac38d344498aa184ea64c3fccd1e : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa13ba6396f245db8d33aae2b348f63f : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/909d81772ace4e3da8eee327f5944b30 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6691cc4fb4b48a5bf3bd8d5608fc52c : SUCCESS in 13m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/abf1c1cffe7f48d18c37f24eaeb40334 : SUCCESS in 10m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/712cd8b8aebe4d5e84952557f7096341 : SUCCESS in 11m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f858080c18584c4aa300719ca4ee0209 : SUCCESS in 10m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27c8b322d2454aa18d6544df822d7dc2 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3e4daf5a5a04aa2ab3b5dd7707817b8 : SUCCESS in 50m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1d73b1f5e6be4ad192585bac4b6c1054 : SUCCESS in 46m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d9da4c07af4dc79d4ba91d47d35553 : SUCCESS in 1h 21m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b6f4eb05c31f40d385728e1640ae104a : FAILURE in 31m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8058c4f894c34672abbc7f2c9f6f561d : FAILURE in 24m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a60d6a8e63dd41e1ba42ccd5651ab893 : SUCCESS in 2h 06m 59s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e32ad03c0bf40a7b8e5dca742cfc95a : FAILURE in 34m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/72a1e158aa2344589f5edd11f7dcc888 : FAILURE in 36m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6adc1a5583f54df7b74ca66cb91da342 : SUCCESS in 1h 02m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/530dab7d51cb47beb457a637256ace3d : SUCCESS in 1h 34m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6be03cd3caee485f9bda78de0b350d93 : SUCCESS in 1h 04m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/baaa4025bb7c41d68387401bc86f7da4 : FAILURE in 32m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46f8170e32d1420e8f56428df6aba8fb : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85a3065f94b218c027d59e9f6f1962ccdf586a9d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-10 16:41:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7386c0f307b91d6a46a4967b42102081da94334c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:40:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76277d97c94548608f44dc1f162810f4 : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4513753c20ef41249162b2d599ab9316 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36457510bc0e4aee9a5b01737b8e3c0b : SUCCESS in 7m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1a1859d81734705960fb09950e7f14a : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38cb04484dff4be3b3639d49205b73c0 : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da87133d057c455692262902528313ec : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cceb3874bb5e4fc7a6c731bdd29db648 : SUCCESS in 16m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/057a79926abd4d5cb7d2228f185b956c : SUCCESS in 8m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fa472b0fba9d4afc8503a02f1ec127e5 : SUCCESS in 10m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/224191c65c3741019f80d7c3d7573233 : SUCCESS in 12m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/587b6c77b032412cad4246e9c39c93ba : SUCCESS in 8m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ed22c1b770a47b79e9492d55983b8ff : SUCCESS in 48m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d30230b2bfc7445fb107b8297824f316 : SUCCESS in 49m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/210f0a91ee6644e6830c253c65cb0e71 : SUCCESS in 1h 28m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3a0e7a552b9a4cd58a7172ec6f607e8e : FAILURE in 39m 10s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3aacc5b0c28c418592e6acf4695ba730 : FAILURE in 27m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/164b7facda1d47358c290143d8e33fe1 : SUCCESS in 1h 45m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/55c5ee595ec5448db71ec60af7a0feae : FAILURE in 40m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81d61ee6f462407a8c101790c11c20a2 : FAILURE in 40m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6169d97b3c184088a8ff0e04fe102f5a : SUCCESS in 1h 04m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1c9a7cfd619f49fdada04410618f277c : SUCCESS in 1h 33m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7edcba3d37fd42a0ac48d084a263f8ce : SUCCESS in 1h 05m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2acd3f6abfcc48ffa478c5ed140cbac3 : FAILURE in 32m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e7d27cbccbc941e9adc6c8529ef992c8 : FAILURE in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a07d28722f207427744b8c67352e4aeed8ae0f44","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-04 18:27:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bbf875b37f279050d709b7666cc550753abe01fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:08:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2c35503bb894cf18f13e549eeb60bdc : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e78e6aa7554bdc953d7a9648980c89 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f03b6bac054ab4987ae4604a1718ec : SUCCESS in 8m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2529e44736154f2c9b27247d4245a9ec : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9c5e59b05874a0d806c8d5d00b8dfe1 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ef6f5dc90714b7aa33c782c4593abb5 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b57cc3dc2a40a2a7862d7403c8be3f : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3722d6224c3f4b9592ab3e31012e9083 : SUCCESS in 6m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/051b71e0cc4d4fa187003dceb79039bb : SUCCESS in 9m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/906c1ed34c6e4dfdb5777d41af5af684 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd8dc5f13b5f4d03ad1e5e07bf1d6e8a : SUCCESS in 9m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/740533764b57444d891c68877aa99fb6 : SUCCESS in 52m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3ba77dca4bec4ff7bd8d611d45a32cfd : SUCCESS in 51m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ea88070664314e5f8c29e2ec8e6ad5e0 : SUCCESS in 2h 12m 18s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/be4d941a85874d61977d154d703b35ac : SUCCESS in 2h 13m 45s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2ab62dabf2a5473e81c5f2690e4b64f6 : FAILURE in 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e08f4ad987ef4c4ea0d24ffcb692b8af : SUCCESS in 2h 09m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd2ae5ca64e741c3ab31bf4a2eb9aca4 : SUCCESS in 2h 33m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/307f2e06f5d74994a45828e735bd6480 : SUCCESS in 1h 39m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10e70657f1144319af3ab70027b71525 : SUCCESS in 1h 02m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/95e846c7703445b388b39278056e2bcd : SUCCESS in 1h 46m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebd0b350a0914f738a50450f612bfd74 : SUCCESS in 1h 14m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e1cff5c36e864d42ba7c3d6a79d33380 : SUCCESS in 1h 09m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e874545e7a844588a12cb9cf6cd7f443 : FAILURE in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad2f63a7efbff9541fb66917c00c219f5c00dba6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-04 07:50:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"681646d5a3ad102f8b81e80524b45a27ab518b88","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-05-06 11:47:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99e58e8591e11ac4e05dd2d7381b304e3956df06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 11:48:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3fdd53a00ba9998a17a5074882e672f499d2cc06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bf479d92b25fbc2c25ab453976e7fe8044015623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:42:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d701a3a6bc8b42679a1b2b9f021026e8 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c2120a12ba0424fb6ca3ddfaa663269 : SUCCESS in 7m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb842993b6bf4b5497e472bce90ea7d9 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74e767804b3f465781dae2781c459144 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7472e3b213a48b58024a39166b7c32e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2184e9f1a2b7405a8593ca344715926d : SUCCESS in 12m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/91c92f9383fd4f8fae4cd0d01adb6d34 : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3cc0bf8aebe4f6e98aecd017fab289f : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37306f7441b64d9493328b17c5189dd2 : SUCCESS in 45m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/33fbe7d7541f455e89e1f5e064a77f0b : SUCCESS in 49m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da5b38386d3346d68f571e6c2361ca10 : SUCCESS in 1h 52m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/14d8e4c8545843e9b012d868207d61b8 : SUCCESS in 1h 07m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5ee01feeed54beeafdc3bb3e7dedc2d : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b1a7149520f4dbee02b87a687521458984bb913f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 13:46:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96677dce17664a7191798a0460e4f382 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b989f9413d202dce3ee7f624afc57168b2d025b7","revisions":{"8508aa632b7631499be32f00973a0753ef5b95b8":{"kind":"REWORK","_number":1,"created":"2020-02-06 21:53:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/706394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/706394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/706394/1"}}},"commit":{"parents":[{"commit":"4eb2fdacec0420081b43f6b0f2dd18eda47d35b2","subject":"Merge \"Use LUKS1 for certfs-ramfs for all distributions\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4eb2fdacec0420081b43f6b0f2dd18eda47d35b2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:53:33.000000000","tz":0},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8508aa632b7631499be32f00973a0753ef5b95b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8508aa632b7631499be32f00973a0753ef5b95b8"}]},"branch":"refs/heads/stable/train"},"b989f9413d202dce3ee7f624afc57168b2d025b7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-04 18:27:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/706394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/706394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/706394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/706394/2"}}},"commit":{"parents":[{"commit":"0c1cb417e904daa75deedaabc72a20d07b138eb9","subject":"Fix uncaught DB exception when trying to get a spare amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c1cb417e904daa75deedaabc72a20d07b138eb9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-02-06 21:50:56.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-03-04 18:27:03.000000000","tz":0},"subject":"Add a periodic image build job","message":"Add a periodic image build job\n\nThis patch adds a periodic job to build the amphora image using the\nreleased version of diskimage-builder (our other jobs use Git master).\nThis job will not publish the produced image, it is only to test the\nbuild.\n\nChange-Id: I611d9cd5a519abc6a330ec2c46a2f5a1fdb89313\n(cherry picked from commit 1f9915037ad68356231691607b461c30b28df3b7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b989f9413d202dce3ee7f624afc57168b2d025b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b989f9413d202dce3ee7f624afc57168b2d025b7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
